)]}'
{"id":"openstack%2Fswift~898679","triplet_id":"openstack%2Fswift~master~Iab48e5abe91635826bbaa33a2b9c73608fad9c19","project":"openstack/swift","branch":"master","attention_set":{"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2025-10-09 09:46:20.000000000","reason":"\u003cGERRIT_ACCOUNT_7847\u003e replied on the change","reason_account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}}},"removed_from_attention_set":{},"hashtags":[],"change_id":"Iab48e5abe91635826bbaa33a2b9c73608fad9c19","subject":"s3api: Sort storage_domain list by length","status":"NEW","created":"2023-10-17 18:11:02.000000000","updated":"2025-11-17 15:45:35.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":3,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"dc355a39b7d23187fdb8d7b0db62ccfc59cba64a","_number":898679,"virtual_id_number":898679,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"tag":"autogenerated:zuul:check","value":1,"date":"2025-07-28 01:46:05.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":-1,"date":"2025-11-17 15:45:35.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-10-13 18:09:24.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-10-17 19:23:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-10-09 09:46:20.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"REVIEWER"},{"updated":"2025-10-13 18:09:24.000000000","updated_by":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"reviewer":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"state":"REVIEWER"}],"messages":[{"id":"2eb88482e55e52944d209b48163e8876d7cf1d2e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-10-17 18:11:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ae52199879e39b0e9e122c21cb846ebace98c2d1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-17 19:23:47.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/84124d5d48f14b9193ee8e191636d35f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/481015e155a641bb8518c6fbcd75de4f : SUCCESS in 9m 54s\n- grenade https://zuul.opendev.org/t/openstack/build/a9e6361b6d5347789e725d4a70e64f3f : FAILURE in 46m 05s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/18f5d32735144b8eb6cf7c008a02b148 : SUCCESS in 1h 00m 19s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/fbe1444e873843dda0bf614be7f3f106 : SUCCESS in 50m 22s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1594122d8a364c5f86d22ce3fa6b7109 : SUCCESS in 58m 25s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ad96354beec040ec9f52581b163d005b : SUCCESS in 19m 14s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/fa4610782ce44abfba327d6126c605ad : SUCCESS in 19m 49s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/02d8349d528147409b7638ab6e66e8b7 : SUCCESS in 20m 02s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/e6aa0bc8f52a4339a096e24a6fb109ec : SUCCESS in 9m 28s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/44c7bb48f5b84d99961133b8d849a4d0 : SUCCESS in 6m 15s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/99b7218d1dcc4f849c6e8acbc664a219 : SUCCESS in 17m 23s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/d4bd28dbb38d4fda9d2ce1ee0946ef56 : SUCCESS in 15m 33s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/fe989a4fbbc44ae3977d0cc74afc91a5 : SUCCESS in 23m 14s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/1e4ca72433dd4973bd201759d11ccd0a : SUCCESS in 12m 54s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/2d5f49ecd3314602be31f063ab96d7b4 : SUCCESS in 24m 42s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/8b15402a751741a29536dd75c95dce49 : SUCCESS in 19m 22s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/b6399278f03e4f5e80b878bdccce85f5 : SUCCESS in 23m 32s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/02935347de9348a2a4f3881b96aaa61a : SUCCESS in 17m 08s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/68445ac18a80476ea217cc28b26b5318 : SUCCESS in 15m 34s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/9ad3405a705145e3ad773a6da318ef8b : SUCCESS in 17m 33s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/4e4efa565b8f476b902a1fafa36d49aa : SUCCESS in 15m 02s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/f145801a8f4748feb12d9a0e20003569 : SUCCESS in 6m 28s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/7e38b8576ad34be19ed92acba63167e4 : SUCCESS in 15m 03s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/992480988a3f4410a3a7ac63fc0f41f4 : SUCCESS in 7m 09s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/f27a3a5262ed4ad48603d69c393fc56b : SUCCESS in 53m 16s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/4235cacf3d244448a0f23a89ca94232f : SUCCESS in 1h 04m 45s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/2a34ebb3d5d74f4398daa05a336f2bf1 : SUCCESS in 1h 02m 28s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/ed24fff070b742ce9dbb469fbb55b9e1 : SUCCESS in 58m 33s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0889c153f6f949549511a1cd838fe957 : SUCCESS in 17m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/404b77bb224a4d99a13e19bf081039d0 : SUCCESS in 5m 06s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/8f15765c3d4245d1ad81295f2583ed47 : SUCCESS in 20m 35s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4e2f5a6c742b4f6a949758f513d6f8a8 : SUCCESS in 59m 45s","accounts_in_message":[],"_revision_number":1},{"id":"be73ae5880613c345c2b19c9753e0845487326f4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-17 20:06:16.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2d1ca1d0433548f8a7bf0a08dbd97b1e\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a1b76ec9aed044af9111fdf8133a961c : SUCCESS in 1h 04m 32s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/80db98b324a84f4f8fc9d991792dd335 : SUCCESS in 1h 12m 56s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/908fd9b1641f4dc3b9889085c2bda8d2 : SUCCESS in 1h 09m 03s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/e48fafa3e45840b9a6f3e460953fa344 : SUCCESS in 1h 05m 27s","accounts_in_message":[],"_revision_number":1},{"id":"743bbd72a9b1192e4192427e351cd221da8e0041","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-05-05 04:01:11.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f3fc9c148743081d4dc2c23417e2d8e58487c1f7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-05 04:44:22.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ff82ed0f0c1444438917416f399a9d1c\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/8d3bc76bc4f54d2184b06e1ff1ac6c60 : SUCCESS in 41m 15s","accounts_in_message":[],"_revision_number":1},{"id":"408aa295ae438cba1e569638ecea6dcdf1b5be96","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-05 05:00:03.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f038957c7dc6470aaeea5cfadb38852a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5b278c4e954e4c8887d9d9a492ab4bf7 : SUCCESS in 7m 38s\n- grenade https://zuul.opendev.org/t/openstack/build/d22f295569b544b6abb1cd3acbe19e48 : SUCCESS in 40m 45s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/ad15899d44bf42b3932094d84802d18e : SUCCESS in 27m 00s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/b63fb0aa5c38482493af22c8df7d43ed : SUCCESS in 37m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cf3b80a768e245128ed6211c7d489d0e : SUCCESS in 44m 59s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/63152354761f4958a1b7fcdcd2f1af5e : SUCCESS in 9m 47s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/243cb3ddaacb4fedb86a44b4e424b75b : SUCCESS in 9m 46s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/dadcee4d04d8433b9bc04f425dcb973f : SUCCESS in 17m 20s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/43f8a4ae06884fe3bd0ebad03f5c017a : SUCCESS in 3m 16s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/bef0934ab80b4e018a1d9ed3a2e9cd4d : SUCCESS in 14m 41s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/a1803f4439c84c9fbf0481dae47e54ba : SUCCESS in 15m 43s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/11541ffab42d40eda4c5831bc9d85eaa : SUCCESS in 16m 45s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/48774fe4546b44699bb9eeb093d29dae : SUCCESS in 28m 18s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/3acb8f5d4dde4f388c65c67132d2dc11 : SUCCESS in 7m 26s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/fe7fb11525d94c1dba84e0130ac9b394 : SUCCESS in 13m 49s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/75419ec4c6c248a2975685c1645551ff : SUCCESS in 7m 52s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/2636296bc46a4f039f5164a8cde2acf6 : SUCCESS in 5m 37s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/b29cebf028a541bc9a3ead871a828f12 : SUCCESS in 13m 13s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/e80481e8e2804f8cbe7dc7d79ff34cd4 : SUCCESS in 5m 58s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/ab53e293238244678fe664966e8bb79f : SUCCESS in 47m 39s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/cd34449978fb4305b54eb22c16f4514f : SUCCESS in 43m 36s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/cb36ec79d0af4831bb2392053cc3a1ab : SUCCESS in 17m 32s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8c113c4e0ee64db490ff3756003c260b : SUCCESS in 20m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ad3d7387a47b4ae9ac7865746108c241 : SUCCESS in 3m 50s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/cb07e1ad78ec4ccaad92d7c33925bda0 : SUCCESS in 20m 59s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/04125cb378ec48549161a3a5c2dc742b : SUCCESS in 48m 08s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ce975ca5b6a24529907fb550e6f70230 : SUCCESS in 54m 00s","accounts_in_message":[],"_revision_number":1},{"id":"b48790f666c1ca990b4a117475dc76328962966c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-07-28 00:38:33.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"982f9b55c177c28ec09bb7094776fe8c3c89ffe4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-28 01:31:39.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/920e7d6dac1745c4a6bfe53b6de681b1\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/ebb551499642490d8d2aaeb16c613d8b : SUCCESS in 50m 37s","accounts_in_message":[],"_revision_number":2},{"id":"af76af9204e600579842ed5a11ff3ac8623302c1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-28 01:46:05.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6509951ca0c54072a9abb8c890d08cbf\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0a86e7626a98489494b82af9756c9ec3 : SUCCESS in 10m 26s\n- grenade https://zuul.opendev.org/t/openstack/build/688e410199b742a687e5c51662b627e4 : SUCCESS in 28m 20s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/88e14ec3b3cb4f1497097d7660a0fc04 : SUCCESS in 56m 57s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/7e0f9742c93644b69e6af1dfe6aa07f6 : SUCCESS in 30m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/db47e40d74314e438aee2d7533231067 : SUCCESS in 1h 03m 57s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8345bd8aaf184eb1bc98c4c309bd0b15 : SUCCESS in 18m 57s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/abf03d02215744998bc48f2752a9cc4b : SUCCESS in 10m 25s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5f62b3e6ea0c43edbe1241747353a985 : SUCCESS in 19m 22s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/d867e68986004460908f8699ca776715 : SUCCESS in 4m 04s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/e666c70e614b4a198b97a14db24ded2c : SUCCESS in 17m 47s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/d2ddc4d35d9641378e4a13d8992d07f3 : SUCCESS in 8m 04s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/b5265493c14e45f0b2f3f9e13dd1dafb : SUCCESS in 10m 26s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/4ccd0c9f6cf74900bf3041c0e75326b4 : SUCCESS in 21m 56s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/ed430e4331c04562aba99ece7d944d56 : SUCCESS in 8m 11s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/b3b339cd27804bc99c3f62eaf661ed10 : SUCCESS in 8m 06s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/bd72e1d85b39481ba8a1666d02e1c6bc : SUCCESS in 8m 08s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/db26e80b934f4e8eabed0d7f1da5ea33 : SUCCESS in 4m 34s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/fec74a8d28044c208fdc3b13906d9865 : SUCCESS in 17m 35s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/7a9b8256a15547cf8ea3c728d847a8cb : SUCCESS in 4m 43s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/e9d39be4a44e44baa2a78d611c84be2d : SUCCESS in 32m 41s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/191f074e74004f509d0e546789efc390 : SUCCESS in 21m 43s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/6f4a0f11fef4470fbab8527b131529d0 : SUCCESS in 22m 08s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a690d377db434539a2ddf2f51dca86af : SUCCESS in 7m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0fdd2dd9560c4b648d494d71539eb1c1 : SUCCESS in 5m 12s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/e67fe1f9c6f842669aac1f51f01b2006 : SUCCESS in 23m 32s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/74256e0f5ca34981b542e31703e23d35 : SUCCESS in 35m 44s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fc4a17564b8c4c01b3a220f31bbeb2d7 : SUCCESS in 39m 00s","accounts_in_message":[],"_revision_number":2},{"id":"a10d03894ea992ee811e5219686d444f90138804","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-10-09 09:46:20.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"850af2f43b84ac275e408c79521635d3317bc90b","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2025-10-13 18:09:24.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"dc355a39b7d23187fdb8d7b0db62ccfc59cba64a","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-11-17 15:45:35.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"515f152d2eafcc0d9a663c78e13005e0ec945de2","revisions":{"2f30b205ffd7ad22b8f0678067768a3ca7629e1b":{"kind":"REWORK","_number":1,"created":"2023-10-17 18:11:02.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/79/898679/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/79/898679/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/79/898679/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/79/898679/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/79/898679/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/79/898679/1"}}},"commit":{"parents":[{"commit":"195f54ecb5c7520b7ae19157be77047a0f4e68ea","subject":"Merge \"Add a swift-reload command\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/195f54ecb5c7520b7ae19157be77047a0f4e68ea"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-10-17 18:04:02.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-10-17 18:04:09.000000000","tz":-420},"subject":"s3api: Sort storage_domain list by length","message":"s3api: Sort storage_domain list by length\n\nPreviously, if your storage_domain list included an entry that was a\nsubdomain of another, the behavior would vary depending on the order\nin the config. For example,\n\n   storage_domain \u003d pool-1.example.com, example.com\n\nwould work as expected, but\n\n   storage_domain \u003d example.com, pool-1.example.com\n\nwould behave as though pool-1.example.com was not in the list at all.\n\nNow, domains are sorted longest to shortest when configs are read,\nensuring consistent behavior.\n\nCloses-Bug: #2031335\nChange-Id: Iab48e5abe91635826bbaa33a2b9c73608fad9c19\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2f30b205ffd7ad22b8f0678067768a3ca7629e1b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2f30b205ffd7ad22b8f0678067768a3ca7629e1b"}]},"branch":"refs/heads/master"},"515f152d2eafcc0d9a663c78e13005e0ec945de2":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":2,"created":"2025-07-28 00:38:33.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/79/898679/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/79/898679/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/79/898679/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/79/898679/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/79/898679/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/79/898679/2"}}},"commit":{"parents":[{"commit":"2e8338240f58f7a5b938fdfa516b665d358e749b","subject":"Merge \"Fix recursion error in account_quota middleware\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2e8338240f58f7a5b938fdfa516b665d358e749b"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-10-17 18:04:02.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-07-28 00:38:15.000000000","tz":-420},"subject":"s3api: Sort storage_domain list by length","message":"s3api: Sort storage_domain list by length\n\nPreviously, if your storage_domain list included an entry that was a\nsubdomain of another, the behavior would vary depending on the order\nin the config. For example,\n\n   storage_domain \u003d pool-1.example.com, example.com\n\nwould work as expected, but\n\n   storage_domain \u003d example.com, pool-1.example.com\n\nwould behave as though pool-1.example.com was not in the list at all.\n\nNow, domains are sorted longest to shortest when configs are read,\nensuring consistent behavior.\n\nCloses-Bug: #2031335\nSigned-off-by: Tim Burke \u003ctim.burke@gmail.com\u003e\nChange-Id: Iab48e5abe91635826bbaa33a2b9c73608fad9c19\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/515f152d2eafcc0d9a663c78e13005e0ec945de2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/515f152d2eafcc0d9a663c78e13005e0ec945de2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"}},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
