)]}'
{"id":"openstack%2Fswift~898964","triplet_id":"openstack%2Fswift~master~Icfc97751c80f4bb4a3373796c01bff6ed5843937","project":"openstack/swift","branch":"master","topic":"p-memcache-set-error","attention_set":{},"removed_from_attention_set":{"34930":{"account":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"last_update":"2023-10-27 04:02:19.000000000","reason":"\u003cGERRIT_ACCOUNT_34930\u003e replied on the change","reason_account":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"}},"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2023-10-30 09:00:00.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Icfc97751c80f4bb4a3373796c01bff6ed5843937","subject":"proxy: emit metric when updating namespaces memcache set fails","status":"MERGED","created":"2023-10-20 13:50:31.000000000","updated":"2023-10-30 09:00:59.000000000","submitted":"2023-10-30 09:00:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":8,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"898964-p-memcache-set-error","meta_rev_id":"6fb4d28b150bfeaab74c6c73b436ccf091db31f3","_number":898964,"virtual_id_number":898964,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-10-30 09:00:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"all":[{"value":2,"date":"2023-10-30 07:47:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":1,"date":"2023-10-27 04:02:19.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"all":[{"value":1,"date":"2023-10-30 07:47:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-10-20 15:23:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-10-20 15:41:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-10-25 22:46:59.000000000","updated_by":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"reviewer":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"state":"REVIEWER"},{"updated":"2023-10-30 07:47:00.000000000","updated_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"reviewer":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"state":"REVIEWER"}],"messages":[{"id":"49df22d1c96aea181be82a78ce1a78dc2c1d1206","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-10-20 13:50:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"aee752147ea7eb5b71728be6b188ecf8a41390c6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-20 15:23:53.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/953094971e454d6f8ee6ef1518608f26\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/140d761d9cce4bc39142eafa77de98a2 : SUCCESS in 1h 08m 01s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4b39c3e0526b4c618f99288b5453aaad : SUCCESS in 1h 19m 21s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/4e42b447140d43e2a4d388f71a37bd43 : TIMED_OUT in 1h 30m 58s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/7c4ce2349dd747219797f847c09be904 : SUCCESS in 1h 28m 12s","accounts_in_message":[],"_revision_number":1},{"id":"e3799a0cf4019979c1cbf6e99b7c65222055f95b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-20 15:41:34.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b165d34383e748d3b5666806d3f19929\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/62d907def37d4f3890229421ef97a1d1 : SUCCESS in 9m 05s\n- grenade https://zuul.opendev.org/t/openstack/build/cbee0cf02ce945a6b6780ecb5de21b88 : SUCCESS in 42m 27s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/0897466d73384a19a388781d609a64cc : SUCCESS in 44m 05s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/01e5c5763ac84d499b0156b504e6fce4 : SUCCESS in 53m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9f05ccd35a614465af4b6219fe257f2e : SUCCESS in 1h 05m 44s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f8073005404745c4aaae3369c69552ed : SUCCESS in 19m 09s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7c8a1c8813cb4b5ea431da7a428fb08c : SUCCESS in 27m 08s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d76e23dfccdc4b65b11742f5663529b7 : SUCCESS in 21m 13s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/b70074e200214e8fa540af5d6e4a7720 : SUCCESS in 9m 11s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/12c1ce1133dd46d2aa8b284dd56d592d : SUCCESS in 6m 52s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/60afffbba72243dd8ffdb9e8a9fdb9c0 : SUCCESS in 17m 49s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/a5867e39e57f4903a8bd5d3b7657e4b1 : SUCCESS in 14m 49s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/5c93047fcc1e454c973e6838af2b8740 : SUCCESS in 15m 14s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/77f77271e263493588fb3737c1af355c : SUCCESS in 20m 00s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/acb65b18efe8416ba77cb7d51e14adc0 : SUCCESS in 16m 38s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/92388ac0b14f461ca455e182e7902595 : SUCCESS in 17m 56s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/0ad300383aa841afa447d619a3e7827e : SUCCESS in 19m 23s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/9d2691f2a5dd4f2b917d5fcf0c787932 : SUCCESS in 20m 58s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/8c0a1493c58f42b4b40f258c42b129ad : SUCCESS in 16m 02s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/b7bdfd59267840e785df3c2a4bdebcde : SUCCESS in 15m 07s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/c8a49072cca94cabaa1c844d742ca613 : SUCCESS in 17m 16s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/220cdab3d04b4a8bba9fba76418a66f7 : SUCCESS in 8m 18s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/2c9d5f477a5e4814bef548caf104c39d : SUCCESS in 14m 46s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/7bf6a77140184c6eb8ac7c9b0b0ae142 : SUCCESS in 7m 30s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/0b56836ce42f4c03a468f54ea4394119 : SUCCESS in 57m 33s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/89323cc2429446cda7489764d804d4b9 : SUCCESS in 1h 01m 49s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/3faa561f65394016acee9f2a04eda530 : SUCCESS in 39m 52s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/97d292fb4bca49b7b4db865333d81b7a : SUCCESS in 59m 02s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9da997a3b6454df4b7c220ac9164c349 : SUCCESS in 18m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d0e9eb3ecf0d4f71aea325d3584bc219 : SUCCESS in 5m 13s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/2ba09b0475074405a80cdb094be58877 : TIMED_OUT in 1h 31m 19s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e01d66d15dbe48dab7cefdb92cd2345b : SUCCESS in 41m 57s","accounts_in_message":[],"_revision_number":1},{"id":"c06f354f263fb503cc356b09bd2191b6cebfefb4","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2023-10-25 22:46:59.000000000","message":"Patch Set 1: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"511ffde2e91c4163272f5fad9aa75e3481926f26","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-10-26 17:24:53.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"9229e7d36a704e55eae18397ab213d8721b6d0b1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-10-26 17:27:37.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"1e8a7fd73d247a8aff35217aec834b9347bd5c61","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-26 17:51:05.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a7e6fd13e87a4ef1a64e1f9598191ba7\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/aa33c3edcd38424d951b373edd789554 : SUCCESS in 14m 59s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2c7a29e9a18346b6b60c6634f21e8bc3 : SUCCESS in 14m 40s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/b6850af628ad4ff9a9aab78077f1c92b : SUCCESS in 22m 24s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/1e7399a68f8844a4830f519006a14bbf : SUCCESS in 20m 51s","accounts_in_message":[],"_revision_number":2},{"id":"feda929d7585829a48c86c61112fbbf01e955ade","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-26 18:45:21.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e04efb17d2074aec89035018435e4e8b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6c7cab964ac74f3f828a69d7ccaad7b3 : SUCCESS in 9m 40s\n- grenade https://zuul.opendev.org/t/openstack/build/247ef9437d834d679399bb95d8e3829b : SUCCESS in 1h 06m 03s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/2ce2ea12452749da8e94a8dfdfd7f710 : SUCCESS in 47m 57s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/53765cc0130144979d321f30056c17a7 : SUCCESS in 39m 07s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c2ffa0d6d8714343adb72d36e42f209a : SUCCESS in 45m 19s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/dd448278060d499a9279575bebe287cf : SUCCESS in 16m 53s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1efa79020afc42daad6ffd27f3964852 : SUCCESS in 26m 01s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/69c3e393c2f24038a70a1c873a916b4b : SUCCESS in 19m 11s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/f8616c56fe9c4f7eb002c0670f7b5fd4 : SUCCESS in 9m 49s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/9bb5a368eee745fe81b0e1a491caa36d : SUCCESS in 5m 28s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/8dab85a9b29348229875feb4a6ad0f75 : SUCCESS in 15m 21s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/7eab76a019664818ae530c0502cc3745 : SUCCESS in 13m 30s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/264707727b9b4a798b68835ba272d13e : SUCCESS in 16m 24s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/1901224323814c1297d8afe3c336fce1 : SUCCESS in 18m 33s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/98c01284aebd43ec8ee7c8b4a5f5d173 : SUCCESS in 16m 46s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/25a07a1fc3924777911a3bd7b01c482f : SUCCESS in 24m 34s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/a17a8ad5e4b942dcbbf21ca99131d883 : SUCCESS in 19m 07s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/ba43dd2d4d0946c392536fc5ab5ae4a9 : SUCCESS in 28m 20s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/53934671eb294b5cbaa79ada73d939a4 : SUCCESS in 14m 52s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/c95edd32119b4ea99a138dfc9824f5ae : SUCCESS in 17m 12s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/a797e5c222684d099b2cfe97f094885a : SUCCESS in 25m 59s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/321065fca1864b7890af1917258c80a7 : SUCCESS in 6m 50s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/fda2c61b32e243759f1ba80ab9a00b40 : SUCCESS in 18m 18s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/b60a3a3c22b64357874ea3359fe76594 : SUCCESS in 8m 31s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/a252af4bb74e4c0fab285ea8693b8432 : SUCCESS in 1h 10m 19s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/f8944cff40bb42bcbbda8a5c632fd06b : SUCCESS in 1h 06m 45s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/a5504b2cb1764eff997d91fedb578026 : SUCCESS in 54m 46s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/a3f7edd7d1f249c89183c116995dc618 : SUCCESS in 58m 07s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/20a281f115d749338b112e382b5ba293 : SUCCESS in 17m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ef3051a2c0694c22871adb0dfd275fe8 : SUCCESS in 5m 52s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/25cacee87b3a4d30aa1bdfebe7201b11 : SUCCESS in 21m 00s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/163e9d7b046d4ff9a498c1c6ce86ab71 : SUCCESS in 45m 20s","accounts_in_message":[],"_revision_number":2},{"id":"69ada62d8d44002fed8924f6deec19daac74fa9d","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2023-10-27 04:02:19.000000000","message":"Patch Set 2: Code-Review+1\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"e7b9de2114978251de447bbf4ea45b5d13ce3dc7","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2023-10-30 07:47:00.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d9e87e6ed4f50cd20272d248cfddeec0d1e30a2f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-30 07:47:33.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"68000b2262604b2cf546aebafb2c17b4196bc911","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-30 09:00:00.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1e7323b151534cdc91fad807063f9692\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/62b94d7ee3014b56b3341c1db38808e4 : SUCCESS in 7m 43s\n- grenade https://zuul.opendev.org/t/openstack/build/65a0be63fbd2407e89ca54e227256daf : SUCCESS in 54m 16s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/db6a97894aab42de899ae4812198add1 : SUCCESS in 44m 54s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/28af39afdc5d4c8cbed1b22a258a127c : SUCCESS in 33m 33s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ea3e2e85df6e4da5807f0dc9011fc8c2 : SUCCESS in 56m 29s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/0aded30ac01042cbaffa890c127d472a : SUCCESS in 8m 20s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/a5a20d998581467ab5984be4eb9d6ab4 : SUCCESS in 5m 14s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/5311ade7ba0e407d8670523fc51d5a59 : SUCCESS in 15m 10s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/48a8d442cde44ba283ba6231ffef8476 : SUCCESS in 14m 37s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/f8db50dc3ce549e4bd50d95cd84e8116 : SUCCESS in 13m 13s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/a161368456254ad9bfad387fecbcc2fb : SUCCESS in 14m 10s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/489ddb4cf9cb4712ba2777646dae0720 : SUCCESS in 17m 10s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/dd9d5fd7ba5b4dd3aa1a6c106c698fe8 : SUCCESS in 16m 03s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/f0125f4a67094e31acd48bf8d73575b4 : SUCCESS in 16m 23s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/1ac78f4d336e40b08e8159f91b558e19 : SUCCESS in 16m 02s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/5224d210ba77465680de5b22e3506fca : SUCCESS in 14m 19s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/15272d51b61446fcbaf6ef9a568d3cc5 : SUCCESS in 14m 48s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/dada1a4a11874624a2b20885e6a2be9f : SUCCESS in 14m 55s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/b7f6287944f24550bee6d9cd71f5050a : SUCCESS in 6m 50s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/a359ce70b54b487eb91edaf1350b55b6 : SUCCESS in 6m 47s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/72751bfb6dc244b0b19b7e1c6a91edd9 : SUCCESS in 54m 47s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/53315e8dc5954620b9df8fec2bdf4e96 : SUCCESS in 1h 00m 00s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/788f612311324565a5984fff37a6d5ce : SUCCESS in 50m 59s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/15c16e1d0e6243d2a8366a31d6990eb8 : SUCCESS in 46m 40s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/95bf3ac4e5a846a0bfc7ec61bf057173 : SUCCESS in 16m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cabab6bd908442139997988940c9b260 : SUCCESS in 5m 24s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/89e62ae684764d89ac7c6c8db4f80d98 : SUCCESS in 51m 05s","accounts_in_message":[],"_revision_number":2},{"id":"4ccf55305e62cff15a8d1e9223448f0634721fa9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-30 09:00:00.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"6fb4d28b150bfeaab74c6c73b436ccf091db31f3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-30 09:00:59.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4700a522270744fc965eaf7136974205\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6d3135365da34980b79628c795695688 : SUCCESS in 42s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/8b97c831502c4b539c60489e6b22b57d : SUCCESS in 35s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/56e3227dc3984f8886b45fb317952182 : SUCCESS in 33s (non-voting)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"965908d7534492377a55e094e822d14948b9f6f3","revisions":{"639ea0adb79571a4ce3d1a679d2fa710d5b9fc4c":{"kind":"REWORK","_number":1,"created":"2023-10-20 13:50:31.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/64/898964/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/64/898964/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/64/898964/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/64/898964/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/64/898964/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/64/898964/1"}}},"commit":{"parents":[{"commit":"195f54ecb5c7520b7ae19157be77047a0f4e68ea","subject":"Merge \"Add a swift-reload command\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/195f54ecb5c7520b7ae19157be77047a0f4e68ea"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-10-20 13:40:06.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-10-20 13:40:06.000000000","tz":60},"subject":"proxy: emit metric when updating namespaces memcache set fails","message":"proxy: emit metric when updating namespaces memcache set fails\n\nPreviously, if a memcache set command had an unexpected response then\nan error was logged but no exception raised from MemcacheRing.set.\nWith this patch, when MemcacheRing.set fails in this way, a\nMemcacheConnectionError will be raised if \u0027raise_on_error\u0027 is\nTrue. This exception is now handled when the proxy tries to set\nupdating namespaces in memcache, and a counter metric emitted with\nname \u0027object.shard_updating.cache.set_error\u0027. If the set is successful\nan \u0027object.shard_updating.cache.set\u0027 counter metric is emitted.\n\nThe logging of MemcacheRing errors is improved to include the memcache\nkey being set, in line with exception logging.\n\nChange-Id: Icfc97751c80f4bb4a3373796c01bff6ed5843937\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/639ea0adb79571a4ce3d1a679d2fa710d5b9fc4c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/639ea0adb79571a4ce3d1a679d2fa710d5b9fc4c"}]},"branch":"refs/heads/master"},"965908d7534492377a55e094e822d14948b9f6f3":{"kind":"REWORK","_number":2,"created":"2023-10-26 17:27:37.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/64/898964/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/64/898964/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/64/898964/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/64/898964/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/64/898964/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/64/898964/2"}}},"commit":{"parents":[{"commit":"195f54ecb5c7520b7ae19157be77047a0f4e68ea","subject":"Merge \"Add a swift-reload command\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/195f54ecb5c7520b7ae19157be77047a0f4e68ea"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-10-20 13:40:06.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-10-26 17:15:08.000000000","tz":60},"subject":"proxy: emit metric when updating namespaces memcache set fails","message":"proxy: emit metric when updating namespaces memcache set fails\n\nPreviously, if a memcache set command had an unexpected response then\nan error was logged but no exception raised from MemcacheRing.set.\nWith this patch, when MemcacheRing.set fails in this way, a\nMemcacheConnectionError will be raised if \u0027raise_on_error\u0027 is\nTrue. This exception is now handled when the proxy tries to set\nupdating namespaces in memcache, and a counter metric emitted with\nname \u0027object.shard_updating.cache.set_error\u0027. If the set is successful\nan \u0027object.shard_updating.cache.set\u0027 counter metric is emitted.\n\nThe logging of MemcacheRing errors is improved to include the memcache\nkey being set, in line with exception logging.\n\nChange-Id: Icfc97751c80f4bb4a3373796c01bff6ed5843937\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/965908d7534492377a55e094e822d14948b9f6f3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/965908d7534492377a55e094e822d14948b9f6f3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
