)]}'
{"id":"openstack%2Fswift~899445","triplet_id":"openstack%2Fswift~master~I288584c131732dd0d90decf4ae4959a1a0cf2bbe","project":"openstack/swift","branch":"master","topic":"s3api-429","attention_set":{},"removed_from_attention_set":{"1179":{"account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"last_update":"2023-10-27 13:06:17.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I288584c131732dd0d90decf4ae4959a1a0cf2bbe","subject":"who\u0027s the boss?","status":"ABANDONED","created":"2023-10-27 03:22:24.000000000","updated":"2023-10-27 13:06:17.000000000","total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"6ff394ec984215990e46d0758e93e76db0b27c1f","_number":899445,"virtual_id_number":899445,"owner":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2023-10-27 04:27:01.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":-1,"date":"2023-10-27 03:31:01.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}],"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-10-27 03:48:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-10-27 04:27:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-10-27 13:06:14.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"CC"}],"messages":[{"id":"a0fa2e49d30eea21c75607a174448fa2ebe57e90","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2023-10-27 03:22:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a981010a47d267a89fc9f77d717fd0690feefd43","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2023-10-27 03:31:01.000000000","message":"Patch Set 1: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b0c995caa70947c04734dc74652d94c160dec77d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-27 03:48:55.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6368e57b8cac43d1a57b55effa0a973c\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/47b80d1e602d4ee1ab61aa1e76967896 : FAILURE in 18m 19s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/aaa11bda38ff46b1a87f9b5341014c6a : FAILURE in 19m 58s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/3ff18a40287645f094f939f8cdc6e30d : SUCCESS in 24m 27s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/1c17b28969ee4512a4cb650f5298665f : SUCCESS in 23m 17s","accounts_in_message":[],"_revision_number":1},{"id":"27f9fe64f209614f7973c39e4b2a544472d8f53a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-27 04:27:01.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/7bb9b72a6bb740f6a0134bd702bab99f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2d09cd2b7d6c4c3980ba1caf29a0ee5d : SUCCESS in 9m 30s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/80f53dd1b90e4654b7420b94b24c7d5a : SUCCESS in 59m 02s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e87ae73a0c534677807bdbbb1398399a : SUCCESS in 18m 28s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/92bf410a2bad4e5abce641279da45033 : SUCCESS in 20m 30s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2c81607e39e246e18a9c14f2c2cd7a95 : SUCCESS in 20m 57s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/32137b5158254233a274aeb6d7b3dcf0 : SUCCESS in 9m 57s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/b829039d7e78467fbb34479707d8efc0 : SUCCESS in 5m 39s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/ddec9021ff5b40a0a8cca89ab753c798 : FAILURE in 17m 31s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/1c64f1afb30e4acebc774da94e50e6a4 : FAILURE in 25m 36s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/4cf7cd8b6df44da38c402c72f1c1c86b : FAILURE in 14m 55s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/40d119c222c84b39aa29c73d911df936 : FAILURE in 25m 44s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/1d66d69cfece44dda8bb12fc4f33a0ad : FAILURE in 25m 25s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/b9573ce5bbbf422b8c2ea398f1c5d027 : SUCCESS in 32m 39s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/80161347e0594c1f87b7c52483d0660d : SUCCESS in 18m 17s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/557da539ae1c42d7b5b8c683a356a496 : SUCCESS in 33m 45s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/c086505a40b74a4fb9689e5c80a46339 : SUCCESS in 17m 09s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/d74565637ca24d41bc3b0cf99b7e5075 : SUCCESS in 16m 27s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/e30fc0a515a54a09aeaf79ce20a685de : SUCCESS in 16m 13s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/6c9da34f5abc4f01a7a80c3f515d2284 : SUCCESS in 56m 43s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/612a257e60014616b5e0b19a741a709e : SUCCESS in 56m 59s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fac6d6e3eaaa418f8500e51e7bdee361 : FAILURE in 22m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fd7d5797bf474a2c8fd1f644f043ffb8 : FAILURE in 4m 34s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/6e34e8af07674de5acae3963d7a446c9 : SUCCESS in 20m 21s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"822b0524e21d899bbb17e5fc07d1c0b6423aeb66","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-10-27 13:06:14.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6ff394ec984215990e46d0758e93e76db0b27c1f","tag":"autogenerated:gerrit:abandon","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-10-27 13:06:17.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a5e3ce256b2a79ebd26761de8808effd027967a1","revisions":{"a5e3ce256b2a79ebd26761de8808effd027967a1":{"kind":"REWORK","_number":1,"created":"2023-10-27 03:22:24.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/45/899445/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/45/899445/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/45/899445/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/45/899445/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/45/899445/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/45/899445/1"}}},"commit":{"parents":[{"commit":"2dc7047bbdde8402ca014627101b5da62dd0d0b9","subject":"Support swift.proxy_logging_status in request env","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2dc7047bbdde8402ca014627101b5da62dd0d0b9"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2023-10-27 03:22:02.000000000","tz":-300},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2023-10-27 03:22:02.000000000","tz":-300},"subject":"who\u0027s the boss?","message":"who\u0027s the boss?\n\nChange-Id: I288584c131732dd0d90decf4ae4959a1a0cf2bbe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a5e3ce256b2a79ebd26761de8808effd027967a1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a5e3ce256b2a79ebd26761de8808effd027967a1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":["label:Workflow\u003dMIN"],"failing_atoms":["label:Workflow\u003dMAX"],"atom_explanations":{}}}]}
