)]}'
{"id":"openstack%2Fswift~900256","triplet_id":"openstack%2Fswift~master~I9c86d1ce9853840d73e2107a6d79b2454c21438a","project":"openstack/swift","branch":"master","attention_set":{},"removed_from_attention_set":{"1179":{"account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"last_update":"2023-11-10 15:44:28.000000000","reason":"\u003cGERRIT_ACCOUNT_1179\u003e replied on the change","reason_account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2023-11-10 16:02:58.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I9c86d1ce9853840d73e2107a6d79b2454c21438a","subject":"docs: Document WSGI server process management","status":"MERGED","created":"2023-11-06 23:36:59.000000000","updated":"2023-11-10 16:07:52.000000000","submitted":"2023-11-10 16:02:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":14,"unresolved_comment_count":4,"has_review_started":true,"submission_id":"900256","meta_rev_id":"30b177d3dac7fc6630e4fbb282ae0f812cf209cf","_number":900256,"virtual_id_number":900256,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-11-10 16:02:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":2,"date":"2023-11-10 15:44:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":1,"date":"2023-11-10 15:44:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-11-06 23:45:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-11-07 21:27:00.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"CC"},{"updated":"2023-11-10 15:44:28.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"}],"messages":[{"id":"31e3e54cf000cf084620a0217088ef7e24b1dbc7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-11-06 23:36:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b3a6141bc149f8826f8a1b76fbd99a2aa29e212b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-06 23:45:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/450654f60ff24d4bb21e24d5e16882f1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d6874f3d3e474409b21aa877fa8d11da : SUCCESS in 7m 15s","accounts_in_message":[],"_revision_number":1},{"id":"e698213adea5850c26fcca151207d0498343a70b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-11-07 16:53:52.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"057aee60133a45b03c17243f0d06a47d9178fdd6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-07 17:10:07.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f31d7e4761c14fca81027690713d7497\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/34c02b10198544bebcd18cec3fc7b7b4 : SUCCESS in 14m 23s","accounts_in_message":[],"_revision_number":2},{"id":"577915e85e48102a88d64dd5d0dd5700c2f73879","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2023-11-07 21:27:00.000000000","message":"Patch Set 2:\n\n(4 comments)","accounts_in_message":[],"_revision_number":2},{"id":"2608049019fb6944fec202612e1b3a3f7738e185","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2023-11-08 17:43:56.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"46afd86a8bdd42b47039075ccca89fe93962621c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-11-09 22:10:13.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"a5b4c0dd796927b99888e9ca0c96eaa1f56f4f8b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-09 22:24:57.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a56913525f484ef3abacf85eaabe4242\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/02bb673add5a432d88bff004c492771f : SUCCESS in 9m 05s","accounts_in_message":[],"_revision_number":3},{"id":"0897d311c43bc84f1513b45fe7f737218c28a156","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-11-09 22:59:26.000000000","message":"Patch Set 3:\n\n(4 comments)","accounts_in_message":[],"_revision_number":3},{"id":"e29382a986efc3744e231acbea7092a4462ee11e","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2023-11-10 15:44:28.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(5 comments)","accounts_in_message":[],"_revision_number":3},{"id":"4293657a1e890901a8c80b35e1f1ddc96a70ac4b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-10 15:44:50.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"abcfbad9f2e7a852164f52a16fdb280e8a924003","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-10 16:02:58.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c116ec93cf304514a439651d884e5c8a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6b1ac43a1c594c35b088df1e4d76bbfd : SUCCESS in 8m 52s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/501d50da1e7c4c5fa879eca4b5cc8ed3 : SUCCESS in 6m 36s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/917733e48ea74c3b8e700bc4bf82df06 : SUCCESS in 8m 24s","accounts_in_message":[],"_revision_number":3},{"id":"fa5e21904b1ed921eeeb7835b5d5dd26f3acc62d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-10 16:02:58.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"30b177d3dac7fc6630e4fbb282ae0f812cf209cf","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-10 16:07:52.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a5d1c78e2f474747b2e3362abd0e1b69\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/05e484b185864d9b9577582a46a24b01 : SUCCESS in 59s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/e0c7a5e38c284cfb976318a52ac43bcf : FAILURE in 31s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/e74fe6cf743f4835ba5d798a93297fb2 : FAILURE in 30s (non-voting)","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"c94e1c15130ef3689ba99337fdab39a80bb3fb29","revisions":{"ed335b272b3d8ed850b1baf432602dfe372589c9":{"kind":"REWORK","_number":1,"created":"2023-11-06 23:36:59.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/56/900256/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/56/900256/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/900256/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/900256/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/900256/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/56/900256/1"}}},"commit":{"parents":[{"commit":"ed15681349ab1394b285a43948682b36613481ce","subject":"Merge \"tests: refactor SLO size/etag sysmeta tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ed15681349ab1394b285a43948682b36613481ce"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-11-06 23:36:45.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-11-06 23:36:45.000000000","tz":-480},"subject":"docs: Document WSGI reload process","message":"docs: Document WSGI reload process\n\nChange-Id: I9c86d1ce9853840d73e2107a6d79b2454c21438a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ed335b272b3d8ed850b1baf432602dfe372589c9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ed335b272b3d8ed850b1baf432602dfe372589c9"}]},"branch":"refs/heads/master"},"c05c2f798d28258a60fa4e91ff2428c9c68dda81":{"kind":"REWORK","_number":2,"created":"2023-11-07 16:53:52.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/56/900256/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/56/900256/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/900256/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/900256/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/900256/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/56/900256/2"}}},"commit":{"parents":[{"commit":"ed15681349ab1394b285a43948682b36613481ce","subject":"Merge \"tests: refactor SLO size/etag sysmeta tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ed15681349ab1394b285a43948682b36613481ce"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-11-06 23:36:45.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-11-07 16:53:34.000000000","tz":-480},"subject":"docs: Document WSGI reload process","message":"docs: Document WSGI reload process\n\nChange-Id: I9c86d1ce9853840d73e2107a6d79b2454c21438a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c05c2f798d28258a60fa4e91ff2428c9c68dda81"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c05c2f798d28258a60fa4e91ff2428c9c68dda81"}]},"branch":"refs/heads/master"},"c94e1c15130ef3689ba99337fdab39a80bb3fb29":{"kind":"REWORK","_number":3,"created":"2023-11-09 22:10:13.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/56/900256/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/56/900256/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/900256/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/900256/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/900256/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/56/900256/3"}}},"commit":{"parents":[{"commit":"c8b19f4fd1b3ef72aa21ec42bd1d0cc32398ec86","subject":"Merge \"Utils: fix Namespace and ShardRange attribute encoding in py2.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c8b19f4fd1b3ef72aa21ec42bd1d0cc32398ec86"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-11-06 23:36:45.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-11-09 22:09:48.000000000","tz":-480},"subject":"docs: Document WSGI server process management","message":"docs: Document WSGI server process management\n\nTalk about graceful shutdowns, seamless reloads, and the swift-reload helper.\n\nChange-Id: I9c86d1ce9853840d73e2107a6d79b2454c21438a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c94e1c15130ef3689ba99337fdab39a80bb3fb29"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c94e1c15130ef3689ba99337fdab39a80bb3fb29"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
