)]}'
{"id":"openstack%2Fswift~900350","triplet_id":"openstack%2Fswift~master~I2cec1cef5eb34874feea9f2922fd13c4f011e410","project":"openstack/swift","branch":"master","topic":"p-proxy-namespaces","attention_set":{},"removed_from_attention_set":{"7233":{"account":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"last_update":"2024-01-05 17:54:54.000000000","reason":"Change was abandoned"},"34930":{"account":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"last_update":"2024-01-05 17:54:54.000000000","reason":"Change was abandoned"},"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2024-01-04 16:33:04.000000000","reason":"\u003cGERRIT_ACCOUNT_7847\u003e replied on the change","reason_account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}},"32761":{"account":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"last_update":"2024-01-05 17:54:54.000000000","reason":"Change was abandoned"},"1179":{"account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"last_update":"2024-01-05 17:54:54.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I2cec1cef5eb34874feea9f2922fd13c4f011e410","subject":"proxy: always use namespaces for updating and listing","status":"ABANDONED","created":"2023-11-07 18:36:33.000000000","updated":"2024-01-05 17:54:54.000000000","total_comment_count":224,"unresolved_comment_count":11,"has_review_started":true,"meta_rev_id":"270e3d7aa78b09e5c4c9a0d2edca33b00cb5138e","_number":900350,"virtual_id_number":900350,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:check","value":1,"date":"2024-01-04 17:44:45.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-11-07 19:53:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-11-07 20:05:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-11-10 19:41:06.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"CC"},{"updated":"2023-11-13 05:45:10.000000000","updated_by":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"reviewer":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"state":"CC"},{"updated":"2023-11-14 04:05:42.000000000","updated_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"reviewer":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"state":"REVIEWER"},{"updated":"2023-11-19 00:57:26.000000000","updated_by":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"reviewer":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"state":"REVIEWER"},{"updated":"2023-11-29 21:56:41.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"},{"updated":"2023-12-28 09:23:51.000000000","updated_by":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"reviewer":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"state":"REVIEWER"}],"messages":[{"id":"cd9149dcaccf6568206903fe027a7d5875d0d4ce","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-11-07 18:36:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4ba37481aad8b3d10b8617221f2888a010493e4f","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-11-07 18:36:49.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"38425671c7b4a8f90d6b3fd63591247a92a8edd8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-07 19:53:03.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d0840fd36b554efa96dee01a168dc355\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d78ae696b90a43859d729100b391595d : FAILURE in 1h 01m 11s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7f3c70b654bc4e628f14957391b043ac : FAILURE in 18m 15s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/c08f799e123e4b82be52411bc054b439 : SUCCESS in 1h 14m 13s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/d94e09cae62648e5ba7f17e9aa4156ad : SUCCESS in 19m 59s","accounts_in_message":[],"_revision_number":1},{"id":"1f7974ed7006f90efa766e5bbdac21a05c7e5658","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-07 20:05:45.000000000","message":"Patch Set 1: Verified-1\n\n(5 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/350e15fb2d554a68a16ebcf1b734b407\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/144fb8ace24f4206971ff0899a16807a : SUCCESS in 9m 37s\n- grenade https://zuul.opendev.org/t/openstack/build/df8883869d6246749fc30ad88290fa90 : SUCCESS in 1h 19m 23s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/12d58b068792492196462ced7b3523b3 : SUCCESS in 1h 01m 29s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/f1d0aaf7e43047488fff543789539a6d : SUCCESS in 42m 54s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e91a6b353b0d45dc8d3ce3ded61d4d3c : SUCCESS in 1h 03m 43s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6231d2c2fedf449180dcf08d8041c86a : SUCCESS in 19m 31s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/74059ce991744cceabffc0c9f53306ae : SUCCESS in 21m 46s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/96bf43b27d5f48caab38d11d697d902a : SUCCESS in 27m 22s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/79df0764c9d94a2c91ba4064a9f89f10 : SUCCESS in 9m 42s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/5f4c95f47d534a3aa5183f5e326e962d : SUCCESS in 6m 52s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/e3aa35d2055343778fae8efb96b80dff : FAILURE in 17m 09s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/e9a1c2ea2fd945acbe74f4e1a73f9d9f : FAILURE in 19m 56s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/f22b06915cf8434e97c9a4426304b5b5 : FAILURE in 20m 58s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/89bbb06481aa4170b8a18b91d9353253 : FAILURE in 16m 51s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/13cdf4d6726342dab6f7af39ee6a83a9 : FAILURE in 14m 59s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/835e39b29ab8435ab2984f8c0b90104b : SUCCESS in 25m 02s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/3c1005e9ddd8413bb5ce28958cbe4506 : SUCCESS in 14m 03s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/97f3ecf893de452d991b6efdaacaae73 : SUCCESS in 20m 13s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/dd4d75f9e1ae42c78ec00af22ad9bd7c : SUCCESS in 15m 08s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/a5f0f61d03e44c84a824f441d0d14ebd : SUCCESS in 16m 34s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/62ad53349b984c84866db1ae335de816 : SUCCESS in 17m 52s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/12b04f4bd6ef4f8f87c399a566750d7f : SUCCESS in 9m 24s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/eddb82d1d9724e1180fcadd3eb0ebe55 : SUCCESS in 19m 45s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/a48b1c85964b468b8eca11e669c12206 : SUCCESS in 7m 50s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/1c1978f577e241d3977c35f1f58202d3 : FAILURE in 1h 01m 30s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/5027c5feba1f42f194cf3db47179b646 : FAILURE in 1h 00m 48s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/4a4a63bf4cbf4240b0b00e40c86b96ee : SUCCESS in 50m 29s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/3edd7b0d2740493c88076078d3dcff5a : SUCCESS in 50m 03s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0c63e3d5b6794c079a1009cb079c921a : FAILURE in 21m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1ad43927c39342a69f4f61db2b24620e : FAILURE in 4m 32s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/48bd37a252514470b30493229ba1160f : SUCCESS in 20m 30s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/73af8d282c5047bb9e77bc90a8657a88 : SUCCESS in 48m 07s","accounts_in_message":[],"_revision_number":1},{"id":"45bbb6ac6c0ae9c0cf20bd6fbb0d8302150e28d4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-11-08 18:00:31.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":2},{"id":"348e580b23e39a6eacba2e1f5a7455a0f6b8957c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-08 19:16:42.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1b01d29d4a9f4bc9833d61386490ab0d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/de1704c38497448d8986e61293bf6bab : SUCCESS in 8m 51s\n- grenade https://zuul.opendev.org/t/openstack/build/1977e83d39a843d0b1f4a093f52d5b3b : SUCCESS in 51m 46s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/f2cd4b5f69934b34a15d466b3c7c313f : SUCCESS in 48m 54s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/57e5eea11ddd4de285b8c9c626e1eef1 : SUCCESS in 55m 05s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5968169b0c3e46a199183efa5a1ed8b9 : SUCCESS in 43m 47s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/37cba54d981546f8b89f942b33bebd42 : SUCCESS in 17m 20s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/11c9d28e2712485c88a936ed0901247b : SUCCESS in 20m 44s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ba0da158835b46af9699d183056a8b19 : SUCCESS in 19m 22s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/51e853e292ab4142aaf9985934ace122 : SUCCESS in 8m 52s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/729b003aa01b4718b17a358d3228b9d8 : SUCCESS in 5m 51s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/4a9ca9f874c944308fd2582c30d9a382 : FAILURE in 21m 28s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/fa814ba329fd48c0b048e0f2e95dea80 : SUCCESS in 18m 21s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/8b98f70105bf42bbb995f1425e64550b : SUCCESS in 15m 40s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/f7a83db85af345f8a3870b23ec4b4ba4 : SUCCESS in 17m 44s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/c9ae762a9d9346008eff62e9132fb823 : SUCCESS in 17m 03s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/36c8b9519aa9446f9028c0882378bb9c : SUCCESS in 17m 17s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/d53fb56dc5a04d37aa336e8ac70ab518 : SUCCESS in 18m 15s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/f95d2e85b93348fea27fdd1888f73b93 : SUCCESS in 16m 59s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/28e8670b06ad4c4da937eac41ad53b6b : SUCCESS in 19m 58s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/d1b78c738e7845b7a8859b825cee1655 : SUCCESS in 14m 32s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/38fa08bb4ec448a0b0c8849ea7bee6ca : SUCCESS in 16m 53s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/28949293c4d544a7a227b37567b17fdc : SUCCESS in 8m 45s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/6cef96f4992f45388dd8f5f0a6252694 : SUCCESS in 13m 08s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/e75ad21b5ba94863901eec7020abd968 : SUCCESS in 7m 34s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/76aaeb7baaf24cdab32a74094bea0159 : FAILURE in 53m 23s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/48096931b6474244927f91fdba52620f : FAILURE in 1h 05m 13s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/fd655530a2eb45ad9c34915e483e895b : SUCCESS in 33m 06s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/ef3ebe2ec72c4726a6a53f21cb3e87a1 : SUCCESS in 35m 03s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4a0a9b842730409da1228deb13c5c620 : FAILURE in 18m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/31108b117452433aad6b0b661d37b88c : SUCCESS in 5m 02s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/d9b45171a402462abae2ed103a666d1d : SUCCESS in 20m 31s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/80fc7714b1cc403f96f3b768c289725f : SUCCESS in 45m 34s","accounts_in_message":[],"_revision_number":2},{"id":"031449c4c94318875664c183a50ffabf055cc89f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-08 19:21:38.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/212732eb05bf4ee1b6a11f29fb4e7529\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a7dbb13fa04b48a2b01b6d53d635e0a1 : SUCCESS in 1h 16m 56s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e473abc2740b4470a6d6cc1e4bb5009b : SUCCESS in 15m 44s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/a7d3190afd6840dd82bdda7c71d1d867 : SUCCESS in 1h 15m 51s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/5ab72426371e498686570bc9ef343847 : SUCCESS in 22m 21s","accounts_in_message":[],"_revision_number":2},{"id":"522e3eec2ee998a266ab6edbe6ce0628d42d131f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-11-09 16:45:01.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"1addbe9cbf1905c037f95819e03db5312ea3a23a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-09 18:02:10.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e960c324db2248fd95b4c3e84076b201\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f885a5e069494ecfb59bb3d711cc5e49 : SUCCESS in 9m 10s\n- grenade https://zuul.opendev.org/t/openstack/build/6fd3e80498eb49f29fe410dd7856cc7a : SUCCESS in 1h 08m 48s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/c81030db041e4e9581993382f17bf4ae : SUCCESS in 1h 02m 48s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/9478468c74af4fd7ae2dd05a0d2a3ea8 : SUCCESS in 39m 54s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b1de7c78a48f44a2b7ed3d965d1ef17e : SUCCESS in 57m 16s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b0f01570aac84b9f949061454f8ddd25 : SUCCESS in 22m 05s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/667dd90373f34c738165eae16c5e9efb : SUCCESS in 22m 14s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ba9630d4ceb34514982c531d0801c626 : SUCCESS in 19m 39s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/b6b242ecd2d94e1aaf60adab5e708723 : SUCCESS in 10m 14s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/b3e02835553c41818d68883546f2600e : SUCCESS in 5m 48s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/cb0e0712d0754aacbe8ac27dfff9e393 : SUCCESS in 20m 03s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/0afe12c9b5c9459dbd112449b4168af0 : SUCCESS in 18m 06s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/25c3c8cb2b5144128dd9c2c593ecfbb6 : SUCCESS in 15m 54s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/688b191ed9ce4202b3c915d40dfe030b : SUCCESS in 16m 27s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/1cc4cb40e3b14ad896e43422f8060c49 : SUCCESS in 18m 45s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/8c3f732101034f228d91792e89fd4af0 : SUCCESS in 18m 56s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/6db66f8bec6a45d880a7c6bbf6c855bc : SUCCESS in 16m 36s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/1bd7b6e4f8b64abeb80d73ec33aa8653 : SUCCESS in 20m 45s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/0d640203546d40db8342fb024fea8ca7 : SUCCESS in 15m 05s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/01b3e4d23136460f9b134b922aad15ac : SUCCESS in 20m 20s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/b443e8c8d098459a98d4109649ef16d7 : SUCCESS in 21m 27s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/80412b4facd34c0199d382328c2a92b4 : SUCCESS in 7m 28s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/a5885a7f53234a2390128eca8ffa22ec : SUCCESS in 20m 16s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/a668972e2df345ddad97ff95ee14f013 : SUCCESS in 7m 50s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/0ce6e14ce07f4e2ca203a5e0eb371b59 : SUCCESS in 1h 02m 38s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/2dbfee548fe44aa6ae6c7874e6a23b04 : SUCCESS in 1h 02m 21s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/ee910ff55f134ac3b007d56cda2e2b23 : SUCCESS in 55m 25s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/b917f9ab283a4e789401a5181f468ec5 : SUCCESS in 52m 55s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ecbdb6eea56f435389d0eda556a9267f : SUCCESS in 15m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6048cb1968414a4caeb1b2268b86f51b : SUCCESS in 4m 51s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/a310a8f3e59c483cb97b792902b7bfb5 : SUCCESS in 21m 21s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/df7f399fdabb4d2384cd775fb42e9361 : SUCCESS in 1h 03m 50s","accounts_in_message":[],"_revision_number":3},{"id":"702e5d5836c1c270c92ea6a9b62d32fa8b2081d5","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-11-09 18:12:54.000000000","message":"Patch Set 3:\n\n(12 comments)","accounts_in_message":[],"_revision_number":3},{"id":"fe8afcb57794ee1fea83bf62f6c15a1b9da1dbb9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-09 18:21:14.000000000","message":"Patch Set 3:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4b7829695c0f43689d91491574bead01\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/6541a448734b4f50bb0ea642353c5678 : TIMED_OUT in 1h 31m 04s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e17cb39fdd2e46fdba7a3ea5f99ee594 : SUCCESS in 18m 53s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/5f580b3a440a4eca9106a184a9677d06 : SUCCESS in 53m 59s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/8d161340bdae47b4a342a20358f179cb : SUCCESS in 21m 51s","accounts_in_message":[],"_revision_number":3},{"id":"856c93b63216874c0c1bbd3e26c9ba886d880b28","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-11-10 19:01:43.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"dbb86be9d2d22e9a4b8c3751daa71acdb9fa6f36","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2023-11-10 19:41:06.000000000","message":"Patch Set 3:\n\n(15 comments)","accounts_in_message":[],"_revision_number":3},{"id":"23a8750d21deca28bc65c91d23bfa8ff4dacb7ec","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-10 19:57:38.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/35523df2e811476e83b9cb6f2152590a\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/fad0cb558da24fd089e5d2bbf989c85c : SUCCESS in 47m 47s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d36f7c811df24b729d5d52853db301ce : SUCCESS in 49m 30s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/c9e4f923274a46b2a40fd632741fb651 : SUCCESS in 53m 43s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/e14656da8fb64b31a55ef0a80456e08c : SUCCESS in 46m 08s","accounts_in_message":[],"_revision_number":4},{"id":"ae370fb3f9c57c975cbce7efc062c9a9f3ce580f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-10 20:16:21.000000000","message":"Patch Set 4: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4e418596ee424fca8ccdbb56d2e6d78b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9f15961a93754ffeb1d7133540090688 : SUCCESS in 11m 46s\n- grenade https://zuul.opendev.org/t/openstack/build/45d52ff35a8d41e28f603c44f5a47729 : SUCCESS in 54m 41s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/240ec63339bc4ca8a8deeb5da2c332f8 : SUCCESS in 54m 15s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/11b974e3e6034e4fb118505cb9de23b7 : SUCCESS in 40m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/94d940fdf7a0420fb73bf0c329b1cb81 : SUCCESS in 55m 10s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d2244b0d70fd47de8a462ef5922fca47 : SUCCESS in 19m 20s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5c3fabe615c54fddb5eb75ff6bf7c27c : SUCCESS in 19m 08s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7848d10907bb45179cf9b589e69b6ca8 : SUCCESS in 18m 52s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/b86dffa8d8a7436f9e821b2a1130f125 : SUCCESS in 9m 09s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/24cda320b3d143fc94a410aa8498f30d : SUCCESS in 5m 20s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/d7391f73568e4c0381c693ceea938858 : FAILURE in 19m 38s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/356375c5e1dc423388c06babf38ee1ea : SUCCESS in 17m 06s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/f887bbe13f074856b140fb39b9082e76 : SUCCESS in 15m 02s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/7ffa2a7e354f4906b242449be67262e8 : SUCCESS in 14m 57s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/014937e710154709b84df442108f4b76 : SUCCESS in 16m 16s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/722d125c5f98489db6a5433adc2c417f : SUCCESS in 18m 32s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/89089b817a7c4a309caa011c41799141 : SUCCESS in 17m 17s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/3336d2089807461ca55fee31ac5faec1 : SUCCESS in 21m 49s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/242d373e1e3e484aafb6d12ae949032d : SUCCESS in 15m 30s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/85d448e6d3b843228f43efaa17a17e56 : SUCCESS in 16m 26s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/1df16b1deb304e80a2c0ef8f37217e05 : SUCCESS in 17m 00s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/0601ec722e574ef78f52ce9dc64d37fa : SUCCESS in 8m 05s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/77d7530205f34d919896caf876e42e9e : SUCCESS in 18m 31s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/99a3f5b1489049c5911c249d6a7ea048 : SUCCESS in 9m 10s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/decbbd3654784b1f86694572c113e390 : SUCCESS in 56m 50s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/def4a53a31854a9594ceca5f49779ef1 : SUCCESS in 58m 24s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/bc18260ea9334cb893a452d8d6e8940a : SUCCESS in 52m 32s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/414481ddc1934ee4a7515f6dc90b4afb : SUCCESS in 41m 08s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e96098ab447a4db6bf97d45054800bca : FAILURE in 19m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3533fa1fbc8849b799f55b2c10c976d3 : FAILURE in 4m 07s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/12e90767c7e4454baf667907a3440e30 : SUCCESS in 26m 05s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cf18154fb5834785befc7aad0b1adbce : SUCCESS in 38m 12s","accounts_in_message":[],"_revision_number":4},{"id":"965cae9a1dc2508b5055ffeabdf7b778c550b49d","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2023-11-13 05:45:10.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"b2b7e69ff27ef1c916ba35696f8e967ce02c7ccd","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-11-13 16:14:58.000000000","message":"Patch Set 4: Workflow-1\n\n(7 comments)","accounts_in_message":[],"_revision_number":4},{"id":"bd2602f4fd6edd48fda54c7d002d112c5195fed7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-11-13 19:18:38.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":5},{"id":"e53338e6fb0496659af9b9402f19ce1b484fea20","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-11-13 19:21:27.000000000","message":"Patch Set 5: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"516a752de9401f8d7eede23e80ab418d17f41172","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-13 20:29:54.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/6348e8826a5344de9edc420fd1c84590\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dd4ec9c850454ee3891cf58bb2c10033 : SUCCESS in 10m 23s\n- grenade https://zuul.opendev.org/t/openstack/build/bb438bd82dda4e51b2f78d9e2e1844bc : SUCCESS in 55m 22s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/37cb6396a3d94eb68e7b970d61a1cb63 : SUCCESS in 59m 00s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/b91bfc06ee824a2e9486b4a36ace21a0 : SUCCESS in 43m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/58ed61a02ae540d988df9f1c70599988 : SUCCESS in 1h 02m 39s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/989d17cc604349a0bb2232b2dbf9c6da : SUCCESS in 20m 56s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7ef9cbe8e49540c9a4b3514aa9a1741f : SUCCESS in 18m 34s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6fa6cd6501284f0f88799ba0fb555821 : SUCCESS in 30m 06s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/3e79c4601c2f4cd8b83c0ddf78e1a92a : SUCCESS in 9m 36s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/495e13cfba2544439dcb7acad6553d94 : SUCCESS in 5m 22s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/8e1365da43c44609ad57accce92b5546 : SUCCESS in 28m 05s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/724f36e0fd45484a9530b1ffe6c8b77b : SUCCESS in 15m 47s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/3805d0cba45045bea6a07de1c103bab7 : SUCCESS in 13m 37s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/c33c452a374143039ccf32cc6081a98a : SUCCESS in 19m 28s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/d9504b8902b4467a854ed074f24ca5ab : SUCCESS in 16m 16s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/33e78a583337484ea33de602f2095cf4 : SUCCESS in 17m 38s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/0573519833bc4db994d8ed1755511057 : SUCCESS in 20m 56s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/cff3581a42b84274a6810986629859a9 : SUCCESS in 18m 07s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/e24d9b5f830442cb94bc192067adeccc : SUCCESS in 17m 12s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/a0a4efc15de4455694f143bc35561f63 : SUCCESS in 16m 37s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/d8ec9389633140a3b055db2b8e5c239a : SUCCESS in 15m 48s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/706819784b6b46fda8d26f665154a5ba : SUCCESS in 6m 49s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/5693682dfa27440db919213716609dec : SUCCESS in 16m 55s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/6f4643f96b084cf9829c02da1a73a37a : SUCCESS in 8m 52s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/583198a2533845608a20bccc8bb47172 : FAILURE in 45m 49s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/e7f27e905e1e44e7a63be9585913b024 : FAILURE in 1h 03m 28s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/28ae8218e52d425cb0e11829cd3d2d68 : SUCCESS in 54m 31s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/be0dfef90292470ebee7a3138e0e1072 : SUCCESS in 52m 56s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/57a7c020239d4c649be101649a6cb519 : SUCCESS in 16m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/85bfa4215993496da0bc03cbb2122ed4 : SUCCESS in 4m 31s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/a58794919919447bb6acab33d1ad1c2f : SUCCESS in 21m 03s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2b41cc7ffcf24580bfa8809a789220d5 : SUCCESS in 50m 34s","accounts_in_message":[],"_revision_number":5},{"id":"9dfc9323506974508268069063463fc003823946","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-13 20:34:14.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f78c95308ab845138c1c2d4633e8eb53\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/522b99951a1a4dcfb96574295ee034b9 : SUCCESS in 16m 48s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5f2a115241754135905b7507ec4668fd : SUCCESS in 56m 28s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/8c85b923e177445181796aa672b06f73 : SUCCESS in 1h 12m 53s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/b7c737f0e10c4b908c5429222c1bcb41 : SUCCESS in 1h 02m 22s","accounts_in_message":[],"_revision_number":5},{"id":"ba206970edf6b09a2789b15b0e820de547ff1f5d","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2023-11-14 04:05:42.000000000","message":"Patch Set 5: Code-Review-1\n\n(9 comments)","accounts_in_message":[],"_revision_number":5},{"id":"a4a09d159688a643870560115801f0dbe9739f03","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2023-11-14 04:23:51.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":6},{"id":"6793cf982b6494a24a3a0afe2c13a3ad5c2c6e49","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2023-11-14 04:25:22.000000000","message":"Patch Set 6: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"d855f2379f2bdf9d61b158350d6f3c334a1ea210","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-14 05:26:29.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9075e800d1f040cb81affd910b038c3e\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ebb31401cfe74125b058bd3578d729ce : SUCCESS in 1h 00m 14s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e7f084f4050f4fa9a14cd9729bc5c6be : SUCCESS in 21m 08s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/b80b7f8d6ef64f30b89b6d71eb77252e : SUCCESS in 57m 47s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/32d40c013ea64f60b42ce9d0cf233948 : SUCCESS in 21m 37s","accounts_in_message":[],"_revision_number":6},{"id":"797025ca1bca4860537bde30f5a773aaaac73e2e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-14 05:39:48.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/215ca0dbd48c4d80b77107e951a4be12\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4d73e04be6f749328b6960cd1f1ea542 : SUCCESS in 8m 12s\n- grenade https://zuul.opendev.org/t/openstack/build/5153a0d7e1ce4dd59c102da565bf15df : SUCCESS in 46m 56s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/db7dda0be73848bdb210fffb151e6cdc : SUCCESS in 1h 07m 53s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/3b4a17dc611b4d02ada876f53817fb78 : SUCCESS in 35m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9526df25432949e3970a81c1143b9795 : SUCCESS in 1h 02m 00s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/05859e745a384b2db3bd388ab2fcc0ab : SUCCESS in 20m 56s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/afda376b0fa64ec192fb0a1ac129521e : SUCCESS in 19m 06s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/09706ba78e264cb5a5ab7cde5f752d06 : SUCCESS in 21m 11s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/d924a886b46c4337a1691d362c0e9ce9 : SUCCESS in 8m 29s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/c024effa5c184e22b86a3d4907d51cda : SUCCESS in 6m 44s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/230b4c32830f48819efa7536143b5c96 : SUCCESS in 15m 12s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/64f43063e59e45079a7bb444f4c50dd0 : SUCCESS in 15m 09s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/426423aa30b54689a703db6042eb5c57 : SUCCESS in 13m 55s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/40b241ed93b44ba0b7646a63318926fd : SUCCESS in 16m 07s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/9078fdbdbc314ced9c323da23b160f5b : SUCCESS in 15m 08s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/dc8026bbb7354e078caae9b6013a5fdf : SUCCESS in 18m 22s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/c662f8c8aac448d1bd1e154a70c9f4eb : SUCCESS in 19m 56s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/e78940c7c3084f7ab85da2151c5b551b : SUCCESS in 17m 19s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/01a733f49f81421593511a3aab98db71 : SUCCESS in 14m 22s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/dd2df55311ef4db1800d91af1aa2f218 : SUCCESS in 19m 02s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/d3c74b8ce73d4417b9dd6c241ce8938c : SUCCESS in 16m 01s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/3cc291b090cd4cdaa79b0c5a8ec7d3a6 : SUCCESS in 6m 36s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/00d27ba8886244a68e4e56dcf7d760a6 : SUCCESS in 17m 21s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/430c2c736bab4fe5af9cebe84c716e26 : SUCCESS in 8m 24s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/27d36255920c4f03990a65dc8fbb307d : FAILURE in 55m 06s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/3a2d62f0425746828d302368b3d513a8 : FAILURE in 55m 28s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/3fd910c6513047f08a7adc3a1af8fe87 : SUCCESS in 52m 23s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/979306733c544b718335cae6f0236880 : SUCCESS in 1h 01m 04s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/869636b6c52f457fa3226d0855379c43 : SUCCESS in 19m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3ffa9d5f9f8147ac92180fac7cae4c49 : SUCCESS in 5m 25s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/cb8ca911968c4bf8b2dff8c5261aaa23 : SUCCESS in 16m 46s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/94992cd81c99487abca7ab2e31b2c26e : SUCCESS in 42m 27s","accounts_in_message":[],"_revision_number":6},{"id":"1e79a831f87e7d90b084a91db2ec05883be62c43","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-11-14 18:42:53.000000000","message":"Patch Set 6:\n\n(15 comments)","accounts_in_message":[],"_revision_number":6},{"id":"064b81658b623ca1fa9262fbbb7a854953e158e9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-11-14 18:46:54.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"fe44d51689bca6fa2730c20128f6c437508b8851","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-11-14 18:47:11.000000000","message":"Patch Set 7: Workflow-1","accounts_in_message":[],"_revision_number":7},{"id":"4702f704a43123ce5c8993dd6928eabdab289ad4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-14 20:04:25.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f3eeabdc8a16412ab8e9c9a3a605e90c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/79500f1eaa9046fd94cbc949580dddf3 : SUCCESS in 7m 50s\n- grenade https://zuul.opendev.org/t/openstack/build/9959c664a5784bdeae8fe70cdd39b98c : SUCCESS in 1h 02m 36s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/f489e4cee583426ea2830622f5dcc711 : SUCCESS in 51m 46s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/ae7fb8badd454754aa0cfe5c51fb652d : SUCCESS in 46m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/529d4bdc836e4ab0b86cd5e6b13198ba : SUCCESS in 51m 44s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0960f187359748e2bd508a89eed47463 : SUCCESS in 27m 56s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/387afed134bc49daabdea6a6185e922c : SUCCESS in 30m 33s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a153217ae304406b86d3f12118c72c4e : SUCCESS in 16m 22s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/bbc51631ac784cdb9984a564b5cc9a9c : SUCCESS in 9m 11s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/f7f1fc9ff28c4364b3ee89a7741677b3 : SUCCESS in 4m 58s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/eb4b1a02ab164febb28f1367fdcc6eb0 : SUCCESS in 15m 36s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/b4b2ae9c318e456bb4882f3d6ceeb9bf : SUCCESS in 17m 18s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/331d7d67b3b54338ac3dd383124b7e5c : SUCCESS in 17m 19s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/1dfa7fb190934b009b276b3c1181a092 : SUCCESS in 17m 55s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/45352a3177cb437097cfb0eea51e9c6a : SUCCESS in 17m 13s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/e6cc1f1982a44dd189c145c49c2fbc87 : SUCCESS in 19m 36s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/57dfef5991224b98a3947cc7ca02bc1c : SUCCESS in 14m 49s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/ac40bb557a604d9d9dab8416cc4f3441 : SUCCESS in 19m 04s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/d443de8e121d4b409f76f56d954d183c : SUCCESS in 16m 52s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/19ce31aac02f47e48e15413234a51d43 : SUCCESS in 16m 07s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/7f327fcc4e6641cab441364d5efef07d : SUCCESS in 16m 53s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/9f254efebf144248bf44360614d57ce7 : SUCCESS in 7m 23s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/112f3df36ad445c29006bc815148cf83 : SUCCESS in 18m 44s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/9b83cead2dd0436bb7726581d52d1648 : SUCCESS in 8m 11s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/280f752e639048698ec6293968bd62d3 : SUCCESS in 57m 34s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/5ab8a1a15cef4d63a987a8035c19c0ef : SUCCESS in 1h 06m 48s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/6c7a50dfe1c44515a0b7ed9117465ba4 : SUCCESS in 46m 32s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/2915f83045b340519a1d4c25f66f2f0b : SUCCESS in 53m 24s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/125b66b7053f4150882fac67f1ec8ee4 : SUCCESS in 29m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/99de9f0fb34a4337bbda5a8993440f89 : SUCCESS in 5m 49s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/3864ce27657e443e92f54b77853fb521 : SUCCESS in 19m 17s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c138cc32ca364644994b2e250ce6e63b : SUCCESS in 1h 03m 30s","accounts_in_message":[],"_revision_number":7},{"id":"0e194471e8899088c8f5bc060726db31230a61dd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-14 20:20:48.000000000","message":"Patch Set 7:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/99cac623bb1f463da0b1a7a1891fb11a\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ac57d07af9fe47a88f69392fcb8165ca : SUCCESS in 1h 30m 38s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8565cfa388a9432db9cdb078c5e3cefe : SUCCESS in 1h 21m 05s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/f60f5198ca484231a28ffe6e27f27928 : SUCCESS in 22m 40s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/17868decc29944cb8acf6a5d618de144 : TIMED_OUT in 1h 31m 19s","accounts_in_message":[],"_revision_number":7},{"id":"88a96c372244340b1dab004f071bd69b063c06d3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2023-11-15 06:20:52.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Verified+1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":8},{"id":"40d7a8886bad234d2e5fd555ae35ec825f8440cb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-15 07:37:15.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/926f89d5392143d79ed61ed3e01747a1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5ca07e53052d441eadf503a0ce48f71a : SUCCESS in 7m 39s\n- grenade https://zuul.opendev.org/t/openstack/build/094bb9df1e6a4657b6c0562f3d39aac5 : SUCCESS in 41m 31s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/ffee971181c64e5caee1075c4fcac1e7 : SUCCESS in 1h 01m 50s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/003637716b7c4f55a60cfa9ad938dea2 : SUCCESS in 55m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/330c90e3c9964b96b17d21fd85d821e6 : SUCCESS in 41m 34s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/60dc03ce0506466f8215036471965e4d : SUCCESS in 16m 12s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/89d4705392e5443888963845558f6153 : SUCCESS in 15m 28s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0bb1b9d3e6664534837d88035f779fda : SUCCESS in 16m 26s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/e28deece1d014ea8b9bbee1dc731dc17 : SUCCESS in 9m 19s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/8b34b342351f46bb958968474aece33e : SUCCESS in 5m 35s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/eccf38c3f65e4764ba6ab2521f2296b7 : SUCCESS in 24m 49s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/2248a3bd8f394780971b13503e2681be : SUCCESS in 15m 24s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/6105da57bb374b11ae46a68ccf5560d7 : SUCCESS in 24m 22s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/ccf34e411e37431a84fe38ee065625af : SUCCESS in 17m 41s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/a8c60a3cd2534efc825aa3de2071260b : SUCCESS in 24m 09s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/90fef17fccc1432eaf886e311f3edff2 : SUCCESS in 17m 03s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/8054056b74d549038b02978b8f6d7ccd : SUCCESS in 25m 13s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/ece88afcec7740dda6fb93fe30b836da : SUCCESS in 26m 03s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/bad169c8e8a044dd9b1286c989945545 : SUCCESS in 15m 59s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/76b1ec9836f84b058a5a813c16e63384 : SUCCESS in 15m 48s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/2ed7d00b62c04f67be40b88d609e7d57 : SUCCESS in 16m 38s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/497a077f556347f698e5584675b8e97b : SUCCESS in 6m 58s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/b2196fae1b48411fa99f405815218809 : SUCCESS in 23m 00s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/221d9fba8ff44200abbce89909bfe7c6 : SUCCESS in 8m 02s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/a8182faae08a405db849ae1c0c0c5618 : SUCCESS in 1h 02m 58s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/bb087e8a052946709af56d65cb76e648 : SUCCESS in 1h 07m 04s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/3faf67e7b17b46eba8d5bedf385ecf30 : SUCCESS in 50m 28s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/22dce070184d46798c797b104c1bd368 : SUCCESS in 51m 05s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b0dc652c93bc44a1921033bb7e38a6b5 : SUCCESS in 14m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0ec208a22afb4bcd874d548c5568e668 : SUCCESS in 4m 34s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/9ec7cea0a124430eaa8e4bfd782883ba : SUCCESS in 18m 06s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/727ede7890ee4dd584b1ac29949740fa : SUCCESS in 1h 08m 26s","accounts_in_message":[],"_revision_number":8},{"id":"00b4d82bea2ef7e8a20ac37b1bb7864c2dc922a6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-15 07:38:19.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/865b715e234048ab94ced8f72893d660\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e499f125cbf142e3847fa81590722900 : SUCCESS in 1h 15m 27s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/92e98fbed1284038bae2f1b10e446226 : SUCCESS in 18m 03s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/4b38c6bebffb4454bcbf1364babbb32d : SUCCESS in 1h 02m 31s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/2e3433ff7ea94b4b8556b53d81cef45d : SUCCESS in 22m 44s","accounts_in_message":[],"_revision_number":8},{"id":"b6529ddfcece3bbbc5fdf4e80fe675d7b880bafd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-11-16 12:02:33.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":9},{"id":"0079df29ba27e06432b188eb4b7cce4a0bb73e36","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-16 12:27:50.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/750c7f151a994bb988e84999b986d5b5\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1f4a434f687446e28334783449a75d29 : SUCCESS in 14m 27s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/59fee32a39cc4b5988cee372423234c5 : SUCCESS in 16m 20s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/ef2efd2eb1ff458c94a7b1f27c748df3 : SUCCESS in 19m 30s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/feb54bb6138041eeac842be00abc4cd4 : SUCCESS in 24m 02s","accounts_in_message":[],"_revision_number":9},{"id":"0a06f83842f59d5bea0b88ae7ac0f2eda12a637e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-16 13:15:47.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/991b262e3b394c15a090fb0f31556354\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e8f0a8ba002644ddb4fb8f14b4e98247 : SUCCESS in 7m 19s\n- grenade https://zuul.opendev.org/t/openstack/build/282148ede49b4a7db2c7ba3bbb6cff6b : SUCCESS in 1h 10m 36s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/0f1b9cffdf354a478034592b97c91094 : SUCCESS in 1h 06m 46s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/218e7b42b3d8478f807135a25d62ead6 : SUCCESS in 1h 00m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/eb6ede7721e34286b1be8e2d00f3418f : SUCCESS in 1h 06m 17s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0ac1bdfbca0f4e10be3b18f57767b65b : SUCCESS in 18m 28s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0d477bd06b1e4add80ac4dcb27b5e836 : SUCCESS in 19m 37s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5e10ae344dde4e5c9d99fc131d8c0cf4 : SUCCESS in 20m 48s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/bdae74b87cc94077b6ed3007eb874e54 : SUCCESS in 7m 27s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/15688c6133ae499ab18c95c1b2ead04e : SUCCESS in 5m 37s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/18f45ddab16c4207b242f635e7d1403a : SUCCESS in 20m 24s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/56a268906a074e5a93f849fa5e48f5ed : SUCCESS in 16m 15s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/bd5afd86503243979adc53e191077ea8 : SUCCESS in 15m 02s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/ebe58726f2b843448028f4547f7a681b : SUCCESS in 16m 14s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/3f881006fb62417e97c05a348862310a : SUCCESS in 16m 52s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/ffd1666843c44b058a5550d1a203c540 : SUCCESS in 19m 51s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/656aed9a51a14b6285624bd5811d1355 : SUCCESS in 17m 37s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/a71bb80d9e43464089a888d50ed64ae5 : SUCCESS in 18m 17s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/7e8779a398624e8ca8e8cb6a76c68df0 : SUCCESS in 16m 02s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/8012d14da73f485cb502ed02288e4574 : SUCCESS in 18m 22s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/b30fcf02aff843d9814dd21c0369fc5c : SUCCESS in 16m 30s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/14cc0554c49647b2b2a722a95868fd43 : SUCCESS in 7m 02s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/0e7e64ca05e9456ba7bf8b5c5ff38095 : SUCCESS in 17m 29s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/4820b70feea04b6c996308fa22d03e3b : SUCCESS in 7m 59s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/5eb9c3473a044f479d523b8b89d7f19d : SUCCESS in 51m 52s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/c6c6b854dc224a94b955a48640433c15 : SUCCESS in 56m 54s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/e4c4e77a6f134dbaa5fa361978f401fb : SUCCESS in 54m 16s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/f41d8a1fc4c24d5d84edd2abb3671db4 : SUCCESS in 43m 05s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/45caaf274c8141a68a1b913e222bad9f : SUCCESS in 18m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c6b288c6d943418b90f336d1bad9849e : SUCCESS in 4m 26s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/6fdcbe04424745cd999c43df5658bf19 : SUCCESS in 21m 08s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c6c4f0b7d09442069f482fa39454343e : SUCCESS in 1h 02m 44s","accounts_in_message":[],"_revision_number":9},{"id":"b7f9ebd72672668fde3117a12f28303b45a985c1","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2023-11-16 22:41:19.000000000","message":"Patch Set 9:\n\n(34 comments)","accounts_in_message":[],"_revision_number":9},{"id":"8a07dc2d7c5b21f900c7204ff9b1dfe6eaffadf1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-11-17 19:01:13.000000000","message":"Uploaded patch set 10.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":10},{"id":"6196bc238a39173104cf5649d47bbada01672d03","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-11-17 19:14:46.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"fa1a6041b5505459e4f67a6473bb020239823432","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-11-17 19:20:30.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"f8e444abc5872fb4ab6da4db0eb9ddece5f68dbe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-17 20:25:04.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/99a736d57e34421f871551da82bb22ba\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/925762a8664541d6a331dc000808f05a : SUCCESS in 8m 59s\n- grenade https://zuul.opendev.org/t/openstack/build/8b2d39cab2f94c79b1b239d8f34d522e : SUCCESS in 41m 53s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/36d7f36639a44c81806ffb1c5ee9c225 : FAILURE in 53m 09s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/da31743cb4ef45e49b61b51b89bc011c : SUCCESS in 52m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9695ff7f2f2b46d6917bdc5ce2216a6f : SUCCESS in 1h 02m 36s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8163eb77c2a8481395309df170e90a17 : SUCCESS in 19m 17s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9df0b21815174f9186c3c6fb5afa832e : SUCCESS in 23m 13s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1cd4211c14384c5fafd5225afa4dd2e5 : SUCCESS in 19m 48s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/f277100123f541d08b68cd28ac23814d : SUCCESS in 9m 54s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/20c6938d716144abb9a4134404697bdd : SUCCESS in 5m 37s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/f41a4b5874c94f1da82cacfa8d5f232d : SUCCESS in 18m 27s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/30722e656aed4eacb7250e80e7095b95 : SUCCESS in 17m 43s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/10ecaab607dc49a791eee8c2823f6d8e : SUCCESS in 15m 13s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/89298c44a73d40599a3cdd6f3b438b25 : SUCCESS in 15m 47s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/fc95b90c6ec84eb6a7237b2b309e4c99 : SUCCESS in 19m 23s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/0d114b7d1a54404fad1f10f07f1c8311 : SUCCESS in 17m 47s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/9fb3e601e13844edafb3f3e29c540c50 : SUCCESS in 21m 53s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/c820f8f4bd2d49f2a233637e98dd1ce5 : SUCCESS in 21m 11s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/479975a8fec546459a5812843c02d5fe : SUCCESS in 15m 37s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/ebf241d6da404cc2b063556c5849ec92 : SUCCESS in 15m 44s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/24ef18b937594dafab66a77f29d3b147 : SUCCESS in 15m 58s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/d8f13ef11a35433792a466e83525c245 : SUCCESS in 7m 08s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/c68ae135f9d64b47950eda589e2aee2d : SUCCESS in 14m 20s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/b83d92c5a28d40f4bc69785f79e944db : SUCCESS in 7m 53s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/a21ec37ebdb249a5ab3c8b0ecec1e603 : SUCCESS in 43m 47s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/a40b2b3066dc46fa9a30c0cb8bca58e9 : SUCCESS in 1h 03m 55s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/33c4881c691c4dadac6e210a4bdcacf7 : SUCCESS in 42m 00s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/2edb6f2f8cd447038be9035b9db3a66b : SUCCESS in 56m 39s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9910d4273b4749ce98d4b855a7e5d8bf : SUCCESS in 16m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/644b769a25c44da3a37e6a1610a09141 : SUCCESS in 5m 37s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/adcb715ea6b744bb913d701fcbe69444 : SUCCESS in 21m 14s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7deea835b2b143dc879126ae049775fe : SUCCESS in 42m 14s","accounts_in_message":[],"_revision_number":11},{"id":"a840ffbbdce3032697859857ff724ab6e50bda22","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-17 20:49:20.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/74905d70175e41d1be6655874b2ea39a\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/6197bda478de433ba6d4fb05f07e703b : SUCCESS in 1h 04m 21s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7edf2aa0eddc44d1a89269a9c53babc3 : SUCCESS in 1h 30m 20s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/8977bc41c93c4e8f91a286f072b9defb : SUCCESS in 31m 40s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/47a64b8c27f54533b449d15f312cd425 : SUCCESS in 1h 07m 13s","accounts_in_message":[],"_revision_number":11},{"id":"0b82552f6a0f3020213e64097cd28b5cdfeaf555","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2023-11-19 00:57:26.000000000","message":"Patch Set 11:\n\n(10 comments)","accounts_in_message":[],"_revision_number":11},{"id":"9407367ca3abf0d598aa255fdfb42bdc7f1523ec","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-11-20 14:38:32.000000000","message":"Patch Set 11:\n\n(28 comments)","accounts_in_message":[],"_revision_number":11},{"id":"eac2c0a866f10723334ad06c0c42aa6296d1b18a","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-11-20 14:47:05.000000000","message":"Patch Set 11:\n\n(5 comments)","accounts_in_message":[],"_revision_number":11},{"id":"bd6a0a42e3b5eff7515813b72297503c43b05820","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-11-20 16:47:37.000000000","message":"Uploaded patch set 12.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":12},{"id":"8f71a03c4f145016cbd227c9b33bd06eacf18405","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-20 18:12:04.000000000","message":"Patch Set 12:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8c9f2deb8e00439d9eca98cde86bd25f\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/648671162dd241eb87a7a052bfc7d9e3 : SUCCESS in 18m 42s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/46c952f2470745559f309e21717e32ab : SUCCESS in 1h 22m 06s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/1e2cabc407884bd5938bb14ebe421001 : SUCCESS in 23m 53s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/853c700b264543b8aa6d7c8d77311521 : SUCCESS in 43m 49s","accounts_in_message":[],"_revision_number":12},{"id":"a98c7bd1308d6104b6f076e84d9a6e855366c7ee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-20 18:19:09.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cdfff3fd5a6b4515afa9375c039b2834\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1edd3c5a39ac4470963a0bf557770a98 : SUCCESS in 8m 25s\n- grenade https://zuul.opendev.org/t/openstack/build/f1e73ba64db5480b918f6e45654553b1 : SUCCESS in 54m 39s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/18573d165cdd48b79089ca170ce49226 : SUCCESS in 45m 46s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/a56dde2cfe1649e6af1cb4a460be61a1 : SUCCESS in 43m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5e2dae1821d24935b892a816c8f296c7 : SUCCESS in 57m 34s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8bf7bd4966df4aa0bffffe6ddffee608 : SUCCESS in 19m 12s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/220806a7ca02439fac3a8954363a3763 : SUCCESS in 20m 41s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/41fd4989954641e29c31e13e5e91bbd8 : SUCCESS in 19m 11s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/938538364cd440f2b8f926e9d8a4a266 : SUCCESS in 9m 14s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/0cfe165154d54be8a50128e090d6263c : SUCCESS in 5m 21s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/86fef9846c9e4fcd89f4b1a76eb05d89 : SUCCESS in 17m 43s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/bbe8742cfbba434cbedc60e84df4128c : SUCCESS in 17m 49s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/f33a7fc7059048f0add5870da503c494 : SUCCESS in 16m 41s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/ae621e883f0a4443af7f858aef45410c : SUCCESS in 15m 46s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/341997b9a599411288f0eb2ef4f997b0 : SUCCESS in 15m 48s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/8ea667cb1de54a96bb98613f824812c6 : SUCCESS in 17m 18s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/0d7a28a568db478ab7b35b1aee690f61 : SUCCESS in 17m 22s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/a9245e1c7dc0444aa02a498e9b9d0540 : SUCCESS in 17m 44s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/71106592a84a48029951ac53e686baa4 : SUCCESS in 15m 46s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/13db063bb8c4401aaef7ed9fe6830dc8 : SUCCESS in 16m 45s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/dd1cb25010fc4bbbadf64aa1650fb32b : SUCCESS in 17m 23s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/c2c98349477742dabc1c9c3319aeb0de : SUCCESS in 7m 19s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/ae061f3ca08a444d9a3bb90bd06928f2 : SUCCESS in 18m 11s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/8959c3cd6f8b4851a9900daa0efb0b30 : SUCCESS in 7m 55s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/d1f8fb02d491457bb1d9bd6084fd9920 : SUCCESS in 1h 13m 48s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/75e2b0cf0e774feaa34f585792d2a6ca : SUCCESS in 1h 23m 14s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/aafcf482116143a58e167c1814a8d80f : SUCCESS in 56m 21s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/41a88fe0c881470fab8eaaeb848835e3 : SUCCESS in 52m 58s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1ef4e20a5a584d3a9ffe497a4e5be6d2 : SUCCESS in 18m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eb3669d20bc141d9a6ea0cd0afa7f0b4 : SUCCESS in 6m 33s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/cfcf1566497044ce81f12364aa4e09f7 : SUCCESS in 20m 48s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e177082052614c129282120d420b5b63 : SUCCESS in 46m 16s","accounts_in_message":[],"_revision_number":12},{"id":"b1acf30be65a527f033878c4c67d5379102da523","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-11-20 18:37:10.000000000","message":"Uploaded patch set 13.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":13},{"id":"028a56aa3a024f4878bd78568c24f8b959cf2785","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-20 19:57:21.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8635d6df96924f71b5e73ecfc25fed67\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7115805dc2c749deac45d2d525458164 : SUCCESS in 10m 55s\n- grenade https://zuul.opendev.org/t/openstack/build/21cd716c56f342fb9a75ff2a9bef163d : SUCCESS in 1h 11m 45s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/2a0974b93ae84c9c97d6825b79baa2ce : SUCCESS in 1h 03m 52s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/5410fc51d0f4436aa13a10b90fbb6bbb : SUCCESS in 1h 00m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/45b8bddb617b47f1a854bac1f489ddd7 : SUCCESS in 1h 06m 44s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/92844c8671434091ae4004dc8de63608 : SUCCESS in 19m 47s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/127a7d94b89445ac9bf85514d3c1901d : SUCCESS in 21m 00s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ec68e1f0aa7d44b38a0a358f83de7715 : SUCCESS in 21m 00s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/75412c88302545febb49528757a81f8d : SUCCESS in 9m 28s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/b0d7fdbdae3d4e858b3d32faede7ed3f : SUCCESS in 6m 04s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/8f38855b10394cdaa8ff49b054a347f9 : SUCCESS in 19m 33s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/e77276c8208540e69ecb030d4e0c60c9 : SUCCESS in 22m 16s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/5dfdbc6218554312a5d0e18dd4880dfc : SUCCESS in 14m 17s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/424ca14e6e3a4ff2b336b01dc6321d8a : SUCCESS in 17m 40s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/e15eb76496fe4b15931d170d938e33ba : SUCCESS in 15m 05s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/a0914f959f0e4fd8b478ff811bf049c0 : SUCCESS in 17m 53s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/049c14642d154fbaac0e9197af3a8dbe : SUCCESS in 19m 48s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/1b6179cc9d644dfab1cb35d31d938dca : SUCCESS in 16m 07s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/0f45aef8e1d541f48a0fa434bc9f2ae7 : SUCCESS in 15m 34s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/65e57d4a03e443c482a69320a3f5d57f : SUCCESS in 17m 51s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/bb8dc3a9eba24ac29afb3a1e30c8e0f4 : SUCCESS in 19m 15s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/fc1abf64b937469e9e31df2a1977c491 : SUCCESS in 9m 57s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/301481d019564ac590277365f5d817f8 : SUCCESS in 15m 35s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/70f2bb1dbaa54bf39c178e42621fd7b6 : SUCCESS in 8m 44s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/241370376a9049448bd1fb4dda29c697 : SUCCESS in 57m 37s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/3e64c621243f4aa5bb86dd4e18461a79 : FAILURE in 1h 07m 59s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/5447f72ec66a49f9b476f9bba3e5a3fe : SUCCESS in 1h 03m 25s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/366f6a0e095048c5bc5b8b76f958b489 : SUCCESS in 52m 13s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e4e247b55f3a417ca008f099707d4d95 : SUCCESS in 22m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c3ec4b42a1744614872159dc1a34fa49 : SUCCESS in 5m 38s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/6dbd6f4fc839423287886482be2ecaf1 : SUCCESS in 23m 46s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5b349c3937464b11915f7b0d6fb2105d : SUCCESS in 1h 10m 33s","accounts_in_message":[],"_revision_number":13},{"id":"a8cf832a8494a10186ddeda066bd5f3942f711ed","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-20 19:59:02.000000000","message":"Patch Set 13:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e0586bf667b3491c8da1eb3366c5c419\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c62ae38857104e37b8ca3735c64a6740 : SUCCESS in 1h 16m 36s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b202282055de4d6b90cf734409b90db9 : SUCCESS in 16m 58s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/aaa64cf7659e4bf9924ce2d107de17be : SUCCESS in 1h 19m 10s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/646d7bc7da9c4fafa07e5784da62f8fd : SUCCESS in 21m 54s","accounts_in_message":[],"_revision_number":13},{"id":"e877fd8369fb354b890544f56888bf141c4d8d01","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-11-21 10:54:11.000000000","message":"Uploaded patch set 14: Patch Set 13 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":14},{"id":"5be90a848689cde50d5b1f703501bc421f57f799","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-21 12:14:27.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/03f01542f23f46899d83cfd5e03aa4da\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/25dee0d62315493c9a51468c5c816579 : SUCCESS in 9m 23s\n- grenade https://zuul.opendev.org/t/openstack/build/d7082e463aee4d609e2445195ed3d178 : SUCCESS in 1h 12m 57s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/8662d482bfaa4d74b3169c6d420a96d5 : SUCCESS in 1h 00m 29s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/046f59f5fcdf40e6b7f4b4b2db2245ba : SUCCESS in 54m 04s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fe161cea11d343778ea4f33c3dc18053 : SUCCESS in 1h 06m 49s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/aef2546adce7424c8a521c60fc798d75 : SUCCESS in 20m 54s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/83efa3944b2e4452bcdc55cb5a4ce900 : SUCCESS in 20m 01s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b13c308e796f4fdc88066830724c7283 : SUCCESS in 20m 28s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/1cf354ce78b341ad8039e10082a2269a : SUCCESS in 8m 41s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/4465411619e04ee883c293f1c628cc10 : SUCCESS in 5m 48s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/6edcc81374594269bbebe302a6321427 : SUCCESS in 17m 39s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/3b7eccee516547b898b692c6c4610b2b : SUCCESS in 17m 11s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/1ecb599ff3cd4ad1bfcfe82555656e30 : SUCCESS in 17m 04s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/5f2043eeed274da5999723dc3ec2407e : SUCCESS in 16m 35s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/382cf49a49c844648036fd2dc6654c32 : SUCCESS in 15m 59s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/424a0f0876ae437d91fd6f76cb9b120d : SUCCESS in 17m 17s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/c225509e2f6b4043838e0675563ed7f1 : SUCCESS in 18m 21s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/009c82b7547c445c81b9522756bcb97f : SUCCESS in 20m 53s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/a65e086942104db59cdf292f25e98db2 : SUCCESS in 15m 33s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/3c592ff9b75041b98c477b0422b4ee68 : SUCCESS in 17m 51s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/d5b086f0174b48efb3e61ceacb8c2843 : SUCCESS in 16m 22s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/a2f3a80d5c894ecba777ba7142f330f7 : SUCCESS in 7m 13s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/7568e4625eca47e69f871572eca248f8 : SUCCESS in 17m 12s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/50211a734d624b908922e62d479bd74d : SUCCESS in 8m 16s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/d8800519deca44249ffaa30f0fc41ba1 : SUCCESS in 57m 36s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/d7543dc9abdc4390a873c949783ebcec : SUCCESS in 1h 02m 08s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/c1e5d0aa6af241aab007d8a580006f47 : SUCCESS in 59m 45s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/dbb96057cee9468fa0281e8ba28455a4 : SUCCESS in 1h 01m 33s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0c19cd263233461584f7f18a6faf7b90 : SUCCESS in 19m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fb26bc70504e484facbbc2682b9952bc : SUCCESS in 4m 56s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/3169f797611f46d99434aa3cd7065b2f : SUCCESS in 21m 09s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8c2c5a78f99343dfb1fd8757c37364d4 : SUCCESS in 1h 06m 28s","accounts_in_message":[],"_revision_number":14},{"id":"4715291e7e1776292958efd933bd1210262d59d2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-21 12:28:00.000000000","message":"Patch Set 14:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0699dc23c4f54a33b50f4ed887254aff\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ccdbc6ad297347ba9a72799eb2a008f1 : SUCCESS in 19m 22s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/563ee2fcd31145c8a4403d7b4a3648a5 : TIMED_OUT in 1h 31m 02s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/e0195e7018734a76a64f365affecde37 : SUCCESS in 23m 08s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/9e1bcac4fdbf473eb1b123f5f0bea5e2 : SUCCESS in 1h 22m 48s","accounts_in_message":[],"_revision_number":14},{"id":"77f36656895fee2aa9cc88b35c0a50e650f93a2b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-11-21 18:37:29.000000000","message":"Uploaded patch set 15.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":15},{"id":"3ba8bf66f794c131815ae097656c5a46607a8c78","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-21 20:05:03.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/65c94578cb23486787ccc11a15cf135a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/04a59094cd614d8ebf0763ed10b3b534 : SUCCESS in 8m 51s\n- grenade https://zuul.opendev.org/t/openstack/build/39d9d0caabd94baab4488bc2b5c14a20 : SUCCESS in 1h 02m 08s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/933b78ed5385401e906e88585271134f : SUCCESS in 1h 15m 48s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/536f0a2ba0714c1ebe8212399ec701a8 : SUCCESS in 50m 23s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/10332a77865d4712b23ef6cb80e64207 : SUCCESS in 1h 00m 37s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/bbe73c9040c847dab51ae59f48805e01 : SUCCESS in 17m 51s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b7f572f65f754a3b9487857de780b4d2 : SUCCESS in 20m 02s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5d912c6b7910415c9f18e24d16cdc642 : SUCCESS in 20m 47s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/66c4c6f684964fc284ac6619b302cdaf : SUCCESS in 8m 20s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/4bd767d71ff143c68a21d4a118eefd27 : SUCCESS in 5m 46s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/83b5d33826dc455d8cd1e8c2cd76dff2 : SUCCESS in 17m 07s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/201ef1b734af4e32a18a3640c6181ccd : SUCCESS in 18m 33s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/9f6b6b0dc63e4f0e8028517e46ceff4a : SUCCESS in 14m 20s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/f3c2bd2ebd0046b287bb559455a043d4 : SUCCESS in 16m 47s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/9642c36e6d69462bb8fd7a097a2712af : SUCCESS in 15m 21s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/4fdb6a28f5d34a9a95f9b0adb86bd4c9 : SUCCESS in 18m 29s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/3b3996e5a4f14932a75ddd7d2d769628 : SUCCESS in 18m 19s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/548142c83748449d87992321b37ab218 : SUCCESS in 17m 34s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/c227eb063426443e996e3e4a9a81759f : SUCCESS in 13m 40s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/a5cb1fa41fee443eb0fbdd30f55bc0cc : SUCCESS in 15m 22s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/4d743eb82a284e6eafd7896273b625a0 : SUCCESS in 15m 44s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/8b55b53120244772aa53e174acecaa6c : SUCCESS in 7m 22s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/c09a2a29c3b143c78743aca6e1c756b2 : SUCCESS in 16m 33s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/5e2417afeff7472da0ca398d3c4a6f53 : SUCCESS in 10m 14s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/ea37141b5a9245ca8e0a5ec640a0b854 : SUCCESS in 59m 15s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/ac71fbf871c543779cb9e5710b9923fe : SUCCESS in 1h 04m 11s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/9a987ef405194493bec00a7ed75913cc : SUCCESS in 42m 06s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/0033466f8c374d62b2e7b33cfd1d03e6 : SUCCESS in 49m 09s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3ec7e64b03cc4ed589bf1d86b18c8bae : SUCCESS in 18m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/496096d6c698480f8394ecdbb991d55e : SUCCESS in 4m 46s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/c36f74c895314cf49e7889ddc19fb680 : SUCCESS in 20m 52s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ecf254ed3b8745c1bafc260ee062dc14 : SUCCESS in 41m 10s","accounts_in_message":[],"_revision_number":15},{"id":"0ddb3b51ea2fc19092f113267e4e1bed2b6ce5c2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-21 20:13:03.000000000","message":"Patch Set 15:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/38b944ce9fae47ae8df88a05646dd9ba\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/24a14967c8954ec793236ebddee9bbfa : TIMED_OUT in 1h 31m 02s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d6cf9dd285274e61a80eec2a8868d984 : SUCCESS in 1h 25m 21s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/102d7b673b3f4facacab444814aed4d4 : TIMED_OUT in 1h 30m 59s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/af87aa1429ec41e28b288570ba5d770c : SUCCESS in 1h 28m 19s","accounts_in_message":[],"_revision_number":15},{"id":"8cca4f513237737d7af9d3e867c518458b53bae2","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2023-11-22 07:05:19.000000000","message":"Patch Set 15:\n\n(6 comments)","accounts_in_message":[],"_revision_number":15},{"id":"8fed0480f9cdce54e3ac249991610b6fb0795fb9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-11-22 11:07:53.000000000","message":"Uploaded patch set 16: Patch Set 15 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":16},{"id":"00a3dbd64db2bb23617e38a7ba546c07625c42a4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-22 12:31:21.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/170ee083bead4d02bf9a003751045189\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3f62c9c6e4364993891df0654aab8a2e : SUCCESS in 9m 17s\n- grenade https://zuul.opendev.org/t/openstack/build/e590824982a14ffab9a0f30f50755d56 : SUCCESS in 1h 03m 53s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/d3decea34b8a4cea8a50f16b0ccc5a34 : SUCCESS in 39m 54s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/bba5238f88cd4a8e94eadcdd8a83bd91 : SUCCESS in 47m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3fac01e6786e435ca0a4bca2c4c2a100 : SUCCESS in 1h 08m 07s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8f5558fb88674502bf8647a6a1ad1fd0 : SUCCESS in 19m 11s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3d048b26fedb436fb219bf5c4d2031f0 : SUCCESS in 18m 00s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3cad83b7ff7f41a3885bfb111bd810f8 : SUCCESS in 19m 01s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/998ba0d633284599b053096c59bd2e2e : SUCCESS in 9m 16s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/c16d8380dae54da19dc528a12d8cdcdb : SUCCESS in 5m 01s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/f3183d533bf04041a49b8cf9673ff1d8 : SUCCESS in 18m 02s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/9fdb950603794db39669ca5d95e6d038 : SUCCESS in 17m 12s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/61e29c80cee74d05914d7f992c99fa3d : SUCCESS in 16m 11s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/4b4cabf4700f4037a1f23c97816375f3 : SUCCESS in 13m 37s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/97baccf6ef1946bebcd235ab2b30d513 : SUCCESS in 16m 24s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/0b1ce30f030c483db702da68928ca2dd : SUCCESS in 17m 04s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/bf086fc878774aa6b797e6ee0338e2b2 : SUCCESS in 20m 07s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/474de47c3b80426c93e9bf50aa364446 : SUCCESS in 17m 03s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/d01412177886435c906ff4bfe9fa69f7 : SUCCESS in 14m 09s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/4a956cdae32a4674abd863552283e47c : SUCCESS in 15m 50s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/b31dfea255544d01bdbf1063df9eec3f : SUCCESS in 16m 24s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/a6edb260483e4c3984d731728cd8443c : SUCCESS in 7m 21s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/1f7e6274205045618ab53ad311e990fc : SUCCESS in 16m 05s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/13064df7072a4c93a7e739049ad38bca : SUCCESS in 8m 15s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/e9c3cb734b2d44ca9539373ebc7be663 : SUCCESS in 1h 04m 17s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/aab4813032384a418759d5753e66b297 : SUCCESS in 1h 07m 45s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/b7d4767a748249ac99dc4ed30f6605b1 : SUCCESS in 35m 43s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/505130820c37436d8ac78d7122efc452 : SUCCESS in 1h 05m 02s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0dc3668c995d468995806a554d81b5be : SUCCESS in 19m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5c6e694774284e0f89b6f47216185a2c : SUCCESS in 5m 05s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/23e4e73d8e5f44388dfc66b98535fe8f : SUCCESS in 21m 15s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c15ec02be7e04d6b8c0008724e588eae : SUCCESS in 54m 12s","accounts_in_message":[],"_revision_number":16},{"id":"ebe11d824b643b5c91546364850080fc7a59ab5b","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-11-22 12:39:23.000000000","message":"Patch Set 16:\n\n(7 comments)","accounts_in_message":[],"_revision_number":16},{"id":"10d7bfca9982c9694eb895cc8d130e936f72b14a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-22 12:41:43.000000000","message":"Patch Set 16:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cda1b483ab814759aa448de8288916c2\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/3983573aade34b5a9c6c0301fada5be7 : TIMED_OUT in 1h 30m 59s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/410f413810ff47688ef1b158ed907e79 : SUCCESS in 54m 06s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/cacf8af919884409a1f86827a8b4c50e : SUCCESS in 1h 11m 09s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/8931250f04f64302bd18803259f5926f : SUCCESS in 22m 56s","accounts_in_message":[],"_revision_number":16},{"id":"7555925beabab1576d868adfc88792aa28218507","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-11-22 12:52:13.000000000","message":"Uploaded patch set 17.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":17},{"id":"3ee36e3cfd94f7a4394bc18503cd49551de59bb9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-22 13:13:17.000000000","message":"Patch Set 17:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/af012e8c4b80439cb9d8e7766138f3ef\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a6d55c206e71460a9437f2d6e64316fc : SUCCESS in 15m 21s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/fdda6fa054e44038b3709c3cb96f68ce : SUCCESS in 15m 13s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/ac425e7a703c407e90b33c40b51f4af9 : SUCCESS in 19m 44s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/61d90b5391394d6abdd518e6b7b25ff5 : SUCCESS in 19m 10s","accounts_in_message":[],"_revision_number":17},{"id":"e0167389bd205fd0748aaf739dea014bfb208c2f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-22 14:18:22.000000000","message":"Patch Set 17: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3c263f31e84247478671bf1e24bfa2b9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ab3597ed34844efb9400f198b94935ff : SUCCESS in 10m 54s\n- grenade https://zuul.opendev.org/t/openstack/build/7748abd6f5fa4aaaa798f3b4fb01c43c : SUCCESS in 1h 08m 46s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/797ac580633f45d4922147c0896a9ac9 : SUCCESS in 1h 03m 22s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/83b8d69ce86c4acc9ecfdf1812c0687d : SUCCESS in 51m 45s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bf8f816f645d431d8464b185b7569d68 : SUCCESS in 1h 09m 51s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3eae1f576d644e49ab61ad6f352b2888 : SUCCESS in 20m 59s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/40b5e38eb82d46eaa6bd31d53f071acf : SUCCESS in 17m 35s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/fa2be3ec6a1044dca12430b8d86dd4d8 : SUCCESS in 18m 35s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/344d936a0a4e44628fd62d21dde07221 : SUCCESS in 10m 12s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/fd077b7b0bba4a938ac7f7df3c967ade : SUCCESS in 5m 26s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/71bd37ec63bb40b78e6e550f0bd61ccc : SUCCESS in 14m 21s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/c59aafc6507e47258d1e3c1c851b5307 : SUCCESS in 16m 48s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/bef424e53bab439fab007c51b77ff777 : SUCCESS in 16m 19s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/d404902c0eb949609560f42f412e4202 : SUCCESS in 15m 15s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/b2dc89442f434bc79a8322d38074a396 : SUCCESS in 16m 53s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/31c1b2d0388347c6a41a9e3887664eab : SUCCESS in 17m 50s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/a89064694ef84b8caba5e7453a16872a : SUCCESS in 18m 24s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/e116161e011d479698a7f1bbc000473f : SUCCESS in 20m 20s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/0af0f2b4a90744fa9204c8f532be1da7 : SUCCESS in 17m 29s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/8dc26a78caf2450da7bc336accf5effa : SUCCESS in 15m 44s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/e92f92b3d50b4b9baebaf9efea0eb772 : SUCCESS in 16m 51s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/da6c208c6fc2473dbadd289dc9558771 : SUCCESS in 7m 17s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/f269e151b4294f118a22081ff9035871 : SUCCESS in 16m 57s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/5f46026c444c4ac288feaac85af0a3f3 : SUCCESS in 8m 04s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/e9ce8d44ac1f4b7f916d993b425a4eb1 : SUCCESS in 1h 18m 32s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/7de1c8d8331942caac5973aee8f9eb8b : SUCCESS in 1h 09m 37s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/1e3d40fe42b943529f6a5d53c7178f6e : SUCCESS in 51m 46s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/bb202d68daed407f8fc374ffcdb2c5e2 : SUCCESS in 51m 16s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b169f408bf7d49cf8ac7d5df77719f30 : SUCCESS in 19m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/680f22ae14d14c64bfdc775303d207a1 : SUCCESS in 5m 25s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/90129e05587f4d9c80b60e42c2aae789 : SUCCESS in 22m 07s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/956ae9aa9d7b48f6b1dfd14d99873803 : SUCCESS in 1h 09m 20s","accounts_in_message":[],"_revision_number":17},{"id":"a8b539bbad3653e175a5cc4606c55a5ab5919bc6","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2023-11-28 05:54:41.000000000","message":"Patch Set 17: Code-Review+1\n\n(3 comments)","accounts_in_message":[],"_revision_number":17},{"id":"8d740c0dfc942f15fb4542c14c422bb463343c17","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2023-11-29 21:56:41.000000000","message":"Patch Set 17: Code-Review+1\n\n(21 comments)","accounts_in_message":[],"_revision_number":17},{"id":"142eaf3e06abb02a1e87ddd7b99f899a5b4c833e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2023-11-29 21:57:11.000000000","message":"Uploaded patch set 18: Patch Set 17 was rebased.\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":18},{"id":"fb2cb07bfb3c79c8c76754259eeb4aad21bd2263","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-29 23:05:15.000000000","message":"Patch Set 18: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8a29e44eaaf44809a74a853f7baa0f5e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c864fb0ac8a549eabc995b85ec42b3aa : SUCCESS in 9m 30s\n- grenade https://zuul.opendev.org/t/openstack/build/b621c72792a041a6881a26fcef15fb82 : SUCCESS in 59m 41s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/98392cb927b44a8baba7103caac5a7ba : SUCCESS in 44m 23s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/79e74483331842a29de4255110e76b52 : SUCCESS in 34m 00s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9fb60c83a09f485dacec2a4ed30f1d1e : SUCCESS in 1h 00m 52s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f72a2e8959644a14b66f21456c3d3a2d : SUCCESS in 18m 07s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0cfe160c7824484abc004295c784ea6f : SUCCESS in 22m 17s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4bff0c82ec14488f8b683f56b0b12bb5 : SUCCESS in 20m 46s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/106e610912ff43b99fa9712db617906c : SUCCESS in 8m 37s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/145eb7d22543465bb5d76e424e8eb15f : SUCCESS in 6m 19s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/af14b92f865e4f80983a19f4b4e95bcc : SUCCESS in 16m 45s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/d8b4dfb74ebd4d48a213066b5ba9f6b1 : SUCCESS in 13m 19s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/9235d13c89b14f27baf5cc7ff1c9a64b : SUCCESS in 14m 51s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/31f10e7a87a94f92b811454b567b9aa8 : SUCCESS in 16m 43s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/b2001768964f408d9825125e310d8d7e : SUCCESS in 11m 58s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/bbd8cb75fce145178c8a7014e2d913f3 : SUCCESS in 17m 57s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/f4466e65a19f42ab98fa8333831e215f : SUCCESS in 18m 35s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/e46da629feeb4d40b59f015975e4e94c : SUCCESS in 18m 46s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/32c2eac35c3249b7bf2af972f6153167 : SUCCESS in 15m 13s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/48c23222f9c74bffb3bf591b6377960b : SUCCESS in 12m 39s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/e921158f8bd44e26b97c25de07195dfa : SUCCESS in 17m 35s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/3e91110babd843cb9c33cd787b0e87e7 : SUCCESS in 7m 19s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/1b20f3aace484666a8ac1820cdda5f2a : SUCCESS in 19m 33s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/5b175e2fc4164827bdbcc78f026627a2 : SUCCESS in 8m 02s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/9eb64a158c6a4718943997dfb152dcd0 : SUCCESS in 57m 52s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/a35772bc93714f219b90faa1d5f75a75 : SUCCESS in 51m 54s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/1f237041399c4091a8df722d02253686 : SUCCESS in 56m 10s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/ec30d1f5ed1a4b75a32d395015e44044 : SUCCESS in 54m 14s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/eab9b9696af44f1fa377a5e1803e2162 : SUCCESS in 17m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e37a881f840348999d519df69a4e644b : SUCCESS in 5m 26s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/5594eb419ca34d7aa0220b7bb744c944 : SUCCESS in 16m 00s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/add820c9498142a69f2fd06f34eb83d2 : SUCCESS in 59m 30s","accounts_in_message":[],"_revision_number":18},{"id":"9195aec74ba2db971ab55d1add2e4f91084ccb72","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-29 23:30:36.000000000","message":"Patch Set 18:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9e59b95158ee47ed8b270cefa7e75bf3\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/de8de605345a4534bbc5f4b425319838 : TIMED_OUT in 1h 31m 00s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1b0d3d9075cd4c88b438ac8800c8f266 : SUCCESS in 18m 10s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/46360e6952f84b83b59aa0727548d787 : SUCCESS in 1h 18m 00s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/03050ec81acf4ba082a2b9f588750ab8 : SUCCESS in 24m 31s","accounts_in_message":[],"_revision_number":18},{"id":"17624fcaccb59d92b89c67c5b3f6a2af4236b312","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2023-11-29 23:41:00.000000000","message":"Patch Set 18: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":18},{"id":"3898c99945588cf57e61f7d08679aa0d041e0e18","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2023-11-30 05:46:11.000000000","message":"Patch Set 18: Code-Review+1\n\n(3 comments)","accounts_in_message":[],"_revision_number":18},{"id":"b2f8ee8fce009398305682a018fcebab7b1e70df","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-11-30 15:39:33.000000000","message":"Uploaded patch set 19.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":19},{"id":"e4037bc2489c8af798352553a7d6e54cae9d873d","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-11-30 15:43:06.000000000","message":"Patch Set 18:\n\n(20 comments)","accounts_in_message":[],"_revision_number":18},{"id":"fccbeb7509bf607f91435368ea29755bc804b8e9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-30 17:13:12.000000000","message":"Patch Set 19: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/96f6f411668d4b0e9e81c43b419e7ac7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cf11df53f92a4eddad6e1cb6b42e78a1 : SUCCESS in 9m 37s\n- grenade https://zuul.opendev.org/t/openstack/build/3b55a54b852845f88dad75d079d51b52 : SUCCESS in 1h 13m 53s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/b7156b7b28464036972fc086178773ea : SUCCESS in 57m 24s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/228d1bcaac9645ff867534c927b22bd3 : SUCCESS in 55m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a659f295981d41dfab619a3670d912b4 : SUCCESS in 59m 40s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7c350e2282fa4a2d84a013beb1d38753 : SUCCESS in 18m 26s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/787a3a57047b4581859dff6bc02bc37f : SUCCESS in 19m 30s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7acce89761e0444cb7c377771b081a8a : SUCCESS in 20m 21s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/6644a35adda84b85b79795c33318c5a9 : SUCCESS in 9m 45s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/0bb08f359ad746ae99c8f97343fec7e8 : SUCCESS in 6m 37s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/2ab6cb4ef8f84838ba0efc04c2df7e86 : FAILURE in 18m 49s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/80ff57fd00c0481ab92384aac8a9dccb : FAILURE in 17m 36s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/7731506c0f4c4c02b6a46cfeaa6d6a9b : SUCCESS in 14m 30s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/ad0734fa659f4bebb08c45c95827c3bb : SUCCESS in 16m 16s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/f32327b6fc6a42ac9c23577804ef9eea : SUCCESS in 15m 09s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/ae0c51cd39b24cee9fed47855b574d41 : SUCCESS in 17m 17s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/396deea31df8425ebb5e6db34c74d7bd : SUCCESS in 16m 32s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/a2522618e8614fbf97a1f290b87d059e : SUCCESS in 17m 14s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/18347c40101049b8959b88540eeb1f41 : SUCCESS in 18m 21s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/4fb75044775c4808b0845423cc6709eb : SUCCESS in 17m 24s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/eba07918e69e497cafb4ecb1365aae3e : SUCCESS in 18m 21s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/325ce876cfa44dc48997750ccce00445 : SUCCESS in 8m 13s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/bcaacbdd749543418f5dd06f9f130acb : SUCCESS in 15m 35s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/ce6289876e274988a41c5e25b642f6df : SUCCESS in 8m 18s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/d9f532d530934ee8aa383b5866d43f6c : SUCCESS in 1h 02m 05s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/fe040083764c40ce92f034831c47884f : SUCCESS in 1h 23m 57s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/9a87dc95935b4702bb675078b412d28b : SUCCESS in 1h 08m 52s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/bb22149b9f10441da3243fa5902fba45 : SUCCESS in 59m 16s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/594545e23c4b4a9d9c72c0346c2c1594 : FAILURE in 16m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/07a834dfaa464a6584fad7192de838f1 : SUCCESS in 5m 02s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/d198d8689ecc499eb8492bb5fa98efb7 : SUCCESS in 21m 16s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0c88121713054174b8996896a880f3f4 : SUCCESS in 1h 06m 48s","accounts_in_message":[],"_revision_number":19},{"id":"1a1f44284b746539bb0fc30902a621cd009b8968","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-30 17:14:21.000000000","message":"Patch Set 19:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/94910ed354dd4dcf8392f7e64760cdc8\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/968f46fa0045427f8f18cf9dfc9d1246 : TIMED_OUT in 1h 31m 09s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/fc93f921f6254262b0563b6ca4fa04ce : SUCCESS in 16m 26s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/fdecad25c87a46d5aacbc9d7f36c5ff6 : TIMED_OUT in 1h 31m 07s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/9647a7223f0346f9beeeaa81056cbfa5 : SUCCESS in 21m 17s","accounts_in_message":[],"_revision_number":19},{"id":"354dada70fc329c6d221615872540859e5a584d5","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2023-11-30 17:49:37.000000000","message":"Patch Set 19:\n\n(1 comment)","accounts_in_message":[],"_revision_number":19},{"id":"07423ec297ca10eeb5281797997b552e206942f6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-11-30 18:20:14.000000000","message":"Uploaded patch set 20.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":20},{"id":"024571ae52b5c742928c2da3ca0047f6642962ac","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-30 19:33:39.000000000","message":"Patch Set 20: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/24f5a0df728745058ea647ec62cf831c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f636f69af46a43059ce6f2ba8c0bfe67 : SUCCESS in 9m 03s\n- grenade https://zuul.opendev.org/t/openstack/build/9eb601abe4834d43ae72856a59121a7c : SUCCESS in 1h 00m 38s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/4dd10cdfa44f4fc99940690bf5c45fcc : SUCCESS in 1h 03m 50s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/f060ba0192574439ab405c269631f0d3 : SUCCESS in 47m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a395e703c90a4bc297edc3e45ab3640e : SUCCESS in 1h 04m 04s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b35c91e0e4474682bcdabab90a2c9ba7 : SUCCESS in 18m 18s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/60447937327a426ca16d41e29b772ef9 : SUCCESS in 18m 34s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c1866822d9594b03ac0c14a980d06dde : SUCCESS in 19m 38s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/24b63b49624a443c9930e2aab97660d9 : SUCCESS in 9m 39s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/5a8f87523121497c91fc655525610878 : SUCCESS in 5m 37s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/1cfd64b9738341c092640048e7805786 : SUCCESS in 18m 57s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/8335fb12c9534a508503ab07d5b4f589 : SUCCESS in 13m 47s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/1705d2af21d5471cb7eceaa0994f121f : SUCCESS in 16m 58s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/5692580046434d67bb4f94f9be7660dd : SUCCESS in 18m 00s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/f7fbb2fe14f9424d88feefc66cd92ce7 : SUCCESS in 17m 13s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/948845299011415e9c551d7cc97573da : SUCCESS in 17m 39s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/08eb5640c37742f986276be2e91ca5dd : SUCCESS in 20m 49s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/aec93e94e7d345aca51525a4d0614d68 : SUCCESS in 19m 18s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/74a3081615544557913d915ece7a3060 : SUCCESS in 16m 43s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/89eb4d5921cd4b018f46d4f6c845bc81 : SUCCESS in 15m 44s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/92c1a9febe534d54aa8bd038c00a9ff1 : SUCCESS in 17m 07s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/ba4dea139a2a4fd88213c21e57e2dbd2 : SUCCESS in 7m 27s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/cbd0404d91a741bf91cc1bedb7ec51c4 : SUCCESS in 17m 41s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/134d6367d79440909a16a8ef74241b1a : SUCCESS in 8m 00s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/a2de90942bc34b0396cfcfe769268b8c : SUCCESS in 59m 53s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/9d6a3ff66c624ba9a88521034d5886ec : SUCCESS in 1h 00m 05s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/49e54e380cdb4faea785a2eb0af2873e : SUCCESS in 54m 10s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/f26acaaf3f5d4b76a4d2ae23c42a590d : SUCCESS in 54m 31s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1fbd467938174345b3c7954f9235ac9a : SUCCESS in 17m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/161f3847581242ab80e3979cae246133 : SUCCESS in 5m 05s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/0ecd897e1e57482686aae723944fc660 : SUCCESS in 20m 55s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5c57e5ab14164299862a8e72d438d215 : SUCCESS in 57m 31s","accounts_in_message":[],"_revision_number":20},{"id":"aa8e94dd818d2ad93244942953ecd5e880dba5d5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-30 20:19:05.000000000","message":"Patch Set 20:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/28f40e41f2534505afbc10ca3b877ca4\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e805c987b4b94a1cbd8ac85e2a3dbb3e : SUCCESS in 19m 57s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/491ec53b562444c1987390effe4a8f93 : SUCCESS in 1h 13m 40s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/a95736cba48c4498873e217ef26f7d90 : SUCCESS in 1h 20m 04s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/3f15e67999d74c71a671e2114d7a4809 : SUCCESS in 1h 12m 28s","accounts_in_message":[],"_revision_number":20},{"id":"03f1e4c09d7e5804e2096d11d2d6ae2a2ef35144","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-12-04 09:30:09.000000000","message":"Uploaded patch set 21.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":21},{"id":"f0a1cd47ccaf1f5ca64fd6db88d62ad3b70e3fbf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-04 10:40:22.000000000","message":"Patch Set 21: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a8050fa40c39481f9cab98182f55cccc\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c98218cd15e14f4490bf82a0df1bc238 : SUCCESS in 8m 55s\n- grenade https://zuul.opendev.org/t/openstack/build/e3d4de1ac0be4dab8e5288a9650b29c5 : SUCCESS in 1h 04m 19s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/b14849bbc0954d9cb0db953ec5f32c8f : SUCCESS in 54m 38s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/619320cab28e43afad736ad3591828f3 : SUCCESS in 41m 45s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cd27b86911fb4537971763b03324275f : SUCCESS in 1h 00m 57s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f973534bd97949e69b3819979deb24de : SUCCESS in 22m 56s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/49740c06ae1b421ebdd79fe4cec91693 : SUCCESS in 25m 26s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f44db89e9d414bfcb83319fd59004c46 : SUCCESS in 27m 36s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/9a20690aacec405199924b0c6361f503 : SUCCESS in 16m 03s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/faa78427561b4a2987930a2289d79891 : SUCCESS in 6m 07s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/d02fc39a82d74c38a6d7202aa8ad5aa5 : SUCCESS in 21m 11s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/54f11d59b99d420aa1993ee85350f12f : SUCCESS in 22m 19s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/b3812242828547bc9d9f27ad13bce380 : SUCCESS in 20m 27s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/a3d28e19127c42c7b7949a11cee3e9ee : SUCCESS in 20m 57s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/52ccc72d2109411b96b2f93b3f454862 : SUCCESS in 15m 48s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/c3051672b6e14c14aebabcf99000ebb2 : SUCCESS in 23m 52s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/0fcfc356eca842919d9c842f64ffe3b1 : SUCCESS in 18m 38s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/9ea72c81f54541e88bcf5c439eb38048 : SUCCESS in 18m 30s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/12d86fc319184eb9a52a2322d2d9cfa3 : SUCCESS in 20m 13s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/689573e7710140a3ab1b791344697dc0 : SUCCESS in 15m 39s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/5247e0fb105d416a9fb1d0781b3d8952 : SUCCESS in 16m 33s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/f2060d308c2a4d97b9658bc6b3bae402 : SUCCESS in 7m 49s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/83f9ef593e74458aaa181a9883e2e583 : SUCCESS in 19m 03s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/b54200d5d9924dfa98723b49793613fc : SUCCESS in 7m 44s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/325cf06c27084560a5fbe79528b8d6d0 : SUCCESS in 1h 06m 06s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/821c28c4b0e54d029e86a3a04599c017 : SUCCESS in 53m 45s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/9c135228c5aa4eb59147beb3e2518ca1 : SUCCESS in 52m 01s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/b2a51eba6e06452886cc6952a0540cb1 : SUCCESS in 57m 37s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/84940350a42d4e1d92f0bdf4d42b27bd : SUCCESS in 17m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2686f764e38a4744833b7d87242c0a9d : SUCCESS in 7m 53s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/b375f91902fc47059da0015e91d8d32b : SUCCESS in 27m 13s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ba9f37d7c2d442ea8ff8f2a44e87ba89 : SUCCESS in 57m 18s","accounts_in_message":[],"_revision_number":21},{"id":"3f09327ede0dac85ac1730be5bfbee686896df9f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-04 11:04:06.000000000","message":"Patch Set 21:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/37de42f8c7634b749be05f0a23d13e04\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ab3646159dbb4064b01fff797283db15 : SUCCESS in 20m 05s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/190dbfd371ba4cd7805cacd105360409 : TIMED_OUT in 1h 30m 57s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/2317f01c6a824129a2a35ebb3eb84e63 : SUCCESS in 23m 46s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/e63440797d92471cae50d0088b15217b : TIMED_OUT in 1h 31m 09s","accounts_in_message":[],"_revision_number":21},{"id":"03a6cb99f27e548ae62980b7b32fb6d8c7aa9311","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2023-12-04 16:39:40.000000000","message":"Patch Set 21:\n\n(7 comments)","accounts_in_message":[],"_revision_number":21},{"id":"18d2f1fff47fe4b4711272ecad469d0bc96a23b8","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-12-04 18:51:22.000000000","message":"Patch Set 21:\n\n(3 comments)","accounts_in_message":[],"_revision_number":21},{"id":"bc9409cedefb43e1f595fb5ec00d984c94ab066b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-12-05 12:30:27.000000000","message":"Uploaded patch set 22.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":22},{"id":"d3f745af5ed21230edd84eba02d53f1a88a42cef","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-12-05 12:31:37.000000000","message":"Patch Set 21:\n\n(2 comments)","accounts_in_message":[],"_revision_number":21},{"id":"f6c3bc6af0a1b7a0ff31f961a314e6a452c2d13f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-05 13:58:01.000000000","message":"Patch Set 22: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ae3f8a728b0741a1ac278750088d420e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c536f2267c674948a4d9aae292925048 : SUCCESS in 10m 25s\n- grenade https://zuul.opendev.org/t/openstack/build/43fa753196804189b5bcff3481d1455d : SUCCESS in 57m 42s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/d58766cc73804af393cb2865417ded9f : SUCCESS in 1h 05m 56s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/21d2268967fe440b845748e1be655e90 : SUCCESS in 52m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5a6faab2d2b64a239c690084fafe564b : SUCCESS in 1h 02m 07s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/336d887b447a48de8ac76769ae14aac8 : SUCCESS in 27m 26s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f8ce1d6afc374c81b38d9ab1fe53074f : SUCCESS in 19m 58s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e5bc26ce3b5b447ab8b62f4bbc3702c7 : SUCCESS in 20m 41s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/3b3cb98d4ef4484798a1060fee554b69 : SUCCESS in 9m 10s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/fac4abdeec5c44b0aee34c3cb296e545 : SUCCESS in 7m 33s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/d304d092bb18429fad348393ee07fb30 : SUCCESS in 18m 19s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/a0528b821ec145a29c5a291bf4468062 : SUCCESS in 17m 21s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/e7e93bed698e4b2cbb3a2c73b29d2f74 : SUCCESS in 12m 15s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/6e39ade7681a48418e0e4f907c3c0cdc : SUCCESS in 16m 22s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/db68d32d768b450a9080d0d68f52fb5e : SUCCESS in 17m 57s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/2b834e30f4ca46dfab2ca026e669d123 : SUCCESS in 18m 26s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/ad02d0cb54844be8917558b10dff56ff : SUCCESS in 16m 27s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/1241d43cc8084096aaaa179cea8513c7 : SUCCESS in 17m 26s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/c6ea353a12f44f4e9c7530dbb7711240 : SUCCESS in 23m 26s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/3574ce81667647b888fc2a8877c4ac83 : SUCCESS in 15m 49s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/758cfd085bfe4febbb6b03c1c52bae80 : SUCCESS in 15m 21s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/e6bb4d7064c4464dbb479c8004331bfa : SUCCESS in 9m 06s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/49b65da80e0b436aa6d44098336cfc0c : SUCCESS in 18m 39s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/d6f6f42b5e2c43e79e7452365a708501 : SUCCESS in 8m 03s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/7568749ed03b4bed990c8a60f31205b5 : SUCCESS in 58m 22s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/cfe3058b9cd74b6597516de604936b73 : SUCCESS in 1h 05m 42s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/fedc586fea674a7483c01144ee625345 : SUCCESS in 58m 26s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/0334182068af419886036af626556af8 : SUCCESS in 1h 00m 41s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/be7a7c867a1d4a3fa2683c922fced938 : SUCCESS in 18m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f3f7ede60aa840319ce14817f3eaa1ad : SUCCESS in 5m 09s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/55983337591e4ab49d3f1d032435996f : SUCCESS in 21m 19s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/facfe5f046d0402190d4cf83933cfaba : SUCCESS in 54m 12s","accounts_in_message":[],"_revision_number":22},{"id":"bdece4dfb5771ca773054684d073678f05cbb94d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-05 14:51:25.000000000","message":"Patch Set 22:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a4aa232a679b4a8fb822584f89ec732a\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/6094c8399bc547f0b6f5fd8f4a18951a : SUCCESS in 1h 00m 41s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/afc86ec63326499baa983f95fc4b4447 : SUCCESS in 1h 07m 55s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/b82837e999534c23b0c0438bb74f9301 : TIMED_OUT in 1h 31m 02s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/8377aec9896a44479530b68a5e912bb9 : SUCCESS in 1h 00m 12s","accounts_in_message":[],"_revision_number":22},{"id":"991815bd3106155ee3db02e0cddb7c11f9e694e4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-12-05 16:21:14.000000000","message":"Uploaded patch set 23: New patch set was added with same tree, parent tree, and commit message as Patch Set 22.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":23},{"id":"975a9efc8c50811e3c5b9671f0668008d11c0466","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-05 17:35:35.000000000","message":"Patch Set 23: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c2626796d47548b2a86c398f6337ef4c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/99685ddca4a6475da491590c7bf5a037 : SUCCESS in 9m 11s\n- grenade https://zuul.opendev.org/t/openstack/build/7cd4e8fe519c4e539333d64f9d03b23f : SUCCESS in 58m 16s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/8b23127741674899a97946ed06a6d34c : SUCCESS in 59m 15s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/4b706f22bdb14b06a7404b23423a60c1 : SUCCESS in 46m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/aa4ef4a143c7484cb64588e8adf7a2bb : SUCCESS in 59m 49s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/af626a5c42c84b6584eee06336254a63 : SUCCESS in 19m 31s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b4aa4f70810b4750b163acd83097dc23 : SUCCESS in 20m 52s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/55577feea76d4037932e8c2b2f50fd19 : SUCCESS in 18m 18s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/7c2b5ca1820a4964b590ddbc08334af8 : SUCCESS in 8m 23s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/dae54c83670f4a27adad602630944773 : SUCCESS in 5m 49s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/2ad78f9abab841e6b50aee71a348c8e6 : SUCCESS in 19m 00s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/4dc1885540364528b8eefe8a0161d383 : SUCCESS in 17m 08s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/3279ec5173c1440e89f57e6d1f19c95f : SUCCESS in 15m 17s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/7c5fdb03177a46149d0a5720200c2886 : SUCCESS in 13m 47s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/b07817fffb1e49238785a88296f2f91a : SUCCESS in 16m 49s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/6faef21b34e74eed9e73beb35cdb0df4 : SUCCESS in 17m 51s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/8533b04ee83a42e1bdefed5903d8f76f : SUCCESS in 20m 30s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/3dd9e077c5fb4cca85241f6bc416423c : SUCCESS in 17m 25s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/01d8fe46ad124c84b433d046ee1a69fc : SUCCESS in 13m 27s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/0c82e2098e3e43e08867a1e0f07a33aa : SUCCESS in 14m 06s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/c141a245b9da424d9d26e68ce31d821f : SUCCESS in 17m 40s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/989d93cd449c493ebfd64ece180e38d4 : SUCCESS in 7m 43s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/63c1a01b7629459786e6d58ca6d5f516 : SUCCESS in 17m 35s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/0c05dd1bc6714b99af0c3a117e758b6f : SUCCESS in 8m 33s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/16f9cf53e6cd4c2c8d7c41e6f4680785 : SUCCESS in 1h 04m 41s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/e7c839cac4184c06839e17dfe90ae21e : SUCCESS in 1h 04m 15s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/f9496789cf2f44ddbfa475d32c20b243 : SUCCESS in 1h 01m 48s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/8bce07c0371241f0bb5af9552d72431a : SUCCESS in 59m 53s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/169b8504f4b14c4990a0f1e9ea44a8c1 : SUCCESS in 30m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dac9543136e940ccb4816e1a8c1b87b3 : SUCCESS in 6m 27s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/4f7bcd1b8cdd4ae8bfa34d7d16172a74 : SUCCESS in 19m 50s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5b65d61897144dcda88e62d545f53fde : SUCCESS in 1h 05m 17s","accounts_in_message":[],"_revision_number":23},{"id":"ba3e81bc5d87caad0a2c1def63009a392cd35859","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-05 18:14:57.000000000","message":"Patch Set 23:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cb2d86486c4542318946fbc33a7b3567\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d9b11426851e45debdb8da2052036a16 : SUCCESS in 55m 30s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ec3885d0911542ea92436d1a4c5a9456 : SUCCESS in 42m 50s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/d006c81b1fa84e9289699d71c8729f45 : SUCCESS in 55m 39s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/08b1ec214213407bb2ee927d79cc25f7 : SUCCESS in 53m 44s","accounts_in_message":[],"_revision_number":23},{"id":"8024004e46c0a2e926590bc35d1a02915df4f8fc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-12-05 18:38:59.000000000","message":"Uploaded patch set 24: Patch Set 23 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":24},{"id":"c8290915d48696e6b6a0e6ed3105a5a57de65101","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-05 19:56:34.000000000","message":"Patch Set 24: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/48122f0c6eb94455a1a20d7a19fd4316\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/79584a06bbb74cd987d9e656b27c609d : SUCCESS in 9m 20s\n- grenade https://zuul.opendev.org/t/openstack/build/6ecea8a3b57c4908a1f737e6268ab1a8 : SUCCESS in 59m 10s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/2e0ad3b0a7e44d3ea8b8d2ad5f35719f : SUCCESS in 1h 02m 52s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/32a9fb8fb5af44dc894c3bbc84864926 : SUCCESS in 58m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/456e6860bb9749f8995a82342e41f568 : SUCCESS in 1h 00m 28s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e094f703311d41118dd807f1b59e77ca : SUCCESS in 21m 22s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a7a2946bbfbe4bccbd5b64299318879f : SUCCESS in 19m 16s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/56c8213775ce4575a62ce8feb1b50004 : SUCCESS in 19m 40s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/01b4983b0d9b49038b55d1593d144970 : SUCCESS in 9m 17s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/c0bf045fb1de49d1b07fd81a8a8bafba : SUCCESS in 5m 38s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/f473198d69f8467ca4685f7839f46525 : SUCCESS in 19m 24s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/88685cdf9ae14e268d14e4203584cc4e : SUCCESS in 16m 40s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/368b6d6095c9473e837cab497206b5e7 : SUCCESS in 17m 47s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/452d4be2baf14a418f63d1a458447fc8 : SUCCESS in 19m 33s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/4b1b82a6c9dc499885e6dceacaecc2fd : SUCCESS in 15m 04s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/45832c73fc9d43bbab12591c00f065ed : SUCCESS in 18m 21s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/66d99b28649c43ffb65829a7ae86e500 : SUCCESS in 17m 15s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/dee736719c9f4ac58055972395828532 : SUCCESS in 16m 45s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/8aba1ac374f949f5b5d8832f49785523 : SUCCESS in 16m 58s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/6dfaecec18a543be9b0616288fca9f15 : SUCCESS in 16m 32s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/7e1c17bbe901465b9bc59bb1f14c098c : SUCCESS in 14m 38s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/430c7a4284284e689f6441a06760f0ed : SUCCESS in 7m 33s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/54e0c6d459e541928f5da6891e799028 : SUCCESS in 17m 09s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/6db80cf718944bca883517f333132a3d : SUCCESS in 8m 25s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/5736002db4164a20af7bbfa9f4cc47f0 : SUCCESS in 1h 02m 19s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/7d20f137c39b45b3844e4b3a8724fbef : SUCCESS in 1h 08m 50s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/19512f2e5f4e42838869b8a578e4f382 : SUCCESS in 54m 19s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/2dcc60a489614615844bc64fe7a7273b : SUCCESS in 57m 02s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ec04f518323841f19834e336bf8db1e6 : SUCCESS in 20m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/de26e679c2e64642a9ba91d7fd98ddc5 : SUCCESS in 4m 58s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/eb5ffa05e46d44c080e195715af90507 : SUCCESS in 22m 32s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/22f391a4f6534e628ec803533b88b8b5 : SUCCESS in 58m 53s","accounts_in_message":[],"_revision_number":24},{"id":"3803fcb8fc1b7a74a4f5d1a10dde9fbf4104c5cc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-05 20:28:53.000000000","message":"Patch Set 24:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5a4360141e61473e93b0797420b69191\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/24395592249548a798f34b53204254f6 : SUCCESS in 56m 53s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ba635d02a1ae47a0b9e6a10af581f30c : SUCCESS in 1h 05m 02s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/01c38e277f664902b1b50df0af82f901 : SUCCESS in 50m 30s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/898dc8be26c84a06bc1cffb2922ba527 : SUCCESS in 46m 54s","accounts_in_message":[],"_revision_number":24},{"id":"597dbf8fe6305bd4ad5f883f7a6a26234c32765d","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2023-12-07 19:32:24.000000000","message":"Patch Set 24:\n\n(1 comment)","accounts_in_message":[],"_revision_number":24},{"id":"9e5570617ee9cde7997b036cef4953a3ee13219f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-12-08 19:24:26.000000000","message":"Uploaded patch set 25.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":25},{"id":"099c344a14de49bf3c02feb0323f9c3c186f61ed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-08 20:37:25.000000000","message":"Patch Set 25: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d5d42ffc78ba48e886317f9f5b4df6d7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b5fadce342ac40f986d5960761845b5d : SUCCESS in 11m 08s\n- grenade https://zuul.opendev.org/t/openstack/build/4a30767bc7fe4b3aa3a4551799d4e536 : SUCCESS in 59m 58s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/538d332f9e364bf99216ec1117e0bb26 : SUCCESS in 56m 57s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/f56d4f41b42e4a96a641c330ee2fae86 : SUCCESS in 51m 45s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/eac0b77f2d4e4316a112402f4bf9f129 : SUCCESS in 52m 27s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6b0e60ad2228478cb45cc3a0da98f58f : SUCCESS in 21m 01s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1a83bbe6f1324544a56a8612407f5c67 : SUCCESS in 21m 12s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/edb01bd3a6a9436fa20e19ea9fdbe623 : SUCCESS in 20m 10s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/2514ff5d76bf44029b3d0f55fa743e75 : SUCCESS in 8m 14s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/c64c9e61e13040118b737b795bcc4393 : SUCCESS in 7m 43s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/5f0cf648cf314055b0d4abd6808e1095 : SUCCESS in 18m 48s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/66bf3880dba84122bd87889268479785 : SUCCESS in 17m 46s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/0c07b8744e7645be903f4c30bdfb63c5 : SUCCESS in 15m 45s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/9f44e139320d48738999e18e0ebeea81 : SUCCESS in 13m 27s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/61203a05026640a2a44796b20a915a71 : SUCCESS in 12m 31s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/61e1580c26d5411095e48e971577bf17 : SUCCESS in 17m 35s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/f5d66c7e583f46c3a14137c1a60bd550 : SUCCESS in 19m 16s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/8a3d1e920562425e87b77b677f3597b5 : SUCCESS in 18m 34s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/865c3bfc668b447cb88123b9daa40466 : SUCCESS in 15m 42s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/873f14ed3fd648dca7b7c19139ddfb5d : SUCCESS in 16m 59s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/0d775d63e7864cd99c6425723001abfa : SUCCESS in 15m 32s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/5c35d5233eab4e9792b8563a82f7d669 : SUCCESS in 8m 31s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/0dbe0a729f824b6d87cc830fdb78d9d8 : SUCCESS in 15m 47s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/ee9c3429df0f4b71a09f4f0f7b411bcb : SUCCESS in 9m 04s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/e3619d8ccd814273aa2bc866c1cf8e1e : SUCCESS in 1h 11m 40s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/0b9a3b45165a498fa77c3668b2207822 : SUCCESS in 1h 00m 17s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/bb71388f528d40be92cfe029c63f3544 : SUCCESS in 50m 09s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/94e0081dd0e148f09215493e0cab27c9 : SUCCESS in 52m 51s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0f609c3261874d899985d78885729bc4 : SUCCESS in 16m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c95263cf6b15441a8b92f2f935642890 : SUCCESS in 6m 46s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/4846a262473c4d619b3af79e2132abb8 : SUCCESS in 27m 46s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2222a85df7324dc19cbac13d4ec9015f : SUCCESS in 57m 56s","accounts_in_message":[],"_revision_number":25},{"id":"930479355077852f818b3e76e8aa5e40ff97bf01","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-08 20:37:40.000000000","message":"Patch Set 25:\n\nThis change depends on a change that failed to merge.\n\nChanges https://review.opendev.org/900740, https://review.opendev.org/895602, https://review.opendev.org/890470, https://review.opendev.org/901477, https://review.opendev.org/902646 are needed.","accounts_in_message":[],"_revision_number":25},{"id":"aaea4c1ffaee1e8c063770437e092a4352904b4d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-12-11 15:15:14.000000000","message":"Uploaded patch set 26: Patch Set 25 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":26},{"id":"ae95760303392435d61f42a95359e8f058e5f8a1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-11 16:27:04.000000000","message":"Patch Set 26:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/82c32c409520481d8bfeb0332bbda8e8\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/8d1805baa5a24c7297080baa407da3a5 : SUCCESS in 17m 11s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/fa5a33cefd214b3a99b1fae8d161ea79 : SUCCESS in 1h 09m 07s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/e5a0726f7e174f7f8993eb736c53daac : SUCCESS in 22m 50s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/c2abbc2cfcb444958fc84462692be377 : SUCCESS in 20m 40s","accounts_in_message":[],"_revision_number":26},{"id":"4918a62eb9ac1fcea750a089bd3418779e7ffe86","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-11 16:38:11.000000000","message":"Patch Set 26: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8b0b8076228f4ce0b73caa1737ccff0a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/88b78060d0294e83ab953da53618a3af : SUCCESS in 10m 01s\n- grenade https://zuul.opendev.org/t/openstack/build/739d077e6993444b8a6233249d52870a : SUCCESS in 1h 08m 28s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/f4827b8631f543188267ae8a366960d3 : SUCCESS in 49m 48s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/5785ee7a5ffb41cea4e583f80e3ccc3c : SUCCESS in 51m 07s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/75166e77ba9d40509ce8f252e1807355 : SUCCESS in 1h 00m 38s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0d57468247e6434a9dc31c99af6eed54 : SUCCESS in 19m 35s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ad9724f3370349308fc470219fb2edf9 : SUCCESS in 27m 38s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/65f2fde345a9486cbd2675b4192b75a9 : SUCCESS in 20m 23s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/a70b9b8813534066a6989756e1dedfd6 : SUCCESS in 9m 03s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/f0d646cceedb4eafb6bcca8cd7a5592c : POST_FAILURE in 5m 30s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/26a8cd440c45499b8abb311e5694482f : SUCCESS in 17m 44s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/66f337d053e446a4afadd6315195051f : POST_FAILURE in 17m 47s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/6e396ed03ab349aa9153a958eb73964f : SUCCESS in 21m 39s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/00c94ecf29224706a8bab4fbe69b5754 : SUCCESS in 18m 35s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/eb9ba6f72b2f4a92b82300f8ff9e31d8 : SUCCESS in 14m 36s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/b935cdada0724493a856570bb90e44b1 : POST_FAILURE in 18m 15s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/b03803b067d641a083a8a7b34c26212b : SUCCESS in 18m 21s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/066c8bd74c634f84922000675745088a : SUCCESS in 21m 25s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/1040176b84fe4b21a778b348e6c2a45b : POST_FAILURE in 14m 29s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/419c4c9a4d29413abfacf741a8d2d020 : SUCCESS in 15m 47s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/51ef46cffe6c4632a1b57ec275952b3f : SUCCESS in 25m 21s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/8efc5b09ad46420b9107f94bbd805c4c : SUCCESS in 7m 41s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/a8330482ddde4509afd127082ddfba18 : SUCCESS in 25m 13s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/9fc97a980ccf4f5cafd9909968973dbb : POST_FAILURE in 8m 29s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/684773db792d41c8bbcc251eec24dc84 : SUCCESS in 51m 37s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/e8a881d56fdb42a9b8ae2e38ef29cfcc : SUCCESS in 57m 08s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/87ca054b57c84c699e4d74d2f7fb7bb9 : SUCCESS in 51m 03s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/cfc5b9d42fe245bd8dfabc4876edc2a4 : SUCCESS in 52m 54s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c6b1cc28f3c04f75ae77ce01594d51b3 : SUCCESS in 17m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9eddf056880341ecab8873ea5540f744 : SUCCESS in 5m 03s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/46aaecb2d907482d80e8885db3a2ac22 : SUCCESS in 19m 28s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/af4fef54cec64614876ad5dffe461523 : SUCCESS in 1h 07m 40s","accounts_in_message":[],"_revision_number":26},{"id":"c8952c82018ac42e382a53e003dc7f5f568b8581","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-12-11 18:59:13.000000000","message":"Uploaded patch set 27.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":27},{"id":"86afd86e6d7ef5804e7deb63480b111a4f70a66a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-11 20:19:05.000000000","message":"Patch Set 27: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f6a5a7b757cb4508875d3b0d4467f7eb\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bf40cfa9485c4e7c806beb3108b4933a : SUCCESS in 9m 33s\n- grenade https://zuul.opendev.org/t/openstack/build/640d81c55cc5496184c999a4b2f4cc28 : SUCCESS in 51m 06s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/d893dd7b3fea4285b811b7f10fdc544e : SUCCESS in 1h 02m 13s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/22458a636a464af2957245fb6e508bf0 : SUCCESS in 59m 10s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8a246da13e98446eb28a50e717592615 : SUCCESS in 1h 10m 20s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c5d34df95fbc46c5b73967ccb6f66047 : SUCCESS in 34m 53s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1b2d015883504c0c8c7caf8fbb0984e1 : SUCCESS in 18m 07s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/402d5b07dd014d52a1cf041e94683e42 : SUCCESS in 22m 59s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/a836e6e4fe9948b1917fdde5bac34b87 : SUCCESS in 9m 35s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/f6aed5a4f6fc483b808bbabbe44945ff : SUCCESS in 6m 50s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/8c0fdf2f68224e4698f223cbd82c064c : SUCCESS in 13m 48s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/2bcd9891809c46098ba269771aefa695 : SUCCESS in 19m 03s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/852641d969d446baafef95af8e37365f : SUCCESS in 14m 33s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/9a8e42f5b61f42df80ed37cdd34efcec : SUCCESS in 18m 23s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/36676bd93b9142d6b91ec692040a29dc : SUCCESS in 22m 33s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/581f59683ead452a90695e342cd26634 : SUCCESS in 19m 30s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/0bc1ac43a26b4679a7855b36db088f77 : SUCCESS in 20m 11s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/6b2c3f94724545b4a8b75d1645c5d9b3 : SUCCESS in 24m 47s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/fb2873c379d64369b03d823b0f4ef46d : SUCCESS in 17m 26s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/d0a2fcb0510642ab9a3b70cb4c80e84e : SUCCESS in 17m 43s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/9d37b1a7eaf9409690f5ee9c9fb957d9 : SUCCESS in 22m 21s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/ee0ae7f24f0b4706b057dee4404bbfdf : SUCCESS in 7m 41s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/2d1166b148354c56b41645ce448a5412 : SUCCESS in 19m 48s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/d1ff18e69f524bb2992ed2a7f3d29093 : SUCCESS in 8m 45s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/39175ac45c954df9b18c34f3814eea34 : SUCCESS in 1h 00m 06s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/7695d3d55fd8450fab00f490f9380c59 : SUCCESS in 1h 05m 10s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/2d2ccbaf4b3b45678b529657fd1d830e : SUCCESS in 46m 45s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/48d404c46f924a35ab9444f016968ec2 : SUCCESS in 51m 03s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/70c4bbacab9f4b0fbfb1753c6a338734 : SUCCESS in 16m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0e29c285133543f3ae9c857b13291dc7 : SUCCESS in 5m 16s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/32b17267087a42aa952d774bb0a81938 : SUCCESS in 25m 22s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f256b09b3d5d42b783c244c91ce1bf5c : SUCCESS in 51m 03s","accounts_in_message":[],"_revision_number":27},{"id":"18cf75614c9b4a16eb790e5eab88ce7c3e240844","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-11 20:24:49.000000000","message":"Patch Set 27:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/42ad0e20e731462a961b388aa2f044ca\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/27db5e1eb3d9442fb6cfcc69a3124f4d : SUCCESS in 17m 51s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/414e81236d8b48778557feb09e2c33fa : SUCCESS in 1h 22m 57s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/33f0ec6e3c004749b3416568eda3d257 : SUCCESS in 24m 57s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/4e557beddfff4833b0224fe87217a091 : SUCCESS in 1h 07m 33s","accounts_in_message":[],"_revision_number":27},{"id":"69255e1e707ef5ee3d617e20900aeef4f0daafe7","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2023-12-14 05:17:38.000000000","message":"Patch Set 27: Code-Review+2\n\n(3 comments)","accounts_in_message":[],"_revision_number":27},{"id":"d8f2a5ff50e7a915d6ba65161463ac68ed94d031","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-12-19 19:06:25.000000000","message":"Patch Set 27:\n\n(3 comments)","accounts_in_message":[],"_revision_number":27},{"id":"6ab4646b2b009bfeb83509efb82970255182bcdd","author":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"date":"2023-12-28 09:23:51.000000000","message":"Patch Set 27: Code-Review+1","accounts_in_message":[],"_revision_number":27},{"id":"6bf784bfcc7abeba9b950e406d534757944eadee","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-01-04 16:29:45.000000000","message":"Uploaded patch set 28.\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":28},{"id":"23a1aef2dbdb36a45b67572ffa1bd0a5f4f75c3f","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-01-04 16:33:04.000000000","message":"Patch Set 28:\n\n(4 comments)","accounts_in_message":[],"_revision_number":28},{"id":"2e32af40eff504ffb063ea01d8973377e24698ae","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-04 17:43:21.000000000","message":"Patch Set 28:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c86f7a20b81744afb133e586f4e7b5dd\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/54608dc3a7904cf2ac09b3ca965c460c : SUCCESS in 19m 08s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8f58008470d7451c8010a10eedab824e : SUCCESS in 17m 31s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/bc8aeae00c634fb98275034efc084313 : SUCCESS in 1h 11m 38s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/6a70c825fc764b5b9b647aab89e1f69c : SUCCESS in 22m 44s","accounts_in_message":[],"_revision_number":28},{"id":"8ccfdfd78dbf6ff5d71d96e72fe75b417c7010ff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-04 17:44:45.000000000","message":"Patch Set 28: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/faa03effb38f44bb846938941617cfc1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a1f6553d69d9430abd2b1c9c8415aa88 : SUCCESS in 8m 15s\n- grenade https://zuul.opendev.org/t/openstack/build/0ab7161dc8a745c2976043ab070b6a5b : SUCCESS in 1h 06m 19s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/a0ff625d231148beb025109895c135b8 : SUCCESS in 1h 01m 05s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/40a8cc3e306246499ad144356dec6f08 : SUCCESS in 36m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/761f4ee5ded542fca245068f2c0fa0cf : SUCCESS in 1h 07m 44s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5686a2298ed44869bbe75d7bc1df890f : SUCCESS in 19m 58s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/67cee88ad3c944f09346c8109cd8043c : SUCCESS in 17m 25s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d7f009614ff1493d929451c45f6ef250 : SUCCESS in 21m 23s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/9d40c8215a924d279e830f7880c23b88 : SUCCESS in 12m 01s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/8a5e52b7d36b45b0928302077537f818 : SUCCESS in 6m 04s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/373316496f4643aeb1fefb3f04022497 : SUCCESS in 14m 34s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/21db2be62ee641b99864d8203cff0d7c : SUCCESS in 18m 32s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/81432a8ecb204f00bc86f3f6c813c8f3 : SUCCESS in 17m 49s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/66554c18b66d4c7898bd113c8b3879ec : SUCCESS in 18m 06s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/9c24ad3abb134d3080f8269837b3abe6 : SUCCESS in 18m 46s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/06ea951f3228445db23f15c3508564df : SUCCESS in 17m 55s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/05d338a2b5a744c383d331965b98add0 : SUCCESS in 18m 02s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/961ee8af17b14d608b60768e2642e3d4 : SUCCESS in 18m 37s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/23f8f5121b304394b0ca08c9eb407463 : SUCCESS in 13m 54s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/c54970f536fd4baab2d2a312a0f7f3e6 : SUCCESS in 15m 00s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/ae34a927d5dc4a928b4ced3220c49b96 : SUCCESS in 15m 41s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/182032072022490aaafd1121ed4d9c43 : SUCCESS in 7m 13s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/206bc9a42ae448d1b855baea10e6ff02 : SUCCESS in 15m 17s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/7f8b6f07a3934d498a34b8192a63f9ca : SUCCESS in 6m 49s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/400740387d884ca6926fd44a95918102 : SUCCESS in 51m 58s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/fe0a75233daf4ae1bebd1c4546171a40 : SUCCESS in 1h 05m 52s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/2ca0ecf6d5fb4536a6c997307c39f7bb : SUCCESS in 49m 27s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/c29ed900c3594ad39debeb723ea97d78 : SUCCESS in 59m 03s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0477fb4399d74d69866617e319432058 : SUCCESS in 20m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aa0a22a2574a4c8ba3814d573159f916 : SUCCESS in 5m 19s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/0d297ad42f7746b4b1e25e3b8eb9d5aa : SUCCESS in 20m 23s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6d5ee102d0f44c8d8db4e42766c4432d : SUCCESS in 46m 43s","accounts_in_message":[],"_revision_number":28},{"id":"270e3d7aa78b09e5c4c9a0d2edca33b00cb5138e","tag":"autogenerated:gerrit:abandon","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-01-05 17:54:54.000000000","message":"Abandoned\n\nsquashed into https://review.opendev.org/c/openstack/swift/+/895602/34","accounts_in_message":[],"_revision_number":28}],"current_revision_number":28,"current_revision":"be14500f77d6e7aaecbc4e5d773b1879f1ee81a0","revisions":{"454625560759519f35c973016a0e4aad17176027":{"kind":"REWORK","_number":1,"created":"2023-11-07 18:36:33.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/50/900350/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/50/900350/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/900350/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/900350/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/900350/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/50/900350/1"}}},"commit":{"parents":[{"commit":"d855d01e2f368ed08039cf9cb2e621f3edd38da1","subject":"Proxy: Use shard-format for auto cont GETs and updating","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d855d01e2f368ed08039cf9cb2e621f3edd38da1"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-11-07 18:17:39.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-11-07 18:36:10.000000000","tz":0},"subject":"proxy: always use namespaces for updating and listing","message":"proxy: always use namespaces for updating and listing\n\nTODO:\n- unit tests are a mess!\n- docstrings\n\nTwo significant changes:\n\n- When handling an object PUT/POST/DELETE, or a container GET *for\n  objects listing*, the proxy should *always* deal in Namespaces when\n  it needs to know about shards. The Namespaces may be read/written\n  from/to memcache. This includes when making GET sub-requests to\n  shards while recursively building an object listing.\n\n  To enable this, the container server and ContainerBroker are\n  modified to support requests for namespaces with query params\n  includes, marker, end_marker and reverse. These are significant the\n  proxy when there is no memcache. (When there is memcache, these\n  params are still sent to the backend but overridden if the container\n  is sharded.)\n\n- The proxy should NOT read or write to memcache when handling a\n  container GET that explicitly requests \u0027shard\u0027 or \u0027object\u0027 record\n  type. A request for \u0027shard\u0027 record type may specify \u0027namespace\u0027\n  format, but this request is unrelated to container listings or\n  object updates and passes directly to the backend.\n\nChange-Id: I2cec1cef5eb34874feea9f2922fd13c4f011e410\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/454625560759519f35c973016a0e4aad17176027"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/454625560759519f35c973016a0e4aad17176027"}]},"branch":"refs/heads/master"},"a04976bca9f58c7580d2b62cd088d11d23d928a8":{"kind":"REWORK","_number":2,"created":"2023-11-08 18:00:31.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/50/900350/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/50/900350/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/900350/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/900350/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/900350/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/50/900350/2"}}},"commit":{"parents":[{"commit":"d855d01e2f368ed08039cf9cb2e621f3edd38da1","subject":"Proxy: Use shard-format for auto cont GETs and updating","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d855d01e2f368ed08039cf9cb2e621f3edd38da1"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-11-07 18:17:39.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-11-08 17:59:51.000000000","tz":0},"subject":"proxy: always use namespaces for updating and listing","message":"proxy: always use namespaces for updating and listing\n\nTODO:\n- unit tests are a mess!\n- docstrings\n\nTwo significant changes:\n\n- When handling an object PUT/POST/DELETE, or a container GET *for\n  objects listing*, the proxy should *always* deal in Namespaces when\n  it needs to know about shards. The Namespaces may be read/written\n  from/to memcache. This includes when making GET sub-requests to\n  shards while recursively building an object listing.\n\n  To enable this, the container server and ContainerBroker are\n  modified to support requests for namespaces with query params\n  includes, marker, end_marker and reverse. These are significant the\n  proxy when there is no memcache. (When there is memcache, these\n  params are still sent to the backend but overridden if the container\n  is sharded.)\n\n- The proxy should NOT read or write to memcache when handling a\n  container GET that explicitly requests \u0027shard\u0027 or \u0027object\u0027 record\n  type. A request for \u0027shard\u0027 record type may specify \u0027namespace\u0027\n  format, but this request is unrelated to container listings or\n  object updates and passes directly to the backend.\n\nChange-Id: I2cec1cef5eb34874feea9f2922fd13c4f011e410\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a04976bca9f58c7580d2b62cd088d11d23d928a8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a04976bca9f58c7580d2b62cd088d11d23d928a8"}]},"branch":"refs/heads/master"},"52ae83fecf9f2a521a66a82d2b44701f8a2082bc":{"kind":"REWORK","_number":3,"created":"2023-11-09 16:45:01.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/50/900350/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/50/900350/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/900350/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/900350/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/900350/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/50/900350/3"}}},"commit":{"parents":[{"commit":"9040ca6b1bae0fcc0752b7e88c5ffb81be0311ca","subject":"sq: refactor test_container","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9040ca6b1bae0fcc0752b7e88c5ffb81be0311ca"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-11-07 18:17:39.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-11-09 16:36:57.000000000","tz":0},"subject":"proxy: always use namespaces for updating and listing","message":"proxy: always use namespaces for updating and listing\n\nTODO:\n- unit tests are a mess!\n- docstrings\n\nTwo significant changes:\n\n- When handling an object PUT/POST/DELETE, or a container GET *for\n  objects listing*, the proxy should *always* deal in Namespaces when\n  it needs to know about shards. The Namespaces may be read/written\n  from/to memcache. This includes when making GET sub-requests to\n  shards while recursively building an object listing.\n\n  To enable this, the container server and ContainerBroker are\n  modified to support requests for namespaces with query params\n  includes, marker, end_marker and reverse. These are significant the\n  proxy when there is no memcache. (When there is memcache, these\n  params are still sent to the backend but overridden if the container\n  is sharded.)\n\n- The proxy should NOT read or write to memcache when handling a\n  container GET that explicitly requests \u0027shard\u0027 or \u0027object\u0027 record\n  type. A request for \u0027shard\u0027 record type may specify \u0027namespace\u0027\n  format, but this request is unrelated to container listings or\n  object updates and passes directly to the backend.\n\nChange-Id: I2cec1cef5eb34874feea9f2922fd13c4f011e410\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/52ae83fecf9f2a521a66a82d2b44701f8a2082bc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/52ae83fecf9f2a521a66a82d2b44701f8a2082bc"}]},"branch":"refs/heads/master"},"d4f5c420f369b9a293db3bca22497e3895605f97":{"kind":"REWORK","_number":4,"created":"2023-11-10 19:01:43.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/50/900350/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/50/900350/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/900350/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/900350/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/900350/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/50/900350/4"}}},"commit":{"parents":[{"commit":"9040ca6b1bae0fcc0752b7e88c5ffb81be0311ca","subject":"sq: refactor test_container","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9040ca6b1bae0fcc0752b7e88c5ffb81be0311ca"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-11-07 18:17:39.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-11-10 19:01:02.000000000","tz":0},"subject":"proxy: always use namespaces for updating and listing","message":"proxy: always use namespaces for updating and listing\n\nTODO:\n- unit tests are a mess!\n- docstrings\n\nTwo significant changes:\n\n- When handling an object PUT/POST/DELETE, or a container GET *for\n  objects listing*, the proxy should *always* deal in Namespaces when\n  it needs to know about shards. The Namespaces may be read/written\n  from/to memcache. This includes when making GET sub-requests to\n  shards while recursively building an object listing.\n\n  To enable this, the container server and ContainerBroker are\n  modified to support requests for namespaces with query params\n  includes, marker, end_marker and reverse. These are significant the\n  proxy when there is no memcache. (When there is memcache, these\n  params are still sent to the backend but overridden if the container\n  is sharded.)\n\n- The proxy should NOT read or write to memcache when handling a\n  container GET that explicitly requests \u0027shard\u0027 or \u0027object\u0027 record\n  type. A request for \u0027shard\u0027 record type may specify \u0027namespace\u0027\n  format, but this request is unrelated to container listings or\n  object updates and passes directly to the backend.\n\nChange-Id: I2cec1cef5eb34874feea9f2922fd13c4f011e410\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d4f5c420f369b9a293db3bca22497e3895605f97"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d4f5c420f369b9a293db3bca22497e3895605f97"}]},"branch":"refs/heads/master"},"8b9925c5a8dfae8c56adb9fb952558b381ce4c0d":{"kind":"REWORK","_number":5,"created":"2023-11-13 19:18:38.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/50/900350/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/50/900350/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/900350/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/900350/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/900350/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/50/900350/5"}}},"commit":{"parents":[{"commit":"9040ca6b1bae0fcc0752b7e88c5ffb81be0311ca","subject":"sq: refactor test_container","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9040ca6b1bae0fcc0752b7e88c5ffb81be0311ca"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-11-07 18:17:39.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-11-13 19:17:34.000000000","tz":0},"subject":"proxy: always use namespaces for updating and listing","message":"proxy: always use namespaces for updating and listing\n\nTODO:\n- check unit test coverage\n- check docstrings\n\nTwo significant changes:\n\n- When handling an object PUT/POST/DELETE, or a container GET *for\n  objects listing*, the proxy should *always* deal in Namespaces when\n  it needs to know about shards. The Namespaces may be read/written\n  from/to memcache. This includes when making GET sub-requests to\n  shards while recursively building an object listing.\n\n  To enable this, the container server and ContainerBroker are\n  modified to support requests for namespaces with query params\n  includes, marker, end_marker and reverse. These are significant the\n  proxy when there is no memcache. (When there is memcache, these\n  params are still sent to the backend but overridden if the container\n  is sharded.)\n\n- The proxy should NOT read or write to memcache when handling a\n  container GET that explicitly requests \u0027shard\u0027 or \u0027object\u0027 record\n  type. A request for \u0027shard\u0027 record type may specify \u0027namespace\u0027\n  format, but this request is unrelated to container listings or\n  object updates and passes directly to the backend.\n\nChange-Id: I2cec1cef5eb34874feea9f2922fd13c4f011e410\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8b9925c5a8dfae8c56adb9fb952558b381ce4c0d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8b9925c5a8dfae8c56adb9fb952558b381ce4c0d"}]},"branch":"refs/heads/master"},"d9a9780436b9b45cec1689551acb9248e1aa42a9":{"kind":"REWORK","_number":6,"created":"2023-11-14 04:23:51.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/50/900350/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/50/900350/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/900350/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/900350/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/900350/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/50/900350/6"}}},"commit":{"parents":[{"commit":"2f1ee006af81149d7b7b546a367bf3daccf4f249","subject":"Proxy: Use shard-format for auto cont GETs and updating","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2f1ee006af81149d7b7b546a367bf3daccf4f249"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-11-07 18:17:39.000000000","tz":0},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2023-11-14 04:23:15.000000000","tz":660},"subject":"proxy: always use namespaces for updating and listing","message":"proxy: always use namespaces for updating and listing\n\nTODO:\n- check unit test coverage\n- check docstrings\n\nTwo significant changes:\n\n- When handling an object PUT/POST/DELETE, or a container GET *for\n  objects listing*, the proxy should *always* deal in Namespaces when\n  it needs to know about shards. The Namespaces may be read/written\n  from/to memcache. This includes when making GET sub-requests to\n  shards while recursively building an object listing.\n\n  To enable this, the container server and ContainerBroker are\n  modified to support requests for namespaces with query params\n  includes, marker, end_marker and reverse. These are significant the\n  proxy when there is no memcache. (When there is memcache, these\n  params are still sent to the backend but overridden if the container\n  is sharded.)\n\n- The proxy should NOT read or write to memcache when handling a\n  container GET that explicitly requests \u0027shard\u0027 or \u0027object\u0027 record\n  type. A request for \u0027shard\u0027 record type may specify \u0027namespace\u0027\n  format, but this request is unrelated to container listings or\n  object updates and passes directly to the backend.\n\nChange-Id: I2cec1cef5eb34874feea9f2922fd13c4f011e410\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d9a9780436b9b45cec1689551acb9248e1aa42a9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d9a9780436b9b45cec1689551acb9248e1aa42a9"}]},"branch":"refs/heads/master"},"0efb8cbde14f1c3e62995f1f00b2effea0b96ee4":{"kind":"REWORK","_number":7,"created":"2023-11-14 18:46:54.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/50/900350/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/50/900350/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/900350/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/900350/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/900350/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/50/900350/7"}}},"commit":{"parents":[{"commit":"2f1ee006af81149d7b7b546a367bf3daccf4f249","subject":"Proxy: Use shard-format for auto cont GETs and updating","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2f1ee006af81149d7b7b546a367bf3daccf4f249"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-11-07 18:17:39.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-11-14 18:45:29.000000000","tz":0},"subject":"proxy: always use namespaces for updating and listing","message":"proxy: always use namespaces for updating and listing\n\nTODO:\n- check unit test coverage\n- check docstrings\n\nTwo significant changes:\n\n- When handling an object PUT/POST/DELETE, or a container GET *for\n  objects listing*, the proxy should *always* deal in Namespaces when\n  it needs to know about shards. The Namespaces may be read/written\n  from/to memcache. This includes when making GET sub-requests to\n  shards while recursively building an object listing.\n\n  To enable this, the container server and ContainerBroker are\n  modified to support requests for namespaces with query params\n  includes, marker, end_marker and reverse. These are significant to\n  the proxy when there is no memcache. (When there is memcache, these\n  params are still sent to the backend but overridden if the container\n  is sharded.)\n\n- The proxy should NOT read or write to memcache when handling a\n  container GET that explicitly requests \u0027shard\u0027 or \u0027object\u0027 record\n  type. A request for \u0027shard\u0027 record type may specify \u0027namespace\u0027\n  format, but this request is unrelated to container listings or\n  object updates and passes directly to the backend.\n\nChange-Id: I2cec1cef5eb34874feea9f2922fd13c4f011e410\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0efb8cbde14f1c3e62995f1f00b2effea0b96ee4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0efb8cbde14f1c3e62995f1f00b2effea0b96ee4"}]},"branch":"refs/heads/master"},"62b4c5993b4615d5d5f6b55fdd5996a0a785cac5":{"kind":"REWORK","_number":8,"created":"2023-11-15 06:20:52.000000000","uploader":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"ref":"refs/changes/50/900350/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/50/900350/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/900350/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/900350/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/900350/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/50/900350/8"}}},"commit":{"parents":[{"commit":"d14bb63bba0e1983a3b1c925fd6f6e85da95c039","subject":"container: add Namespace marker/end_marker/include/reverse support.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d14bb63bba0e1983a3b1c925fd6f6e85da95c039"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-11-07 18:17:39.000000000","tz":0},"committer":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2023-11-15 06:20:02.000000000","tz":-480},"subject":"proxy: always use namespaces for updating and listing","message":"proxy: always use namespaces for updating and listing\n\nTODO:\n- check unit test coverage\n- check docstrings\n\nTwo significant changes:\n\n- When handling an object PUT/POST/DELETE, or a container GET *for\n  objects listing*, the proxy should *always* deal in Namespaces when\n  it needs to know about shards. The Namespaces may be read/written\n  from/to memcache. This includes when making GET sub-requests to\n  shards while recursively building an object listing.\n\n  To enable this, the container server and ContainerBroker are\n  modified to support requests for namespaces with query params\n  includes, marker, end_marker and reverse. These are significant to\n  the proxy when there is no memcache. (When there is memcache, these\n  params are still sent to the backend but overridden if the container\n  is sharded.)\n\n- The proxy should NOT read or write to memcache when handling a\n  container GET that explicitly requests \u0027shard\u0027 or \u0027object\u0027 record\n  type. A request for \u0027shard\u0027 record type may specify \u0027namespace\u0027\n  format, but this request is unrelated to container listings or\n  object updates and passes directly to the backend.\n\nChange-Id: I2cec1cef5eb34874feea9f2922fd13c4f011e410\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/62b4c5993b4615d5d5f6b55fdd5996a0a785cac5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/62b4c5993b4615d5d5f6b55fdd5996a0a785cac5"}]},"branch":"refs/heads/master"},"45d8510a56e9a4fd40cfab3949f78956dac1d80a":{"kind":"REWORK","_number":9,"created":"2023-11-16 12:02:33.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/50/900350/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/50/900350/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/900350/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/900350/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/900350/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/50/900350/9"}}},"commit":{"parents":[{"commit":"b67c88714a4da72235f61ffa0bd2cf024bc0a76f","subject":"container: add Namespace marker/end_marker/include/reverse support.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b67c88714a4da72235f61ffa0bd2cf024bc0a76f"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-11-07 18:17:39.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-11-16 12:01:30.000000000","tz":0},"subject":"proxy: always use namespaces for updating and listing","message":"proxy: always use namespaces for updating and listing\n\nTwo significant changes:\n\n- When handling an object PUT/POST/DELETE, or a container GET *for\n  objects listing*, the proxy should *always* deal in Namespaces when\n  it needs to know about shards. The Namespaces may be read/written\n  from/to memcache. This includes when making GET sub-requests to\n  shards while recursively building an object listing.\n\n  To enable this, the container server and ContainerBroker are\n  modified to support requests for namespaces with query params\n  includes, marker, end_marker and reverse. These are significant to\n  the proxy when there is no memcache. (When there is memcache, these\n  params are still sent to the backend but overridden if the container\n  is sharded.)\n\n- The proxy should NOT read or write to memcache when handling a\n  container GET that explicitly requests \u0027shard\u0027 or \u0027object\u0027 record\n  type. A request for \u0027shard\u0027 record type may specify \u0027namespace\u0027\n  format, but this request is unrelated to container listings or\n  object updates and passes directly to the backend.\n\nChange-Id: I2cec1cef5eb34874feea9f2922fd13c4f011e410\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/45d8510a56e9a4fd40cfab3949f78956dac1d80a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/45d8510a56e9a4fd40cfab3949f78956dac1d80a"}]},"branch":"refs/heads/master"},"7f54b63a1eb7ca7c5b85a62acf7701e18d29b65d":{"kind":"REWORK","_number":10,"created":"2023-11-17 19:01:13.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/50/900350/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/50/900350/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/900350/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/900350/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/900350/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/50/900350/10"}}},"commit":{"parents":[{"commit":"874efb7173cfe6dc640282ed286a9fbb7e854b2b","subject":"proxy: move _get_shard_ranges to ObjectController","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/874efb7173cfe6dc640282ed286a9fbb7e854b2b"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-11-17 16:09:41.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-11-17 18:53:34.000000000","tz":0},"subject":"proxy: always use namespaces for updating and listing","message":"proxy: always use namespaces for updating and listing\n\nThis patch attempts to remove confusion as to when the proxy\ncontrollers are handling Namespace objects (or dict representations of\nnamespaces) vs ShardRanges.\n\nWith this patch, the proxy no longer uses ShardRanges or their dict\nrepresentation when dealing with namespaces that have been cached for\nupdating or listing states. Up-to-date container servers will return\nmore compact namespaces, and full format shard range dicts returned\nfrom older container servers are parsed as Namespaces by the proxy.\n\nMuch of this is simply about using consistent and accurate\nnames. However, in some circumstances, ShardRange objects were\npreviously instantiated when not absolutely necessary: these are\nreplaced with Namespace objects.\n\nThe ObjectController would previously specify that the backend should\nreturn full format ShardRanges when asked for updating state shard\nranges *with an includes parameter*. This isn\u0027t necessary with a\nmodern backend which now supports the \u0027includes\u0027 parameter with\n\u0027namespace\u0027 format requests. This wasn\u0027t strictly necessary with a\nlegacy backend, which would have returned full format ShardRanges\nanyway if an \u0027includes\u0027 parameter was sent.\n\nThe ContainerController._get_from_shards method needs a list of\nnamespaces to build a sharded listing. It is currently passed a\nresponse which has a list of either namespaces or full shard range\ndicts, depending on what the backend server returned. The list of\ndicts is converted to a list of ShardRange objects. This is convenient\nbecause the ShardRange conveniently splits the namespace name into\naccount and container parts, but it is also confusing given that the\nlist will often be of namespace dicts.\n\nThis patch refactors _get_from_shards to clarify that it does not\nrequire ShardRange objects. The method is now passed a list of\nnamespaces, which is parsed from the response body before the method\nis called.\n\nSome unit tests are also refactored to be more realistic when mocking\n_get_from_shards.\n\nChange-Id: I2cec1cef5eb34874feea9f2922fd13c4f011e410\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7f54b63a1eb7ca7c5b85a62acf7701e18d29b65d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7f54b63a1eb7ca7c5b85a62acf7701e18d29b65d"}]},"branch":"refs/heads/master"},"934fe827cb7120834769f2b623da9476d91138eb":{"kind":"REWORK","_number":11,"created":"2023-11-17 19:14:46.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/50/900350/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/50/900350/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/900350/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/900350/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/900350/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/50/900350/11"}}},"commit":{"parents":[{"commit":"1bc35ca77f09840562cb27364a0cc3e9827ccc4a","subject":"proxy: move _get_shard_ranges to ObjectController","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1bc35ca77f09840562cb27364a0cc3e9827ccc4a"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-11-17 16:09:41.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-11-17 19:11:44.000000000","tz":0},"subject":"proxy: always use namespaces for updating and listing","message":"proxy: always use namespaces for updating and listing\n\nThis patch attempts to remove confusion as to when the proxy\ncontrollers are handling Namespace objects (or dict representations of\nnamespaces) vs ShardRanges.\n\nWith this patch, the proxy no longer uses ShardRanges or their dict\nrepresentation when dealing with namespaces that have been cached for\nupdating or listing states. Up-to-date container servers will return\nmore compact namespaces, and full format shard range dicts returned\nfrom older container servers are parsed as Namespaces by the proxy.\n\nMuch of this is simply about using consistent and accurate\nnames. However, in some circumstances, ShardRange objects were\npreviously instantiated when not absolutely necessary: these are\nreplaced with Namespace objects.\n\nThe ObjectController would previously specify that the backend should\nreturn full format ShardRanges when asked for updating state shard\nranges *with an includes parameter*. This isn\u0027t necessary with a\nmodern backend which now supports the \u0027includes\u0027 parameter with\n\u0027namespace\u0027 format requests. This wasn\u0027t strictly necessary with a\nlegacy backend, which would have returned full format ShardRanges\nanyway if an \u0027includes\u0027 parameter was sent.\n\nThe ContainerController._get_from_shards method needs a list of\nnamespaces to build a sharded listing. It is currently passed a\nresponse which has a list of either namespaces or full shard range\ndicts, depending on what the backend server returned. The list of\ndicts is converted to a list of ShardRange objects. This is convenient\nbecause the ShardRange conveniently splits the namespace name into\naccount and container parts, but it is also confusing given that the\nlist will often be of namespace dicts.\n\nThis patch refactors _get_from_shards to clarify that it does not\nrequire ShardRange objects. The method is now passed a list of\nnamespaces, which is parsed from the response body before the method\nis called.\n\nSome unit tests are also refactored to be more realistic when mocking\n_get_from_shards.\n\nChange-Id: I2cec1cef5eb34874feea9f2922fd13c4f011e410\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/934fe827cb7120834769f2b623da9476d91138eb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/934fe827cb7120834769f2b623da9476d91138eb"}]},"branch":"refs/heads/master"},"38a21feddd570a1abb9c41119a76d7b02930245c":{"kind":"REWORK","_number":12,"created":"2023-11-20 16:47:37.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/50/900350/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/50/900350/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/900350/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/900350/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/900350/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/50/900350/12"}}},"commit":{"parents":[{"commit":"612e601b0c1bd36109a28747eec8d3d1f60c52dd","subject":"proxy: move _get_shard_ranges to ObjectController","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/612e601b0c1bd36109a28747eec8d3d1f60c52dd"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-11-17 16:09:41.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-11-20 14:32:52.000000000","tz":0},"subject":"proxy: always use namespaces for updating and listing","message":"proxy: always use namespaces for updating and listing\n\nThis patch attempts to remove confusion as to when the proxy\ncontrollers are handling Namespace objects (or dict representations of\nnamespaces) vs ShardRanges.\n\nWith this patch, the proxy no longer uses ShardRanges or their dict\nrepresentation when dealing with namespaces that have been cached for\nupdating or listing states. Up-to-date container servers will return\nmore compact namespaces, and full format shard range dicts returned\nfrom older container servers are parsed as Namespaces by the proxy.\n\nMuch of this is simply about using consistent and accurate\nnames. However, in some circumstances, ShardRange objects were\npreviously instantiated when not absolutely necessary: these are\nreplaced with Namespace objects.\n\nThe ObjectController would previously specify that the backend should\nreturn full format ShardRanges when asked for updating state shard\nranges *with an includes parameter*. This isn\u0027t necessary with a\nmodern backend which now supports the \u0027includes\u0027 parameter with\n\u0027namespace\u0027 format requests. This wasn\u0027t strictly necessary with a\nlegacy backend, which would have returned full format ShardRanges\nanyway if an \u0027includes\u0027 parameter was sent.\n\nThe ContainerController._get_from_shards method needs a list of\nnamespaces to build a sharded listing. It is currently passed a\nresponse which has a list of either namespaces or full shard range\ndicts, depending on what the backend server returned. The list of\ndicts is converted to a list of ShardRange objects. This is convenient\nbecause the ShardRange conveniently splits the namespace name into\naccount and container parts, but it is also confusing given that the\nlist will often be of namespace dicts.\n\nThis patch refactors _get_from_shards to clarify that it does not\nrequire ShardRange objects. The method is now passed a list of\nnamespaces, which is parsed from the response body before the method\nis called.\n\nSome unit tests are also refactored to be more realistic when mocking\n_get_from_shards.\n\nChange-Id: I2cec1cef5eb34874feea9f2922fd13c4f011e410\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/38a21feddd570a1abb9c41119a76d7b02930245c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/38a21feddd570a1abb9c41119a76d7b02930245c"}]},"branch":"refs/heads/master"},"e4de470b690ca40491ae178764e2cc2d4e8e8018":{"kind":"REWORK","_number":13,"created":"2023-11-20 18:37:10.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/50/900350/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/50/900350/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/900350/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/900350/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/900350/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/50/900350/13"}}},"commit":{"parents":[{"commit":"2ba8ef1a715fd9cf695733a5583b1afd00bb5543","subject":"DNM proxy: GET shards probe test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2ba8ef1a715fd9cf695733a5583b1afd00bb5543"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-11-17 16:09:41.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-11-20 18:36:42.000000000","tz":0},"subject":"proxy: always use namespaces for updating and listing","message":"proxy: always use namespaces for updating and listing\n\nThis patch attempts to remove confusion as to when the proxy\ncontrollers are handling Namespace objects (or dict representations of\nnamespaces) vs ShardRanges.\n\nWith this patch, the proxy no longer uses ShardRanges or their dict\nrepresentation when dealing with namespaces that have been cached for\nupdating or listing states. Up-to-date container servers will return\nmore compact namespaces, and full format shard range dicts returned\nfrom older container servers are parsed as Namespaces by the proxy.\n\nMuch of this is simply about using consistent and accurate\nnames. However, in some circumstances, ShardRange objects were\npreviously instantiated when not absolutely necessary: these are\nreplaced with Namespace objects.\n\nThe ObjectController would previously specify that the backend should\nreturn full format ShardRanges when asked for updating state shard\nranges *with an includes parameter*. This isn\u0027t necessary with a\nmodern backend which now supports the \u0027includes\u0027 parameter with\n\u0027namespace\u0027 format requests. This wasn\u0027t strictly necessary with a\nlegacy backend, which would have returned full format ShardRanges\nanyway if an \u0027includes\u0027 parameter was sent.\n\nThe ContainerController._get_from_shards method needs a list of\nnamespaces to build a sharded listing. It is currently passed a\nresponse which has a list of either namespaces or full shard range\ndicts, depending on what the backend server returned. The list of\ndicts is converted to a list of ShardRange objects. This is convenient\nbecause the ShardRange conveniently splits the namespace name into\naccount and container parts, but it is also confusing given that the\nlist will often be of namespace dicts.\n\nThis patch refactors _get_from_shards to clarify that it does not\nrequire ShardRange objects. The method is now passed a list of\nnamespaces, which is parsed from the response body before the method\nis called.\n\nA bug is fixed in _get_from_shards: it now always sets the\nX-Backend-Record-Type header to \u0027object\u0027, rather than sometimes\nleaving the value as \u0027shard\u0027.\n\nSome unit tests are also refactored to be more realistic when mocking\n_get_from_shards.\n\nChange-Id: I2cec1cef5eb34874feea9f2922fd13c4f011e410\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e4de470b690ca40491ae178764e2cc2d4e8e8018"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e4de470b690ca40491ae178764e2cc2d4e8e8018"}]},"branch":"refs/heads/master"},"7cb6b2bb06f85f760e54c2ba45ada51a95690709":{"kind":"TRIVIAL_REBASE","_number":14,"created":"2023-11-21 10:54:11.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/50/900350/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/50/900350/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/900350/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/900350/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/900350/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/50/900350/14"}}},"commit":{"parents":[{"commit":"3c21853fbf7ebb08d5af2d66bdb17693cf249495","subject":"DNM proxy: GET shards probe test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3c21853fbf7ebb08d5af2d66bdb17693cf249495"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-11-17 16:09:41.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-11-21 10:30:32.000000000","tz":0},"subject":"proxy: always use namespaces for updating and listing","message":"proxy: always use namespaces for updating and listing\n\nThis patch attempts to remove confusion as to when the proxy\ncontrollers are handling Namespace objects (or dict representations of\nnamespaces) vs ShardRanges.\n\nWith this patch, the proxy no longer uses ShardRanges or their dict\nrepresentation when dealing with namespaces that have been cached for\nupdating or listing states. Up-to-date container servers will return\nmore compact namespaces, and full format shard range dicts returned\nfrom older container servers are parsed as Namespaces by the proxy.\n\nMuch of this is simply about using consistent and accurate\nnames. However, in some circumstances, ShardRange objects were\npreviously instantiated when not absolutely necessary: these are\nreplaced with Namespace objects.\n\nThe ObjectController would previously specify that the backend should\nreturn full format ShardRanges when asked for updating state shard\nranges *with an includes parameter*. This isn\u0027t necessary with a\nmodern backend which now supports the \u0027includes\u0027 parameter with\n\u0027namespace\u0027 format requests. This wasn\u0027t strictly necessary with a\nlegacy backend, which would have returned full format ShardRanges\nanyway if an \u0027includes\u0027 parameter was sent.\n\nThe ContainerController._get_from_shards method needs a list of\nnamespaces to build a sharded listing. It is currently passed a\nresponse which has a list of either namespaces or full shard range\ndicts, depending on what the backend server returned. The list of\ndicts is converted to a list of ShardRange objects. This is convenient\nbecause the ShardRange conveniently splits the namespace name into\naccount and container parts, but it is also confusing given that the\nlist will often be of namespace dicts.\n\nThis patch refactors _get_from_shards to clarify that it does not\nrequire ShardRange objects. The method is now passed a list of\nnamespaces, which is parsed from the response body before the method\nis called.\n\nA bug is fixed in _get_from_shards: it now always sets the\nX-Backend-Record-Type header to \u0027object\u0027, rather than sometimes\nleaving the value as \u0027shard\u0027.\n\nSome unit tests are also refactored to be more realistic when mocking\n_get_from_shards.\n\nChange-Id: I2cec1cef5eb34874feea9f2922fd13c4f011e410\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7cb6b2bb06f85f760e54c2ba45ada51a95690709"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7cb6b2bb06f85f760e54c2ba45ada51a95690709"}]},"branch":"refs/heads/master"},"ee29faaa2d7dde2764a32938e904a0173dfcd25e":{"kind":"REWORK","_number":15,"created":"2023-11-21 18:37:29.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/50/900350/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/50/900350/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/900350/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/900350/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/900350/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/50/900350/15"}}},"commit":{"parents":[{"commit":"df4f66bc62bd18fa289759c1309d2776bca0a04f","subject":"DNM proxy: GET shards probe test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/df4f66bc62bd18fa289759c1309d2776bca0a04f"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-11-17 16:09:41.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-11-21 18:35:39.000000000","tz":0},"subject":"proxy: always use namespaces for updating and listing","message":"proxy: always use namespaces for updating and listing\n\nThis patch attempts to remove confusion as to when the proxy\ncontrollers are handling Namespace objects (or dict representations of\nnamespaces) vs ShardRanges.\n\nWith this patch, the proxy no longer uses ShardRanges or their dict\nrepresentation when dealing with namespaces that have been cached for\nupdating or listing states. Up-to-date container servers will return\nmore compact namespaces, and full format shard range dicts returned\nfrom older container servers are parsed as Namespaces by the proxy.\n\nMuch of this is simply about using consistent and accurate\nnames. However, in some circumstances, ShardRange objects were\npreviously instantiated when not absolutely necessary: these are\nreplaced with Namespace objects.\n\nThe ObjectController would previously specify that the backend should\nreturn full format ShardRanges when asked for updating state shard\nranges *with an includes parameter*. This isn\u0027t necessary with a\nmodern backend which now supports the \u0027includes\u0027 parameter with\n\u0027namespace\u0027 format requests. This wasn\u0027t strictly necessary with a\nlegacy backend, which would have returned full format ShardRanges\nanyway if an \u0027includes\u0027 parameter was sent.\n\nThe ContainerController._get_from_shards method needs a list of\nnamespaces to build a sharded listing. It is currently passed a\nresponse which has a list of either namespaces or full shard range\ndicts, depending on what the backend server returned. The list of\ndicts is converted to a list of ShardRange objects. This is convenient\nbecause the ShardRange conveniently splits the namespace name into\naccount and container parts, but it is also confusing given that the\nlist will often be of namespace dicts.\n\nThis patch refactors _get_from_shards to clarify that it does not\nrequire ShardRange objects. The method is now passed a list of\nnamespaces, which is parsed from the response body before the method\nis called.\n\nA bug is fixed in _get_from_shards: it now always sets the\nX-Backend-Record-Type header to \u0027object\u0027, rather than sometimes\nleaving the value as \u0027shard\u0027.\n\nSome unit tests are also refactored to be more realistic when mocking\n_get_from_shards.\n\nChange-Id: I2cec1cef5eb34874feea9f2922fd13c4f011e410\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ee29faaa2d7dde2764a32938e904a0173dfcd25e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ee29faaa2d7dde2764a32938e904a0173dfcd25e"}]},"branch":"refs/heads/master"},"139c51e53307a8a9cb6c15f6ac4a67df492bc68c":{"kind":"TRIVIAL_REBASE","_number":16,"created":"2023-11-22 11:07:53.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/50/900350/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/50/900350/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/900350/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/900350/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/900350/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/50/900350/16"}}},"commit":{"parents":[{"commit":"f0194063b8e8923f49bdfe1d3511934e190e29ea","subject":"DNM proxy: GET shards probe test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f0194063b8e8923f49bdfe1d3511934e190e29ea"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-11-17 16:09:41.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-11-22 11:07:31.000000000","tz":0},"subject":"proxy: always use namespaces for updating and listing","message":"proxy: always use namespaces for updating and listing\n\nThis patch attempts to remove confusion as to when the proxy\ncontrollers are handling Namespace objects (or dict representations of\nnamespaces) vs ShardRanges.\n\nWith this patch, the proxy no longer uses ShardRanges or their dict\nrepresentation when dealing with namespaces that have been cached for\nupdating or listing states. Up-to-date container servers will return\nmore compact namespaces, and full format shard range dicts returned\nfrom older container servers are parsed as Namespaces by the proxy.\n\nMuch of this is simply about using consistent and accurate\nnames. However, in some circumstances, ShardRange objects were\npreviously instantiated when not absolutely necessary: these are\nreplaced with Namespace objects.\n\nThe ObjectController would previously specify that the backend should\nreturn full format ShardRanges when asked for updating state shard\nranges *with an includes parameter*. This isn\u0027t necessary with a\nmodern backend which now supports the \u0027includes\u0027 parameter with\n\u0027namespace\u0027 format requests. This wasn\u0027t strictly necessary with a\nlegacy backend, which would have returned full format ShardRanges\nanyway if an \u0027includes\u0027 parameter was sent.\n\nThe ContainerController._get_from_shards method needs a list of\nnamespaces to build a sharded listing. It is currently passed a\nresponse which has a list of either namespaces or full shard range\ndicts, depending on what the backend server returned. The list of\ndicts is converted to a list of ShardRange objects. This is convenient\nbecause the ShardRange conveniently splits the namespace name into\naccount and container parts, but it is also confusing given that the\nlist will often be of namespace dicts.\n\nThis patch refactors _get_from_shards to clarify that it does not\nrequire ShardRange objects. The method is now passed a list of\nnamespaces, which is parsed from the response body before the method\nis called.\n\nA bug is fixed in _get_from_shards: it now always sets the\nX-Backend-Record-Type header to \u0027object\u0027, rather than sometimes\nleaving the value as \u0027shard\u0027.\n\nSome unit tests are also refactored to be more realistic when mocking\n_get_from_shards.\n\nChange-Id: I2cec1cef5eb34874feea9f2922fd13c4f011e410\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/139c51e53307a8a9cb6c15f6ac4a67df492bc68c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/139c51e53307a8a9cb6c15f6ac4a67df492bc68c"}]},"branch":"refs/heads/master"},"6af59026dd9c2d0f126eaf07047a25a9d555d58b":{"kind":"REWORK","_number":17,"created":"2023-11-22 12:52:13.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/50/900350/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/50/900350/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/900350/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/900350/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/900350/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/50/900350/17"}}},"commit":{"parents":[{"commit":"f0194063b8e8923f49bdfe1d3511934e190e29ea","subject":"DNM proxy: GET shards probe test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f0194063b8e8923f49bdfe1d3511934e190e29ea"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-11-17 16:09:41.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-11-22 12:47:35.000000000","tz":0},"subject":"proxy: always use namespaces for updating and listing","message":"proxy: always use namespaces for updating and listing\n\nThis patch attempts to remove confusion as to when the proxy\ncontrollers are handling Namespace objects (or dict representations of\nnamespaces) vs ShardRanges.\n\nWith this patch, the proxy no longer uses ShardRanges or their dict\nrepresentation when dealing with namespaces that have been cached for\nupdating or listing states. Up-to-date container servers will return\nmore compact namespaces, and full format shard range dicts returned\nfrom older container servers are parsed as Namespaces by the proxy.\n\nMuch of this is simply about using consistent and accurate\nnames. However, in some circumstances, ShardRange objects were\npreviously instantiated when not absolutely necessary: these are\nreplaced with Namespace objects.\n\nThe ObjectController would previously specify that the backend should\nreturn full format ShardRanges when asked for updating state shard\nranges *with an includes parameter*. This isn\u0027t necessary with a\nmodern backend which now supports the \u0027includes\u0027 parameter with\n\u0027namespace\u0027 format requests. This wasn\u0027t strictly necessary with a\nlegacy backend, which would have returned full format ShardRanges\nanyway if an \u0027includes\u0027 parameter was sent.\n\nThe ContainerController._get_from_shards method needs a list of\nnamespaces to build a sharded listing. It is currently passed a\nresponse which has a list of either namespaces or full shard range\ndicts, depending on what the backend server returned. The list of\ndicts is converted to a list of ShardRange objects. This is convenient\nbecause the ShardRange conveniently splits the namespace name into\naccount and container parts, but it is also confusing given that the\nlist will often be of namespace dicts.\n\nThis patch refactors _get_from_shards to clarify that it does not\nrequire ShardRange objects. The method is now passed a list of\nnamespaces, which is parsed from the response body before the method\nis called.\n\nA bug is fixed in _get_from_shards: it now always sets the\nX-Backend-Record-Type header to \u0027object\u0027, rather than sometimes\nleaving the value as \u0027shard\u0027.\n\nSome unit tests are also refactored to be more realistic when mocking\n_get_from_shards.\n\nChange-Id: I2cec1cef5eb34874feea9f2922fd13c4f011e410\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6af59026dd9c2d0f126eaf07047a25a9d555d58b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6af59026dd9c2d0f126eaf07047a25a9d555d58b"}]},"branch":"refs/heads/master"},"8bd68b0cfc7d63d26f8abe4c041f9673202033ee":{"kind":"TRIVIAL_REBASE","_number":18,"created":"2023-11-29 21:57:11.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/50/900350/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/50/900350/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/900350/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/900350/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/900350/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/50/900350/18"}}},"commit":{"parents":[{"commit":"37086f56f42e7c968e60090a1ceab1f591e3b070","subject":"sq: do not require auto_shard","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/37086f56f42e7c968e60090a1ceab1f591e3b070"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-11-17 16:09:41.000000000","tz":0},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2023-11-29 21:39:23.000000000","tz":-360},"subject":"proxy: always use namespaces for updating and listing","message":"proxy: always use namespaces for updating and listing\n\nThis patch attempts to remove confusion as to when the proxy\ncontrollers are handling Namespace objects (or dict representations of\nnamespaces) vs ShardRanges.\n\nWith this patch, the proxy no longer uses ShardRanges or their dict\nrepresentation when dealing with namespaces that have been cached for\nupdating or listing states. Up-to-date container servers will return\nmore compact namespaces, and full format shard range dicts returned\nfrom older container servers are parsed as Namespaces by the proxy.\n\nMuch of this is simply about using consistent and accurate\nnames. However, in some circumstances, ShardRange objects were\npreviously instantiated when not absolutely necessary: these are\nreplaced with Namespace objects.\n\nThe ObjectController would previously specify that the backend should\nreturn full format ShardRanges when asked for updating state shard\nranges *with an includes parameter*. This isn\u0027t necessary with a\nmodern backend which now supports the \u0027includes\u0027 parameter with\n\u0027namespace\u0027 format requests. This wasn\u0027t strictly necessary with a\nlegacy backend, which would have returned full format ShardRanges\nanyway if an \u0027includes\u0027 parameter was sent.\n\nThe ContainerController._get_from_shards method needs a list of\nnamespaces to build a sharded listing. It is currently passed a\nresponse which has a list of either namespaces or full shard range\ndicts, depending on what the backend server returned. The list of\ndicts is converted to a list of ShardRange objects. This is convenient\nbecause the ShardRange conveniently splits the namespace name into\naccount and container parts, but it is also confusing given that the\nlist will often be of namespace dicts.\n\nThis patch refactors _get_from_shards to clarify that it does not\nrequire ShardRange objects. The method is now passed a list of\nnamespaces, which is parsed from the response body before the method\nis called.\n\nA bug is fixed in _get_from_shards: it now always sets the\nX-Backend-Record-Type header to \u0027object\u0027, rather than sometimes\nleaving the value as \u0027shard\u0027.\n\nSome unit tests are also refactored to be more realistic when mocking\n_get_from_shards.\n\nChange-Id: I2cec1cef5eb34874feea9f2922fd13c4f011e410\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8bd68b0cfc7d63d26f8abe4c041f9673202033ee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8bd68b0cfc7d63d26f8abe4c041f9673202033ee"}]},"branch":"refs/heads/master"},"584bb2c3d8432e42d53bfd8325f205044fc19911":{"kind":"REWORK","_number":19,"created":"2023-11-30 15:39:33.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/50/900350/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/50/900350/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/900350/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/900350/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/900350/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/50/900350/19"}}},"commit":{"parents":[{"commit":"e4c8d3de8946c65e76797cec9b069e4852074faf","subject":"proxy: set x-backend-record-type\u003dobject with object listing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e4c8d3de8946c65e76797cec9b069e4852074faf"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-11-17 16:09:41.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-11-30 14:33:50.000000000","tz":0},"subject":"proxy: always use namespaces for updating and listing","message":"proxy: always use namespaces for updating and listing\n\nThis patch attempts to remove confusion as to when the proxy\ncontrollers are handling Namespace objects (or dict representations of\nnamespaces) vs ShardRanges.\n\nWith this patch, the proxy no longer uses ShardRanges or their dict\nrepresentation when dealing with namespaces that have been cached for\nupdating or listing states. Up-to-date container servers will return\nmore compact namespaces, and full format shard range dicts returned\nfrom older container servers are parsed as Namespaces by the proxy.\n\nMuch of this is simply about using consistent and accurate\nnames. However, in some circumstances, ShardRange objects were\npreviously instantiated when not absolutely necessary: these are\nreplaced with Namespace objects.\n\nThe ObjectController would previously specify that the backend should\nreturn full format ShardRanges when asked for updating state shard\nranges *with an includes parameter*. This isn\u0027t necessary with a\nmodern backend which now supports the \u0027includes\u0027 parameter with\n\u0027namespace\u0027 format requests. This wasn\u0027t strictly necessary with a\nlegacy backend, which would have returned full format ShardRanges\nanyway if an \u0027includes\u0027 parameter was sent.\n\nThe ContainerController._get_from_shards method needs a list of\nnamespaces to build a sharded listing. It is currently passed a\nresponse which has a list of either namespaces or full shard range\ndicts, depending on what the backend server returned. The list of\ndicts is converted to a list of ShardRange objects. This is convenient\nbecause the ShardRange conveniently splits the namespace name into\naccount and container parts, but it is also confusing given that the\nlist will often be of namespace dicts.\n\nThis patch refactors _get_from_shards to clarify that it does not\nrequire ShardRange objects. The method is now passed a list of\nnamespaces, which is parsed from the response body before the method\nis called.\n\nSome unit tests are also refactored to be more realistic when mocking\n_get_from_shards.\n\nChange-Id: I2cec1cef5eb34874feea9f2922fd13c4f011e410\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/584bb2c3d8432e42d53bfd8325f205044fc19911"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/584bb2c3d8432e42d53bfd8325f205044fc19911"}]},"branch":"refs/heads/master"},"0cfd884b6c4b175b26203d5384a964b62c302b1d":{"kind":"REWORK","_number":20,"created":"2023-11-30 18:20:14.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/50/900350/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/50/900350/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/900350/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/900350/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/900350/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/50/900350/20"}}},"commit":{"parents":[{"commit":"b1c2dc1d6da2c0ae2ba6f128c7cf3b21c9f09fb5","subject":"proxy: set x-backend-record-type\u003dobject with object listing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b1c2dc1d6da2c0ae2ba6f128c7cf3b21c9f09fb5"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-11-17 16:09:41.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-11-30 18:19:09.000000000","tz":0},"subject":"proxy: always use namespaces for updating and listing","message":"proxy: always use namespaces for updating and listing\n\nThis patch attempts to remove confusion as to when the proxy\ncontrollers are handling Namespace objects (or dict representations of\nnamespaces) vs ShardRanges.\n\nWith this patch, the proxy no longer uses ShardRanges or their dict\nrepresentation when dealing with namespaces that have been cached for\nupdating or listing states. Up-to-date container servers will return\nmore compact namespaces, and full format shard range dicts returned\nfrom older container servers are parsed as Namespaces by the proxy.\n\nMuch of this is simply about using consistent and accurate\nnames. However, in some circumstances, ShardRange objects were\npreviously instantiated when not absolutely necessary: these are\nreplaced with Namespace objects.\n\nThe ObjectController would previously specify that the backend should\nreturn full format ShardRanges when asked for updating state shard\nranges *with an includes parameter*. This isn\u0027t necessary with a\nmodern backend which now supports the \u0027includes\u0027 parameter with\n\u0027namespace\u0027 format requests. This wasn\u0027t strictly necessary with a\nlegacy backend, which would have returned full format ShardRanges\nanyway if an \u0027includes\u0027 parameter was sent.\n\nThe ContainerController._get_from_shards method needs a list of\nnamespaces to build a sharded listing. It is currently passed a\nresponse which has a list of either namespaces or full shard range\ndicts, depending on what the backend server returned. The list of\ndicts is converted to a list of ShardRange objects. This is convenient\nbecause the ShardRange conveniently splits the namespace name into\naccount and container parts, but it is also confusing given that the\nlist will often be of namespace dicts.\n\nThis patch refactors _get_from_shards to clarify that it does not\nrequire ShardRange objects. The method is now passed a list of\nnamespaces, which is parsed from the response body before the method\nis called.\n\nSome unit tests are also refactored to be more realistic when mocking\n_get_from_shards.\n\nChange-Id: I2cec1cef5eb34874feea9f2922fd13c4f011e410\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0cfd884b6c4b175b26203d5384a964b62c302b1d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0cfd884b6c4b175b26203d5384a964b62c302b1d"}]},"branch":"refs/heads/master"},"8212898c56041425d373364a36108cdd2adda343":{"kind":"REWORK","_number":21,"created":"2023-12-04 09:30:09.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/50/900350/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/50/900350/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/900350/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/900350/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/900350/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/50/900350/21"}}},"commit":{"parents":[{"commit":"b1c2dc1d6da2c0ae2ba6f128c7cf3b21c9f09fb5","subject":"proxy: set x-backend-record-type\u003dobject with object listing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b1c2dc1d6da2c0ae2ba6f128c7cf3b21c9f09fb5"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-11-17 16:09:41.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-12-04 09:29:23.000000000","tz":0},"subject":"proxy: always use namespaces for updating and listing","message":"proxy: always use namespaces for updating and listing\n\nThis patch attempts to remove confusion as to when the proxy\ncontrollers are handling Namespace objects (or dict representations of\nnamespaces) vs ShardRanges.\n\nWith this patch, the proxy no longer uses ShardRanges or their dict\nrepresentation when dealing with namespaces that have been cached for\nupdating or listing states. Up-to-date container servers will return\nmore compact namespaces, and full format shard range dicts returned\nfrom older container servers are parsed as Namespaces by the proxy.\n\nMuch of this is simply about using consistent and accurate\nnames. However, in some circumstances, ShardRange objects were\npreviously instantiated when not absolutely necessary: these are\nreplaced with Namespace objects.\n\nThe ObjectController would previously specify that the backend should\nreturn full format ShardRanges when asked for updating state shard\nranges *with an includes parameter*. This isn\u0027t necessary with a\nmodern backend which now supports the \u0027includes\u0027 parameter with\n\u0027namespace\u0027 format requests. This wasn\u0027t strictly necessary with a\nlegacy backend, which would have returned full format ShardRanges\nanyway if an \u0027includes\u0027 parameter was sent.\n\nThe ContainerController._get_from_shards method needs a list of\nnamespaces to build a sharded listing. It is currently passed a\nresponse which has a list of either namespaces or full shard range\ndicts, depending on what the backend server returned. The list of\ndicts is converted to a list of ShardRange objects. This is convenient\nbecause the ShardRange conveniently splits the namespace name into\naccount and container parts, but it is also confusing given that the\nlist will often be of namespace dicts.\n\nThis patch refactors _get_from_shards to clarify that it does not\nrequire ShardRange objects. The method is now passed a list of\nnamespaces, which is parsed from the response body before the method\nis called.\n\nSome unit tests are also refactored to be more realistic when mocking\n_get_from_shards.\n\nChange-Id: I2cec1cef5eb34874feea9f2922fd13c4f011e410\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8212898c56041425d373364a36108cdd2adda343"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8212898c56041425d373364a36108cdd2adda343"}]},"branch":"refs/heads/master"},"e3de45d302f9305c963df84c0a17f701ace7e8a8":{"kind":"REWORK","_number":22,"created":"2023-12-05 12:30:27.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/50/900350/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/50/900350/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/900350/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/900350/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/900350/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/50/900350/22"}}},"commit":{"parents":[{"commit":"f1c1fd892c8f4ebf00a0f985ec6aaf2a8e6fdd8f","subject":"container: add Namespace marker/end_marker/include/reverse support.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f1c1fd892c8f4ebf00a0f985ec6aaf2a8e6fdd8f"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-11-17 16:09:41.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-12-05 12:29:16.000000000","tz":0},"subject":"proxy: always use namespaces for updating and listing","message":"proxy: always use namespaces for updating and listing\n\nThis patch attempts to remove confusion as to when the proxy\ncontrollers are handling Namespace objects (or dict representations of\nnamespaces) vs ShardRanges.\n\nWith this patch, the proxy no longer uses ShardRanges or their dict\nrepresentation when dealing with namespaces that have been cached for\nupdating or listing states. Up-to-date container servers will return\nmore compact namespaces, and full format shard range dicts returned\nfrom older container servers are parsed as Namespaces by the proxy.\n\nMuch of this is simply about using consistent and accurate\nnames. However, in some circumstances, ShardRange objects were\npreviously instantiated when not absolutely necessary: these are\nreplaced with Namespace objects.\n\nThe ObjectController would previously specify that the backend should\nreturn full format ShardRanges when asked for updating state shard\nranges *with an includes parameter*. This isn\u0027t necessary with a\nmodern backend which now supports the \u0027includes\u0027 parameter with\n\u0027namespace\u0027 format requests. This wasn\u0027t strictly necessary with a\nlegacy backend, which would have returned full format ShardRanges\nanyway if an \u0027includes\u0027 parameter was sent.\n\nThe ContainerController._get_from_shards method needs a list of\nnamespaces to build a sharded listing. It is currently passed a\nresponse which has a list of either namespaces or full shard range\ndicts, depending on what the backend server returned. The list of\ndicts is converted to a list of ShardRange objects. This is convenient\nbecause the ShardRange conveniently splits the namespace name into\naccount and container parts, but it is also confusing given that the\nlist will often be of namespace dicts.\n\nThis patch refactors _get_from_shards to clarify that it does not\nrequire ShardRange objects. The method is now passed a list of\nnamespaces, which is parsed from the response body before the method\nis called.\n\nSome unit tests are also refactored to be more realistic when mocking\n_get_from_shards.\n\nChange-Id: I2cec1cef5eb34874feea9f2922fd13c4f011e410\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e3de45d302f9305c963df84c0a17f701ace7e8a8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e3de45d302f9305c963df84c0a17f701ace7e8a8"}]},"branch":"refs/heads/master"},"559c3cc58514a3e3c367b8e51e419e7de0868395":{"kind":"NO_CHANGE","_number":23,"created":"2023-12-05 16:21:14.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/50/900350/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/50/900350/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/900350/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/900350/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/900350/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/50/900350/23"}}},"commit":{"parents":[{"commit":"d5c6dd2aff02006551a5af03862ceab29b10f360","subject":"proxy: set x-backend-record-type\u003dobject with object listing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d5c6dd2aff02006551a5af03862ceab29b10f360"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-11-17 16:09:41.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-12-05 16:14:11.000000000","tz":0},"subject":"proxy: always use namespaces for updating and listing","message":"proxy: always use namespaces for updating and listing\n\nThis patch attempts to remove confusion as to when the proxy\ncontrollers are handling Namespace objects (or dict representations of\nnamespaces) vs ShardRanges.\n\nWith this patch, the proxy no longer uses ShardRanges or their dict\nrepresentation when dealing with namespaces that have been cached for\nupdating or listing states. Up-to-date container servers will return\nmore compact namespaces, and full format shard range dicts returned\nfrom older container servers are parsed as Namespaces by the proxy.\n\nMuch of this is simply about using consistent and accurate\nnames. However, in some circumstances, ShardRange objects were\npreviously instantiated when not absolutely necessary: these are\nreplaced with Namespace objects.\n\nThe ObjectController would previously specify that the backend should\nreturn full format ShardRanges when asked for updating state shard\nranges *with an includes parameter*. This isn\u0027t necessary with a\nmodern backend which now supports the \u0027includes\u0027 parameter with\n\u0027namespace\u0027 format requests. This wasn\u0027t strictly necessary with a\nlegacy backend, which would have returned full format ShardRanges\nanyway if an \u0027includes\u0027 parameter was sent.\n\nThe ContainerController._get_from_shards method needs a list of\nnamespaces to build a sharded listing. It is currently passed a\nresponse which has a list of either namespaces or full shard range\ndicts, depending on what the backend server returned. The list of\ndicts is converted to a list of ShardRange objects. This is convenient\nbecause the ShardRange conveniently splits the namespace name into\naccount and container parts, but it is also confusing given that the\nlist will often be of namespace dicts.\n\nThis patch refactors _get_from_shards to clarify that it does not\nrequire ShardRange objects. The method is now passed a list of\nnamespaces, which is parsed from the response body before the method\nis called.\n\nSome unit tests are also refactored to be more realistic when mocking\n_get_from_shards.\n\nChange-Id: I2cec1cef5eb34874feea9f2922fd13c4f011e410\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/559c3cc58514a3e3c367b8e51e419e7de0868395"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/559c3cc58514a3e3c367b8e51e419e7de0868395"}]},"branch":"refs/heads/master"},"a6e9bb549c6acf98d4d4092565a1ac545d4f02e8":{"kind":"TRIVIAL_REBASE","_number":24,"created":"2023-12-05 18:38:59.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/50/900350/24","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/50/900350/24","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/900350/24 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/900350/24 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/900350/24 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/50/900350/24"}}},"commit":{"parents":[{"commit":"ac8c61b66f6596de9b4e66b2e03fef3348df29b1","subject":"container: add Namespace marker/end_marker/include/reverse support.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ac8c61b66f6596de9b4e66b2e03fef3348df29b1"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-11-17 16:09:41.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-12-05 18:23:41.000000000","tz":0},"subject":"proxy: always use namespaces for updating and listing","message":"proxy: always use namespaces for updating and listing\n\nThis patch attempts to remove confusion as to when the proxy\ncontrollers are handling Namespace objects (or dict representations of\nnamespaces) vs ShardRanges.\n\nWith this patch, the proxy no longer uses ShardRanges or their dict\nrepresentation when dealing with namespaces that have been cached for\nupdating or listing states. Up-to-date container servers will return\nmore compact namespaces, and full format shard range dicts returned\nfrom older container servers are parsed as Namespaces by the proxy.\n\nMuch of this is simply about using consistent and accurate\nnames. However, in some circumstances, ShardRange objects were\npreviously instantiated when not absolutely necessary: these are\nreplaced with Namespace objects.\n\nThe ObjectController would previously specify that the backend should\nreturn full format ShardRanges when asked for updating state shard\nranges *with an includes parameter*. This isn\u0027t necessary with a\nmodern backend which now supports the \u0027includes\u0027 parameter with\n\u0027namespace\u0027 format requests. This wasn\u0027t strictly necessary with a\nlegacy backend, which would have returned full format ShardRanges\nanyway if an \u0027includes\u0027 parameter was sent.\n\nThe ContainerController._get_from_shards method needs a list of\nnamespaces to build a sharded listing. It is currently passed a\nresponse which has a list of either namespaces or full shard range\ndicts, depending on what the backend server returned. The list of\ndicts is converted to a list of ShardRange objects. This is convenient\nbecause the ShardRange conveniently splits the namespace name into\naccount and container parts, but it is also confusing given that the\nlist will often be of namespace dicts.\n\nThis patch refactors _get_from_shards to clarify that it does not\nrequire ShardRange objects. The method is now passed a list of\nnamespaces, which is parsed from the response body before the method\nis called.\n\nSome unit tests are also refactored to be more realistic when mocking\n_get_from_shards.\n\nChange-Id: I2cec1cef5eb34874feea9f2922fd13c4f011e410\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a6e9bb549c6acf98d4d4092565a1ac545d4f02e8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a6e9bb549c6acf98d4d4092565a1ac545d4f02e8"}]},"branch":"refs/heads/master"},"44da8e77c66fab43e3d6b996ee55faba95b0dfa2":{"kind":"REWORK","_number":25,"created":"2023-12-08 19:24:26.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/50/900350/25","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/50/900350/25","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/900350/25 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/900350/25 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/900350/25 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/50/900350/25"}}},"commit":{"parents":[{"commit":"fb66723cb77971b0cbe09623a577a5f5b98259cd","subject":"container: add Namespace marker/end_marker/include/reverse support.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fb66723cb77971b0cbe09623a577a5f5b98259cd"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-11-17 16:09:41.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-12-08 19:20:13.000000000","tz":0},"subject":"proxy: always use namespaces for updating and listing","message":"proxy: always use namespaces for updating and listing\n\nThis patch attempts to remove confusion as to when the proxy\ncontrollers are handling Namespace objects (or dict representations of\nnamespaces) vs ShardRanges.\n\nWith this patch, the proxy no longer uses ShardRanges or their dict\nrepresentation when dealing with namespaces that have been cached for\nupdating or listing states. Up-to-date container servers will return\nmore compact namespaces, and full format shard range dicts returned\nfrom older container servers are parsed as Namespaces by the proxy.\n\nMuch of this is simply about using consistent and accurate\nnames. However, in some circumstances, ShardRange objects were\npreviously instantiated when not absolutely necessary: these are\nreplaced with Namespace objects.\n\nThe ObjectController would previously specify that the backend should\nreturn full format ShardRanges when asked for updating state shard\nranges *with an includes parameter*. This isn\u0027t necessary with a\nmodern backend which now supports the \u0027includes\u0027 parameter with\n\u0027namespace\u0027 format requests. This wasn\u0027t strictly necessary with a\nlegacy backend, which would have returned full format ShardRanges\nanyway if an \u0027includes\u0027 parameter was sent.\n\nThe ContainerController._get_from_shards method needs a list of\nnamespaces to build a sharded listing. It is currently passed a\nresponse which has a list of either namespaces or full shard range\ndicts, depending on what the backend server returned. The list of\ndicts is converted to a list of ShardRange objects. This is convenient\nbecause the ShardRange conveniently splits the namespace name into\naccount and container parts, but it is also confusing given that the\nlist will often be of namespace dicts.\n\nThis patch refactors _get_from_shards to clarify that it does not\nrequire ShardRange objects. The method is now passed a list of\nnamespaces, which is parsed from the response body before the method\nis called.\n\nSome unit tests are also refactored to be more realistic when mocking\n_get_from_shards.\n\nChange-Id: I2cec1cef5eb34874feea9f2922fd13c4f011e410\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/44da8e77c66fab43e3d6b996ee55faba95b0dfa2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/44da8e77c66fab43e3d6b996ee55faba95b0dfa2"}]},"branch":"refs/heads/master"},"7a60c5a342207c21d1bae1a6f91729adb89f70aa":{"kind":"TRIVIAL_REBASE","_number":26,"created":"2023-12-11 15:15:14.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/50/900350/26","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/50/900350/26","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/900350/26 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/900350/26 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/900350/26 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/50/900350/26"}}},"commit":{"parents":[{"commit":"c5ea203be8f91f270cb1b82404c43316d15360e8","subject":"container: add Namespace marker/end_marker/include/reverse support.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c5ea203be8f91f270cb1b82404c43316d15360e8"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-11-17 16:09:41.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-12-11 15:08:26.000000000","tz":0},"subject":"proxy: always use namespaces for updating and listing","message":"proxy: always use namespaces for updating and listing\n\nThis patch attempts to remove confusion as to when the proxy\ncontrollers are handling Namespace objects (or dict representations of\nnamespaces) vs ShardRanges.\n\nWith this patch, the proxy no longer uses ShardRanges or their dict\nrepresentation when dealing with namespaces that have been cached for\nupdating or listing states. Up-to-date container servers will return\nmore compact namespaces, and full format shard range dicts returned\nfrom older container servers are parsed as Namespaces by the proxy.\n\nMuch of this is simply about using consistent and accurate\nnames. However, in some circumstances, ShardRange objects were\npreviously instantiated when not absolutely necessary: these are\nreplaced with Namespace objects.\n\nThe ObjectController would previously specify that the backend should\nreturn full format ShardRanges when asked for updating state shard\nranges *with an includes parameter*. This isn\u0027t necessary with a\nmodern backend which now supports the \u0027includes\u0027 parameter with\n\u0027namespace\u0027 format requests. This wasn\u0027t strictly necessary with a\nlegacy backend, which would have returned full format ShardRanges\nanyway if an \u0027includes\u0027 parameter was sent.\n\nThe ContainerController._get_from_shards method needs a list of\nnamespaces to build a sharded listing. It is currently passed a\nresponse which has a list of either namespaces or full shard range\ndicts, depending on what the backend server returned. The list of\ndicts is converted to a list of ShardRange objects. This is convenient\nbecause the ShardRange conveniently splits the namespace name into\naccount and container parts, but it is also confusing given that the\nlist will often be of namespace dicts.\n\nThis patch refactors _get_from_shards to clarify that it does not\nrequire ShardRange objects. The method is now passed a list of\nnamespaces, which is parsed from the response body before the method\nis called.\n\nSome unit tests are also refactored to be more realistic when mocking\n_get_from_shards.\n\nChange-Id: I2cec1cef5eb34874feea9f2922fd13c4f011e410\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7a60c5a342207c21d1bae1a6f91729adb89f70aa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7a60c5a342207c21d1bae1a6f91729adb89f70aa"}]},"branch":"refs/heads/master"},"9360a785c5936ea4af627cc3426cb3f79ca0577d":{"kind":"REWORK","_number":27,"created":"2023-12-11 18:59:13.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/50/900350/27","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/50/900350/27","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/900350/27 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/900350/27 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/900350/27 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/50/900350/27"}}},"commit":{"parents":[{"commit":"c5ea203be8f91f270cb1b82404c43316d15360e8","subject":"container: add Namespace marker/end_marker/include/reverse support.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c5ea203be8f91f270cb1b82404c43316d15360e8"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-11-17 16:09:41.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-12-11 18:40:40.000000000","tz":0},"subject":"proxy: always use namespaces for updating and listing","message":"proxy: always use namespaces for updating and listing\n\nThis patch attempts to remove confusion as to when the proxy\ncontrollers are handling Namespace objects (or dict representations of\nnamespaces) vs ShardRanges.\n\nWith this patch, the proxy no longer uses ShardRanges or their dict\nrepresentation when dealing with namespaces that have been cached for\nupdating or listing states. Up-to-date container servers will return\nmore compact namespaces, and full format shard range dicts returned\nfrom older container servers are parsed as Namespaces by the proxy.\n\nMuch of this is simply about using consistent and accurate\nnames. However, in some circumstances, ShardRange objects were\npreviously instantiated when not absolutely necessary: these are\nreplaced with Namespace objects.\n\nThe ObjectController would previously specify that the backend should\nreturn full format ShardRanges when asked for updating state shard\nranges *with an includes parameter*. This isn\u0027t necessary with a\nmodern backend which now supports the \u0027includes\u0027 parameter with\n\u0027namespace\u0027 format requests. This wasn\u0027t strictly necessary with a\nlegacy backend, which would have returned full format ShardRanges\nanyway if an \u0027includes\u0027 parameter was sent.\n\nThe ContainerController._get_from_shards method needs a list of\nnamespaces to build a sharded listing. It is currently passed a\nresponse which has a list of either namespaces or full shard range\ndicts, depending on what the backend server returned. The list of\ndicts is converted to a list of ShardRange objects. This is convenient\nbecause the ShardRange conveniently splits the namespace name into\naccount and container parts, but it is also confusing given that the\nlist will often be of namespace dicts.\n\nThis patch refactors _get_from_shards to clarify that it does not\nrequire ShardRange objects. The method is now passed a list of\nnamespaces, which is parsed from the response body before the method\nis called.\n\nSome unit tests are also refactored to be more realistic when mocking\n_get_from_shards.\n\nChange-Id: I2cec1cef5eb34874feea9f2922fd13c4f011e410\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9360a785c5936ea4af627cc3426cb3f79ca0577d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9360a785c5936ea4af627cc3426cb3f79ca0577d"}]},"branch":"refs/heads/master"},"be14500f77d6e7aaecbc4e5d773b1879f1ee81a0":{"kind":"REWORK","_number":28,"created":"2024-01-04 16:29:45.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/50/900350/28","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/50/900350/28","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/900350/28 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/900350/28 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/900350/28 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/50/900350/28"}}},"commit":{"parents":[{"commit":"103846d66b209898fad3bf08ae53043d642a92cb","subject":"sq: support x-backend-override-deleted for namespace GET","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/103846d66b209898fad3bf08ae53043d642a92cb"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-11-17 16:09:41.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-01-04 16:24:49.000000000","tz":0},"subject":"proxy: always use namespaces for updating and listing","message":"proxy: always use namespaces for updating and listing\n\nThis patch attempts to remove confusion as to when the proxy\ncontrollers are handling Namespace objects (or dict representations of\nnamespaces) vs ShardRanges.\n\nWith this patch, the proxy no longer uses ShardRanges or their dict\nrepresentation when dealing with namespaces that have been cached for\nupdating or listing states. Up-to-date container servers will return\nmore compact namespaces, and full format shard range dicts returned\nfrom older container servers are parsed as Namespaces by the proxy.\n\nMuch of this is simply about using consistent and accurate\nnames. However, in some circumstances, ShardRange objects were\npreviously instantiated when not absolutely necessary: these are\nreplaced with Namespace objects.\n\nThe ObjectController would previously specify that the backend should\nreturn full format ShardRanges when asked for updating state shard\nranges *with an includes parameter*. This isn\u0027t necessary with a\nmodern backend which now supports the \u0027includes\u0027 parameter with\n\u0027namespace\u0027 format requests. This wasn\u0027t strictly necessary with a\nlegacy backend, which would have returned full format ShardRanges\nanyway if an \u0027includes\u0027 parameter was sent.\n\nThe ContainerController._get_from_shards method needs a list of\nnamespaces to build a sharded listing. It is currently passed a\nresponse which has a list of either namespaces or full shard range\ndicts, depending on what the backend server returned. The list of\ndicts is converted to a list of ShardRange objects. This is convenient\nbecause the ShardRange conveniently splits the namespace name into\naccount and container parts, but it is also confusing given that the\nlist will often be of namespace dicts.\n\nThis patch refactors _get_from_shards to clarify that it does not\nrequire ShardRange objects. The method is now passed a list of\nnamespaces, which is parsed from the response body before the method\nis called.\n\nSome unit tests are also refactored to be more realistic when mocking\n_get_from_shards.\n\nChange-Id: I2cec1cef5eb34874feea9f2922fd13c4f011e410\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/be14500f77d6e7aaecbc4e5d773b1879f1ee81a0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/be14500f77d6e7aaecbc4e5d773b1879f1ee81a0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
