)]}'
{"id":"openstack%2Fswift~900518","triplet_id":"openstack%2Fswift~master~I3786d185a93ca272bf1ebd0369d27b490d537574","project":"openstack/swift","branch":"master","topic":"extend-slo-attrs","attention_set":{},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2023-11-10 11:09:32.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I3786d185a93ca272bf1ebd0369d27b490d537574","subject":"sq? rename SloAttrs-\u003eRespAttrs","status":"ABANDONED","created":"2023-11-09 12:49:54.000000000","updated":"2023-11-10 11:09:32.000000000","total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"00b8169a3244eb911ccc4783c34c9b71743ccbe5","_number":900518,"virtual_id_number":900518,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2023-11-09 13:55:48.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-11-09 16:59:59.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-11-09 13:15:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-11-09 13:55:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-11-09 16:59:59.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"},{"updated":"2023-11-09 20:24:43.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"CC"}],"messages":[{"id":"637c355b1c162b68ce90ababbe82a00e3ecfae4f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-11-09 12:49:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"10de2056fb6b7fae5d6d1ab9bd45731298fb044c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-09 13:15:19.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f61eaf221eee4078945f55132a49e51e\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a686b6bf23ca44e693092438d8127d83 : SUCCESS in 15m 35s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6ba324c23d36412ab2249f18d9219e2b : SUCCESS in 17m 18s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/e32bdf065d064c55a7bd8e5b8283819c : SUCCESS in 24m 08s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/4c86becb01aa43a3917aa8d770fae609 : SUCCESS in 20m 47s","accounts_in_message":[],"_revision_number":1},{"id":"c5f707b154ed186a46755b841a36ddf5e23edf66","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-09 13:55:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9afe4a53a2444225aba2acd7bfcfa986\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4598fbbd36444a60ab6582817eadb275 : SUCCESS in 8m 36s\n- grenade https://zuul.opendev.org/t/openstack/build/0b12f6aef88e4adcbd508e6d923ea86a : SUCCESS in 49m 01s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/5f79ade1758442e294cc318c1879b58e : SUCCESS in 48m 21s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/5d60a8fa813f48a09683b25210b0286b : SUCCESS in 40m 14s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/979c065f6de1456390dd3eaf2767b770 : SUCCESS in 50m 04s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ef16dec8d83b425da5f46af9e48fb007 : SUCCESS in 18m 12s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f8875b52f50a4ad1bbcb62c682d14979 : SUCCESS in 26m 19s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/fa24072ea4724363911b9567000e173b : SUCCESS in 21m 55s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/8f537752ec924cd686a3a69fc7a508a2 : SUCCESS in 9m 41s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/70a152c67c584992911bb0028116cbb1 : SUCCESS in 6m 24s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/b1b5369c34be47c78f5c70cd09311512 : SUCCESS in 17m 03s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/f0b666ebbf9f459fb812abe5cf8acb7c : SUCCESS in 18m 02s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/04b3f07de5a841f9811a1b7347890cf1 : SUCCESS in 16m 50s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/5c05af1a0ad741e08b70ab6fcde0fea8 : SUCCESS in 25m 40s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/a5db778a921d4f8f911d869821e7158d : SUCCESS in 18m 01s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/0d53f6dac6c34e268940c71347ee0c1b : SUCCESS in 25m 39s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/6aa3cb916fd3497aaf1d8252a31a7e6d : SUCCESS in 18m 17s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/5d2a553728304f44a7fa3b09a232505a : SUCCESS in 20m 43s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/9d6b409db56e43cc8f1f6b869213ad8f : SUCCESS in 16m 43s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/1457c3e0b3d642129c0d04a2a02472fa : SUCCESS in 23m 38s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/9365d666507b44899d05a5949d36e900 : SUCCESS in 17m 13s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/e3ebd624018741c1a46c5330ec84988a : SUCCESS in 6m 45s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/609e74e5f44d42e89650c9add84b6b9f : SUCCESS in 15m 51s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/bda72d3a3541400f83af23e8353a9ae2 : SUCCESS in 7m 11s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/c7c487db662b4c7984c20dcbd69a742e : SUCCESS in 47m 18s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/151f32158a2e4a329a57bf7e4093588a : SUCCESS in 57m 38s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/0b23be7d25624ce8ada408eb3602a957 : SUCCESS in 43m 02s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/f8f1f8b53af947419dad6585b07f55a2 : SUCCESS in 43m 36s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2080a635fd164eafbd9025d49fe7c240 : SUCCESS in 16m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ee23f77c4d864db089edd013340aaa6f : SUCCESS in 4m 20s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/f7bc7f4c33e94ab8afea391d59678127 : SUCCESS in 20m 02s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/39257ffc3d8247a482c2a2fecb8b23b9 : SUCCESS in 44m 22s","accounts_in_message":[],"_revision_number":1},{"id":"f8f03940f34fb80f8b80badfe0261e382df064ad","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-11-09 16:59:59.000000000","message":"Patch Set 1: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"49952ecf074b20869ac607f4de7d46252bc2c23d","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2023-11-09 20:24:43.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"00b8169a3244eb911ccc4783c34c9b71743ccbe5","tag":"autogenerated:gerrit:abandon","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-11-10 11:09:32.000000000","message":"Abandoned\n\nsquashed to parent","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ea4334f4c4a93d43be4d9354038d7a76af8e413e","revisions":{"ea4334f4c4a93d43be4d9354038d7a76af8e413e":{"kind":"REWORK","_number":1,"created":"2023-11-09 12:49:54.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/18/900518/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/18/900518/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/900518/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/900518/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/900518/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/18/900518/1"}}},"commit":{"parents":[{"commit":"16ba57fbfbde518d162e9d02846d494946650e15","subject":"slo: refactor GET/HEAD response handling","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/16ba57fbfbde518d162e9d02846d494946650e15"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-11-09 12:37:04.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-11-09 12:39:24.000000000","tz":0},"subject":"sq? rename SloAttrs-\u003eRespAttrs","message":"sq? rename SloAttrs-\u003eRespAttrs\n\nChange-Id: I3786d185a93ca272bf1ebd0369d27b490d537574\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ea4334f4c4a93d43be4d9354038d7a76af8e413e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ea4334f4c4a93d43be4d9354038d7a76af8e413e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
