)]}'
{"id":"openstack%2Fswift~900546","triplet_id":"openstack%2Fswift~master~I18b98525e9d2d6e7c775fc55bb60bc6ac5820509","project":"openstack/swift","branch":"master","topic":"p-proxy-namespaces","attention_set":{},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2023-11-14 04:24:59.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I18b98525e9d2d6e7c775fc55bb60bc6ac5820509","subject":"sq: refactor test_container","status":"ABANDONED","created":"2023-11-09 16:45:01.000000000","updated":"2023-11-14 04:24:59.000000000","total_comment_count":10,"unresolved_comment_count":5,"has_review_started":true,"meta_rev_id":"eab2e26216df2ec90e382650e94cea5fe9b6c6ea","_number":900546,"virtual_id_number":900546,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2023-11-09 17:54:47.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-11-09 17:54:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-11-14 04:14:03.000000000","updated_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"reviewer":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"state":"CC"}],"messages":[{"id":"d4a1bcc56f8e87ccc7448af4b99982951a44f814","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-11-09 16:45:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fa99fe5642433046dabb4a76506caa04be1e8460","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-09 17:54:47.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e687603f994b4e6ab7cbe698b0316e43\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bf1d83f08c4b4426802c7d81e87ea02a : SUCCESS in 9m 47s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/4b20dd8bfd3041b380ec5211e5a7bd8b : SUCCESS in 49m 32s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/83eb451af5b44c9fb99973125f6b9ab9 : SUCCESS in 17m 21s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f794997ccca04f25806408427618f9ab : SUCCESS in 18m 45s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5e64413f50404c96bc184d18b5e49456 : SUCCESS in 18m 08s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/146e2e87d2364a9a993ce306f69507fa : SUCCESS in 9m 18s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/7487c055cf6a44da98cbb2b87fc1049e : SUCCESS in 9m 20s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/5df1df79548c4a3096b7ab44c7f5e6c2 : SUCCESS in 18m 41s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/86beaf1fcc47494cb52477edf2cc64e3 : SUCCESS in 19m 38s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/9e3b4f44fe0645c18f75c3f672200951 : SUCCESS in 18m 12s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/94569cf5dde748fd8378b858c3d05820 : SUCCESS in 18m 43s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/838cf51eb6ce492688e07c54548b296b : SUCCESS in 20m 03s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/baf638d6437845749f196a0132af2ceb : SUCCESS in 19m 00s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/f7d4b65a5bc843a89e000444b40a6e30 : SUCCESS in 19m 27s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/aa90954651dd4446826753239da07b5d : SUCCESS in 19m 48s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/25c59ceaffde4671a7180b309c357659 : SUCCESS in 17m 11s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/9e7c6c55d5944b2489ef5383d79c578d : SUCCESS in 16m 29s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/3d2903f7643a4425bb3e0f5a7f45e0a0 : SUCCESS in 22m 19s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/5ad7eeba68b6474ab32cae47f4cfa062 : SUCCESS in 52m 08s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/20974624f349438b8f2b52fa34c60d5c : SUCCESS in 1h 02m 05s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/78e39d8b227f43aebec5d1d55fed49fa : SUCCESS in 16m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c7845b33e3674893b4fe966831ad5c95 : SUCCESS in 4m 59s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/5e90997a24374bb988da516a03289099 : SUCCESS in 26m 43s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d3a6478017a17dc20d2280e1f50bb9063ce6afc3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-09 18:19:41.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f8cb53e7430c4aaebf9f38e0c7127dcd\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/506cc0d013f24621a4bb200e565cccc3 : SUCCESS in 17m 55s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e1549d73c8c04277a53f6223168b1f2f : SUCCESS in 18m 55s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/2f1fd2c4fca343a090f1a7fda5b33d6c : FAILURE in 1h 29m 30s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/fd680337b1eb492b9abd84afab551376 : SUCCESS in 21m 01s","accounts_in_message":[],"_revision_number":1},{"id":"3e01f187cc5ba3f98d63de204ba7c12cb3db88a6","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-11-09 18:21:10.000000000","message":"Patch Set 1:\n\n(7 comments)","accounts_in_message":[],"_revision_number":1},{"id":"cd92ea6e3aeaae643bf8f6b42466001f99f38caa","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2023-11-14 04:14:03.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"4e4ab733e37f9bfada7652f116d2cf97a4117a27","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2023-11-14 04:24:47.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"eab2e26216df2ec90e382650e94cea5fe9b6c6ea","tag":"autogenerated:gerrit:abandon","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2023-11-14 04:24:59.000000000","message":"Abandoned\n\nsqaushed into parent","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9040ca6b1bae0fcc0752b7e88c5ffb81be0311ca","revisions":{"9040ca6b1bae0fcc0752b7e88c5ffb81be0311ca":{"kind":"REWORK","_number":1,"created":"2023-11-09 16:45:01.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/46/900546/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/46/900546/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/46/900546/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/46/900546/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/46/900546/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/46/900546/1"}}},"commit":{"parents":[{"commit":"d855d01e2f368ed08039cf9cb2e621f3edd38da1","subject":"Proxy: Use shard-format for auto cont GETs and updating","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d855d01e2f368ed08039cf9cb2e621f3edd38da1"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-11-08 18:25:44.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-11-09 16:36:47.000000000","tz":0},"subject":"sq: refactor test_container","message":"sq: refactor test_container\n\n- Separate the tests that need to be run with and without legacy\n  backend responses.\n\n- Use helper methods to define headers expected from backend servers.\n\nChange-Id: I18b98525e9d2d6e7c775fc55bb60bc6ac5820509\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9040ca6b1bae0fcc0752b7e88c5ffb81be0311ca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9040ca6b1bae0fcc0752b7e88c5ffb81be0311ca"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
