)]}'
{"id":"openstack%2Fswift~900626","triplet_id":"openstack%2Fswift~master~Ia514cd9604eb2ad53f71ecd519a2b91bcda68b99","project":"openstack/swift","branch":"master","topic":"extend-slo-attrs","attention_set":{},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2023-11-28 22:02:46.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Ia514cd9604eb2ad53f71ecd519a2b91bcda68b99","subject":"WIP: slo: don\u0027t hide segments in RespAttrs","status":"ABANDONED","created":"2023-11-10 14:59:22.000000000","updated":"2023-11-28 22:02:46.000000000","total_comment_count":2,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"a2790b242e804fa0470263ec8519985bfd5ad604","_number":900626,"virtual_id_number":900626,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:check","value":1,"date":"2023-11-10 16:19:21.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":1,"date":"2023-11-10 15:17:32.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-11-10 15:17:32.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"},{"updated":"2023-11-10 15:58:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-11-10 16:19:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"be90a605f1eda7c496d37450b8293a7913621d89","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-11-10 14:59:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f85a98b8bb35914c247176b74783dc65cde8ea73","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2023-11-10 15:17:32.000000000","message":"Patch Set 1: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"7f40fe7065ca1a76a1220e18f777971b32c80104","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-10 15:58:13.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/710c0774fcb34cb59e070b44dab20747\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e1572de90d2541a28aa311598bd82d82 : SUCCESS in 55m 55s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5688fbb057ba414bbc1f6dc9c977bf06 : SUCCESS in 19m 12s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/a01e575a8d284aff84c1b2f87ea05a1a : SUCCESS in 45m 17s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/8eacdc15f80d429a84d88c980741d4fd : SUCCESS in 22m 04s","accounts_in_message":[],"_revision_number":1},{"id":"17d440262eb6c66c84bd180dc607194cc016b3c6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-10 16:19:21.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fdb8c7bc31a141c29537d4810fb904cd\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0438448cb66a42089c876ef71da5f9c8 : SUCCESS in 9m 46s\n- grenade https://zuul.opendev.org/t/openstack/build/95096eb9beea4d19b56b80627390be38 : SUCCESS in 48m 15s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/68b08e0546414f21a7db28e8e7acc8f7 : SUCCESS in 54m 25s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/33fcd9c5293f4a62a88095fdeadb67db : SUCCESS in 40m 01s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ccea53edfe304a9ab2a696841c345478 : SUCCESS in 1h 04m 15s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/57e82354b7ea4b4eb86437e1978d639a : SUCCESS in 15m 26s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e4c6e984e34b446bbbcfc5211e87ee59 : SUCCESS in 21m 22s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8fed3e5ff84b4b76aa024893200f5fad : SUCCESS in 25m 38s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/df9e1d41b380460d87a57579f2e4458a : SUCCESS in 8m 57s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/2ae0c7a3304b4c068cfcb711ccadb4c9 : SUCCESS in 5m 13s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/38510036cdb14872b942ace97aebcc9a : SUCCESS in 23m 29s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/70680fe5d55f417c9290c906332ce78f : SUCCESS in 19m 55s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/2ac6354b03a649238870ea690f4a81c6 : SUCCESS in 11m 27s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/24a98771aa36427ca62140f7297dfb02 : SUCCESS in 16m 54s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/8adb744af6c74230b2b8f18be6a75362 : SUCCESS in 18m 27s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/995a38f23c4244ddacea4d65f4cf9c61 : SUCCESS in 18m 48s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/9d5e65c750324a979a7d9704040c8dbe : SUCCESS in 20m 28s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/b806e080a2b14ea08f9827f45d94a728 : SUCCESS in 19m 23s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/1c39808a516e4500bbf68ad51c0077e5 : SUCCESS in 14m 56s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/9fe315dfe5b94b5db6b85566b07459e6 : SUCCESS in 19m 20s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/bdde0ae33512430c9397ed06aa10dcb6 : SUCCESS in 16m 47s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/e59b83fb47584ff4b06425c00b5c4295 : SUCCESS in 7m 54s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/229e689d75ef49eeafbe2563195102fa : SUCCESS in 17m 08s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/0db3e243aac5426e9b15d3854688f924 : SUCCESS in 8m 41s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/3a147145ec844b149ab413582485a829 : SUCCESS in 57m 05s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/340f5a9bba8f437f86d21b3f6eacb10b : SUCCESS in 1h 13m 36s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/52226ab99a404520be2a2dd0f46f3435 : SUCCESS in 55m 20s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/7f93dcaf2e3d496d94f9123cade63529 : SUCCESS in 57m 56s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/71f5daa0c6de454e892b22a25f6a0e40 : SUCCESS in 17m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5c8f8df6ebe24fa6ba0c79a836eb9fb7 : SUCCESS in 4m 50s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/bb6435a6761e4f579c5d39940df26cb0 : SUCCESS in 22m 28s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1527dad18d8445c7a67455305694fe15 : SUCCESS in 36m 04s","accounts_in_message":[],"_revision_number":1},{"id":"a2790b242e804fa0470263ec8519985bfd5ad604","tag":"autogenerated:gerrit:abandon","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2023-11-28 22:02:46.000000000","message":"Abandoned\n\nsquashed 893578: slo: refactor GET/HEAD response handling | https://review.opendev.org/c/openstack/swift/+/893578","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"eef2cfa3cf6db421263080617d0860cff06cdc35","revisions":{"eef2cfa3cf6db421263080617d0860cff06cdc35":{"kind":"REWORK","_number":1,"created":"2023-11-10 14:59:22.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/26/900626/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/26/900626/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/26/900626/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/26/900626/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/26/900626/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/26/900626/1"}}},"commit":{"parents":[{"commit":"49cdb8da3a02d432f9977d0229265c991a076516","subject":"slo: part-number\u003dN query parameter support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/49cdb8da3a02d432f9977d0229265c991a076516"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-11-10 14:57:16.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-11-10 14:57:16.000000000","tz":0},"subject":"WIP: slo: don\u0027t hide segments in RespAttrs","message":"WIP: slo: don\u0027t hide segments in RespAttrs\n\nwhen we need segments, parse the segments and then use the segments all\nin one place.\n\nChange-Id: Ia514cd9604eb2ad53f71ecd519a2b91bcda68b99\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/eef2cfa3cf6db421263080617d0860cff06cdc35"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/eef2cfa3cf6db421263080617d0860cff06cdc35"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
