)]}'
{"id":"openstack%2Fswift~900633","triplet_id":"openstack%2Fswift~master~I3057cf13e8fc5ff923c92c1c1ec6a4e596c915c3","project":"openstack/swift","branch":"master","topic":"inherit-children","attention_set":{},"removed_from_attention_set":{"1179":{"account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"last_update":"2023-11-30 23:18:52.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I3057cf13e8fc5ff923c92c1c1ec6a4e596c915c3","subject":"wip: WOMM; checking gate","status":"ABANDONED","created":"2023-11-10 16:17:13.000000000","updated":"2023-11-30 23:18:52.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"5e7e086ba962d6edf5fb45f6e697020b547d8ad0","_number":900633,"virtual_id_number":900633,"owner":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2023-11-10 17:31:51.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-11-10 17:08:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-11-10 17:31:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"be2e9bc8d922222c2dffb23de62f3af71a19a07d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2023-11-10 16:17:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"15c56a8b4905b6a227d6a16acd2b9644a8ee1972","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-10 17:08:26.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dce63a6984624c70b635aba41a255bcd\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/5b759f6da803404e96c644c6839676f1 : SUCCESS in 19m 11s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4fd9cdcaf7434b10bbf9d06a8bcc0981 : SUCCESS in 17m 35s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/609b4be7f8c14ff5b94ab0309cc0240b : SUCCESS in 49m 12s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/1a9b99aff29e4dc2ade47f6853b4627b : SUCCESS in 23m 40s","accounts_in_message":[],"_revision_number":1},{"id":"37cf9846500ba3d974e3372374c0dd8a9af029bd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-10 17:31:51.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/2100fc46cbf54de3a8ef2d77706f3b7e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/45e141847dc341fb8da4b3009fa4f07a : SUCCESS in 8m 06s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/f41f424187cd4cbfb7839c1719125bfa : SUCCESS in 1h 06m 36s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8ae0a7ed2dba45548f5d2f3500e0f80e : SUCCESS in 18m 51s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5568eed16d9c4fe98655a8f952e165fc : SUCCESS in 22m 41s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/149211ab26aa4b8b9dab4ff67b98119b : SUCCESS in 21m 55s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/b7448c03b5654f3bbf865b88cf09fc47 : SUCCESS in 9m 07s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/a06c6940df634686b45d8e1b8a99c7be : SUCCESS in 5m 29s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/cf1690389add4de082216c2f86c66a53 : SUCCESS in 21m 49s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/d17ee0972f2b4886a0fd15fa7cf50870 : SUCCESS in 21m 04s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/8353ad03f95b4d10a6a83ccdd5cc4194 : SUCCESS in 22m 44s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/3c7be83067fb4d2a8d1f3c82c5ebbda7 : SUCCESS in 21m 07s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/9e1dfadb197c494b8f3f97b7dee2ead4 : SUCCESS in 14m 42s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/dc0f854c3e3c4ff5a97a47135fbc1a75 : SUCCESS in 24m 33s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/f4361a6fc3854497990a6e926cc20375 : SUCCESS in 17m 38s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/4ed659667ef5479982fa06375aa246f5 : SUCCESS in 18m 55s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/0b10d91a76324701b6807bcc41c3f89e : SUCCESS in 17m 00s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/f92451e8fbb746b8a0eb502182135181 : SUCCESS in 15m 48s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/54d2aa0373ee452d9eb04e0b3158eecb : SUCCESS in 17m 50s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/e56a5d8178da427b8cfb15514e3da98a : SUCCESS in 59m 12s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/d585498b66bb4daf8581099cd4b03b53 : SUCCESS in 58m 36s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/af690b07f21543f2b4ce54c034926867 : SUCCESS in 19m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/242a1bdab2ba409fb56a4ce0d97cfa06 : FAILURE in 4m 50s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/729e3e33312e4ce98ee634a287e35448 : SUCCESS in 21m 02s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5e7e086ba962d6edf5fb45f6e697020b547d8ad0","tag":"autogenerated:gerrit:abandon","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-11-30 23:18:52.000000000","message":"Abandoned\n\nSquashed into parent","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f99c5f5cec9a003ff2698e2365c63a8a49fb2b3e","revisions":{"f99c5f5cec9a003ff2698e2365c63a8a49fb2b3e":{"kind":"REWORK","_number":1,"created":"2023-11-10 16:17:13.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/33/900633/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/33/900633/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/33/900633/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/33/900633/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/33/900633/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/33/900633/1"}}},"commit":{"parents":[{"commit":"c8e775a94eb1350c1650e91f7615ae83b2e5081f","subject":"wsgi: Reap stale workers (after a timeout) following a reload","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c8e775a94eb1350c1650e91f7615ae83b2e5081f"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2023-11-10 16:17:05.000000000","tz":-360},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2023-11-10 16:17:05.000000000","tz":-360},"subject":"wip: WOMM; checking gate","message":"wip: WOMM; checking gate\n\nChange-Id: I3057cf13e8fc5ff923c92c1c1ec6a4e596c915c3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f99c5f5cec9a003ff2698e2365c63a8a49fb2b3e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f99c5f5cec9a003ff2698e2365c63a8a49fb2b3e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
