)]}'
{"id":"openstack%2Fswift~900638","triplet_id":"openstack%2Fswift~master~I47669e8c3909141d1bcceae62dd0b4c3999228da","project":"openstack/swift","branch":"master","topic":"env-reexec-test","attention_set":{"1179":{"account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"last_update":"2023-11-10 18:30:39.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I47669e8c3909141d1bcceae62dd0b4c3999228da","subject":"send more info over wsgi re-exec","status":"NEW","created":"2023-11-10 17:05:20.000000000","updated":"2023-11-30 23:24:55.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":6,"unresolved_comment_count":3,"has_review_started":true,"meta_rev_id":"a6f88cabdd93ed8795921e6ae03218ef628c2de3","_number":900638,"virtual_id_number":900638,"owner":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2023-11-10 18:30:39.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-11-10 17:34:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-11-10 18:30:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-11-13 18:24:31.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"CC"}],"messages":[{"id":"7400c2f7984c3e2b44e33ff662a3ff620fb55b25","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2023-11-10 17:05:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c2be334c6045dca4806407db5153375835cc25b1","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2023-11-10 17:19:06.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"05ec04c72df9aff48b0793460fe2f50abe8f4842","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-10 17:34:26.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/99baf2e33c1340ac8e9a294df35ef63e\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/537a0ab5e49e442dacd97e98504c7760 : SUCCESS in 16m 20s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/62f1161e88bb42dfa54f16bbfb3942fa : SUCCESS in 17m 09s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/f290c28f2e9c4737bddd6d89e5df0077 : SUCCESS in 28m 01s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/6ad91731189e4f0c900ac983ccc7bf95 : SUCCESS in 25m 25s","accounts_in_message":[],"_revision_number":1},{"id":"11705011c9720339fbf5b92c6f6f95be93829ed5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-10 18:30:39.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d0d5fb270968429eb11528c676df897c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8388c12498564babbe077fb4763e6ee7 : SUCCESS in 9m 26s\n- grenade https://zuul.opendev.org/t/openstack/build/057139ae52ae49de96f88b2ae0b1634c : SUCCESS in 53m 15s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/261bdfceb46a4a2c9b183add0c1d7592 : SUCCESS in 48m 36s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/4f8829d3ede14a9380dddd7d08dbdc75 : SUCCESS in 34m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6e9c0eb9e79043e086dbf69794f65739 : SUCCESS in 50m 34s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/37f0741d301941eda8630e3c38c8ae84 : SUCCESS in 19m 06s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/529f47da215742cbba295c0cdc70c1ef : SUCCESS in 18m 49s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e7872cb8193b45c7a99e67159ad37365 : SUCCESS in 20m 44s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/77bf3b27761441b4adf9c2318ff45a89 : SUCCESS in 8m 29s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/4c060a8658964f809ebc916e9d672008 : SUCCESS in 7m 14s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/c4681e454b4f4f43a94c8fd4538e8203 : SUCCESS in 16m 36s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/dddcb9386e1a44fbb64fdf80caed5012 : SUCCESS in 22m 04s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/e0ac4dd324704abcb74d6e73c32909a8 : SUCCESS in 15m 32s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/e61857cea34f4b68882b9d975f7308bb : SUCCESS in 14m 56s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/a31a6ea344304005881833d826e40c04 : SUCCESS in 16m 56s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/8d71ec0fb1d5478a9ccba541032ce6ef : SUCCESS in 18m 02s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/d2696aac94d5442ca1d0830c2462167a : SUCCESS in 27m 45s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/f4566b4e385a48099b8048eaf44267b6 : SUCCESS in 22m 21s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/23bc3e2f5e3b4f51a9a2e9721e565769 : SUCCESS in 15m 46s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/f92cea65d2c54359a3ccbdfcc10e26e6 : SUCCESS in 17m 18s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/77c927b7237d41d9b680467c2118bee9 : SUCCESS in 17m 29s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/0c080bfd274d441b9bea347355b097da : SUCCESS in 8m 04s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/89a31dab955c4c24b83ec4a5c8eb90e7 : SUCCESS in 17m 02s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/d9c99cd0b2ba4db5bcc9e1da6d164a70 : SUCCESS in 9m 34s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/5aedfd7625d944c09f77aa06ad182d47 : SUCCESS in 57m 44s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/66710110ed35478caf09a27783d3393a : SUCCESS in 53m 44s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/83a085aeb76144bb9713862f0d27b927 : FAILURE in 12m 35s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/2d72674aa62646fba1a398ab921cfd1a : SUCCESS in 40m 21s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5ee5a0571159456eadc9d398b9440cdc : SUCCESS in 16m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5a7f0eb70e0849c7af4f7dec137c3bd3 : SUCCESS in 4m 49s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/b2c785464ea44ecb8a839d78e6fadfba : SUCCESS in 21m 38s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/98fc77ddb8cc453aa16c8db21a125b1e : SUCCESS in 1h 04m 42s","accounts_in_message":[],"_revision_number":1},{"id":"0746f3cfc3bcb36010270de0d3e2e958996aed42","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-11-13 18:24:31.000000000","message":"Patch Set 1:\n\n(4 comments)","accounts_in_message":[],"_revision_number":1},{"id":"a6f88cabdd93ed8795921e6ae03218ef628c2de3","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-11-30 23:24:55.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"07dfa89946f9178db2a3aeec80c163f611635123","revisions":{"07dfa89946f9178db2a3aeec80c163f611635123":{"kind":"REWORK","_number":1,"created":"2023-11-10 17:05:20.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/38/900638/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/38/900638/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/38/900638/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/38/900638/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/38/900638/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/38/900638/1"}}},"commit":{"parents":[{"commit":"d93ce6333311777a31eb07f95c36f08257a5dce3","subject":"Merge \"Add mailmap entry for my new mail\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d93ce6333311777a31eb07f95c36f08257a5dce3"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2023-11-10 17:05:12.000000000","tz":-360},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2023-11-10 17:05:12.000000000","tz":-360},"subject":"send more info over wsgi re-exec","message":"send more info over wsgi re-exec\n\nChange-Id: I47669e8c3909141d1bcceae62dd0b4c3999228da\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/07dfa89946f9178db2a3aeec80c163f611635123"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/07dfa89946f9178db2a3aeec80c163f611635123"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
