)]}'
{"id":"openstack%2Fswift~900690","triplet_id":"openstack%2Fswift~master~I9ec451033f303be89d0f08c057078be9216a737c","project":"openstack/swift","branch":"master","topic":"p-proxy-namespaces","attention_set":{"1179":{"account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"last_update":"2023-11-10 20:46:42.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I9ec451033f303be89d0f08c057078be9216a737c","subject":"get_from_shards(namespaces)","status":"NEW","created":"2023-11-10 19:33:56.000000000","updated":"2023-11-10 20:46:42.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"03f5a084c248af70c1e2cc5cbf855c740424c94a","_number":900690,"virtual_id_number":900690,"owner":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2023-11-10 20:46:42.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-11-10 20:40:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-11-10 20:46:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f2440d74acbe6ab5634c8579e92e605bbfd48e41","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2023-11-10 19:33:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b9f0721756c42046f1960bf82b37ed284c892d2f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-10 20:40:12.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/241453712b144a2a9a68723b1d2910e9\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c6b2ec466e56490a84b16d7dc2d6cf2f : FAILURE in 14m 46s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/59eb7ac203d7441d9117aa90a537ed8c : FAILURE in 1h 03m 53s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/f035602e689b4520a93e2d196b1149b9 : SUCCESS in 23m 17s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/e4053759341443b6b467bf6a3eb0739e : SUCCESS in 53m 25s","accounts_in_message":[],"_revision_number":1},{"id":"03f5a084c248af70c1e2cc5cbf855c740424c94a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-10 20:46:42.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/80396a24b903451abb6d6062dd2680bc\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c7ecc2d8651c440e90ea84c4fe2f7799 : SUCCESS in 9m 04s\n- grenade https://zuul.opendev.org/t/openstack/build/bef28d7bb4434bdf99e95ac15ef9c429 : SUCCESS in 41m 08s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/48692562eb70456e932e879a8540128b : SUCCESS in 1h 04m 35s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/21846e8440fa4c9299a7ced2386c0016 : SUCCESS in 32m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/556732313307470bbb20240aae010868 : SUCCESS in 1h 01m 57s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/47fcafb2b811418cacd2fb309c38f87a : SUCCESS in 19m 46s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8d1d4c53394c4303abd3fb27c437a741 : SUCCESS in 19m 22s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5f9b74c5ff1041908b6ec4a478fbd766 : SUCCESS in 19m 07s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/9fac63a213a345ee81c9cbec7a02dbdb : SUCCESS in 9m 32s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/25576ebd1a5f4035973594308808ed52 : SUCCESS in 6m 17s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/c63747dead9c40ae9421d777cd34ebd8 : FAILURE in 17m 55s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/8853a862417e408f973496e8fa3a1ff0 : FAILURE in 16m 49s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/35fa1036a87e481d8ae66e67ac66e0fa : FAILURE in 16m 52s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/8724035d5aa64d4ea67afe85dc7ff11e : FAILURE in 16m 25s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/f5e53d2efe3443f5ae575640e5c3109d : FAILURE in 15m 55s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/25fc63d65b9b4dd781922a088835f438 : SUCCESS in 18m 21s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/896ecacffacb418c8c2274b4fdd04dee : SUCCESS in 18m 20s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/9afcff097234423d96abc0aa9072992b : SUCCESS in 17m 50s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/62e05f17ecef4265a1f2cb5b56043352 : SUCCESS in 16m 19s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/cb8f8bf6df154a00aa90bc6d902caf48 : SUCCESS in 15m 32s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/d686c2e9edb54546b5948da7b4fd2319 : SUCCESS in 15m 33s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/5976b668cc574c1aa182dd6737a0c09f : SUCCESS in 7m 23s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/4d9a398821654b8a8c9e01c7016b9da1 : SUCCESS in 17m 14s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/ede2046afc8a4eb9ad8fad508d1d0ccd : SUCCESS in 7m 11s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/68b3d757c7b5433cb12e63c97d42eca3 : FAILURE in 1h 02m 57s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/fe570c20dd6b43d4997dffe6cfb6dcf5 : FAILURE in 1h 05m 28s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/43cb6c6208b14413a92b508c9d6266a4 : SUCCESS in 1h 00m 20s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/275a7eefaa754c9ba4f100e775fcc170 : SUCCESS in 55m 36s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2d778c56dcab4d4093829514a4945d6a : FAILURE in 18m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bdb1a0eaf5294362990a0d2d0884e0ba : FAILURE in 4m 29s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/590bd54864274d2e840e42cdba8c337f : SUCCESS in 21m 23s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/70e2094468404c4cbab88e6e375ed99e : SUCCESS in 1h 00m 06s\n\nWarning:\n  Comments left for invalid file test/unit/proxy/controllers/test_container.py","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7d5a419f3eaff21e6fcb632bed277f607dd370dc","revisions":{"7d5a419f3eaff21e6fcb632bed277f607dd370dc":{"kind":"REWORK","_number":1,"created":"2023-11-10 19:33:56.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/90/900690/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/90/900690/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/900690/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/900690/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/900690/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/90/900690/1"}}},"commit":{"parents":[{"commit":"d4f5c420f369b9a293db3bca22497e3895605f97","subject":"proxy: always use namespaces for updating and listing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d4f5c420f369b9a293db3bca22497e3895605f97"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2023-11-10 19:33:48.000000000","tz":-360},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2023-11-10 19:33:48.000000000","tz":-360},"subject":"get_from_shards(namespaces)","message":"get_from_shards(namespaces)\n\nChange-Id: I9ec451033f303be89d0f08c057078be9216a737c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7d5a419f3eaff21e6fcb632bed277f607dd370dc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7d5a419f3eaff21e6fcb632bed277f607dd370dc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
