)]}'
{"id":"openstack%2Fswift~900814","triplet_id":"openstack%2Fswift~master~I52036f39884292dd7b316156965c6ff77acc252b","project":"openstack/swift","branch":"master","topic":"container_namespace_new_testcases","attention_set":{},"removed_from_attention_set":{"1179":{"account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"last_update":"2023-11-15 13:02:01.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I52036f39884292dd7b316156965c6ff77acc252b","subject":"squash: be explicit about testing sr_and_ns","status":"ABANDONED","created":"2023-11-13 18:03:22.000000000","updated":"2023-11-15 13:02:01.000000000","total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"c12080a031a89229ca924a4c920705e8b0064fa1","_number":900814,"virtual_id_number":900814,"owner":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2023-11-13 19:11:54.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-11-13 19:11:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-11-14 05:10:47.000000000","updated_by":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"reviewer":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"state":"CC"}],"messages":[{"id":"b4249e71423baafc5b2402ab11d42bb3f6e62691","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2023-11-13 18:03:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"467fba809bd2e63aa1a70fa278d1f1de986dae6d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-13 19:11:54.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/7da4c346767d4cfb8e846759ac5f6ffa\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2b05ccfc2e5f417281ffa0c4fd6268ca : SUCCESS in 9m 49s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/3904fd9e15c54c7a81b58fa35e62aadf : SUCCESS in 1h 00m 39s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7b2189d210d040afbd8baa3f99c75e6f : SUCCESS in 20m 24s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c42bb3f37fb24a179dc53e8889154cad : SUCCESS in 19m 14s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4aa5d4afa3fd4d169e055cd731e92e67 : SUCCESS in 19m 51s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/3729546feeb5432aa5d14188a5fd5405 : SUCCESS in 9m 10s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/9157062cb6734ec09f45daf2ee4281c5 : SUCCESS in 6m 02s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/8be5d297984a452cafa885ea3a81b7ef : FAILURE in 27m 27s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/716ebbed15554b79b68cadbc897d2f99 : SUCCESS in 18m 03s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/b870e4d2b0f543cda20bf830f5a04dc7 : SUCCESS in 26m 14s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/8d8624f1dd31437bb408648884b2c34a : SUCCESS in 17m 24s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/c1fa3c5ad08a4b57aa634bf6a5546672 : SUCCESS in 22m 50s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/88490e79e42b4fa6b483cb22d399d49b : SUCCESS in 24m 31s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/0abecc0a536745bdbcb124c18cff229b : SUCCESS in 19m 52s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/0f709a360b5046b9965641ea30d9ca3e : SUCCESS in 19m 03s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/9f0ec6ff18804a0e87ff1388cc4af410 : SUCCESS in 21m 59s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/9b5b660d141f4d2ea86e840fc8e25b58 : SUCCESS in 18m 09s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/a3ce1699c0814a8ab17d2a3c709d851e : SUCCESS in 16m 44s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/3c3cb83778bd492bb5e45161903f68d8 : SUCCESS in 44m 28s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/a67fee5409f14482a9dce026e94e3333 : SUCCESS in 55m 24s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d11e6d5ebab5435e85e11a15134cf7ab : FAILURE in 24m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eddae32807774da08ea260639f34c008 : FAILURE in 4m 50s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/0c0cdfbc896d410ebcbc19ee80aa7edb : SUCCESS in 20m 54s (non-voting)\n\nWarning:\n  Comments left for invalid file test/unit/proxy/controllers/test_container.py","accounts_in_message":[],"_revision_number":1},{"id":"8bfb6ccc98e669bc82bfbd1514dd9eab6c68ecba","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-13 19:38:11.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/566203a75acc4266964d8be1ebd4a632\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ec0f023650f24444ba04ae2144db5eb8 : TIMED_OUT in 1h 31m 07s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/33aad8926b7c43e2b84b248a92703f2e : TIMED_OUT in 1h 31m 00s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/3e54334ef13443f6944fbc605a375119 : SUCCESS in 22m 18s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/5b377b66b39b4f04b90a9bc905d1d209 : TIMED_OUT in 1h 31m 07s","accounts_in_message":[],"_revision_number":1},{"id":"428907bf73d285c565b9a45426e0508b6319b7f6","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2023-11-14 05:10:47.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"c12080a031a89229ca924a4c920705e8b0064fa1","tag":"autogenerated:gerrit:abandon","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-11-15 13:02:01.000000000","message":"Abandoned\n\nThis is great but I think it was already squashed","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"44f9f454fb7c25484610d3032eb3a900912b69a8","revisions":{"44f9f454fb7c25484610d3032eb3a900912b69a8":{"kind":"REWORK","_number":1,"created":"2023-11-13 18:03:22.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/14/900814/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/14/900814/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/14/900814/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/14/900814/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/14/900814/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/14/900814/1"}}},"commit":{"parents":[{"commit":"261ff41c7ea4917f030f656cfafeba4badda09cd","subject":"Container: add more test cases for the new Namespace interface.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/261ff41c7ea4917f030f656cfafeba4badda09cd"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2023-11-13 18:03:09.000000000","tz":-360},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2023-11-13 18:03:09.000000000","tz":-360},"subject":"squash: be explicit about testing sr_and_ns","message":"squash: be explicit about testing sr_and_ns\n\nChange-Id: I52036f39884292dd7b316156965c6ff77acc252b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/44f9f454fb7c25484610d3032eb3a900912b69a8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/44f9f454fb7c25484610d3032eb3a900912b69a8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
