)]}'
{"id":"openstack%2Fswift~900956","triplet_id":"openstack%2Fswift~master~Ic9f4a0ea6b8e18e1624c516890ab69884a299773","project":"openstack/swift","branch":"master","hashtags":[],"change_id":"Ic9f4a0ea6b8e18e1624c516890ab69884a299773","subject":"docs: Simplify reload-process-tree SVGs","status":"MERGED","created":"2023-11-14 23:15:32.000000000","updated":"2024-08-27 23:01:10.000000000","submitted":"2024-08-27 23:00:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"900956","meta_rev_id":"8aae1d0ffe09a04ce19c1752a45bdf56e292d22e","_number":900956,"virtual_id_number":900956,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-08-27 23:00:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-08-27 21:51:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-08-27 21:51:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-11-14 23:30:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9a8db07571b648da11478302614acb149f386442","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-11-14 23:15:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d47daa1a79a1fb1f54bf3c835cf7b5a32d02567f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-14 23:30:34.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6b9dfc0aec84429cab9f83ef5b847366\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d399943fa1d64466b04732e459806f01 : SUCCESS in 9m 16s","accounts_in_message":[],"_revision_number":1},{"id":"42c4a09b64d707c179e0ae51660bafdf7e06c72a","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-11-14 23:36:29.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"79005bdf70ab896f5782a3a0762ae71fc8304968","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-05-02 15:38:40.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"4c5cf579517fb4db8935a242d5e03dfcf109e9da","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-02 15:53:06.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6a4544bb9dc54f7d882a8e9babe9eff8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8698d943c9244dbd94004fa2321d890f : SUCCESS in 8m 54s","accounts_in_message":[],"_revision_number":2},{"id":"99e3b197e4997a53a4904e7e1cdc5b0d270e2c86","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-05-02 16:03:02.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"36b3cc0425e670018b555594a8aceccc57e94fdb","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-08-27 20:59:26.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"8954df3ebe0cc09a4cf6f3c10e39aed714c685ea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-27 21:12:05.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/566d9ff3ccc74203b024e5a110c1a9ff\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ed11c9dd34914bc9ada882a350511d0e : SUCCESS in 12m 12s","accounts_in_message":[],"_revision_number":2},{"id":"9741b71e03c44761354eee8f1fb3ca7317213e96","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-08-27 21:23:53.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"db7b10672080fabd70435c2dc3c24405f94dbf41","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-27 21:37:21.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e32632aff2c844a9bf3371c68b6bff26\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8ab9596e551149108e76230bb4a69540 : SUCCESS in 8m 09s","accounts_in_message":[],"_revision_number":3},{"id":"f8fdc4abfcd990b2a3eea9b6c2513a3a31755e5a","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-08-27 21:51:50.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"02f957eebe569d6de7c2b6cc4365c45c4640c342","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-27 21:52:15.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"286386156a346b660497d99a663f01638b4f78a0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-27 23:00:03.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8953df56716744ed803bd093311dd821\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aacc138219444f67b50803ecb851e2d1 : SUCCESS in 8m 09s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/b953bc59abbb4e0bab5463478c65b3fb : SUCCESS in 6m 44s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/b475c5b1098d4c59a500240c6224bb08 : SUCCESS in 7m 56s","accounts_in_message":[],"_revision_number":3},{"id":"d000593da190e06fff55a5f1e299fb386509341a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-27 23:00:04.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"8aae1d0ffe09a04ce19c1752a45bdf56e292d22e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-27 23:01:10.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6ccc97d7ca644237a914f0567b9a3698\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7b47fae91f5f49ba8059dbebba408af4 : SUCCESS in 46s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/00fc161205dc4cc0978093564e775b7c : FAILURE in 27s (non-voting)","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"2e11ac32dac6ecdfc890adb4b1fa75dfd4eb5419","revisions":{"6beec9aca0383ead46ec505d748e5ad74908c63f":{"kind":"REWORK","_number":1,"created":"2023-11-14 23:15:32.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/56/900956/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/56/900956/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/900956/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/900956/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/900956/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/56/900956/1"}}},"commit":{"parents":[{"commit":"2115cba46f9a323e0ff10bbd3df156c5dbdf1c41","subject":"Add abstract sockets for process notifications","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2115cba46f9a323e0ff10bbd3df156c5dbdf1c41"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-11-14 22:57:06.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-11-14 22:57:06.000000000","tz":-480},"subject":"docs: Simplify reload-process-tree SVGs","message":"docs: Simplify reload-process-tree SVGs\n\nThose were so unwieldy as to be uneditable before.\n\nChange-Id: Ic9f4a0ea6b8e18e1624c516890ab69884a299773\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6beec9aca0383ead46ec505d748e5ad74908c63f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6beec9aca0383ead46ec505d748e5ad74908c63f"}]},"branch":"refs/heads/master"},"c106e9516ff69f3b8005560a74c7b4261f9efd4f":{"kind":"REWORK","_number":2,"created":"2024-05-02 15:38:40.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/56/900956/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/56/900956/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/900956/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/900956/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/900956/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/56/900956/2"}}},"commit":{"parents":[{"commit":"761d9196772bc150fd4923fb04367b919a521d59","subject":"tests: Use mock.patch more","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/761d9196772bc150fd4923fb04367b919a521d59"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-11-14 22:57:06.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-05-02 15:38:27.000000000","tz":-420},"subject":"docs: Simplify reload-process-tree SVGs","message":"docs: Simplify reload-process-tree SVGs\n\nThose were so unwieldy as to be uneditable before.\n\nChange-Id: Ic9f4a0ea6b8e18e1624c516890ab69884a299773\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c106e9516ff69f3b8005560a74c7b4261f9efd4f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c106e9516ff69f3b8005560a74c7b4261f9efd4f"}]},"branch":"refs/heads/master"},"2e11ac32dac6ecdfc890adb4b1fa75dfd4eb5419":{"kind":"REWORK","_number":3,"created":"2024-08-27 21:23:53.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/56/900956/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/56/900956/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/900956/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/900956/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/900956/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/56/900956/3"}}},"commit":{"parents":[{"commit":"761d9196772bc150fd4923fb04367b919a521d59","subject":"tests: Use mock.patch more","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/761d9196772bc150fd4923fb04367b919a521d59"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-11-14 22:57:06.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-08-27 21:23:39.000000000","tz":-420},"subject":"docs: Simplify reload-process-tree SVGs","message":"docs: Simplify reload-process-tree SVGs\n\nThose were so unwieldy as to be uneditable before.\n\nChange-Id: Ic9f4a0ea6b8e18e1624c516890ab69884a299773\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2e11ac32dac6ecdfc890adb4b1fa75dfd4eb5419"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2e11ac32dac6ecdfc890adb4b1fa75dfd4eb5419"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
