)]}'
{"id":"openstack%2Fswift~900957","triplet_id":"openstack%2Fswift~master~Ic93c4669cb2a57c14152d8d17220dcf3515bbc3a","project":"openstack/swift","branch":"master","attention_set":{},"removed_from_attention_set":{"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2025-03-11 20:39:57.000000000","reason":"\u003cGERRIT_ACCOUNT_15343\u003e replied on the change","reason_account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}},"hashtags":[],"change_id":"Ic93c4669cb2a57c14152d8d17220dcf3515bbc3a","subject":"wsgi: Avoid the need for a separate socket-closer child","status":"NEW","created":"2023-11-14 23:15:32.000000000","updated":"2026-02-13 23:57:03.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":3,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"c00a932e869c2709f43eb96903c6aebfa6bf8c5c","_number":900957,"virtual_id_number":900957,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2026-02-13 23:57:03.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-11-15 00:29:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-11-15 00:40:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8747d79abc0df38bdb48173fae7b13ff900f834b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-11-14 23:15:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"31a659290f22fa7b94a7c51a88ba8a290c7ab7df","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-15 00:29:54.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8e5cf924045d48f7986998aaa3f923d2\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b8e339ba5c024736850b38c354529c1f : SUCCESS in 14m 32s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/046b6fc6599a4657b4afd0fd6bc075d0 : SUCCESS in 1h 12m 14s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/4d4f578046db4a6086ccc401eb144c4c : SUCCESS in 27m 00s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/6ca01b9ed64447bbb5b0d8a3097ebb87 : SUCCESS in 1h 11m 38s","accounts_in_message":[],"_revision_number":1},{"id":"7c81cfa7bcaaf57d690bab250f593ed9853ead1b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-15 00:40:30.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a5879e7e516d4d6ba61a1bc99ce0fed7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b5188fa2ceec48b0807a7ea9ae9be2f4 : SUCCESS in 7m 55s\n- grenade https://zuul.opendev.org/t/openstack/build/db4aa9bd287148afba5813e99a5b31f4 : SUCCESS in 42m 36s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/24641b75243846c8bb925290e423a2da : SUCCESS in 42m 51s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/9f92c73972c8440c989eab05eaf5222c : SUCCESS in 32m 01s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/475018367e9f4c15a189396a4feacb97 : SUCCESS in 1h 12m 06s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1c00117d55d44ac49ae20a0ad0205a27 : SUCCESS in 19m 37s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/142bff5342a34321bfdfee5e670aae4f : SUCCESS in 19m 33s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5d5d69f7d21d4aa4a73754a15ecb4904 : SUCCESS in 19m 03s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/5ef4870facd041bd94399c7269cf8e38 : SUCCESS in 9m 16s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/e3e9058a54594061a1167127f41093bd : SUCCESS in 6m 26s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/d230aea33d0141bc92414ccd0e88da68 : SUCCESS in 19m 28s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/d9d6491d0a7b414da2706bd765c37608 : SUCCESS in 15m 35s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/e58b033eb5e446ed975f071d74852c4d : SUCCESS in 19m 48s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/950278e55d4a40868ea009ce201b1547 : SUCCESS in 15m 04s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/ed4f6169e3384b1b8012bd7558ae1799 : SUCCESS in 20m 45s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/d04d452fd846490b8656d5ade00ed0a8 : SUCCESS in 21m 53s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/6da60b7d6231443594d45eac2b25b32e : SUCCESS in 21m 45s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/90e385e6a1ab4621a2a1e6773dc42cfb : SUCCESS in 16m 03s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/30eb97b586924c80a387c68c0681d9d1 : SUCCESS in 14m 31s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/c3da0db0db484e49b725b76d5167c1b7 : SUCCESS in 13m 33s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/3b198b49c11a4c0fbd844e07dc922c23 : SUCCESS in 13m 47s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/1a201aa0f33d40f0a9008a14b274b5a3 : SUCCESS in 7m 37s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/b5f26dd1da864ce484e6eac1322246a5 : SUCCESS in 13m 41s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/fb24599789224dd3a35e0dcb2b6067f2 : SUCCESS in 7m 05s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/9177527008124654a429f47a08e02000 : SUCCESS in 1h 05m 01s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/90377b4d830341238dd94b6344246028 : SUCCESS in 1h 17m 31s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/c5b1869fcf8347559312a8b97c80e941 : SUCCESS in 58m 47s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/c7dd6f89550444ceb29816887423c3c1 : SUCCESS in 51m 06s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ebc4d15d7168491fa957fb9aad31786c : SUCCESS in 13m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6f0adfa4a0cc4417b2094143775b12ae : SUCCESS in 4m 53s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/5c1ecd80f1bc4d37b97804bcea2361ca : SUCCESS in 22m 03s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6c94ac131d0444ae99d13564aacd0b2a : SUCCESS in 1h 02m 29s","accounts_in_message":[],"_revision_number":1},{"id":"473d784edcde917bf01f7126c4a2a33ed063c4e9","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-11-27 22:55:55.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b95701e0e41877dd28cd73994c5fa26638ca4663","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-12-04 01:20:23.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"a9e4b362f943381e814aa3a9426db706dfc19840","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-04 02:38:16.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c3b59b136f3b49c9a5a6c4665cf6cfdc\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/05afc4106ec64d19987f80624b9193c8 : SUCCESS in 4m 45s\n- grenade https://zuul.opendev.org/t/openstack/build/741d8017800a4bf4b48c33137bfefcc0 : SUCCESS in 45m 09s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/1d66b62cf82340d5b46bc4c1e658e05b : SUCCESS in 42m 11s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/01fb711996bd41d89d7cbe8f2244ef09 : SUCCESS in 54m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fa5bb5cda9db4662b24e13ab6338056f : SUCCESS in 1h 11m 56s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/51667f57df884fec99ef5648da79ef3d : SUCCESS in 19m 00s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/270fa4f3e5b2418bbcee39d8c192cd22 : SUCCESS in 21m 34s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d543c170b7c749f7bdaa98f4b51c3f15 : SUCCESS in 19m 41s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/84be368fae8f4ebcacbce3fdf57b04c9 : SUCCESS in 6m 00s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/a1c6849d8b8740aca6855022623df520 : SUCCESS in 16m 58s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/cf5ed03d7fd9493292b7f18c1eb9f567 : SUCCESS in 18m 28s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/c5c454e03d2c43e2b47a95d6b73364cf : SUCCESS in 17m 14s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/3c8f356246e34c72bf90d4ce030db9bf : SUCCESS in 32m 05s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/0f6e32a6c9bd4019a0e8e5ec4f3e2bec : SUCCESS in 15m 25s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/48d5541810934c8d97a97209f7fe1bfe : SUCCESS in 14m 15s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/e907384246904d4291c72809448a9e2f : SUCCESS in 16m 13s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/ff4e6d9bd44c44bdb6700bb067dcb99f : SUCCESS in 12m 03s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/858d5b296d3f4a809de3245182ad703a : SUCCESS in 14m 22s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/5200a390fef342a4a53b996b29d97a8a : SUCCESS in 12m 51s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/3e08b55641884a14bcddb728be364fa6 : SUCCESS in 5m 24s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/e68b40c0df44488f9e88f9292f2a3fb2 : SUCCESS in 16m 26s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/4223514332eb4f58b7cd1f22c22a3212 : SUCCESS in 6m 43s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/c87d1cd654e64cb1a93e1e9b007fb047 : SUCCESS in 45m 21s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/89af93b8c387414b80abcbf88ec833e7 : SUCCESS in 34m 31s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/4aaf5bf21609405991d399952c8e372e : SUCCESS in 51m 42s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a1ef0a0117cd4bbbb671785a6ce43406 : SUCCESS in 20m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ef5f15c8c95c4d6cbdf102b2e58cda60 : SUCCESS in 2m 43s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/d58d5979a8894c3989e8e7d492755d68 : SUCCESS in 18m 01s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/30ac64ab8914484b9e0e03c7f6fd1993 : SUCCESS in 48m 59s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7d240de575b445e2a504d28425d62563 : SUCCESS in 43m 45s","accounts_in_message":[],"_revision_number":2},{"id":"b16bdd2ecbb247f6246192028db834448c6894d4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-04 03:13:12.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9b4d1a82bbb8468eb26f596597fcfd2d\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/3ef0957d585b414599aea63dbccd44c6 : SUCCESS in 1h 50m 52s","accounts_in_message":[],"_revision_number":2},{"id":"a16ad11a199e20f9e197f38a55536d01502b1ea9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-03-07 23:28:10.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"b0c291375c936208a070275c09bad4ca5eb1965b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-08 00:14:02.000000000","message":"Patch Set 3:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bacfa13248cc4014a3f17ded09490295\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/1cbd4fcb9bb44a098092a969c2f2116d : FAILURE in 44m 20s","accounts_in_message":[],"_revision_number":3},{"id":"4262152b33fda47f58eaeda7e0b69606c98a9074","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-08 00:37:19.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d0ccb5c2766048ccb071d6f4c77f2c16\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c3df5da83b854d68ba23fcc36e067320 : SUCCESS in 11m 53s\n- grenade https://zuul.opendev.org/t/openstack/build/aa58769463784a4d86185764f93e969c : SUCCESS in 57m 25s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/ff7ae30c3f824bb79b1b92545da8b5be : SUCCESS in 54m 28s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/686464fac4784a679d353dc117810a04 : SUCCESS in 51m 01s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ea565f11898f4080aa448aed3a2ff4fd : SUCCESS in 1h 08m 32s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f08a0f016d7a473e90351bf40bfcb455 : SUCCESS in 19m 09s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/114fa439b8454817a3b47ee802461abf : SUCCESS in 19m 32s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8ad07399aaf6482287f9763b52ac2a01 : SUCCESS in 11m 06s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/225bf4c82e874d55b48d1ea591f1ffdc : SUCCESS in 5m 52s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/69081cac152d433eb1f7832d2443007d : FAILURE in 7m 32s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/248d57ef198848d6a07872addc19e3ef : FAILURE in 20m 47s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/d8066c545b584fab92646a4ca0cd33a6 : FAILURE in 10m 22s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/71aa64013b5f49dc82677a74c2554a6f : SUCCESS in 12m 10s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/8b1244262143491cb59fe86792bcc2a6 : SUCCESS in 13m 10s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/de1de304f3f44995ba79954a6a27efcd : SUCCESS in 13m 18s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/c5c2992187914044a38f053e1719af2c : SUCCESS in 4m 58s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/082e18e88b544d8b8cf1f62c3c13b86a : SUCCESS in 13m 27s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/00d0fb2381464b6997a5978713d6293a : SUCCESS in 7m 03s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/1cfe882cf5104dcdbf1274f346456cca : SUCCESS in 54m 49s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/0e15cced09334fb4a953ed054ace54ba : SUCCESS in 29m 02s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/bf45afbc70014d8bbdd27bd98dba1b53 : SUCCESS in 54m 16s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8cb5b6a193eb4f8693fa99a74a067bcc : FAILURE in 16m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/761971d5ab5e4f31b032cb41e8873816 : FAILURE in 3m 08s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/dd536c3c79e64c1c9cf2f0c0412c59fb : SUCCESS in 10m 21s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9e1868863c5d408abd58624f2e345a90 : SUCCESS in 38m 25s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b41fd918c05046afba858f9164b2a5d8 : SUCCESS in 34m 08s","accounts_in_message":[],"_revision_number":3},{"id":"657d279bb83d64d45ff8278be31920e2ae858e9f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-03-10 19:59:51.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"ad9bfdaf40abd43be15a99acd33d5815590aeddc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-10 20:43:49.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9649e041d2e34934aeb1db66f1c76049\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/d92f4d28e7a84c428fed198a501c3cc2 : SUCCESS in 42m 09s","accounts_in_message":[],"_revision_number":4},{"id":"7a30fa2b5e942fc54c7a367a1c115573dc9c6fd6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-10 21:13:27.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d9914e3f2f2d422ea072d7da4711cd4f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/954164ccbf944ea7acdce6a13812be2b : SUCCESS in 8m 20s\n- grenade https://zuul.opendev.org/t/openstack/build/eca802ab7828423cb1304ae56989a514 : SUCCESS in 33m 07s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/25d04af89e034f2e91ab2e7d8aade3ab : SUCCESS in 58m 22s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/5219ade922ca409ea176559a87cc42ec : SUCCESS in 39m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c91f86ac460049bfa61d1e063cf8ded2 : SUCCESS in 39m 00s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8e9e367c683f413893b292aefcdbbca5 : SUCCESS in 19m 58s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1bda10459bee40409e010b305697e00a : SUCCESS in 10m 55s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0cd50fe4930444eea3eae69f9440950f : SUCCESS in 23m 37s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/b04fadbf2f7d495095eb564cf58ce919 : SUCCESS in 6m 32s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/185083e0913543bca7e1c43712ba43bb : SUCCESS in 19m 22s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/00f98915a2e443b8a65edc4f20e7c3bd : SUCCESS in 8m 25s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/6e8b4a55d0134c318df1140fd32a7478 : SUCCESS in 21m 26s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/e26cc49e5b634bb5b68913295ccecc73 : SUCCESS in 13m 45s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/089ccb713bb740aca2385b66f9bfd9f6 : SUCCESS in 12m 32s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/ec3f10d9352642f7804c7939cba097e6 : SUCCESS in 17m 43s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/e0497e8fe35c4d2c84e42fb999e34e11 : SUCCESS in 6m 09s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/2e7fedfac2c148e4bcd74a0566b15ccc : SUCCESS in 8m 04s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/c02067d6737d4edcaded8f1036145fad : SUCCESS in 7m 24s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/1e3d948a19874dea8c014bd595543db2 : SUCCESS in 47m 34s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/ccc76a55c17343c89ec388c362849cba : SUCCESS in 49m 47s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/0c808bde7ee4498481148202f31c5f39 : SUCCESS in 32m 56s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/21de323c24ea4c09888bd931adfc10ce : SUCCESS in 19m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b0778842e0d44e70a9eda4350fa73dd5 : FAILURE in 5m 25s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/76d9be8508b44f2f9d1375b058fa2d60 : SUCCESS in 20m 06s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c662ebb4d0714865b83d64d5846e416a : SUCCESS in 29m 49s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fb8fe600569347b1a9178dc889a4348e : SUCCESS in 1h 06m 56s","accounts_in_message":[],"_revision_number":4},{"id":"2664ce6aba2d28caebac11d2d40fe7dbe0f01337","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-03-10 22:07:10.000000000","message":"Patch Set 5: Published edit on patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"4dcc4ece247160f314008954ef6e14ea3343ef33","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-10 23:11:39.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/75e22ebe7196498ba99a553eb1cde28d\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/6c28e7d8659e4e4b9ec69d040fe3f12b : SUCCESS in 1h 02m 29s","accounts_in_message":[],"_revision_number":5},{"id":"d80948bd760bcda45a08c9ecc6bcc89965f2c3ed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-10 23:17:23.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eb63f16fc07248ef806d7fcc128b08c1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/787c9043f99e4b60814d2fe28f7bc4a6 : SUCCESS in 8m 16s\n- grenade https://zuul.opendev.org/t/openstack/build/96a963581f624191b600e04ad27e2213 : SUCCESS in 1h 03m 28s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/489a0275bb674401bd5cc626a933b8b8 : SUCCESS in 42m 37s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/978c3b816ca24fa4b37ec0c1145e18c7 : SUCCESS in 36m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ce69d6dce1264b48a8dfd787ba50bf0c : SUCCESS in 34m 07s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/70756ea8ca974f37a1a98c67f1038667 : SUCCESS in 19m 08s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/fa2f2e657f764bb29e5a1b760c8241dd : SUCCESS in 12m 08s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4ef1fd9533cb415bbe1a30bd3b097153 : SUCCESS in 20m 57s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/6ed53b3bead64d37b0cfce5c34e031e3 : SUCCESS in 6m 52s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/89760b2b6574445d963afd198b8f54c8 : SUCCESS in 15m 38s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/4cc22fee2dd247879c2bfb3cd00efde5 : SUCCESS in 21m 33s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/3d4687ee678141498a166735adbbcefc : SUCCESS in 21m 51s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/f3198390e3424d579a487ab8326026e3 : SUCCESS in 14m 19s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/0c9c4668c0bd47db9b741c41e7b0187f : SUCCESS in 15m 34s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/dbbaac74708045bcb2abccabcd1fc981 : SUCCESS in 15m 41s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/0ac3c1bd2fbd407c9127c4f04d5fc6f6 : SUCCESS in 14m 26s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/1781c4a9ea3c4121a330bc92f3912ed7 : SUCCESS in 17m 17s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/4836dc4cc3af4e7c9d3e39427a82bfc9 : SUCCESS in 8m 39s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/19610070ae964c57a5623f2442ccce90 : SUCCESS in 56m 37s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/25a94d1b5e244d098216cb64da52b26f : SUCCESS in 52m 43s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/dbe98b22e3ed4b3d8f5576432a5081a4 : SUCCESS in 35m 07s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/84d5e61dc428435e9c1f1e424914e02c : SUCCESS in 14m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8d834756dd98433288b52ccd8eafcb46 : SUCCESS in 2m 30s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/77722a9d0f974c6caa847a2e47f055c1 : SUCCESS in 19m 14s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d80c8f21dccb4bc1abdd16928e9dfea2 : SUCCESS in 1h 01m 37s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/825db141132049eabf38973b8c117b55 : SUCCESS in 42m 59s","accounts_in_message":[],"_revision_number":5},{"id":"cbc5ee4137fee87bcc4c8f640fb665990103235c","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-03-11 20:39:57.000000000","message":"Patch Set 5:\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"4feacb5a1b3a1355300d8791eb0421cdec6dd726","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-02-13 22:42:58.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased. Commit message was updated.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"48357cfd1f3a8db9b5a255b3d077a9436a8bb511","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-13 23:32:06.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5d5ec67fda7843bfa7080ea2375e8fdc\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/6aa6a2b24cae4a9997c3597325b073c6 : SUCCESS in 46m 23s","accounts_in_message":[],"_revision_number":6},{"id":"c00a932e869c2709f43eb96903c6aebfa6bf8c5c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-13 23:57:03.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cf57f5f70e2c4d7c9a8cf766ded4d919\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/03c40be5833b4fe0b6795c127cb262ac : SUCCESS in 9m 54s\n- grenade https://zuul.opendev.org/t/openstack/build/729d032c46a845e890877ae1d18fe66b : SUCCESS in 28m 56s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/6d8008dc1b554f409392332a87a3121a : SUCCESS in 58m 12s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/64dfa43a4143401193ce33b9f7e369d4 : SUCCESS in 41m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f183d43b8767445ab3dc49986382071e : SUCCESS in 1h 08m 44s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/de992c6e6c6340cd8664c315e850c1ee : SUCCESS in 9m 57s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b52c2b70e4324300bd0626393719cc64 : SUCCESS in 11m 25s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3a22a564f058499590afd40ae0083dda : SUCCESS in 18m 42s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/582428a5d1e847158473f5271e547567 : SUCCESS in 6m 26s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/b4d7a9b2a88847f2b9698c66255163d0 : SUCCESS in 24m 06s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/92347cce4f9d447a97a11589be71fc17 : SUCCESS in 20m 22s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/f7a598da295b482fadc5d87ab2c4e356 : SUCCESS in 10m 25s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/92db17b9392c4ebd8cc5dc8772026852 : SUCCESS in 29m 02s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/7734d39b581b4f4dbcba4feef1550773 : SUCCESS in 14m 27s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/9fc73396631f49cfaea83ed0fbb7a06b : SUCCESS in 7m 48s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/d783cd4f7ae645689e27fd01ecfa5fd2 : SUCCESS in 8m 54s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/c2a515f6fd4e4adcb44161344190a8e9 : SUCCESS in 10m 31s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/c4801652bfd040dfae2dd5725357ebdb : SUCCESS in 12m 54s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/7e570f844e1a4a59a1557839cce985e0 : SUCCESS in 8m 29s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/f58aaf1c71e543fa9162260f7045e13c : SUCCESS in 33m 34s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/29e769a63451401db085dcee6d8e4123 : SUCCESS in 50m 36s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/19f39654115343afa3988541ae1e6229 : SUCCESS in 56m 21s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a6fd471585a84c36b6cd01f6a142bf98 : SUCCESS in 28m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d27596d2230f4737a6e8c9807c26a93b : SUCCESS in 4m 08s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/1dbb6f1702024951b67b2b48838be19c : SUCCESS in 14m 59s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a8a63e78ab2c4135b6891693cad8d393 : SUCCESS in 47m 22s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"245ba9e721f365fe269e920638b1b654b4f9a004","revisions":{"05beaa510fb3f645c662773df192b3b08d190323":{"kind":"REWORK","_number":1,"created":"2023-11-14 23:15:32.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/57/900957/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/57/900957/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/57/900957/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/57/900957/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/57/900957/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/57/900957/1"}}},"commit":{"parents":[{"commit":"6beec9aca0383ead46ec505d748e5ad74908c63f","subject":"docs: Simplify reload-process-tree SVGs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6beec9aca0383ead46ec505d748e5ad74908c63f"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-11-14 20:05:23.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-11-14 23:07:57.000000000","tz":-480},"subject":"wsgi: Avoid the need for a separate socket-closer child","message":"wsgi: Avoid the need for a separate socket-closer child\n\nChange-Id: Ic93c4669cb2a57c14152d8d17220dcf3515bbc3a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/05beaa510fb3f645c662773df192b3b08d190323"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/05beaa510fb3f645c662773df192b3b08d190323"}]},"branch":"refs/heads/master"},"ff44b3bd56f6e04af3e3382ced61fe41c262af51":{"kind":"REWORK","_number":2,"created":"2024-12-04 01:20:23.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/57/900957/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/57/900957/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/57/900957/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/57/900957/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/57/900957/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/57/900957/2"}}},"commit":{"parents":[{"commit":"5f344732fcb907e8e359a8e9f921f84586c4b082","subject":"Add abstract sockets for process notifications","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5f344732fcb907e8e359a8e9f921f84586c4b082"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-11-14 20:05:23.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-12-04 01:19:55.000000000","tz":-480},"subject":"wsgi: Avoid the need for a separate socket-closer child","message":"wsgi: Avoid the need for a separate socket-closer child\n\nChange-Id: Ic93c4669cb2a57c14152d8d17220dcf3515bbc3a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ff44b3bd56f6e04af3e3382ced61fe41c262af51"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ff44b3bd56f6e04af3e3382ced61fe41c262af51"}]},"branch":"refs/heads/master"},"1cb91e36ca367952adcb7448c97570a36c0f74a7":{"kind":"REWORK","_number":3,"created":"2025-03-07 23:28:10.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/57/900957/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/57/900957/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/57/900957/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/57/900957/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/57/900957/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/57/900957/3"}}},"commit":{"parents":[{"commit":"8f2fd0bc1c78bf7cbbae58c69e52dd9ade68a047","subject":"Merge \"tests: Fix keystoneclient import\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8f2fd0bc1c78bf7cbbae58c69e52dd9ade68a047"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-11-14 20:05:23.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-03-07 23:26:50.000000000","tz":-480},"subject":"wsgi: Avoid the need for a separate socket-closer child","message":"wsgi: Avoid the need for a separate socket-closer child\n\nChange-Id: Ic93c4669cb2a57c14152d8d17220dcf3515bbc3a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1cb91e36ca367952adcb7448c97570a36c0f74a7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1cb91e36ca367952adcb7448c97570a36c0f74a7"}]},"branch":"refs/heads/master"},"4cdddd1106d6d04fd7f029bbb524e3963dc55102":{"kind":"REWORK","_number":4,"created":"2025-03-10 19:59:51.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/57/900957/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/57/900957/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/57/900957/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/57/900957/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/57/900957/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/57/900957/4"}}},"commit":{"parents":[{"commit":"8f2fd0bc1c78bf7cbbae58c69e52dd9ade68a047","subject":"Merge \"tests: Fix keystoneclient import\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8f2fd0bc1c78bf7cbbae58c69e52dd9ade68a047"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-11-14 20:05:23.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-03-10 19:59:44.000000000","tz":-420},"subject":"wsgi: Avoid the need for a separate socket-closer child","message":"wsgi: Avoid the need for a separate socket-closer child\n\nChange-Id: Ic93c4669cb2a57c14152d8d17220dcf3515bbc3a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4cdddd1106d6d04fd7f029bbb524e3963dc55102"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4cdddd1106d6d04fd7f029bbb524e3963dc55102"}]},"branch":"refs/heads/master"},"ab9790a6d5f8f3e06b86b5f27a9f92f193851416":{"kind":"REWORK","_number":5,"created":"2025-03-10 22:07:10.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/57/900957/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/57/900957/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/57/900957/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/57/900957/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/57/900957/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/57/900957/5"}}},"commit":{"parents":[{"commit":"8f2fd0bc1c78bf7cbbae58c69e52dd9ade68a047","subject":"Merge \"tests: Fix keystoneclient import\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8f2fd0bc1c78bf7cbbae58c69e52dd9ade68a047"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-11-14 20:05:23.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-03-10 22:07:10.000000000","tz":0},"subject":"wsgi: Avoid the need for a separate socket-closer child","message":"wsgi: Avoid the need for a separate socket-closer child\n\nChange-Id: Ic93c4669cb2a57c14152d8d17220dcf3515bbc3a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ab9790a6d5f8f3e06b86b5f27a9f92f193851416"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ab9790a6d5f8f3e06b86b5f27a9f92f193851416"}]},"branch":"refs/heads/master"},"245ba9e721f365fe269e920638b1b654b4f9a004":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":6,"created":"2026-02-13 22:42:58.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/57/900957/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/57/900957/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/57/900957/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/57/900957/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/57/900957/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/57/900957/6"}}},"commit":{"parents":[{"commit":"c7bee623658b464072783096b98adaddfb1005bc","subject":"Merge \"s3api: Support for object expiration time\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c7bee623658b464072783096b98adaddfb1005bc"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-11-14 20:05:23.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-02-13 22:42:47.000000000","tz":-480},"subject":"wsgi: Avoid the need for a separate socket-closer child","message":"wsgi: Avoid the need for a separate socket-closer child\n\nChange-Id: Ic93c4669cb2a57c14152d8d17220dcf3515bbc3a\nSigned-off-by: Tim Burke \u003ctim.burke@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/245ba9e721f365fe269e920638b1b654b4f9a004"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/245ba9e721f365fe269e920638b1b654b4f9a004"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
