)]}'
{"id":"openstack%2Fswift~901095","triplet_id":"openstack%2Fswift~master~I877981acda14aefaadb06bb13a233078212cd545","project":"openstack/swift","branch":"master","topic":"first_sync_and_big_check","attention_set":{"1179":{"account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"last_update":"2024-12-17 22:24:11.000000000","reason":"\u003cGERRIT_ACCOUNT_7233\u003e replied on the change","reason_account":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"}}},"removed_from_attention_set":{"7233":{"account":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"last_update":"2024-12-17 22:24:11.000000000","reason":"\u003cGERRIT_ACCOUNT_7233\u003e replied on the change","reason_account":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"}}},"hashtags":[],"change_id":"I877981acda14aefaadb06bb13a233078212cd545","subject":"db_replication: big containers that haven\u0027t sunk should rsync","status":"NEW","created":"2023-11-16 03:11:36.000000000","updated":"2024-12-18 07:51:17.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":10,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"7f77ade2cd32632fc3265f0669b4cfa243fce70d","_number":901095,"virtual_id_number":901095,"owner":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"tag":"autogenerated:zuul:check","value":1,"date":"2024-12-18 06:51:28.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}],"REVIEWER":[{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-11-16 03:41:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-11-16 04:50:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-12-17 16:33:52.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"CC"}],"messages":[{"id":"1e2179798a60a486e3907c7fee4dbbbde1621bde","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2023-11-16 03:11:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"134e763802ba0f1dc42a682bab96db76c81d76cc","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2023-11-16 03:12:03.000000000","message":"Patch Set 1: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f53629399c2aa38981bf320e72ef88bb784659e6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-16 03:41:08.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9eb67205ae94445283c5b737546138c0\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/7f201eec0d70439f84e2893a9c3edb6f : FAILURE in 17m 01s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f410aac686bb499f95deb3c43ac6d190 : FAILURE in 15m 38s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/8f3305ff0e35451eb51cd8f41156fae7 : SUCCESS in 28m 27s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/b5379d6da84a4c7381a2a6a00f6f435b : SUCCESS in 21m 44s","accounts_in_message":[],"_revision_number":1},{"id":"be6828944366e66881646772c82c1e621e856a63","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-16 04:50:47.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8555d5a950864159930d5afde1061791\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/039872581fee45aa84d7489fcebf20b5 : SUCCESS in 9m 20s\n- grenade https://zuul.opendev.org/t/openstack/build/7f42d33d3dc0436a8e0a90e6a6ea576d : SUCCESS in 55m 10s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/b2ba1b9df873430293009709673f0065 : SUCCESS in 53m 08s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/e963a301e3354ab2b9b19730fda1707f : SUCCESS in 55m 11s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1a7efa01125d40d49812a94ad93f20e6 : SUCCESS in 1h 02m 23s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b03f24c14fa74ac5b559d3d89a93f776 : SUCCESS in 20m 31s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1f7800fc7c234eccb1e7246f25c2e3d8 : SUCCESS in 18m 51s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/740795600e3b4c6285cc9cba8754056f : SUCCESS in 17m 27s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/924bfa05bdce497e952e7f28553af6c1 : SUCCESS in 9m 12s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/45c1cb6934094996a29027d01321d4b6 : SUCCESS in 5m 52s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/6cd1c3962d674eacb0b12407c2349586 : FAILURE in 21m 37s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/cdd9dedf52574aefb55128368c1ae869 : FAILURE in 16m 19s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/add39b2476cf452eaf4dc716989af517 : FAILURE in 15m 58s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/ffc148e73e444d49ad2826d4ec9dd4ae : FAILURE in 15m 48s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/d8792320bc9f45d8a66a161c62909e3b : FAILURE in 18m 07s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/bef1c4bbc6974f2aa03e7c54bdf35ae0 : SUCCESS in 19m 10s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/75d3e4f3f8fc4ad3bb4135eeaab2c6fe : SUCCESS in 18m 48s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/479689f167bc4d27adf8730217047050 : SUCCESS in 19m 06s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/1b7c86c3c3c74607a8762ef34f1d3b2a : SUCCESS in 15m 13s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/0737f1c54eb74f5a8add944323eee221 : SUCCESS in 18m 53s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/a5faa9b8f21e4a6cb57c3df8f23be54d : SUCCESS in 16m 21s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/3270304a13134f1dbff5e38300e54335 : SUCCESS in 7m 23s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/9f66c8a9f8a24ee183a14266cc460c26 : SUCCESS in 15m 25s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/63c99c93ce3444d1a63b2f58752c0dca : SUCCESS in 8m 04s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/acf375482d5b4e7e9a535e3cceda2bb0 : SUCCESS in 1h 03m 58s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/b9fd642500a144ebb1f3a908acad7e0a : SUCCESS in 54m 43s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/c48b975a40a4469f91d2431eab604d0b : SUCCESS in 58m 02s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/bd5bd06ed0cb4d8fb81917bc0a53162b : SUCCESS in 53m 20s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c042a93df7a9486f81ac5c88027b57ba : FAILURE in 14m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e3b5a1b97777488b9f8a1b556dfbb9ff : SUCCESS in 4m 32s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/f64f16b42ebe4671a8eedee650160e09 : SUCCESS in 22m 34s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/15c8b1d7892440b7a01976a6e44b241c : SUCCESS in 38m 36s","accounts_in_message":[],"_revision_number":1},{"id":"c5e4881b4eab2aa22b88a8f1d92fe5495617cfb0","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2023-11-16 23:00:44.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6fc7aa14547c7bf5997c59ec4ed268d3e1c1e01f","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2023-11-17 03:13:29.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9748c7e639488534f48ed9ad92e627ac2bbdc4b3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2024-12-09 05:40:03.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":2},{"id":"16a097fc1bd5a78869ffe1b16653420e1539c1d5","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2024-12-09 05:55:23.000000000","message":"Patch Set 1: Code-Review-1 Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"50b10ffe78b993c65bbb8873146545bd46671f15","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2024-12-09 05:56:59.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ee5ab43dbca77453f073f07adcf20f90cb09994f","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2024-12-09 05:57:44.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b28bdf827d5a7db932746d8b4daa686a5dcf8166","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-09 06:54:16.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a7587b31cff74c28bb2f704bddef00c5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a15cd8e858f84e3e9a6ddede0600ca85 : SUCCESS in 7m 12s\n- grenade https://zuul.opendev.org/t/openstack/build/4abe5a9053014796abee553749c80d3c : SUCCESS in 58m 04s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/a883eda8727749a0b4dd90fa2bf4dc26 : SUCCESS in 48m 11s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/14cc6c583cfa41b9913ce96e5e6df4e2 : SUCCESS in 46m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9df03041f16046dd9e5512cd13005f1e : SUCCESS in 58m 23s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/77c18b0588204428b1e91c60b3e66df5 : SUCCESS in 23m 24s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/84c427e6b6aa4097aa0b0653de4dd03d : SUCCESS in 12m 44s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/afa3136615ce416a8b7d5690c072f6f2 : SUCCESS in 11m 52s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/d12e8f1e1a52454e95877ea97f1cf546 : SUCCESS in 5m 06s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/d20b727263bb4660afdb3f795dc6fc50 : SUCCESS in 20m 20s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/beea597b47334304a71848dd3d1b6390 : SUCCESS in 10m 45s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/aa0a4a2992884781ac43396ddddba362 : SUCCESS in 12m 34s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/5cd5c38325d3468abf0d615667b2bc3d : SUCCESS in 12m 58s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/2fd152b78fd649ffbeea7fc087fd971d : SUCCESS in 18m 17s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/d35057c5c7f74646b92d097588b23354 : SUCCESS in 9m 38s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/c839e826eb1c449c827df828adabff41 : SUCCESS in 17m 20s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/60785ed863704786a0eacb8b664abcbd : SUCCESS in 9m 40s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/bf899c2d472846e8a4ec1d0b232caf2c : SUCCESS in 15m 11s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/7300bb88dfdf4227ba5495042e8fe902 : SUCCESS in 9m 40s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/f2451f8c238a4415ad83d32c4a591f28 : SUCCESS in 7m 06s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/c694ccb0c54446d9a30a323293c9f3cf : SUCCESS in 15m 46s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/45289f25d78c4191a7515423cbcaf103 : SUCCESS in 7m 18s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/86b89dc38256405d81eb34070e38a05a : SUCCESS in 1h 12m 30s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/074555df5d8a4b919929c070cb26a59b : SUCCESS in 52m 41s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/d2bb7faa05234746abf41a455257a360 : SUCCESS in 48m 09s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c382139418fa44daba767e5e9c518399 : SUCCESS in 18m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c85fc2ab48064303bffc7c101016dd89 : SUCCESS in 3m 50s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/5f0049259934447f9a5e9e0edae496b9 : SUCCESS in 17m 31s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1ca900b861634799bb77dcb6dc408719 : SUCCESS in 44m 10s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/87e80649d9cd40868e6cb5858ba4d863 : SUCCESS in 52m 56s","accounts_in_message":[],"_revision_number":2},{"id":"ef6b6c84891984af5dda098fec566d0622b15a06","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-09 07:45:22.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3bb65c8d990d40afbfe7dee2f1c15b30\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/91cdb61a42e7435c98791219cd98cfa6 : FAILURE in 2h 03m 34s","accounts_in_message":[],"_revision_number":2},{"id":"3ed2b2cefc2585cb22bd422a768f4a1c572c0499","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-12-17 16:33:52.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"9f89c81ad8db822ffc9ced31e857e736a40d733b","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2024-12-17 22:24:11.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"2f8a0cb7bd05f898e386b7485d1daf533d8fae86","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2024-12-18 03:58:15.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"9481f732a3fee2c109a76d817a0e196c51ee493a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2024-12-18 05:54:38.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"0d276ae698428d93426cb2d338db227f35471029","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-18 06:51:28.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f7df27d7067b4da789fb1c8ed0429d7f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d2b267f0a0a84be39545856ce01ac733 : SUCCESS in 8m 09s\n- grenade https://zuul.opendev.org/t/openstack/build/6843dba9e6444a88b26f56466db41c81 : SUCCESS in 43m 36s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/889a7c47885448e8b3276ab3a60b13bd : SUCCESS in 44m 02s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/d225bd2fba1346dcaaf9fd15499726f4 : SUCCESS in 43m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fdfd28f9c33341d8bcf6ba78210d4e59 : SUCCESS in 49m 51s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9b92569eb6454c42b9bcf8d247153deb : SUCCESS in 22m 15s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/eb28690db6ae4ab0bcb05d155d4d1397 : SUCCESS in 15m 56s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ea41835ae794411aa6d48b8c76ec2df7 : SUCCESS in 13m 19s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/05a29591bccc4217963e3fc9ce9f9bd6 : SUCCESS in 4m 03s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/ae44c78dd3e64193a13aae303085a421 : SUCCESS in 9m 41s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/2a9a07e5fb3d42cba01f220dc28deb09 : SUCCESS in 19m 03s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/f19c29e36a664cb880b5e9f4cfaa2737 : SUCCESS in 10m 34s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/01219ef40e484eab80e157d820dd3cf2 : SUCCESS in 27m 32s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/b6897795136f4356b5c781d39efa875e : SUCCESS in 10m 43s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/d4ead7b2d35a4467bff7cccee7f7c4e1 : SUCCESS in 17m 59s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/b4fb211bca664cdd93dbe736d6615250 : SUCCESS in 10m 07s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/005be9a9c3784f65bdce5fdbe4827054 : SUCCESS in 14m 50s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/6d3251533c6b4165953efe6bc3612ca7 : SUCCESS in 15m 21s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/a7d8cf8eaee04e1a9cfefa13dffe630a : SUCCESS in 14m 31s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/07616dcf741b4a3eb7f688215993e413 : SUCCESS in 5m 18s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/e723c4ca9aa44acfaee01dc37e8b33d9 : SUCCESS in 17m 46s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/f2000fe6e2bb417e8030efc61b9df919 : SUCCESS in 7m 02s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/95736eca03fc4268a14c1086776302d9 : SUCCESS in 36m 58s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/abb4ba526d9e4012809308c268a18dd5 : SUCCESS in 51m 29s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/d4e398b7dd6843fca321e5db95a21405 : SUCCESS in 50m 04s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f428f18e81a0426b886ce413269f7750 : SUCCESS in 21m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2447ae5c810c46f780d22f3da1026767 : SUCCESS in 4m 24s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/75f23a85e77e44f58fe7a9676e527986 : SUCCESS in 18m 02s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9dac467ea4364a14bf0134473830c111 : SUCCESS in 38m 57s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f099f366bbd140919eadc305eb32f519 : SUCCESS in 49m 21s","accounts_in_message":[],"_revision_number":3},{"id":"7f77ade2cd32632fc3265f0669b4cfa243fce70d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-18 07:51:17.000000000","message":"Patch Set 3:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ed0100d6b50c43b78f508af8128d3038\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/65dcab03de1447edb2cd5f7f68802994 : FAILURE in 1h 53m 57s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"6628cbf75f16015d39aa80882664ba129c6cb53d","revisions":{"1882301c67b6b89224a29d251c0f343c1e566355":{"kind":"REWORK","_number":1,"created":"2023-11-16 03:11:36.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/95/901095/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/95/901095/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/95/901095/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/95/901095/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/95/901095/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/95/901095/1"}}},"commit":{"parents":[{"commit":"b0d0c494389a3b72e3558dea160e0c335160daa5","subject":"Merge \"slo: refactor GET/HEAD response handling\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b0d0c494389a3b72e3558dea160e0c335160daa5"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2023-11-16 03:07:28.000000000","tz":660},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2023-11-16 03:07:28.000000000","tz":660},"subject":"db_replication: big containers that haven\u0027t sunk should rsync","message":"db_replication: big containers that haven\u0027t sunk should rsync\n\nWhen the sharder cleaves we get large shards in handoffs. The first replica\nwill replicate fine, as it gets a 404. But the second+ replicas would never\nhave sunk but there are a lot of objects to move. If there are more the\nmax.diffs then the shard sits on the handoff node.\nWe want move the shard handoffs faster.\n\nThis change adds a new check, if the remote sync point is 0, ie it\u0027s\nnever sunk to that location before yet the container has more then\nmax_diff rows to send then fall back to rsync.\n\nChange-Id: I877981acda14aefaadb06bb13a233078212cd545\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1882301c67b6b89224a29d251c0f343c1e566355"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1882301c67b6b89224a29d251c0f343c1e566355"}]},"branch":"refs/heads/master"},"0c61936c497bd790ac65e61cb64cbecc00182b01":{"kind":"REWORK","_number":2,"created":"2024-12-09 05:40:03.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/95/901095/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/95/901095/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/95/901095/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/95/901095/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/95/901095/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/95/901095/2"}}},"commit":{"parents":[{"commit":"b4579b414e51db88bfbcae5d3d678ed8fe1a7941","subject":"Merge \"CI: point antelope job at 2023.1-eom tag\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b4579b414e51db88bfbcae5d3d678ed8fe1a7941"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2023-11-16 03:07:28.000000000","tz":660},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2024-12-09 05:33:43.000000000","tz":660},"subject":"db_replication: big containers that haven\u0027t sunk should rsync","message":"db_replication: big containers that haven\u0027t sunk should rsync\n\nWhen the sharder cleaves we get large shards in handoffs. The first\nreplica will replicate fine, as it gets a 404. But the second+ replicas\nwould also never have sunk but there are a lot of objects to move. If\nthere are more the max.diffs, which there always is, then the shard sits\non the handoff node waiting for container replicator cycles to slowly\nmove all the data off the handoff location.  This in production could\ntake days.\nWorse, if one of these other replicas have rows that the already\npushed replica doesn\u0027t then we can be missing objects in the listing for\ndays.\n\nThis change adds a new check, if the remote sync point is 0, ie it\u0027s\nnever sunk to that location before, yet the container has more then\nper_diff * max_diff rows (more then diff_cap) to send then fall back to\nrsync_then_merge.  This will allow a cleaved shard to be moved off the\nhandoff location as quickly as possible.\n\nChange-Id: I877981acda14aefaadb06bb13a233078212cd545\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0c61936c497bd790ac65e61cb64cbecc00182b01"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0c61936c497bd790ac65e61cb64cbecc00182b01"}]},"branch":"refs/heads/master"},"6628cbf75f16015d39aa80882664ba129c6cb53d":{"kind":"REWORK","_number":3,"created":"2024-12-18 05:54:38.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/95/901095/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/95/901095/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/95/901095/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/95/901095/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/95/901095/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/95/901095/3"}}},"commit":{"parents":[{"commit":"b4579b414e51db88bfbcae5d3d678ed8fe1a7941","subject":"Merge \"CI: point antelope job at 2023.1-eom tag\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b4579b414e51db88bfbcae5d3d678ed8fe1a7941"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2023-11-16 03:07:28.000000000","tz":660},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2024-12-18 05:54:03.000000000","tz":660},"subject":"db_replication: big containers that haven\u0027t sunk should rsync","message":"db_replication: big containers that haven\u0027t sunk should rsync\n\nWhen the sharder cleaves we get large shards in handoffs. The first\nreplica will replicate fine, as it gets a 404. But the second+ replicas\nwould also never have sunk but there are a lot of objects to move. If\nthere are more the max.diffs, which there always is, then the shard sits\non the handoff node waiting for container replicator cycles to slowly\nmove all the data off the handoff location.  This in production could\ntake days.\nWorse, if one of these other replicas have rows that the already\npushed replica doesn\u0027t then we can be missing objects in the listing for\ndays.\n\nThis change adds a new check, if the remote sync point is 0, ie it\u0027s\nnever sunk to that location before, yet the container has more then\nper_diff * max_diff rows (more then diff_cap) to send then fall back to\nrsync_then_merge.  This will allow a cleaved shard to be moved off the\nhandoff location as quickly as possible.\n\nChange-Id: I877981acda14aefaadb06bb13a233078212cd545\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6628cbf75f16015d39aa80882664ba129c6cb53d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6628cbf75f16015d39aa80882664ba129c6cb53d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
