)]}'
{"id":"openstack%2Fswift~901335","triplet_id":"openstack%2Fswift~master~Id79c156432350c11c52a4004d69b85e9eb904ca6","project":"openstack/swift","branch":"master","topic":"p-proxy-namespaces","attention_set":{},"removed_from_attention_set":{"7233":{"account":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"last_update":"2024-02-01 09:12:48.000000000","reason":"\u003cGERRIT_ACCOUNT_7233\u003e replied on the change","reason_account":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"}},"34930":{"account":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"last_update":"2024-01-31 16:46:45.000000000","reason":"\u003cGERRIT_ACCOUNT_34930\u003e replied on the change","reason_account":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"}},"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2024-02-01 11:59:47.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Id79c156432350c11c52a4004d69b85e9eb904ca6","subject":"proxy: only use listing shards cache for \u0027auto\u0027 listings","status":"MERGED","created":"2023-11-17 19:01:13.000000000","updated":"2024-02-01 12:00:50.000000000","submitted":"2024-02-01 11:59:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":75,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"901335-p-proxy-namespaces","meta_rev_id":"296a332b638c2244370fe575462b1879f74f9553","_number":901335,"virtual_id_number":901335,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-02-01 11:59:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"all":[{"value":2,"date":"2024-02-01 09:12:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":1,"date":"2024-01-31 16:46:45.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"all":[{"value":1,"date":"2024-02-01 09:12:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-11-17 20:19:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-11-19 01:01:40.000000000","updated_by":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"reviewer":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"state":"CC"},{"updated":"2024-01-16 06:35:44.000000000","updated_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"reviewer":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"state":"REVIEWER"},{"updated":"2024-01-25 00:32:03.000000000","updated_by":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"reviewer":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"state":"REVIEWER"}],"messages":[{"id":"809c63ae734d2b1e3c7de8cb1ad4ee67c603c02a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-11-17 19:01:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6ece2b59690a97d6c8c0a9dc50a283e6a74e629f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-11-17 19:14:46.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"8bc917311927fae512faa451c064ee1a8cd87f7e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-17 20:19:44.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b253abe10ed8499d9bd8b4f39c82d291\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f5d2a560cd484e59a00f293c43b1f3db : SUCCESS in 8m 37s\n- grenade https://zuul.opendev.org/t/openstack/build/41c42928f0af4d949d3a2dc5c681a44f : SUCCESS in 45m 13s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/c11485984c074787bf580fb83c4f9165 : FAILURE in 37m 20s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/54493d9227d9484ea0065472f313afd8 : SUCCESS in 36m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a0b2757936914dec97e491bf45fd401b : SUCCESS in 54m 15s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d7f6e92df1364695bdc5cbb7882bc58f : SUCCESS in 19m 16s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/cc743a244e3744f1b622876415c59b54 : SUCCESS in 18m 23s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b8c5f49016064608b65afed81946203a : SUCCESS in 24m 05s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/b11863973db341f7a7f69941717613bb : SUCCESS in 10m 09s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/60fd70b4856743cf9b71e5953529606c : SUCCESS in 5m 30s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/5bc24fda801f4e68b3b8056c083d5c5b : SUCCESS in 20m 13s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/c024e6dbd8b84401bbcfdb78e40742f3 : SUCCESS in 21m 22s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/09575f4f45264248941dbbcfc75fe39d : SUCCESS in 20m 27s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/c4188343dfb74ba48751d7b8ad481ac1 : SUCCESS in 13m 49s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/f2f980019d434e808f967b4c413843ba : SUCCESS in 22m 02s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/ec9266fd61154c49a7fd028eebc32268 : SUCCESS in 22m 08s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/30c2cbfe540e44f7997757a89cd17689 : SUCCESS in 18m 20s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/60e4fcdb02444760b2e009316836cdc7 : SUCCESS in 15m 35s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/4ac51f0353d94e9b989711b002c3777b : SUCCESS in 20m 48s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/cbca605e6a38404d86756719e8dd1636 : SUCCESS in 17m 32s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/aabb3a3527cb4a1d8165b9c1dc0fbc6e : SUCCESS in 15m 33s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/fc2b5af5cf04415bbd91afbea0660040 : SUCCESS in 7m 57s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/e9d1d1caa70e47c1b0221e20f9a043a0 : SUCCESS in 20m 18s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/0a1523045b59415995c67af86584b44c : SUCCESS in 10m 17s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/b662550868b24feab594d6b768cef5ce : SUCCESS in 52m 57s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/1cb4c432a74e4938b713314e0a89b841 : SUCCESS in 58m 35s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/04931919bad8497d856a85fa373cdfc6 : SUCCESS in 40m 30s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/9bb52b53385540cfa1357b07271fd704 : SUCCESS in 47m 51s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/eb6e2163ee8e4780afba935056c3ab91 : SUCCESS in 24m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/743b73c714c841949c07875395d1e5c1 : SUCCESS in 5m 55s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/47061ab6c2c948249b33137b0c806906 : SUCCESS in 24m 32s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1e37f41bdad3491a944a8ea4837bfba2 : SUCCESS in 56m 46s","accounts_in_message":[],"_revision_number":2},{"id":"0318f7815a83a74f8e59eb4ffd56332236408e6f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-17 20:20:36.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/185b9d3dde0f4c488fbc52bbd214315c\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c8695784e41d4f3e86807c0a88aed39a : SUCCESS in 24m 20s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6765ed1f08d64eb9a54e2db59c687d21 : SUCCESS in 57m 03s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/130b515fd5914cf5a6058275b6331f56 : SUCCESS in 30m 52s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/13e68b9bd36e4529853b4aca7a3fc566 : SUCCESS in 1h 01m 43s","accounts_in_message":[],"_revision_number":2},{"id":"f838e91c28b97b02bca161a433654bc683e3df34","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2023-11-19 01:01:40.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"172a4d591fb94e0d1eb8bcbcfbf9f4b0fb8a1157","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-11-20 14:48:10.000000000","message":"Patch Set 2:\n\n(4 comments)","accounts_in_message":[],"_revision_number":2},{"id":"5b7df78897d7242cca97041c256b345d6b62f7b5","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-11-20 16:41:50.000000000","message":"Patch Set 2:\n\n(5 comments)","accounts_in_message":[],"_revision_number":2},{"id":"1d891d692c88432e787aae6d8b468c3149660fdb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-11-20 16:47:37.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"3f2fcb9d6da55bde1b2e05c60e2c21e117470786","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-20 17:40:12.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a38518c392934b27a9155c192d539c04\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/26180b7f76fe4479a9ac0e7ff100d673 : SUCCESS in 18m 31s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7b2f5324005a4f1e8a6eee508c4093f3 : SUCCESS in 49m 47s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/5656d1d2659b45aab492787fa428edda : SUCCESS in 46m 32s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/289a8cf9524a4debbd18b816803e16af : SUCCESS in 21m 36s","accounts_in_message":[],"_revision_number":3},{"id":"068027ac452e9b89b26468bab9148ad3b82fb0a2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-20 18:05:49.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cb429b5a52c441328f372d0c85e6a00f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/062dddba4ce44a1c96441f1951f10765 : SUCCESS in 9m 25s\n- grenade https://zuul.opendev.org/t/openstack/build/c82fec9f8d374ef895384f673f652ae8 : SUCCESS in 52m 54s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/7581e4e66c094681999d551f7c01ef2a : SUCCESS in 48m 39s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/f3dce69f1baf40fbb74302aeee1faea4 : SUCCESS in 49m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cfbb0d29ceca4cf4ba9659b20a3bca2a : SUCCESS in 1h 11m 36s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4e1300260233429f9f6521b3a0eec7eb : SUCCESS in 20m 50s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c7f70e3e589245368f3a1f1385e532ed : SUCCESS in 20m 19s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e20ff81d649047dea6456299be8c477d : SUCCESS in 20m 46s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/ac371d703b03406f87b718c4330f0cba : SUCCESS in 9m 30s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/4d1089ff21964f36975373fbabd3df0d : SUCCESS in 5m 58s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/327f385e57ce4f4fadfccc797b7877d4 : SUCCESS in 18m 43s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/eebbed92861b4ae0934e48111e31d788 : SUCCESS in 16m 15s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/7e63ba5620364a9dbd031c6638db91d5 : SUCCESS in 15m 53s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/3bd28feb50f34a09846344328ae94b84 : SUCCESS in 17m 04s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/a829e8c3626d453392dc988bc2551c73 : SUCCESS in 16m 17s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/ae8e7b70aa0f4676a143cd78bf1c5201 : SUCCESS in 18m 00s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/6f1ba626ae0f41739199821c737f3049 : SUCCESS in 20m 15s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/66cd922f32e94201a38bc6af82b88810 : SUCCESS in 17m 42s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/07b62f2bf7bd4b668c57a6e58099960d : SUCCESS in 16m 51s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/733515c294774e12bc104814fb4cfb84 : SUCCESS in 16m 10s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/60fbaa2ebc0c471fad0e7fe914e51b71 : SUCCESS in 17m 39s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/82ddb024d9454e97b9ff137216a6b5bf : SUCCESS in 7m 33s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/49500b949a094c6cb61486ca0c382768 : SUCCESS in 16m 30s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/3ee8fd6cf5444a0b8df72042b2207b42 : SUCCESS in 8m 34s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/979b6026097e4763b9c05adc81bb7df0 : SUCCESS in 58m 29s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/9570504592e34c08a06b4697caf9c26b : SUCCESS in 58m 56s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/447ce5eff28b4062b6a5723c771a4ce5 : SUCCESS in 55m 26s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/6c208143de3249868eb70ca7a5aaee30 : SUCCESS in 40m 06s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2b62114edb1c414781a17f0354b57e2e : SUCCESS in 16m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/56bf7b7d51ff4f7b9149852faece6739 : SUCCESS in 4m 19s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/dce9d98f5cbe40bc9d60716f08049d0c : SUCCESS in 21m 24s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5f5af0881d484cb6a6927cb85396e499 : SUCCESS in 1h 07m 33s","accounts_in_message":[],"_revision_number":3},{"id":"7b3c88c4ca33200ba392d3023a7807e102c0e870","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-11-20 18:37:10.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"257dbf1277999c33a318b0df977d357f42d0f1bc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-20 19:56:12.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d94a8bb23f4c46b5956cfee4be029103\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/33cba30528af41a4ad30b677689f5622 : SUCCESS in 9m 04s\n- grenade https://zuul.opendev.org/t/openstack/build/fea4ac890260444aa76bb8a789542c2e : SUCCESS in 1h 06m 30s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/b6e7215816fc4230a2e3764dedd426de : SUCCESS in 1h 08m 56s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/a6007580bb8442148565d34c58473459 : SUCCESS in 53m 55s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3184a6e1ce61403daabc2590b5a4432c : SUCCESS in 1h 04m 11s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/24cb277ad9e54a788943453c4be688ed : SUCCESS in 20m 52s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/945196a7115b45e88a7aa7efd8ae2975 : SUCCESS in 19m 09s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/03394a8808b848ea8a95cadca71c6fed : SUCCESS in 17m 44s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/829a5e89f1e84899968300bdc8421a69 : SUCCESS in 8m 56s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/2cd7f8ccbca9473a8e26fca06523c0b7 : SUCCESS in 7m 11s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/c81158d1272149368f83187eba48a005 : SUCCESS in 17m 04s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/cee9bfebac2f42c1bfafcc9d2bc0d896 : SUCCESS in 15m 01s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/63a39b1baeb54dca933306ec7eb9481c : SUCCESS in 13m 39s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/573379776dac40919d7779884aa61df4 : SUCCESS in 16m 08s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/b56f1071d9904a3e9d51f4768b693b9b : SUCCESS in 14m 38s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/42ffc9027a90407fbfa60e2c86f64a1d : SUCCESS in 18m 05s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/cdca801a48154cc0af3165a54884dff9 : SUCCESS in 17m 29s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/f01ad770772346a1a2d4c2ba91676bf9 : SUCCESS in 19m 34s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/4fb24744d1794045a6b2efef4ed7707a : SUCCESS in 13m 50s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/d1db9ed5e61c4c55b86f8f9db4c1830c : SUCCESS in 15m 05s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/992be760e18e44cd90d53a992b26cd86 : SUCCESS in 14m 50s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/4f5e936ee2c44d7e81325bb305011d06 : SUCCESS in 6m 41s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/6cb65a2e91cb4af991e966f7ad52d605 : SUCCESS in 20m 11s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/5adc06f21a764435b48959322ea8240f : SUCCESS in 7m 34s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/74b7cbc5d7364cb783010d9cd5b65cff : SUCCESS in 54m 59s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/d48d7db7c4a14134a79ef03cd2946d1b : SUCCESS in 59m 39s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/da09d8331c9c443e9389d48988d1fb76 : SUCCESS in 53m 10s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/3d388b4908294af8a9b455647bb684c4 : SUCCESS in 49m 36s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e7b4d97bebf3433abba6f9da764a8712 : SUCCESS in 19m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1b9d2e8169c64d8e841aebc5c34a49c0 : SUCCESS in 4m 39s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/a40c66bbc5d1440c80fac85ea98486ba : SUCCESS in 21m 27s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/20c5c8b400c743609d09be5ef4e379ed : SUCCESS in 1h 07m 19s","accounts_in_message":[],"_revision_number":4},{"id":"bffd451b2375b301ca8142813e5fcc557fa84a22","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-20 20:02:56.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ecd8fae57dd84fdcbdeaec3b664ffa71\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d785d5c5cc6e4700a9e22db37b124b54 : SUCCESS in 1h 04m 21s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e1cc7eec23e84e33b2c75c7470b952a0 : SUCCESS in 19m 01s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/54059b946852495e810db6e8c4ad5509 : SUCCESS in 1h 23m 03s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/44853dad0c31453990ccf60bf0b27d6a : SUCCESS in 25m 23s","accounts_in_message":[],"_revision_number":4},{"id":"0900ba74236df87533ad288190e7dba24ad91bec","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-11-21 10:54:11.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"28497807624e4ccca1b7a8801154da9023e39bc2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-21 12:05:23.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a9e930803c264e2fbe931c671f425941\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/03a5f0c1d38a449cb957c7ce401286a3 : SUCCESS in 8m 43s\n- grenade https://zuul.opendev.org/t/openstack/build/e4f4d7c3be63419d9fa1c40d94cded72 : SUCCESS in 1h 00m 50s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/93bc62303d31444480b3de97f3673a4a : SUCCESS in 56m 36s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/249ef7c1e88143da85ccb2c16c91da1a : SUCCESS in 55m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d954cb5d8520486284d30dc380fb0df0 : SUCCESS in 1h 03m 48s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9196e95a1b4c42c0b2e4b918804f417b : SUCCESS in 17m 22s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b319b9046c9747f28dc6838ed5607d61 : SUCCESS in 20m 51s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b3d54a8502cd425688f6422e30b34254 : SUCCESS in 21m 04s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/83774d6dbbaf40f7801d763a7b2fd41d : SUCCESS in 9m 53s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/c6f0f79471eb4663963ca4ed4f459840 : SUCCESS in 6m 01s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/b060c1f030a34e7698a88adeb91eaf95 : SUCCESS in 18m 10s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/571e1369062c40d4a6908a2cb9e6d601 : SUCCESS in 16m 58s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/d59afec10ff84e328672009e37b53625 : SUCCESS in 18m 05s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/c00f954e7ed74d0cbfb1e2bbd3056873 : SUCCESS in 16m 03s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/42cd3e33e22f404fac4a3029bb1569ac : SUCCESS in 17m 55s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/03ad04b82285433eb77574615fb2f891 : SUCCESS in 16m 50s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/0aef3148751e4fdcb486d36a003a0b73 : SUCCESS in 18m 06s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/91651b0f723d430093f06a89dce15784 : SUCCESS in 17m 33s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/b20e0768e43e4a3f96a3c1b0bda3d4e7 : SUCCESS in 15m 38s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/cafb2d3afbe341c4a154f724a2d5d78c : SUCCESS in 17m 56s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/4d69e3cf6e8f4fc39e8297dde7d9436c : SUCCESS in 16m 14s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/2ed755987d6d49eb85bb935ce2d3b846 : SUCCESS in 6m 57s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/7f8b4ab52a7a4c68ab2af7d360b9ab0e : SUCCESS in 16m 54s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/4a484ace250c4f058ade62aeb579d68b : SUCCESS in 7m 36s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/8dc8aee2192d4b0f9963453587a361b3 : SUCCESS in 55m 49s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/540b47423b204a21bb3a40b9cb7c7c30 : SUCCESS in 1h 01m 00s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/8e41ff7bf634456e9c04d304c273f0d1 : SUCCESS in 55m 00s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/663e7a731dab4ef6ade240a11c5752d8 : SUCCESS in 53m 49s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/80a4012d1973422abca6fcf6cfa5eefa : SUCCESS in 20m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8d5f763ba2fe4ecea63cd621daa80513 : SUCCESS in 5m 17s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/2fbdf0c8a193481e870ecd321d87f789 : SUCCESS in 20m 05s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/779e45d130c94a0981b91ead0a93f849 : SUCCESS in 1h 02m 49s","accounts_in_message":[],"_revision_number":5},{"id":"76daef0c2184796b256318ec46b678fafb08934a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-21 13:20:11.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5fc9789263e64861b4055223159ac9e1\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/71177418e1c342ab9af6366a7c3fbb5b : SUCCESS in 18m 40s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d2978804e5a54657af2e4af478986a96 : SUCCESS in 59m 06s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/c5a02a2e044a4b85bc89a2497c6b2a01 : SUCCESS in 1h 09m 21s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/65e1b1799e584252aa755ef05a834a10 : SUCCESS in 23m 17s","accounts_in_message":[],"_revision_number":5},{"id":"a335accc8d2820e54d68315088c164a7fa5def93","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-11-21 18:37:29.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"f8278a4ad21e316ea4a7f1b067db8af4ac2d1551","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-21 20:33:47.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7b84df0b92304c639e5bce5add8371fc\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/246c76a614874a68b14fcfdf906c5dfc : SUCCESS in 9m 26s\n- grenade https://zuul.opendev.org/t/openstack/build/189619190234464092e69df8d614731c : SUCCESS in 1h 03m 12s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/0aeec066c4f7486a8620a3c578d3f19f : SUCCESS in 1h 04m 00s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/04b01b06bfbf496db5044f698a0d7d14 : SUCCESS in 48m 51s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8d0db4994c0e4669bfc165fbffea4075 : SUCCESS in 40m 07s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/28bbaf65044a4617b9efb85f297720c4 : SUCCESS in 20m 15s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9233dc0ef1dd4b948f51c0e90080918f : SUCCESS in 21m 05s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/51b84131842d4146bc62f10e2f424357 : SUCCESS in 19m 56s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/1deb4de38e0248f5bfbef9170ed78fd1 : SUCCESS in 9m 50s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/fa72b7e6892c4cbcae8ec504d276469e : SUCCESS in 5m 46s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/f525bea9e2ea4063b15c6f26bb82cbcf : SUCCESS in 18m 44s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/f0f51cd50db64110872a63861cecd515 : SUCCESS in 15m 53s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/f2bb3168f7ec448fb8e5f9fb1c62b51e : SUCCESS in 17m 17s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/afb8d0e7eaee4e16b80f4e18efde90a8 : SUCCESS in 17m 14s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/e17df3d671044565bd43ecec9c55e8f9 : SUCCESS in 16m 45s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/96144e10a55f4e12ba1646a22b2c353b : SUCCESS in 17m 54s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/10431eac394743228b84ecb6685c19b2 : SUCCESS in 17m 26s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/7668daa567bb4caea7b964437e87547f : SUCCESS in 19m 13s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/9f20e1f7994a42aa8b66579555434868 : SUCCESS in 15m 48s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/a766749a62a745b0bef482091c4f4d69 : SUCCESS in 17m 14s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/b8a881ef1d11441592062733ef524a96 : SUCCESS in 14m 55s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/08e222c8440647daa91b00050146bb2b : SUCCESS in 6m 56s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/c15d288ac2ad40668523d30d94a9413e : SUCCESS in 15m 35s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/52820758327a4e62bf53f928009d53c3 : SUCCESS in 7m 05s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/20a49a909cee43c183cc1e846ab27e4b : SUCCESS in 56m 48s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/6ee97861a6ea4c58a94876b7aafdfef0 : SUCCESS in 1h 18m 11s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/124bc90545c44b62939a8d238acb2c8e : SUCCESS in 54m 52s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/9fa3cb1fce3942bba39a16a43fec36ec : SUCCESS in 59m 27s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5dba9e8868ba4cf9b52bb824dbb2624b : SUCCESS in 17m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/73ab325afeca41b2a646c1a37c9bcdd9 : SUCCESS in 5m 07s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/e5d122d6ec72452193095449d7280d4d : SUCCESS in 16m 00s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/34c9422225bb41c1b1f087f61d46909b : SUCCESS in 1h 08m 51s","accounts_in_message":[],"_revision_number":6},{"id":"9d0e6d831024ae7ff192af7cac814ea1647fcade","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-21 21:03:02.000000000","message":"Patch Set 6:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b4c1f42dd35d4942b1d7c5cb58fd0dd8\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/3298ff7b5f354e19b0104ada61e3d138 : SUCCESS in 1h 12m 25s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7721e895ca1b4b1c973a7e89f9d7b916 : TIMED_OUT in 1h 31m 02s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/1477585745e744758ab5201ace112b3c : SUCCESS in 1h 09m 46s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/58cea399b38041c0bc2d16619ab005b5 : SUCCESS in 1h 24m 43s","accounts_in_message":[],"_revision_number":6},{"id":"d43b0debd8d04486272c72e2e1ea871c6064c62d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-11-22 11:07:53.000000000","message":"Uploaded patch set 7: Patch Set 6 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":7},{"id":"7cd7944adfe85a20e7a8659f7f4eced1854aaf18","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-22 12:47:38.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3b5d328f0be348fda34b76cc78dfabb3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/510b4058017141899b828a15d9ea1b8e : SUCCESS in 8m 52s\n- grenade https://zuul.opendev.org/t/openstack/build/4b03cbc9204145e59ec8cd39d1e5ea1e : SUCCESS in 44m 49s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/f8fb5a3bfeb14dcdbfafe20227fc62ae : SUCCESS in 1h 03m 08s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/178728efda8746569ba83f2c97af89f6 : SUCCESS in 44m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8d98f0f10bfc442a9f514694cb91e6b3 : SUCCESS in 1h 03m 25s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1bd84d89b324478fbc8488b956612bcc : SUCCESS in 17m 08s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9309da7a0f864a2eae4b09c6dca25d06 : SUCCESS in 17m 29s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6bd202b085ff4561aa0e059b9c83cba1 : SUCCESS in 18m 19s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/90930506e8cc4a9d9ea7f3df2775fdec : SUCCESS in 9m 32s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/14305a98cf1e47cc8f246f9e88a4f3ab : SUCCESS in 5m 47s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/072ef11bbc66499ebf45132c1c8ab670 : SUCCESS in 17m 36s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/441c0f4b1f82469a91b76c68982ec49b : SUCCESS in 18m 12s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/f9c6aa945c0342b491395da1ac4e33e6 : SUCCESS in 15m 56s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/50d0cec1f5cf4332846194793dea0fbd : SUCCESS in 16m 01s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/ed20e6eac12b461880ac84c9f5c4e749 : SUCCESS in 18m 36s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/d1af6d7f05354303b27d50caf2d4e71f : SUCCESS in 18m 03s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/a3b804a82b044549a02c71939639100a : SUCCESS in 16m 55s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/bd59a767e92749a8ba2fb92d514cc263 : SUCCESS in 19m 25s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/ce92aececb484f03a8e085bbe054cbb4 : SUCCESS in 17m 32s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/ea20a4528efb4015a593aedf3613ccf4 : SUCCESS in 16m 13s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/4c0f2c48064147509f97fe6b9df2b8be : SUCCESS in 16m 52s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/ebe2c75a5cb44e2eb80a7eb9795801d6 : SUCCESS in 7m 03s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/7e47c6dbcf07480f871c888e77cde450 : SUCCESS in 16m 05s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/236611f3523243e78240196ab8eb1de0 : SUCCESS in 7m 54s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/5005243ca2c6464983c6aa84e7e2bf84 : SUCCESS in 53m 08s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/311aee6999ba4562a120e90337031d2f : SUCCESS in 1h 00m 11s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/c9b423eeafe04be8bf17398d27ecec96 : SUCCESS in 57m 08s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/4332aca62c364918a08dea0fb7458a8b : SUCCESS in 49m 53s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d0dc5f30defc45c8b9eb3a4df576b75b : SUCCESS in 18m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fbdb3bc1f4bd4982ba9e8f1790ab3c47 : SUCCESS in 4m 39s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/34ded3d96fb546d49a05b24cc82fb7f1 : SUCCESS in 20m 54s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a6362914253b424b94449d62e535e559 : SUCCESS in 48m 01s","accounts_in_message":[],"_revision_number":7},{"id":"5f7222d083ea91f2e1cd2e3bb52c8b743bed5bc1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-11-22 12:52:13.000000000","message":"Uploaded patch set 8: Patch Set 7 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":8},{"id":"080b53e6ad6fc9a97c62ad9f22b07be6ee927b4c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-22 14:09:23.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d8bc5c8013644fd59463ac47cc6d3b4f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d41ec97382184045a7c3654368bca954 : SUCCESS in 8m 30s\n- grenade https://zuul.opendev.org/t/openstack/build/f63df9585721478d8de879ac59e5499c : SUCCESS in 1h 01m 21s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/0af6b21f294a42319828bd502ae4cddc : SUCCESS in 1h 00m 15s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/f3b2433da4104829a439dc78d0b9a360 : SUCCESS in 43m 58s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2b5d2189c451404b9ddc30e4b6c5925d : SUCCESS in 1h 07m 49s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7be72e4139c843128478cc26d0dc4fef : SUCCESS in 21m 08s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/da0183ec441246e8aec31b43b7347b39 : SUCCESS in 19m 01s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/eef686d2b22d4804b9e65c076d8fe001 : SUCCESS in 25m 34s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/cb7c1a673f6641e48408f35ae7beaf32 : SUCCESS in 9m 28s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/1eff5bebb3df40adb982c2e99ea1d66c : SUCCESS in 7m 01s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/d9f0b56d96d941979db41d3e69c0dac3 : SUCCESS in 16m 15s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/803f34cfbea04ac2acba752e8692a04b : SUCCESS in 17m 30s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/0e17854c368e4bea831f6506d6102326 : SUCCESS in 13m 05s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/a9dd1803a40d4db7bf5dec4f12803344 : SUCCESS in 17m 40s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/ee37c28dd4da4c5abe7bdb7d7d676ece : SUCCESS in 18m 36s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/a9cf3f20ec7041399aaa287c58a2f6c1 : SUCCESS in 17m 36s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/10608af121e9486ba5ddde8e57b14e3e : SUCCESS in 19m 42s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/510b72c065cd41d6b66cec42516afc8d : SUCCESS in 19m 41s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/66a536fca9484a0086f92f0471348a28 : SUCCESS in 16m 02s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/f1c2be4adf3a413d8680271909b7343b : SUCCESS in 17m 22s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/8a62bbaf10e941fd863a1b57cb10e891 : SUCCESS in 17m 49s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/9f1ee00a53d241a4a22967dc6a7350d6 : SUCCESS in 7m 38s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/48aa2dfa185348c5b471cb2ab7bb0ae3 : SUCCESS in 18m 24s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/e448958c7f6c4cea9a3f59adcd8691c5 : SUCCESS in 7m 27s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/6fb326e98353467691358e1681c75ed7 : SUCCESS in 56m 51s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/b424129945a044e2874203b8c5158347 : SUCCESS in 1h 00m 46s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/4e86b8bc8d2049cebd69549b18efbcea : SUCCESS in 42m 18s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/5ca24911ca604ae1932a9476b620a5fc : SUCCESS in 40m 40s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b125fdeffb7d48c4b7a7f25eee804a6d : SUCCESS in 16m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6a86c638b46849e185dd38a5dfc417c1 : SUCCESS in 4m 58s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/53e3ee8c6e054fd187c656ba7a72881f : SUCCESS in 20m 47s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/94876668932d4e6f8c000784075380bc : SUCCESS in 1h 03m 31s","accounts_in_message":[],"_revision_number":8},{"id":"f55a6cfa3f8b79cb34efbf4b8c3bdd5a24016d57","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-22 14:20:44.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5b24e78e622a465598328172dccc1297\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/72652c2598ee47efa46e1d677b902f00 : SUCCESS in 1h 25m 40s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/df94fc4c2afe47afaf6d64aa0ad05361 : SUCCESS in 16m 17s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/d7eeedda322444149f884b84f111e6ba : SUCCESS in 1h 15m 11s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/9856e7ab73584af4a982f2a577dae866 : SUCCESS in 22m 56s","accounts_in_message":[],"_revision_number":8},{"id":"26afecc7612ff5015a225c62b3bb855660485713","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-11-30 16:22:51.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":9},{"id":"532c8a3577a8e3ddaad22a65e79fe92ea0a76fe4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-30 17:35:52.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/587378324efa4894af4b5dd003665a94\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1973f297500f4725a6cc3c973a35bf82 : SUCCESS in 9m 31s\n- grenade https://zuul.opendev.org/t/openstack/build/c08a5fa4f27c4d95b6cd1c8378f9284d : SUCCESS in 46m 18s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/f7528fc6feff45529b5b3f707bc7740a : SUCCESS in 1h 05m 54s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/83c80d93e5bd4f6d95fee4fb7ed45559 : SUCCESS in 58m 51s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b99a83294aa24bb3946c61ad27f054bc : SUCCESS in 1h 04m 42s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ead8ee6315d54752ab8f0e20896bd61f : SUCCESS in 16m 40s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5eaf5685f1cf4d1fb5fe6000fab26f2d : SUCCESS in 21m 18s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3f08c76e573549a585b5545cb2cde969 : SUCCESS in 19m 40s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/df1145588046476da253f86119cc96f0 : SUCCESS in 8m 59s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/64733e4d4e7249319b6eaf93d634560e : SUCCESS in 5m 51s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/3e45453bb79e4b8c8c4c0e30f5e47087 : FAILURE in 18m 18s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/d0983c63d725457e93e8695f9a3b5409 : FAILURE in 21m 02s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/25023f816f8e4e6193dd3fc69a079954 : SUCCESS in 14m 12s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/84c156b8987c4cc99835ba7bd9a99778 : SUCCESS in 18m 25s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/2bdf4a988fe44bbfb1b8db9af454515a : SUCCESS in 17m 08s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/a2924e3a240d4e028ac2f44b12d87098 : SUCCESS in 23m 30s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/fb012d6f7a0e4ffcadf14d2175e04bae : SUCCESS in 17m 31s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/b50e6e23e8a24d34a84cb909fb132271 : SUCCESS in 19m 02s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/24cd70c7441a45a396a4cd72153b3194 : SUCCESS in 17m 03s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/6c44a2ff37094eaeb9fb380951d59742 : SUCCESS in 18m 10s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/6637ca88fa5d406fb01d98d50f5a5d7c : SUCCESS in 27m 31s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/eba3fe922a314728bf9a0965a5b2dbd1 : SUCCESS in 7m 57s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/e8070e5d04eb4b7eac16a1f7083ca500 : SUCCESS in 18m 28s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/915a1b691c344de9abc05a355db5e142 : SUCCESS in 11m 02s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/ee2c9d87e27740d78f3fa89ebf2d5ed0 : SUCCESS in 1h 08m 22s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/de26953012d749028ccc680052ec397f : FAILURE in 53m 25s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/1a635a7fa5584f54b86a49b4ee2eb694 : SUCCESS in 54m 57s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/d29037590fd541b08bf1768c263b5e7e : SUCCESS in 1h 02m 52s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/98cdd76d39a24a90a0427f79a1d5a53f : FAILURE in 17m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/57f9ef0a1f19433a821e783f454e453a : SUCCESS in 5m 11s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/c3997e839b25442f8c556e3c0e98c65d : SUCCESS in 21m 04s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8823fd01efc74e77b7f7494b292439cf : SUCCESS in 47m 06s","accounts_in_message":[],"_revision_number":9},{"id":"e0eedd126c69f3e6a9e041ee4494ce70b4f1bf6c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-30 17:57:30.000000000","message":"Patch Set 9:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/54b969940e47460eb0d4838047edcd5d\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/fd0d09fefc904c6ba82af0b9c80780da : SUCCESS in 1h 25m 13s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b477ce1063034f56a7009f012afb8abd : SUCCESS in 14m 46s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/0938dfe5bbb44adba3a454b0da579d30 : TIMED_OUT in 1h 31m 04s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/ab65622b80df4379a37cbd94e6145df2 : SUCCESS in 20m 54s","accounts_in_message":[],"_revision_number":9},{"id":"6cc44d8651272910cd7c25c674fb66b29373887c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-11-30 18:20:14.000000000","message":"Uploaded patch set 10: Patch Set 9 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":10},{"id":"c20b79707910bcfbc686da35a5c0917881964e71","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-30 19:40:53.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/050c843c855d4461983f02752aa7c639\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a463515461914408a0d18e442c9af93c : SUCCESS in 9m 06s\n- grenade https://zuul.opendev.org/t/openstack/build/ed42c7cc44154f00bfcb7e529a5aa9a4 : SUCCESS in 1h 10m 35s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/51fb7b91bcae4b70b77fc9544d94f068 : SUCCESS in 1h 11m 33s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/bb44f9a4d0de4c95bb4f31a3a54911e9 : SUCCESS in 58m 33s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/51e8729b968a440399f39f855f7337ab : SUCCESS in 1h 12m 16s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9b3aeb65406e4184a587b25ed4626fa7 : SUCCESS in 18m 50s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/863cd33145aa433f9139a71da702d34e : SUCCESS in 19m 15s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/582a79f9083c4df39e3b2f22c8fe7045 : SUCCESS in 29m 35s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/b92fff9495634b13ab8f7eb1d46f79f6 : SUCCESS in 10m 13s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/2ea795e6df5b4258943016fd0f37ac23 : SUCCESS in 6m 20s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/3abef9fd284843508c62724f1d3479dc : SUCCESS in 19m 23s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/f99b19f2e52b4510bf7c7b78dd318609 : SUCCESS in 20m 51s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/d388ac85bf30483b9fb524a767fb99af : SUCCESS in 16m 13s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/2ca97b6a1a4e4523929f503241bdee0c : SUCCESS in 19m 41s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/b9197a86971f4e1798667cae627d011c : SUCCESS in 17m 29s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/17b255bf90a747ef886b5312e8c1d556 : SUCCESS in 17m 42s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/5313ac0cbd06421cab55f9f24d3db2fb : SUCCESS in 19m 36s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/aea5c8d002c44631944c084aec21c2ee : SUCCESS in 18m 46s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/75d0599d66c34782961324419b598fd6 : SUCCESS in 15m 05s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/0fee00e4556e4b5ba2ef1561feafaeae : SUCCESS in 17m 29s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/c434f2ad38f54c939783a0dfa7cd44a6 : SUCCESS in 17m 33s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/02e3ba24b70c48b999f823db48067248 : SUCCESS in 7m 42s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/0609a169c4be49b4bd7c0d1ae22a260c : SUCCESS in 18m 16s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/28feb722398d43ac9d11a09b0769908d : SUCCESS in 9m 10s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/c0ab3f119c9c4ddf86a3739c8199d2bc : SUCCESS in 57m 12s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/c66010e11d6f469ba7a9c42be31b8a09 : SUCCESS in 1h 00m 12s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/21006f4ee52d4ead9c2d56aca7e7647b : SUCCESS in 57m 29s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/71563383c34746db8456cddac68e1da5 : SUCCESS in 57m 07s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f6b24b9ff63545d6be8472c72cad889d : SUCCESS in 18m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/80fd3a185f844c0890875aef97a213b7 : SUCCESS in 4m 58s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/330f4f4ba2164f62bc786d3312ac8b47 : SUCCESS in 21m 41s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a3391258f55f4b60bdbc68f34c61de50 : SUCCESS in 44m 09s","accounts_in_message":[],"_revision_number":10},{"id":"a78b55c4bfd038401a275651d1b9bacdf11e2b88","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-30 21:26:45.000000000","message":"Patch Set 10:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4e996042fc21497787d0aa95bac237f1\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/22999f193fc641bdbf7c035bedd65226 : TIMED_OUT in 1h 31m 10s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/cc29672f06824c688ef9c20ac2b4948b : TIMED_OUT in 1h 31m 29s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/a840d66fb404426d92fc18141659a9ef : TIMED_OUT in 1h 31m 00s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/04908559fa5743578baac6ad3e96ee59 : SUCCESS in 1h 30m 37s","accounts_in_message":[],"_revision_number":10},{"id":"110fb3db4ce9b087e6896491dbfa906b4839fd8d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-12-04 09:30:09.000000000","message":"Uploaded patch set 11: New patch set was added with same tree, parent tree, and commit message as Patch Set 10.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":11},{"id":"0eaa21ff5c56d827743608b64a075d504eae4a44","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-04 10:48:10.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3b11c320a68e4ef1bf595239c88062d1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2f1a190e395240039b5e77c81337bd0a : SUCCESS in 10m 04s\n- grenade https://zuul.opendev.org/t/openstack/build/91966e3df1004ab2ad85877b880eb86f : SUCCESS in 1h 04m 38s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/2fb1adf5b5e042cd8c19672369b11ee7 : SUCCESS in 59m 36s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/e60efab1544047fb8d125e7fde725d64 : SUCCESS in 49m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e69577c55348475e9ae2dec9d73e0c30 : SUCCESS in 1h 07m 55s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d724cb5321f04ee58d3380c382963a30 : SUCCESS in 19m 44s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/708bd012a93741e8acbb139cc2bf80ab : SUCCESS in 24m 18s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c633f61a6500489dbb10bee0b52c9820 : SUCCESS in 18m 22s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/6f0288fed9c544f4acbbf0dd20b6c4e0 : SUCCESS in 9m 17s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/a8039bbce81d4bcba7879bd3fc182653 : SUCCESS in 9m 53s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/04245610f990438b8902d99024f86ad9 : SUCCESS in 19m 50s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/a482734796984892a4f04226a4889b60 : SUCCESS in 21m 39s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/efda80ee596049d686e713bbbab75a3e : SUCCESS in 18m 51s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/18bfc3da3a0944d9a2474ddbcb7833b9 : SUCCESS in 16m 02s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/c88d7da9cfc34d27b92ff2aaeb4bec9b : SUCCESS in 16m 06s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/5fc388dc701945268c6ea2ca13a54f1e : SUCCESS in 18m 13s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/56b71cfa13774dccbde16fdd4992de6b : SUCCESS in 18m 17s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/5c7e64e1b69d415aba86ef785df1f3bd : SUCCESS in 18m 17s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/64653adeaf0246109bf34d2a9104b02c : SUCCESS in 17m 02s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/1ce7ce492cfe424082a52802dd547395 : SUCCESS in 15m 35s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/e61249ee717d47389d0e1289937dc470 : SUCCESS in 18m 44s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/4eaeb56f98bb40ebad7a5bba91a78921 : SUCCESS in 6m 55s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/08a6c4d983bd41f8a5572dc27a24754c : SUCCESS in 24m 38s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/56ce4b4f16a94c4c89ca6d0c978dc2c0 : SUCCESS in 7m 39s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/1458a8ce1b1d4e29a8bb975ea0b584af : SUCCESS in 1h 09m 56s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/c58a9aa3b58442989e3c3aa11d8e8265 : SUCCESS in 57m 15s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/097e1c66f9b14032a3740722c1ba288e : SUCCESS in 59m 52s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/a23137b71e31449a8e6de2d0e7cc903e : SUCCESS in 39m 14s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/423f98bbc2f249478c3186b10b478057 : SUCCESS in 18m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8c138c586ec24ee5877d20633aec84ec : SUCCESS in 4m 48s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/385c3c9cce5e4f58ad7d01af21d6f312 : SUCCESS in 20m 17s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f8a1a8e3e8ac464bb15d3514d462593c : SUCCESS in 43m 58s","accounts_in_message":[],"_revision_number":11},{"id":"2904934c0cf121f20548b61e165de8e1c85489a6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-04 11:03:51.000000000","message":"Patch Set 11:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2f22eef88fcb498396daa7283313c8a8\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/5c46971b416c4e6e8f140e582dbd8073 : SUCCESS in 20m 59s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/39c68521d3df4a2dabae3b3b60150206 : TIMED_OUT in 1h 30m 57s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/1e028a1f12734a139f4893072de2a234 : SUCCESS in 25m 02s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/dedf0ac20d574bc9a66eb45c51a309a6 : SUCCESS in 1h 10m 58s","accounts_in_message":[],"_revision_number":11},{"id":"908d749237fb634fc0208773235cb25611398923","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-12-05 13:58:14.000000000","message":"Uploaded patch set 12: Patch Set 11 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":12},{"id":"18156d1297f7f2e5ff9b8f30d4b03372decc2145","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-05 15:02:23.000000000","message":"Patch Set 12:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5617af6bd6af43e9993b81bc57149feb\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/daa86689ec9a4cea8e66ff17dbcdd3e3 : SUCCESS in 15m 22s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/279f07dc1fb84f3d849feafa85872837 : SUCCESS in 18m 46s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/0d1d4d98ce744686ad3f72d689c6f753 : SUCCESS in 20m 26s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/89581b078e204192b76f433dca860035 : SUCCESS in 1h 01m 29s","accounts_in_message":[],"_revision_number":12},{"id":"50e780b4e19290e60b6b19c553732d624f62860a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-05 15:16:16.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d6a98cc99d0042109f2704d1c4bcbce9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/446eeae4c5ee41a19ec017fe3209bba0 : SUCCESS in 10m 44s\n- grenade https://zuul.opendev.org/t/openstack/build/f618c20b747d45b1b9aaff4f0d2e9c3a : SUCCESS in 49m 52s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/0fd9a1070709487993b45ab4d8e1a032 : SUCCESS in 1h 01m 44s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/3c3c50a8cb3d4db29652b615a8db2b80 : SUCCESS in 50m 58s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/961dc222613844f8aca08aa48ec367e2 : SUCCESS in 54m 04s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/51284b5ae1d445a78a112a2b6fce4766 : SUCCESS in 22m 09s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/01a87abd91a04615ab89a72755e6d92d : SUCCESS in 21m 20s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/816e3d2230824f1290d48812e1eb9117 : SUCCESS in 21m 27s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/4d3cab04693b471bab0428df773141b6 : SUCCESS in 8m 30s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/90ed19967b724a3297c7caa34fffc2f6 : SUCCESS in 6m 14s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/2cb8e9d1bf0d486d840d55c6c8fdd2bf : SUCCESS in 17m 16s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/cb06c50d22344d3983b6d641d966ecf0 : SUCCESS in 18m 48s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/fbc6c44829c94a4b96807fd3b9636c48 : SUCCESS in 16m 20s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/d1f8fea076ad4d8ea58e7f9371834eb3 : SUCCESS in 16m 29s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/1cdc5f1d2ecc43868b7a76cdd11f64da : SUCCESS in 23m 03s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/9de9690f09dc4e35b92c9e0e700e0e59 : SUCCESS in 21m 52s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/2c267195c9a348dd965bf7cefa4dff9f : SUCCESS in 19m 28s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/d1b231ea8ffe4e0c8b15efee981f275f : SUCCESS in 18m 34s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/02fc84f5dd324d18bc24132e0242fa8e : SUCCESS in 22m 01s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/91026fae58a040178159a6cede71b9c2 : SUCCESS in 17m 29s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/82bd58ade7134268be646433d4dccea1 : SUCCESS in 21m 52s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/1c86802d83c64b4fa5de66e169dde501 : SUCCESS in 7m 29s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/d62f2ee443884e1f8e4571904a039c5c : SUCCESS in 19m 47s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/2f643f094fc14542be6a4a5f116c9977 : SUCCESS in 8m 20s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/4e44ac0cf6a244e3a9ffd4395b825894 : SUCCESS in 1h 08m 08s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/1fbc7b47b02c4f2cbdb5c9484f7b22a3 : SUCCESS in 1h 03m 11s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/2dcb53a134fd406cb193599480336aa4 : SUCCESS in 50m 28s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/041ecd99decd489d98bf0a390628fa92 : SUCCESS in 39m 27s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/370d089a73dd423d80cccbf82465604a : SUCCESS in 18m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0fa1f6221d6e42d6a1e51d241ac15301 : SUCCESS in 5m 44s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/5df6cdf98b0045eeb971bc7195e3046c : SUCCESS in 26m 10s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/66f625b4631d47ee8fd5ef6addaaaf4a : SUCCESS in 1h 09m 11s","accounts_in_message":[],"_revision_number":12},{"id":"c302c2ff2f0c6e138304c03bba9c3751622185a0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-12-05 16:21:14.000000000","message":"Uploaded patch set 13: New patch set was added with same tree, parent tree, and commit message as Patch Set 12.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":13},{"id":"75f0dedb9326fe135f612316c3d5e55bf1a55831","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-05 17:31:02.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f5ca85d171104f9c8a3395f479036e1c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/04ce1770eb6c4401acc1ac1a8a4581b5 : SUCCESS in 9m 11s\n- grenade https://zuul.opendev.org/t/openstack/build/da0e5e1f9f2c45119ae2469e9ee18bc1 : SUCCESS in 47m 53s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/a3abe1accc524030bb21f81ffa96c9c6 : SUCCESS in 1h 01m 06s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/9e4977a1d6aa450fbfb6d94b0f8f0820 : SUCCESS in 36m 38s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8946e5ce79964abb944ce42fbb0f1b19 : SUCCESS in 1h 02m 57s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7282ca66cd8848158f3c690ce4e6ec29 : SUCCESS in 19m 29s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/fbeee80e8b9b4a308d9e887b6b2e9422 : POST_FAILURE in 31m 54s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4fba2d0d00474bbbbcf44b22a0e6aa07 : SUCCESS in 20m 22s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/95d9abf5439f4ac1875793464212e180 : SUCCESS in 9m 16s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/1cbbeb74c3804fc1b54f4b99fdee5a58 : SUCCESS in 5m 48s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/d93d47246d624250acaa20f5b77f110b : SUCCESS in 16m 31s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/c817b7e004df47598242d74dc2fec5e3 : SUCCESS in 16m 19s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/b5caaba1728f453b89a99fe0f8505881 : SUCCESS in 15m 05s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/041bd019a84e42458b819a58a70bc59f : SUCCESS in 18m 05s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/5e08995583c34019afb601ab165b2311 : SUCCESS in 13m 38s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/41ee44caafb4405e9bde903f74414a82 : SUCCESS in 16m 38s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/e631a2eb3fbd4cee92ed69891a143c45 : SUCCESS in 17m 32s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/ba11cd3c89ff496babe949c13841c47e : SUCCESS in 23m 41s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/084b2aad582142debc8447a0bc7f6ddb : SUCCESS in 15m 02s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/21fb05d1e5d94385a0c9323df4c87a9b : SUCCESS in 18m 13s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/3e27852eadf7438288df2eb707b7cd21 : SUCCESS in 17m 03s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/bea9646427194763abfd814a759dbc9b : SUCCESS in 7m 19s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/2a684f8ffaab413ba43524f91f5cdf71 : SUCCESS in 15m 17s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/5041a3b888a049168edacc58824850ca : SUCCESS in 7m 48s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/3ca01a4196de49af8e0f0bbd5afaeb6d : SUCCESS in 56m 38s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/68ce852963964bbba30e03f382c0e439 : SUCCESS in 50m 49s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/62e331c5271c4d40b269a91c6bb53f7a : SUCCESS in 38m 57s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/e912a3a1a9f244c6bd2ee5fd4950aebc : SUCCESS in 48m 29s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6aa13cc48ccd41fea578d0f9d2cd9a38 : SUCCESS in 17m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fce2e8848b914213a2dc7fca07b4b0bd : SUCCESS in 6m 10s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/5fd56bd26db54817b06f898dddfff0f3 : SUCCESS in 22m 53s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a078933caffd46b58c26403d0dccf5c0 : SUCCESS in 1h 03m 22s","accounts_in_message":[],"_revision_number":13},{"id":"7efbc8560ac82e8382e30b2096493b3b65a741c3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-12-05 18:38:59.000000000","message":"Uploaded patch set 14.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":14},{"id":"6f65ad07c2896c8c92964d49ec5b3c9d73e572fd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-05 20:42:44.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7dda9ff185094872be347f198604dda0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cfde32845f8b45f88fd318905cc3681f : SUCCESS in 9m 12s\n- grenade https://zuul.opendev.org/t/openstack/build/020d9c2ea29441559270800c05e4ab77 : SUCCESS in 1h 52m 01s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/48fe67c2de8a4870b08e2dc7c9507f20 : SUCCESS in 54m 25s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/b3fa4c4f38af4ee2bfadf04a8b02462f : SUCCESS in 49m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/dca7a1f942ef424eaa85bfaa9dbaf116 : SUCCESS in 59m 49s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b7cd56c3572745d3a21574bd16cd3495 : SUCCESS in 19m 24s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d2c968bed5ba472a9f499004038ad81c : SUCCESS in 18m 05s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e0a384e831b14ef4837a5d94cdbcea25 : SUCCESS in 19m 53s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/6686ab6385ba4a119b3429d4674a5307 : SUCCESS in 9m 01s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/df402677c3554853baffd421fca72d7e : SUCCESS in 6m 03s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/f0c50acff5db47a49c9cdcc52d5ff615 : SUCCESS in 18m 05s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/200a7a40f24b4cbc9773ba2f5ac9f67a : SUCCESS in 16m 35s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/8c9f6eb466854b02a5c72f784a854b1e : SUCCESS in 28m 46s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/00400dc46c91414ea4776d0a3876b845 : SUCCESS in 13m 43s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/e55561ba72d8495fb29be57f680c1a53 : SUCCESS in 16m 51s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/8ef94fa043764b88a27d4fd1908ce3df : SUCCESS in 18m 23s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/feb89b87c6404caeaee8d2880fb9cd00 : SUCCESS in 17m 09s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/30ed085255a3437c959d1540cbe2cf67 : SUCCESS in 18m 05s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/a27b94f31dc74f0aac1b2cdf3d399e92 : SUCCESS in 16m 51s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/811a4bae9fec4883bb6614870a168686 : SUCCESS in 14m 37s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/bb2f1e02f751412b8d56a128cf4000c8 : SUCCESS in 14m 47s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/fc0bdb8c2ecf47d5879a6d247812b6b0 : SUCCESS in 8m 09s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/c94f7b814c3d49f6bf2c087ea9083810 : SUCCESS in 19m 55s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/99592a91e9994882a82e21aefae2024f : SUCCESS in 8m 24s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/7cfc4351a4d948719e250ef5e5c8c358 : SUCCESS in 59m 39s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/2c34e00114404168bc4b221e9d07d5b4 : SUCCESS in 1h 47m 26s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/1d584776c23342f583c48ee2c2d4d776 : SUCCESS in 1h 02m 15s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/2141adef4324449dbb8a1208381d1f5c : SUCCESS in 56m 37s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d6fee9cbbc074eea97418c18496bf404 : SUCCESS in 20m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/196b62caed684048b1d20389572cd068 : SUCCESS in 5m 11s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/c70116a89ffb40f695250515c3a0c66a : SUCCESS in 22m 17s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/18ca2efffa7647e9a8a367b011c66fb3 : SUCCESS in 1h 02m 54s","accounts_in_message":[],"_revision_number":14},{"id":"1aca19126ecb93ad538bcee5a28e52aa4d741456","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-05 20:52:11.000000000","message":"Patch Set 14:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/93ebaca892ae4cdc96b953195a75ccb0\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/6c2ff3c5978849899fe62d16cef171de : SUCCESS in 58m 22s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e7b67518ae5b4759b9f549a72418e590 : SUCCESS in 53m 01s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/cb600e9aa3aa4164b93fd3c30ffae824 : SUCCESS in 1h 23m 20s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/6fdf576260d3460e8895e2f80150f308 : SUCCESS in 1h 01m 47s","accounts_in_message":[],"_revision_number":14},{"id":"876e8cbab541f7082bbb8b2697b05d7558202849","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-12-11 18:59:13.000000000","message":"Uploaded patch set 15.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":15},{"id":"337d4c8af7645874b34bef976b3ff6aae5e167c0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-11 20:13:04.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/13282b479f594049bbad5a8e615b3b0a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/46494f15a9004c70a938382e3ba82154 : SUCCESS in 8m 02s\n- grenade https://zuul.opendev.org/t/openstack/build/ce3c3d5437904bb6835fc414d00e42d8 : SUCCESS in 1h 07m 18s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/c0fbfb42c4dd4508b0aa860081793605 : SUCCESS in 44m 06s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/5e77f1841f744dc0ad860580a77ce0ea : SUCCESS in 1h 05m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e70692675d03460c88aa5586618cc620 : SUCCESS in 1h 04m 48s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/856bc7ecf8634ef4858c6067d1a71f98 : SUCCESS in 19m 45s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f4cf33a3b45346508979488e3bb8a0f8 : SUCCESS in 22m 53s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/00109c9ae2c34e6b933b9d9f5eb8c742 : SUCCESS in 20m 29s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/83d048eaee8648d38fa0b8e006b4a301 : SUCCESS in 8m 42s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/569b49438e8348a0a014d98a7a0b894f : SUCCESS in 6m 17s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/e2dab1e26a184a9aa230ff45f6c97b24 : SUCCESS in 21m 14s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/3a6587ca79e74c2abaf6ef8be19e4644 : SUCCESS in 18m 05s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/1ce87b6cef944b99bc469eee98159b9c : SUCCESS in 17m 43s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/4dc2d62f956849199ed410a828c24b03 : SUCCESS in 24m 03s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/6f0837ff08b94dc69e67d25ac14e4065 : SUCCESS in 18m 50s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/d903663ec3904e25bcc24bf62773140d : SUCCESS in 18m 30s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/e7c134b28c574d4cbd54b9e6761259f0 : SUCCESS in 15m 53s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/697c3c3cc54b480c94af64fc0478cb4c : SUCCESS in 24m 24s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/f15eba9ca3564809a24b0fc4c309d4ec : SUCCESS in 17m 17s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/60982437999c46b18c3549138ea3b3ca : SUCCESS in 19m 22s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/3d58a3d0a10a46abb163384159468136 : SUCCESS in 17m 14s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/03d9ff8f7e0f476d966705a53c452ce8 : SUCCESS in 9m 51s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/e7d07aa02a824c95a3ee314134643e52 : SUCCESS in 16m 51s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/d56e110a5e12457f821bc7eb3e07e702 : SUCCESS in 8m 03s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/8c5ff6b7e71644d5a6dbb2f963e92e11 : SUCCESS in 58m 10s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/3f7865d5af5f487a98c5b2ec86103928 : SUCCESS in 49m 19s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/2278f6bd0103433cbf0713b58d5ecc0f : SUCCESS in 38m 47s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/46dafcc5c1094f2d86ad6d364f047ecf : SUCCESS in 56m 48s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b06bac54a2ca422d84f18076de8d47a3 : SUCCESS in 19m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/37ba53fd512e4800b5a7f422b926d20f : SUCCESS in 4m 35s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/ac288c711b64449abc99c3d06e5e96cb : SUCCESS in 25m 48s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7484a51d63d44a1fb1fc0f27f578a192 : SUCCESS in 1h 06m 26s","accounts_in_message":[],"_revision_number":15},{"id":"03dc429413a77efcfe44c46c2e5506b618f85884","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-11 20:33:35.000000000","message":"Patch Set 15:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/67b7c48ab7a94bbdb58a3b0744f91ce9\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e87389ef00f541a79d9292b00362af21 : SUCCESS in 20m 48s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/912807892da24908bc55c31c0048d780 : SUCCESS in 1h 21m 10s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/0de7ff1dd3ce4674a6ce576f1b6ee2fc : SUCCESS in 24m 13s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/0c2edc337e594872b4ff819698c51100 : TIMED_OUT in 1h 31m 07s","accounts_in_message":[],"_revision_number":15},{"id":"ed4fab39b332466e2b714f6bcb2338b6a0d7c56b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-01-08 15:01:05.000000000","message":"Uploaded patch set 16: Patch Set 15 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":16},{"id":"18e1dc50293b4bd98f493b5d2994d17e8e345c44","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-08 16:24:28.000000000","message":"Patch Set 16: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/aad921aa2bb44eaea64b7e5697e30209\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dea2ccad608a487096712d47f773a41a : SUCCESS in 8m 35s\n- grenade https://zuul.opendev.org/t/openstack/build/3857a570169b4c298a694a7240c8878a : SUCCESS in 52m 18s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/74f7e8713eb7434b92a92db28b43d3e0 : SUCCESS in 1h 00m 19s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/1a4720d85e344a1e92686757ec6797da : SUCCESS in 42m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b73ab8f7c8754c858ea380c66487905d : SUCCESS in 1h 03m 39s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d1789b1ff8fa4123b19e93a5c8debc9b : SUCCESS in 19m 21s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/90db6654c46e490c962b52ce88efc989 : SUCCESS in 20m 04s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ec8a3295502b4e2596dc6b309d62fd28 : SUCCESS in 19m 52s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/b2b8b3f8a6d24d64b48b460224f27cda : SUCCESS in 11m 47s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/146ae73636654f4b8357839432b555db : SUCCESS in 6m 26s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/8aad9745d90b4dbabb0e070f6bfa2124 : FAILURE in 17m 58s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/51cf9f7d0b864bbeb5809d1c711476e2 : SUCCESS in 15m 53s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/0a59ba618e1b466298324bf0c52adb8f : SUCCESS in 14m 37s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/0061643f6d6243d38d2a930bd4936e39 : SUCCESS in 17m 31s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/4b9c4900e62f4043a01cadae1237542d : SUCCESS in 13m 37s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/d8921e2e347945ca9f866efcdf15ef46 : SUCCESS in 16m 49s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/054e3d282a7c40abbfc6caacd9a58cd0 : SUCCESS in 20m 54s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/f9c56efed9e84fc6b16a737e7fdf5137 : SUCCESS in 18m 50s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/c0ab91a26ffd4c269f0c2881180b436f : SUCCESS in 13m 42s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/dd60ad47f71c4958a5f94a38e617adf3 : SUCCESS in 14m 20s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/903d491db5c747b4a8b786eae9807438 : SUCCESS in 16m 11s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/c4010e586fcf4a32881f5345cfc44504 : SUCCESS in 7m 21s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/870b16f59c7c486583fbb7fd0588c43c : SUCCESS in 24m 31s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/442bce1a881d4453aabe68fc33f4264b : SUCCESS in 8m 14s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/bb06b0e525eb48fbbdc258c5c6d37540 : SUCCESS in 1h 06m 01s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/ee6d974075c0448ea5611d47a81b0553 : SUCCESS in 1h 17m 16s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/0145883d9c844a35ba4060b1673d7fc6 : SUCCESS in 57m 45s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/69d58a66341349199fa17c0a2da2d2d1 : SUCCESS in 1h 00m 19s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/158b903e2beb459ea8ce9672c89a56e9 : FAILURE in 23m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eee1bad26de64f6c8f77cdaa35ed2dc1 : SUCCESS in 4m 23s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/17d49a3bc84a4a44aa53fc350821730e : SUCCESS in 20m 28s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ca7b04fd2cf140408b10a84f47f0728b : SUCCESS in 1h 22m 05s","accounts_in_message":[],"_revision_number":16},{"id":"11fbfa710d8b41ea02bca8635c9608187164a94f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-08 16:34:05.000000000","message":"Patch Set 16:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/af57ec0aacc44b738e812daa9a4b12fb\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9855e3f55686451d8520345f6726ba22 : SUCCESS in 53m 04s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/26fd16d9513942d9909c3c45b54abac1 : SUCCESS in 1h 30m 15s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/8e184666f2f74ef18b01e642d8e02020 : SUCCESS in 29m 01s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/18983d27edb7474e816a5bb78c3cdcbc : SUCCESS in 1h 22m 40s","accounts_in_message":[],"_revision_number":16},{"id":"c1b962bf770e158cde6bc81f28e03a098894dac8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-01-08 18:10:41.000000000","message":"Uploaded patch set 17: Patch Set 16 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":17},{"id":"7ca435f92c1132f93783adc12d724e937efdc05f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-08 18:39:23.000000000","message":"Patch Set 17:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1db76ec08cd34035875742d4b2eea108\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/cea8589fe3764930b07807fe8f99b76e : SUCCESS in 18m 12s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/24f90c8a85fa4777924f312410e70e8c : SUCCESS in 18m 15s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/ecb64df2fc1e4d74b063c7e396705050 : SUCCESS in 27m 34s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/045a6bab7f6949c98195e6acc0ad7909 : SUCCESS in 23m 46s","accounts_in_message":[],"_revision_number":17},{"id":"e7331a5ea809894022773daa4e3b8be7b3dd1075","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-08 19:23:27.000000000","message":"Patch Set 17: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/910fe0e382564635b7271d08e9d1dc40\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/35f61b5ae9d945a18e13d2aae278bbd7 : SUCCESS in 9m 19s\n- grenade https://zuul.opendev.org/t/openstack/build/804556a471784ec4a0366606b3f9c2fc : SUCCESS in 1h 00m 00s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/4d43c8ed604d46e4910ce89665aa753d : SUCCESS in 1h 04m 35s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/5d1ea9fbe2a9452cb7cd56e0f6decb16 : SUCCESS in 49m 04s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/aae9f39cb63f4e3985a2cb011babef2b : SUCCESS in 58m 38s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f023a37ba85c46cfa14347d2f33a9ce9 : SUCCESS in 19m 09s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/15f55433a6304eddbe5876df1240afa2 : SUCCESS in 16m 44s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/042310823e854cf4b80f02c198f92040 : SUCCESS in 21m 54s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/f8134631c9c348358358ebe95d5a30dd : SUCCESS in 10m 30s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/437a93fb0abc493e8600b5e363fe496a : SUCCESS in 5m 31s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/d0ec7bbe272849cdb99e99b26333774b : SUCCESS in 15m 47s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/5056e0a017a5412b9c48f7c5de0964fd : SUCCESS in 17m 07s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/9768b31a04584d67901471f0425e1f40 : SUCCESS in 15m 57s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/7db1f2e3617947a0bb8965a8f93a68db : SUCCESS in 15m 23s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/98672ccdcc8d4f98ba43227ea41da134 : SUCCESS in 14m 33s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/fcf2fdc42eea4aada13371876bef924a : SUCCESS in 16m 22s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/38b3d90a7d6943619ff122a0bd81d7eb : SUCCESS in 16m 38s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/1a1b130c8c3f481798aaf4b23defc395 : SUCCESS in 20m 46s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/bb4b1a195ed14476a0f64d8129e59680 : SUCCESS in 14m 31s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/90ab43ab93494e218968212e74c76f27 : SUCCESS in 14m 42s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/831ddb42072349e6b83a036b98c1c490 : SUCCESS in 18m 11s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/5a535e3d6ba34f74adcae28a661b7776 : SUCCESS in 7m 32s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/ec253091932641d989a53803728943dd : SUCCESS in 15m 05s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/37194b068f9149e9afd62a68b462e969 : SUCCESS in 8m 12s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/715181b6f6f442cfb7b9a9b3625fe298 : SUCCESS in 54m 59s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/4e187cdf1bf94d6ca8e7914be2b7b9db : SUCCESS in 58m 43s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/05c7decf3bc94d3b8a4b3858935b1980 : SUCCESS in 57m 15s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/4932b213936a4e3bb030db7cb8827002 : SUCCESS in 50m 45s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/92b734a99de94b4a878fdcb607448d13 : SUCCESS in 17m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b707c3f9a21d438b9bf0f70557c496fc : SUCCESS in 5m 02s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/dc63c19be6474e71896a4407faddb286 : SUCCESS in 19m 46s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0f7d2333ee6c4e67bbfab186a9b27bad : SUCCESS in 55m 30s","accounts_in_message":[],"_revision_number":17},{"id":"ba3521e40f3c8b1e5d753d19e9e7857c11526c63","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-01-10 14:57:30.000000000","message":"Uploaded patch set 18: Patch Set 17 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":18},{"id":"fbede201e8da4693c2b09b98981347d3e7864ab5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-10 15:59:06.000000000","message":"Patch Set 18:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7ce84d909e5c40ea859a19a0bfa92549\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/297a669ff20140379d8b27d68e951db6 : SUCCESS in 59m 17s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7260c226f44348b5baac369292fad927 : SUCCESS in 56m 23s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/383da65bf0394f289e3b16907095c13b : SUCCESS in 19m 00s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/556d620b9d444499914b0f380c7bb2dc : SUCCESS in 55m 10s","accounts_in_message":[],"_revision_number":18},{"id":"cf0fc5b12305f4068afaae1e8973d3698780c402","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-10 16:18:26.000000000","message":"Patch Set 18: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/05c166ccfe644c9595ff475768f3ad32\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7b197c27e3b44f51a177ae61586e107d : SUCCESS in 9m 52s\n- grenade https://zuul.opendev.org/t/openstack/build/42b825a9690f417a8373855d68996915 : SUCCESS in 1h 14m 21s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/364f7d1263344546902457dbb6401630 : SUCCESS in 1h 05m 01s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/b3856fc926ab490497a35384113dd767 : SUCCESS in 42m 05s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2541031b2adc426b882bc7169e90ba0e : SUCCESS in 22m 22s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a0fc692c20ce43b0807932cc1c894c95 : SUCCESS in 17m 26s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d7ea863d409e41fe8b580e8e1131ba19 : SUCCESS in 18m 08s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7746e909e8284b658734fa708b92fac0 : SUCCESS in 22m 37s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/825b99fc57ae4fb494cc03eb71f31616 : SUCCESS in 10m 25s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/c2c137c4f432416eb8feee042034d256 : SUCCESS in 5m 33s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/88dfb14ac8a2449099fe742abe6d5a96 : SUCCESS in 16m 40s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/ffc634f8d588464c877c96db0066c95f : SUCCESS in 17m 58s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/40e060168a3441749f64f994b152ad21 : SUCCESS in 17m 36s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/e8ca3c0036f74a638ba30b3af5d934b6 : SUCCESS in 16m 39s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/5504d5ef14444393bc65ef7f6ec2aa85 : SUCCESS in 13m 26s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/d36ae35c3e014f8494449b4196fb7c20 : SUCCESS in 17m 58s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/850e463429f546f2a93e83c09219d3be : SUCCESS in 17m 25s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/3299622f71fb40d3ba85c6e2811b314d : SUCCESS in 17m 35s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/12b653ccd5d148fc9cdcb281cd480533 : SUCCESS in 16m 29s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/17493f0bc793403cafdde3c34c276a5e : SUCCESS in 19m 15s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/3db15c9a27b34713b15c136dfaff17dc : SUCCESS in 15m 05s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/174b9a98923649069729376cd09243c8 : SUCCESS in 7m 30s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/03409f394ed54eb796510b61dca7ac65 : SUCCESS in 17m 36s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/8b887ca38f694e59b8b17e0cda237e1b : SUCCESS in 7m 11s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/15f69062cc064cac98c403f95083b3d2 : SUCCESS in 56m 48s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/544e025e4c5c4b579c8be6aea860912d : FAILURE in 53m 26s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/31b9143c4ab74073afe564017a711056 : SUCCESS in 53m 34s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/316ad40e80484fb0a1fc6e9add514acf : SUCCESS in 50m 00s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1cf785d3ae7b4e379c36585b3a537208 : SUCCESS in 14m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e49d9543fba14a2298a68ec396c8ee72 : SUCCESS in 5m 03s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/d0551cdb2e844904801bdbc0ffed4f80 : SUCCESS in 22m 00s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f4c6110e4f4f432a9fd407c2029d456e : SUCCESS in 56m 15s","accounts_in_message":[],"_revision_number":18},{"id":"f6dcc22058c054fec6aaee2b4c34d5ee7dd70913","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-01-11 10:51:52.000000000","message":"Uploaded patch set 19: Patch Set 18 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":19},{"id":"b2fe0c4791b9f70ee50d816b7808cd846abcff3a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-11 11:58:54.000000000","message":"Patch Set 19:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c9c6a226a6f8463b831dfcc7cacb4e00\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/8a008c392532436b8345509dceca6d0b : SUCCESS in 23m 56s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/25cccb45af644ad987101d33aaca64b6 : SUCCESS in 23m 39s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/a2a50cbbf7ce44a8a269de97e0af10af : SUCCESS in 27m 00s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/1b610467c1af4b258d0ef2de76d8a6ba : SUCCESS in 1h 04m 13s","accounts_in_message":[],"_revision_number":19},{"id":"43a0d54f91e3c92d07f490be1c279f23d4b2f7b1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-11 12:06:53.000000000","message":"Patch Set 19: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9ad9d7c21abd4723868893793b3c792f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1e731e25799d4268bdcfeb8a7b6c9493 : SUCCESS in 10m 33s\n- grenade https://zuul.opendev.org/t/openstack/build/f071041b3732460dbf76e88fec57865d : SUCCESS in 1h 01m 53s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/e64f3c7acdcb4eff89d313c25f13ea24 : SUCCESS in 59m 46s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/c537224bd55d4264a7c682a92b191ed9 : SUCCESS in 51m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f205da8c81104a2b996b26b482571e43 : SUCCESS in 31m 35s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/10c6663964a949739284a1ca3fc6ffe1 : SUCCESS in 19m 27s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f94974caff8349d793eafe76bb00e064 : SUCCESS in 19m 48s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f574447c525641c1b0aa2d71d8481792 : SUCCESS in 19m 00s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/41fc43f3a0644078827a08cea87ee476 : SUCCESS in 11m 49s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/37612385dd314a908a0438c34aaec0e9 : SUCCESS in 6m 06s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/9ce8863ff0ab4ca6821eb8064eb9ad55 : SUCCESS in 19m 13s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/b290a5da992c4b33971d55ae4606d0fd : SUCCESS in 18m 14s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/03f961a2da914402af4ca90a6ca90a74 : SUCCESS in 17m 42s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/957dce85e2fa4adca36cb452f4a2eeb3 : SUCCESS in 14m 22s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/42cd9d80d76b475fbf481b4104d8faaf : SUCCESS in 15m 28s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/2538a9f4d5fe49b494e107e28c837329 : SUCCESS in 15m 05s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/f1a2dbd765cb41eb8200e3675a9fbe14 : SUCCESS in 18m 31s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/c3045ed52e2f4df1944d94e1cb6a63ba : SUCCESS in 20m 19s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/cc5a5d55757643d1b37bdc9320455d5e : SUCCESS in 18m 11s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/bc7c8becef0844a8ad457a115ae72c2c : SUCCESS in 16m 08s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/04c8201e04bf44cb851d9e5f01d16c7c : SUCCESS in 19m 30s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/8cae652eeb8b4ba097386de8e11a3ae0 : SUCCESS in 7m 43s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/241eea2930f147b3809055a71c493f03 : SUCCESS in 16m 41s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/831fcce06ad74f07a16f14e5b634c16d : SUCCESS in 10m 18s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/360b6838eb7e4abc8c65afc41bc4d444 : SUCCESS in 53m 24s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/ab84f5dddcc84c37907cda52221f159e : FAILURE in 1h 03m 44s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/678b7714974d41419106340f86d0fd5f : SUCCESS in 59m 08s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/2d96658379294caaab239b2cddc5bc85 : SUCCESS in 56m 13s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a3abd8248a3246b9ae620465ca0332d5 : SUCCESS in 22m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aba42507c7ff4b8b8fba7939f1010f52 : SUCCESS in 5m 54s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/cbd9f114a418432ea0c7112e3e9e4d3d : SUCCESS in 24m 36s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9d73fed61e2d497fabe4b1c16f9cb2ef : SUCCESS in 1h 05m 32s","accounts_in_message":[],"_revision_number":19},{"id":"53032e1f16204cdbd3d471d13dbc3022972704cb","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-01-11 13:35:56.000000000","message":"Patch Set 19:\n\n(1 comment)","accounts_in_message":[],"_revision_number":19},{"id":"2464c238b993abfef0db773e6644411ea6f846cb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-11 15:09:45.000000000","message":"Patch Set 19:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/195e3eee67b34684b582a919f3b393d8\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c6f36dd71fd8480d994770af917abfe2 : SUCCESS in 1h 22m 50s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/20d4b264d5b5431fb10f54a9d7e4aa6c : TIMED_OUT in 1h 30m 54s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/a8cb5249a27345c5805b92f9428237de : SUCCESS in 46m 30s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/6c0a4a0c0fab4ddeb7f9e51370d8d567 : SUCCESS in 21m 49s","accounts_in_message":[],"_revision_number":19},{"id":"cd59dd0035529e416b2f2c7fd77d85a3899dd983","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-11 15:33:00.000000000","message":"Patch Set 19: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f5b53bec974f4404a6a85cba0f444d07\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b1c8551ca75c450891ec52e4d634d5f8 : SUCCESS in 8m 47s\n- grenade https://zuul.opendev.org/t/openstack/build/bd9e9a2893cb423bbf0b34f14002ffe0 : SUCCESS in 1h 05m 10s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/3f917f3bc959444d8bd0797b9ab52e77 : SUCCESS in 58m 45s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/54e4b9ca60b34d1ebbc61824fd34ac16 : SUCCESS in 51m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bfba8627c3774202bf11a9e23ba9763d : SUCCESS in 30m 18s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5f38463d62244622895137d1a820512a : SUCCESS in 16m 45s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e7022484a366418f999a94b11166a329 : SUCCESS in 18m 21s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/93842818ff4347ee88ab6461a4ee5c35 : SUCCESS in 21m 27s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/e9aadf3e95ad4cb4acc303ed3b53239f : SUCCESS in 10m 25s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/babb5fabc18547fabca985a1876bde42 : SUCCESS in 5m 41s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/23cb7aab1a7e4c9e8196a91a5f8d55ee : SUCCESS in 18m 23s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/5df8b241a39840439eebfc3f4bf4e010 : SUCCESS in 16m 42s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/84e0bf7a80cd4f92975a6ccf88d7c67e : SUCCESS in 16m 19s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/5a3e22f3e28c46b280b6405fb3c32cbe : SUCCESS in 17m 21s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/242c7d0cd01d4f76af7696c0f0d6d629 : SUCCESS in 18m 53s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/eba78049cbbb4df3b2a225a967ae21a4 : SUCCESS in 17m 57s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/4675a04b8fc44a8ba6d54bd7b31d5419 : SUCCESS in 18m 40s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/3396359aaaca42d0940ea00c0441ec70 : SUCCESS in 19m 37s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/88626585b384443f8fe25f575f73ef91 : SUCCESS in 17m 15s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/11681c08b9a4482cb51f87640cbd750f : SUCCESS in 16m 49s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/45b2b922262f4b93b4b4d100b262437b : SUCCESS in 18m 02s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/2c8924e61fe8479cb06a1ad7711bb739 : SUCCESS in 6m 51s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/a0e521cfa1294752b28ad11d047f8b7f : SUCCESS in 15m 29s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/3813ab7a3efa42de9eef535dc7219353 : SUCCESS in 7m 53s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/ac1eee28e44942b28172e3a11f8c2315 : SUCCESS in 57m 20s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/80c5371932104c75ab3f166bd217cc8b : SUCCESS in 1h 07m 31s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/eb11caaa47d242b0aff817c43ad5d73c : SUCCESS in 53m 17s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/4fb3c903d96241f984409c10bea03bb0 : SUCCESS in 54m 14s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c8093d30f9224f6aa9beb6f20edb66b7 : SUCCESS in 18m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a513f4eb2cfd49fa8bbbe6ac10b489ea : SUCCESS in 5m 05s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/8084e8e97d1b4050a7eb7b2d99a04013 : SUCCESS in 22m 00s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c59c3e673af94978bc0a190d1f094c7c : SUCCESS in 47m 17s","accounts_in_message":[],"_revision_number":19},{"id":"2d1d60e52c61993fd6c22768ff5aef37fc503951","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2024-01-16 06:35:44.000000000","message":"Patch Set 19: Code-Review-1\n\n(8 comments)","accounts_in_message":[],"_revision_number":19},{"id":"a3d1b4fdc750ba9b45a1461347d449548d387ebb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-01-16 10:20:31.000000000","message":"Uploaded patch set 20: Commit message was updated.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":20},{"id":"771d8dd6ac5c49dd35748b47acc396431c173b60","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-01-16 10:20:40.000000000","message":"Patch Set 19:\n\n(3 comments)","accounts_in_message":[],"_revision_number":19},{"id":"6c88c76b13a3edf3629bb7f56f34e7449f9307fd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-16 11:15:04.000000000","message":"Patch Set 20:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/de3298226a024736b49eb382241ff831\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b3e430f091bd4535a89b1b810503eb15 : SUCCESS in 16m 44s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8c69c181e4a14bf8abceb23a757d35a9 : SUCCESS in 15m 50s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/f5bf7b294cb44af49acbef76142f4ef4 : SUCCESS in 21m 07s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/9a4f36e038b1445c9908982a81c80db4 : SUCCESS in 52m 52s","accounts_in_message":[],"_revision_number":20},{"id":"6737e31d892a308178f1898d62d639fa35264c3e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-16 11:43:34.000000000","message":"Patch Set 20: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8df47fbbf2e3471a8ad55b4fda303f55\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3a301dc32f7646199c5b1e123a32f563 : SUCCESS in 9m 21s\n- grenade https://zuul.opendev.org/t/openstack/build/fc04280b894a418091f40403b0f03288 : SUCCESS in 56m 45s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/9c5442b67aa8465881937792cca1d5c1 : SUCCESS in 49m 09s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/9c3a3eef030e4ea4892d0a2491bc7b0b : SUCCESS in 49m 05s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7e4adf72d2294d5cbcc25750ec29959e : SUCCESS in 27m 49s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4ccafc8deed54c8da578badd73c3a3cb : SUCCESS in 18m 00s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/cc29158f56364ad3b3024509e7598d24 : SUCCESS in 15m 54s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d379d038b2ed46a7b98dbcccc87dfef4 : SUCCESS in 21m 58s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/346bcc369f084ae98bd1cc1bb787d55d : SUCCESS in 10m 55s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/035c8e2691ff45df9815687bf5f69782 : SUCCESS in 6m 55s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/3f02233364ab44f5bfc9fe5e7e2a5694 : SUCCESS in 21m 06s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/55fc767ce3bb4f7fa22923d33370099d : SUCCESS in 19m 04s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/465a9e70c526478fa0d6db2507b2651e : SUCCESS in 16m 20s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/61d10be68ec343268571ebbda771b5b2 : SUCCESS in 17m 52s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/0c0f5c02fe2d4520ae21fdcdb008db22 : SUCCESS in 17m 43s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/4d0d7d75244d46e89d99a63fbe8733c2 : SUCCESS in 18m 00s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/9f7a11e7955044238378bacd5e5ff277 : SUCCESS in 19m 07s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/da65eb1081c54c65b62c2acb30eb896b : SUCCESS in 20m 28s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/2e278066deee49259737329c62f3006c : SUCCESS in 15m 25s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/82f0e35aaca643f9b6469975f32faac5 : SUCCESS in 15m 49s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/8ef7e8dbf86844e8993550290f5622d9 : SUCCESS in 17m 09s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/7228eccc35544bc7873ff30aab07eb73 : SUCCESS in 8m 32s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/ec4585501dc6457a9bc2554a9093c685 : SUCCESS in 19m 53s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/21a32f7ec97448da9c704eef5fcb0d05 : SUCCESS in 7m 26s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/a52dfbd4ddac48daa874ae6a15070611 : SUCCESS in 1h 02m 07s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/7a56f5611d3845d78547dac743c6a116 : SUCCESS in 1h 11m 28s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/7a32d0b6869b41ca8e9c33867bc93c04 : SUCCESS in 59m 09s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/d4dd9002b6b44dc1ad4e13b333cbfe67 : SUCCESS in 53m 58s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4dbd088b7da54d64b911b851f0a8b940 : SUCCESS in 19m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d4f8f9f723d543c2a78678da96addee6 : SUCCESS in 4m 58s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/ba2839eb571447b3a2c32c11982dd13f : SUCCESS in 27m 02s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f2db9a634bde45feae3c1dc6d07d52da : SUCCESS in 1h 03m 18s","accounts_in_message":[],"_revision_number":20},{"id":"9aa8847cbfb25b19dcf6355434b53ef2910919c8","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2024-01-18 05:40:04.000000000","message":"Patch Set 20:\n\n(4 comments)","accounts_in_message":[],"_revision_number":20},{"id":"8b0cc3b1cc559565441ae0c4171ff4308337c822","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2024-01-18 06:33:44.000000000","message":"Patch Set 20:\n\n(1 comment)","accounts_in_message":[],"_revision_number":20},{"id":"4d44b84ee630f6ee2a135c9a1b57db5032a204bb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-01-18 11:14:13.000000000","message":"Uploaded patch set 21.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":21},{"id":"a0fa4dd98b141c700dfe30c6083eee515a583fb4","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-01-18 11:22:09.000000000","message":"Patch Set 20:\n\n(2 comments)","accounts_in_message":[],"_revision_number":20},{"id":"532985c37ecd0d9cfbbb53a32c6f6bfd2a34fcb1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-18 12:27:40.000000000","message":"Patch Set 21: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1118771a61ae43b9a9b7d4883fb57c60\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d412c23857404643b4cc0e5419037c3f : SUCCESS in 11m 25s\n- grenade https://zuul.opendev.org/t/openstack/build/ed177a0d9b32445ca563710e02c5ff4d : SUCCESS in 55m 50s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/a4a126d332e347e696d8e5a0977e648c : SUCCESS in 56m 12s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/f5cf93287c524c43a611a498504cb3d9 : SUCCESS in 47m 37s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0e1cc0eac2ac45bf9426cf258ac35af0 : SUCCESS in 30m 04s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2d91b1622acf4e228eccac896db5b591 : SUCCESS in 21m 08s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6dc4e8430aad4cafb78fc4990849d06b : SUCCESS in 19m 53s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e42c16300dbb4ac5b39900aa2cc8d5b8 : SUCCESS in 22m 27s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/3f265885b90542e290e342c94576e5f4 : SUCCESS in 13m 50s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/3bd2c0e631c8414ca1bc6a996af7b928 : SUCCESS in 5m 34s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/a00d87b363f8492ca07f20b77829dfbb : SUCCESS in 15m 11s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/312f3bda02c547bca0b0936a34a4dd97 : SUCCESS in 16m 17s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/88f102e2760b48b4af2e92429a1382e9 : SUCCESS in 19m 15s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/3263d1e523184c9b9ce9d5806679003d : SUCCESS in 16m 50s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/7bb31acc8c3241b6a799f257327b9816 : SUCCESS in 16m 53s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/bd3778a6b6b04b589f896446d9c84742 : SUCCESS in 21m 29s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/40b7cda0d49143819ad7e6a84b0c7328 : SUCCESS in 19m 04s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/ac42862c139949feaa26008d9eced684 : SUCCESS in 18m 53s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/3ce5adeddef948d8b8190fe0f1a92415 : SUCCESS in 18m 05s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/3257130de2164b7d8fe3da366cf9c4b9 : SUCCESS in 15m 45s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/74faeb8c810649248053a2540e13725b : SUCCESS in 16m 22s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/4225be8e3ed44ac0a5ac9ea30dbbcb1a : SUCCESS in 7m 56s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/58fd521d2e194707a7cefefd9f823140 : SUCCESS in 16m 48s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/e6dcce6da23d4c7e853b7966e58900ad : SUCCESS in 7m 35s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/dd822f06f4be4ea79dd301899125df1f : SUCCESS in 52m 07s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/1c2f6392dd1544c4a615aaaca740e2fe : SUCCESS in 1h 05m 27s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/be9c84c37ed34bdc863b50f0a17583e4 : SUCCESS in 43m 24s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/6772dd98a40e4b52b294c0aae460105c : SUCCESS in 55m 30s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6573145899d64a92ac485f8c9e9c1f9d : SUCCESS in 20m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/956088bcda6d421d926d890b35f08d24 : SUCCESS in 4m 49s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/52d10965298b4da8a7a0dc9e1c96c99f : SUCCESS in 27m 57s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/571fbd2635274dc094c6f41dff51d4ab : SUCCESS in 38m 58s","accounts_in_message":[],"_revision_number":21},{"id":"8a7488726c6b2b607f613fc52bfd344ff2601e3c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-18 12:47:44.000000000","message":"Patch Set 21:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/98ca6ddf21eb41e3b61855b02303ca3e\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/280a643f657747f8a90e03404f8c999a : SUCCESS in 14m 59s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/115d4fa7acdd47a69241849f6e61c9fc : SUCCESS in 15m 27s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/4ccd349d77174ca1a01ec05363709068 : SUCCESS in 19m 21s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/59446eb4ff0d4f09accfa8b83d65a46a : TIMED_OUT in 1h 31m 03s","accounts_in_message":[],"_revision_number":21},{"id":"72560586573075d85ce92b342da54a6887cc3e4c","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2024-01-22 06:48:45.000000000","message":"Patch Set 21:\n\n(3 comments)","accounts_in_message":[],"_revision_number":21},{"id":"e5692de260d797d17b2539c317267e1483cabb09","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-01-22 09:46:28.000000000","message":"Patch Set 21:\n\n(1 comment)","accounts_in_message":[],"_revision_number":21},{"id":"98414533bc5427e485f02858232c01fb55cf6731","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2024-01-23 05:18:19.000000000","message":"Patch Set 21: Code-Review+2\n\n(3 comments)","accounts_in_message":[],"_revision_number":21},{"id":"732064dcdf65268fa1f21dfeea8c4c3b6c626ec9","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2024-01-23 05:39:18.000000000","message":"Patch Set 21:\n\n(4 comments)","accounts_in_message":[],"_revision_number":21},{"id":"e836825f12cf643c9f9c74ee51252dddc3e865e2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-01-23 13:02:06.000000000","message":"Uploaded patch set 22.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":22},{"id":"91c75b2e810ed58f10fff779e90ded2291291692","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-01-23 13:10:13.000000000","message":"Uploaded patch set 23.","accounts_in_message":[],"_revision_number":23},{"id":"678bc3605c42d4f11c6f023606be6ad2ea16eee7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-23 14:21:27.000000000","message":"Patch Set 23:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2154f6ba48b7438b993084ba32d2138d\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/60914ff54ed64dd7b548d28ad959b2a5 : SUCCESS in 1h 08m 10s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b1ad26ce1ddd4288837a0ee859c3cfad : SUCCESS in 1h 09m 40s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/8939e867a1b7419881a26fadb4f7a58f : SUCCESS in 58m 37s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/73838e3eedee4839bd908bb03b442fc1 : SUCCESS in 1h 01m 53s","accounts_in_message":[],"_revision_number":23},{"id":"88d4266f1b45fa18db84ca702055ed0dc47f5cba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-23 14:29:52.000000000","message":"Patch Set 23: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bcc1442d578643f685adade818effdc3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f8c7b62a4ed446ebae3c2f2c18e31076 : SUCCESS in 9m 35s\n- grenade https://zuul.opendev.org/t/openstack/build/ded22793c112439d9fc4161df2e341a5 : SUCCESS in 52m 52s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/ff992387838149d3ba50da1a7d5d6e6c : SUCCESS in 45m 36s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/53a9277614804cf380e85a762c4c8fae : SUCCESS in 53m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/787b45ad4ad447ba9e94667af56e25f7 : SUCCESS in 39m 10s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/89821f4e97ea4e978b91344af335fcb2 : SUCCESS in 20m 53s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5a7564e44a2a46fb82235ba4d4c7c228 : SUCCESS in 21m 18s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/196d7c8213be4e3995879195331827e4 : SUCCESS in 20m 03s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/41fd9775ac5a44b79b8f2def4ca6ee97 : SUCCESS in 10m 48s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/9376e437ec9e402ba5afd408f9409cec : SUCCESS in 5m 31s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/762922b7b3714d62b9063b97bf74304f : SUCCESS in 18m 52s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/3ebda42e92744d9fbf44a9d6caae8be9 : SUCCESS in 13m 44s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/a787ab2e56f54869bd0df0a0d765e5f3 : SUCCESS in 17m 26s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/341ab5087a9a42a29a88a6c08235f3d7 : SUCCESS in 15m 43s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/06df07f219034d8bbb39190c260c22f7 : SUCCESS in 15m 22s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/7636e06b82db483885241cfd7ef15a4e : SUCCESS in 19m 57s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/f0f99e0146664e93a0cc5d6e60e56594 : SUCCESS in 20m 04s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/236f06545ae04485bf1d8b81768afc12 : SUCCESS in 20m 55s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/5e200e95585b46d38e392a57a2792c8c : SUCCESS in 17m 00s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/35611b05298d4a7ba737a499ef3d1303 : SUCCESS in 14m 00s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/d27f0c78695342c4932847c122f4972f : SUCCESS in 18m 16s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/618b191c0a334ecbab9654d9b06e2857 : SUCCESS in 6m 52s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/09063354c13641c19f16feda4acfdd46 : SUCCESS in 19m 25s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/e38a10cc6ce346418f70db4c3a84d69a : SUCCESS in 7m 54s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/c3f3c253cd54408f98abcd24e7364d68 : SUCCESS in 58m 47s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/3ff072ebba2048558d4186f6944965c8 : SUCCESS in 1h 11m 45s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/afd82ff21ce64fb8b737e18770b8f1a9 : SUCCESS in 58m 11s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/35f0f87cabc64682b4a1b8b6afc97219 : SUCCESS in 46m 58s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/aa1efa112b4946d0b7d748887cd1b6cd : SUCCESS in 18m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/506a153431ba459688f03649fba5fe8c : SUCCESS in 4m 40s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/513b0f1480b64998a88fb9e5768bc2ae : SUCCESS in 19m 42s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/99d7edd11ebd446983de8665da10102d : SUCCESS in 1h 08m 58s","accounts_in_message":[],"_revision_number":23},{"id":"681cf48b1ea4be61dda005194ea434dbb0c8326a","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2024-01-24 06:40:51.000000000","message":"Patch Set 23:\n\n(8 comments)","accounts_in_message":[],"_revision_number":23},{"id":"4786e8ddc68169c7da714429f4e78cb4aa9dfe2b","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2024-01-24 16:42:41.000000000","message":"Patch Set 23:\n\n(1 comment)","accounts_in_message":[],"_revision_number":23},{"id":"cbd604601a3097690472d496c3c5cc8f7a856bcf","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-01-24 16:43:48.000000000","message":"Patch Set 23:\n\n(8 comments)","accounts_in_message":[],"_revision_number":23},{"id":"d8d0df4d795b4a0dfbd3e23c9e46ee233e907e4f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-01-24 16:50:04.000000000","message":"Uploaded patch set 24.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":24},{"id":"3a0febffa7b0952e4a525a55bbddae2014bbb673","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-24 18:17:44.000000000","message":"Patch Set 24:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/87e8f9b689324beeabd00142f8faf909\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0e13ee34e77f4dc3ab6807aa922c0f6b : SUCCESS in 1h 01m 33s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e3769c3b1d6b457f991bf689d434ed47 : SUCCESS in 15m 22s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/c0f0b9e1914649a790266029deff7109 : SUCCESS in 1h 25m 51s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/19c0a62317b04ae4bceb85959faaa1d3 : SUCCESS in 16m 48s","accounts_in_message":[],"_revision_number":24},{"id":"38b5b3b2498a3d95c452530585440a237bea1729","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-24 18:27:20.000000000","message":"Patch Set 24: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6926770c664e4922a4d82d033d0ac6d0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7fdacb18887b45f9b3e3e9a632d9d85a : SUCCESS in 8m 40s\n- grenade https://zuul.opendev.org/t/openstack/build/6f9706abd3f74388bc9ec91bd8c9cb1d : SUCCESS in 51m 15s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/8e85cf3c05244321953fe6e899f5c51d : SUCCESS in 1h 02m 29s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/a7bbca30b3ed43b79c33c998496982c7 : SUCCESS in 50m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1b0acbc5884d40c2ba3a78d313de9c82 : SUCCESS in 39m 06s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/fe98f781b6d342bdaf371bc355ce1574 : SUCCESS in 20m 15s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4cf1e2e51d9e480db3230a640385671c : SUCCESS in 17m 35s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/69690038bfc341f5b85ac10a8b927d94 : SUCCESS in 19m 04s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/be7ae2447e77484ab09646e9517c75cf : SUCCESS in 9m 37s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/5c743b04356d4292a7e8314bc556f5d8 : SUCCESS in 7m 31s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/c723e4a3ea454df3b4bebaae8ad1df30 : SUCCESS in 17m 43s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/4de66efd86c74e82bfa1cc1415d5bbfd : SUCCESS in 19m 12s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/e43be3e4456b412ab2fcff787222dd52 : SUCCESS in 16m 22s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/b299d26cdf5f4f92bdbf11d24fd6381c : SUCCESS in 16m 56s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/7fa3de9cc25e4a91b4b26d15c1866f42 : SUCCESS in 16m 01s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/9ce3e682e0a54ab5b28baa41f2ef9ff1 : SUCCESS in 17m 29s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/145f63eb0b0f44b98f8fbb5dd59079fd : SUCCESS in 18m 25s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/565a253a55c043e7bace517a081d4bcf : SUCCESS in 21m 16s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/5a6e70f8585d4c57ac034ac22402ec74 : SUCCESS in 15m 56s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/81cca1f74f6241fb96fd788d8aa6548b : SUCCESS in 16m 57s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/c60f666772b9468c85042203a0115926 : SUCCESS in 19m 08s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/91c11bf3314a4d07a8feb5c581fb6672 : SUCCESS in 8m 48s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/ef319d54ae9a430399e67d86f8743dbc : SUCCESS in 17m 34s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/7ef53e7c6c864197bd765aa966e85402 : SUCCESS in 7m 37s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/623eb1567c8c4750b7dfde726fa07f64 : SUCCESS in 1h 01m 28s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/045313a27e414881916082d1dc45f272 : SUCCESS in 1h 04m 02s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/010dec267aa64ce48f058e5f59ba420c : SUCCESS in 57m 58s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/a749ab27ea8442b280d2ddeab0bb571b : SUCCESS in 53m 30s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2812d448e784471d8c5e1f0d9f731109 : SUCCESS in 21m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/28faecbb48a44c8d8e954d8283b231e2 : SUCCESS in 5m 45s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/6655644d9ac040afbd8b532887abef4e : SUCCESS in 20m 53s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/997e2eeee1754e18ab42ae673bd9c23b : SUCCESS in 51m 54s","accounts_in_message":[],"_revision_number":24},{"id":"bae6a4a5a70cf84981f6eddd3cfdad87537c0efd","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2024-01-25 00:32:03.000000000","message":"Patch Set 24: Code-Review+1\n\n(8 comments)","accounts_in_message":[],"_revision_number":24},{"id":"656812c33ec8931577590fb9bed9f3d43de5ad25","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-01-25 12:20:15.000000000","message":"Uploaded patch set 25.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":25},{"id":"737430e95e37919520fafa8637cb47a0763433f0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-25 12:58:35.000000000","message":"Patch Set 25:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cbc6c6790dfb4a669304e863c76ad53f\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/eee1ac5bf9dc48a5989e26ae29077b88 : SUCCESS in 27m 11s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5e6acb859cb447188fb7a981d23aeeb2 : SUCCESS in 28m 11s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/a35a170d71854eed8231e668f7f4d359 : SUCCESS in 33m 29s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/37b823a7043b410ea020d086401b9606 : SUCCESS in 33m 27s","accounts_in_message":[],"_revision_number":25},{"id":"16a7ebda1f6aa383cc1b924e4e52ceb80480df1b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-25 13:34:51.000000000","message":"Patch Set 25: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/08072f11202642408a4fc146ac9efe62\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3f6defebf503481b898b2bc251b21d2a : SUCCESS in 6m 57s\n- grenade https://zuul.opendev.org/t/openstack/build/4329c9667bd244d29f834f9f28c752cd : SUCCESS in 56m 35s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/095a4c738cc044e8ab17bb2cc95be917 : SUCCESS in 36m 28s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/beee7195f2b44116ba04dfcb83d47ae8 : SUCCESS in 30m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1d714191ca1d4d87a32114545438e276 : SUCCESS in 37m 16s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/10275a89052c4b37aef1614f42d80494 : SUCCESS in 19m 17s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/86b87a2ca36c4c5e8ab00d51e62773f3 : SUCCESS in 21m 06s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/38fd189816214bd48d6f7417a96f06f4 : SUCCESS in 20m 47s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/b02dac120b584c9693bdd8c4953a2b61 : SUCCESS in 12m 38s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/d975ba38d1424c94828a53c08718d48b : SUCCESS in 5m 49s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/5b35d8841fbe4e51965168da067c9d6f : SUCCESS in 18m 57s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/ea3f607b5189458db199fd5936fa8cbd : SUCCESS in 17m 49s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/d8f545a5a7564e4bbad9ae335132f381 : SUCCESS in 17m 02s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/996d8eb9137a4fad80b8422e9ff8bd42 : SUCCESS in 17m 55s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/a57cb889f7c443658908c45ab31df536 : SUCCESS in 18m 57s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/01617939ead5439696edb6bc7252218e : SUCCESS in 17m 01s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/1ab85c12dca949ddbe2aee93eed305c1 : SUCCESS in 19m 23s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/4e107fc6d7bb4f0589efd956e5be47f6 : SUCCESS in 18m 34s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/c310c724c191492aa63f34a899c9d40f : SUCCESS in 15m 59s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/0496aa4f87bf469f8e4bb95a4937cd8d : SUCCESS in 15m 32s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/26468e6738464123b717a7a42ef2b77c : SUCCESS in 16m 42s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/d0bc8252728a435b9a2e56e14d3486e9 : SUCCESS in 7m 29s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/d2dab33e5f6c465ebd1e55976e7303de : SUCCESS in 14m 50s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/d0b0cb4797934323917af8c42552705b : SUCCESS in 7m 46s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/c161c054788b4fc7ae91902640c2b244 : SUCCESS in 51m 24s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/5d3caa134a62414387a604b8f1aaf5f3 : FAILURE in 1h 00m 45s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/1104a25e274b41d48e53dadf1604d144 : SUCCESS in 37m 56s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/5c42e4bd90a247f5a5cc93ed779e49c0 : SUCCESS in 57m 38s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/761b58fb32eb4dc39efb59db383dbd43 : SUCCESS in 18m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/440e7caa27a0422192976f709ea8c884 : SUCCESS in 5m 05s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/412290cd0b8842d596b0e8da58d1b1fe : SUCCESS in 21m 06s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/455e6ad56df6441f852ccf07a4ca0456 : SUCCESS in 1h 07m 01s","accounts_in_message":[],"_revision_number":25},{"id":"34caab6883c358d1ed4e85fff848dd36a4eeaecd","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-01-25 17:26:59.000000000","message":"Patch Set 25:\n\n(5 comments)","accounts_in_message":[],"_revision_number":25},{"id":"d4aa5df57b03841f2df3a26ac32192647beddf33","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-25 18:19:58.000000000","message":"Patch Set 25:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c603f389de2f46a181c28d08cfbaeece\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/db0328e84d434917906a94d8d608b3c2 : SUCCESS in 48m 46s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f774424713d54306922d54fc640e48de : SUCCESS in 17m 56s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/59d374bf8b3f4c29bf8ea91be516b9e9 : SUCCESS in 51m 18s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/9757164f39c74820be69d9d9fb943d71 : SUCCESS in 21m 51s","accounts_in_message":[],"_revision_number":25},{"id":"31f1aac12d364fbdf2980a643d7c7416801249b1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-25 19:03:53.000000000","message":"Patch Set 25: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/024090968f25448c93820208075f96ab\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2a197bd57f464783b4f263c75c2c0c5e : SUCCESS in 9m 04s\n- grenade https://zuul.opendev.org/t/openstack/build/3d5191c938e246919e700ea920665f45 : SUCCESS in 1h 18m 15s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/7d7b31053d3d41d38a163dc1235dcc29 : SUCCESS in 1h 08m 11s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/62ae4400158c4e289ebca8816dbf2825 : SUCCESS in 54m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/483f260c062f435cb00d9e76d884253f : SUCCESS in 37m 18s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ee7f523ff3d24de19ee352f74979bebe : SUCCESS in 18m 30s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/13070e7ab20347809f75a531005de7af : SUCCESS in 22m 27s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c029553f84844343970852dcfc5fb09c : SUCCESS in 22m 23s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/bcc732b922f84fda89061eca9e6cd367 : SUCCESS in 9m 29s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/d1b905725d83447c8f31cf4c40ed9aa0 : SUCCESS in 5m 43s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/7a7de54ed0f84da49bc2d3059012286d : SUCCESS in 18m 27s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/51c37ca712e542c1ac8a2f22ade77dde : SUCCESS in 17m 46s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/0912101bc7eb4409bce895e55b012b4e : SUCCESS in 17m 52s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/1b976d8fbd494402983edc2df4acace5 : SUCCESS in 19m 17s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/0c108d75cf0f41f79efb4b51fcf74fa6 : SUCCESS in 16m 45s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/7c3303c213024bab884f41aa5e4ea844 : SUCCESS in 20m 02s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/615346bf080647368fc620b8868ae821 : SUCCESS in 19m 01s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/dea77ae2964e4a109362878e68e06050 : SUCCESS in 21m 55s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/deccecf3994d49f1a9c10fa28453f31f : SUCCESS in 15m 28s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/ae2441d612444879bb6544d1c698fda6 : SUCCESS in 20m 39s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/0083a5940681462ca3dbfc505555aa2e : SUCCESS in 18m 01s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/98929ab561814c1bae9f59592cb13185 : SUCCESS in 7m 34s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/204e0e79a8d7416a86056f6a1476bda4 : SUCCESS in 16m 14s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/f4bc9953cb1c4c159953d69cd2a48001 : SUCCESS in 7m 26s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/a52b40d11bd74b598def1daf5e586fde : SUCCESS in 1h 00m 47s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/9ae85aba1af3445198403668dc38625d : SUCCESS in 1h 08m 57s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/ea1f0ef81f0b40edad29ee6fd68aa4b1 : SUCCESS in 1h 03m 03s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/369260f4541f4019b21efcd54a418e9d : SUCCESS in 58m 25s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8253bf0d788244a1b2f3afbe247ef227 : SUCCESS in 20m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5f59f22a54b645adabfd7b87267e2a81 : SUCCESS in 4m 44s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/925b3e2e08b44360a6554a7242e05d87 : SUCCESS in 20m 28s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/65171ba305754e9fbee8aad1e6d242f7 : SUCCESS in 1h 06m 00s","accounts_in_message":[],"_revision_number":25},{"id":"accceaa6dfa4fe8b8810b63288a82aab42e2bb8c","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2024-01-25 23:57:46.000000000","message":"Patch Set 25: Code-Review+1","accounts_in_message":[],"_revision_number":25},{"id":"ec2fda12747bea153fe0755548f625ece7c4ca81","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-01-29 16:04:03.000000000","message":"Patch Set 25:\n\n(1 comment)","accounts_in_message":[],"_revision_number":25},{"id":"6605914faf566271ab59a2ed7d1f01f8deee5a40","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-01-29 17:07:22.000000000","message":"Patch Set 25:\n\n(1 comment)","accounts_in_message":[],"_revision_number":25},{"id":"c56dbab8f03d298307ee07be14718357be1a8c3b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-01-31 11:03:11.000000000","message":"Uploaded patch set 26.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":26},{"id":"fe8cf5205faf52a29bbce261079a0264d35467d4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-31 11:25:32.000000000","message":"Patch Set 26:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a3f3cef975f744338de957ff7211c0ba\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/82363ad0ce09482891d57fb63861b26e : SUCCESS in 17m 24s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9e862f51707a430b94728d25567aaa9d : SUCCESS in 17m 53s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/3a9d308dbc9d41b397d76bb6be1ced60 : SUCCESS in 19m 30s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/3b070612c76d431bbb9e34b32e1a7ad7 : SUCCESS in 21m 10s","accounts_in_message":[],"_revision_number":26},{"id":"223cc42aca5948b5cea2d56f3ccfc99a710b4008","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-31 12:39:29.000000000","message":"Patch Set 26: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/df3475f121244d7c8ae99bbfd8bf93db\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4df09f7ca44c4ea5b48318d0e708501b : SUCCESS in 7m 19s\n- grenade https://zuul.opendev.org/t/openstack/build/320e03dd23e447a4aded393f4fbd5fdd : SUCCESS in 44m 45s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/d2ca6b5624fc4b478d54ea6c4359468c : SUCCESS in 37m 33s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/4a7cdb19b9a241628bf59afc0d7e127e : SUCCESS in 33m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/85bbe474adac44039e58576a6958343c : SUCCESS in 30m 22s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a15fb6c5439b4e0b9d0c2d542717d327 : SUCCESS in 18m 49s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/58023270da4c424c8885cb9f88ef2641 : SUCCESS in 20m 47s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/009fcc55b5a54a9e9f0b47211c0127a8 : SUCCESS in 20m 53s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/247e5d8d37ce425d807c1adca4dbf046 : SUCCESS in 12m 34s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/d5bace3aac7f46a59056d6360bca195f : SUCCESS in 6m 22s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/3f8c6836ab194363b01da80f463f9de9 : SUCCESS in 17m 29s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/789878171f0e408f921c21b3dfe84a21 : SUCCESS in 17m 34s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/a6620e7f0897419eada6d1d392c18b1e : SUCCESS in 16m 42s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/a01a3dc9f58c4be38f5b8efff5049446 : SUCCESS in 16m 48s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/3d5cf2f71fc742e2be9a211c0a4d5f07 : SUCCESS in 16m 41s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/35a37a79928944b4ab850f537e0cfa0d : SUCCESS in 17m 47s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/5a00a2b6c33849f185dad86c39a76613 : SUCCESS in 18m 42s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/639d83c4b4c84c83853e62f026dfa6ca : SUCCESS in 19m 08s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/3b555bf331c34eb2a4dd1ef6ff54356d : SUCCESS in 17m 32s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/ce0a4ae85ba0410fa58e389871eb2b5d : SUCCESS in 16m 06s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/37cd0ec341414981876f94698276cc5a : SUCCESS in 19m 26s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/5c955e329ac742bda980405eec54fc61 : SUCCESS in 6m 47s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/10d96efd0309488499e3c9759a747071 : SUCCESS in 15m 02s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/ba0c441a091a40ebac76059a53502d02 : SUCCESS in 8m 08s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/4ed060945e1c4917b6df7536b1f8564b : SUCCESS in 56m 58s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/37484c10cf7b42df8e9db32093d688dd : SUCCESS in 53m 09s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/fdb4885266a84b4e8d6757193d3b92b5 : SUCCESS in 59m 35s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/0c27cede70a241409b611924270b5c3b : SUCCESS in 55m 13s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/dc5b4bacb9aa45e5b3766105990e9ee5 : SUCCESS in 15m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/814347a0e5b2481aa6718dd1316cfcbd : SUCCESS in 5m 01s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/d12cb6f7b9454d61b6b8458b73f4c1e6 : SUCCESS in 20m 29s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0b4ab7d26c5f4d889665a34e3076372c : SUCCESS in 54m 24s","accounts_in_message":[],"_revision_number":26},{"id":"9675d9252059db09a8fea09bf7fe5319ea220531","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2024-01-31 16:46:45.000000000","message":"Patch Set 26: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":26},{"id":"08eec64b8f6f7134465ae4d47f913a54d0f066a9","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2024-02-01 09:12:48.000000000","message":"Patch Set 26: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":26},{"id":"2bd1022596904278978bb2c1434623571f33a52e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-01 09:13:27.000000000","message":"Patch Set 26: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":26},{"id":"8ff7f2d509c3e996cb293f153b13df24238e63cb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-01 11:59:47.000000000","message":"Patch Set 26: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3cec172f0f114dc883ad37cb4cd58736\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/080884d1764943b3b749c0fc367cf592 : SUCCESS in 7m 05s\n- grenade https://zuul.opendev.org/t/openstack/build/73a0bcd9151c48b982c36f163d675d7e : SUCCESS in 1h 08m 23s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/033311863eed438080e1107690c63a53 : SUCCESS in 48m 14s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/bd7731154b5842bcb7efaed1bc990d6e : SUCCESS in 38m 38s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1dddc88734a74ecfb6c3a66ddaa76658 : SUCCESS in 23m 33s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/510d537aed554ac0aa425323e64d807b : SUCCESS in 11m 17s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/2bbe483aa4f94add9e5db6278c045b1f : SUCCESS in 6m 53s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/a58464b62afd48feb7fc52877236c5f7 : SUCCESS in 18m 30s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/0d1e2e4c423447148cac2b8967500517 : SUCCESS in 17m 36s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/9bd6dda5a7f341c890cfe6de6785a373 : SUCCESS in 16m 58s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/aca376ab05e04f968addac72eba87b2f : SUCCESS in 17m 50s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/9fea4b3db47148e99d7d30549a02dbd6 : SUCCESS in 16m 12s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/7799b9fcbd9c4d649b211fce5b432fa7 : SUCCESS in 20m 45s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/21b44061adb940a9b0fa42f68beb05a4 : SUCCESS in 19m 41s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/ded29e8938c44ce394102cdb187eb858 : SUCCESS in 20m 36s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/e2d466d27d7d4ff09e34a7b8bfa06e9c : SUCCESS in 18m 16s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/c19955dde3d94676ba4ddc04f5aa5091 : SUCCESS in 15m 08s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/a1fa72bf09c74270bf4de22dca58d954 : SUCCESS in 16m 13s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/e0abb174c35c4b009de51e6de53a86ee : SUCCESS in 7m 11s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/b0a146c131374fdb8c293f994534659d : SUCCESS in 8m 14s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/a59d9b273c1344e6a27f10f95f4b9a99 : SUCCESS in 58m 11s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/911ceb9e981d4a22aa5e7f3204b4dd02 : SUCCESS in 1h 07m 36s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/99807ba6871142ce9f25bb2fc32b58ce : SUCCESS in 53m 08s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/183a2dfdac4f4938956ffc612433126f : SUCCESS in 35m 54s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0dcbcf52e0054397be01665f47265f57 : SUCCESS in 22m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d1ce9491096c489faeb1441b55bb2fa5 : SUCCESS in 5m 03s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5410e444ee0d418a8052990b98ca0812 : SUCCESS in 1h 01m 14s","accounts_in_message":[],"_revision_number":26},{"id":"1c0cacfe22bb8bc6d7f480cc78ae6424ea1eeb2a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-01 11:59:47.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":26},{"id":"296a332b638c2244370fe575462b1879f74f9553","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-01 12:00:50.000000000","message":"Patch Set 26:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c9bf2a2a9e3d4e8391ca904d3319fe0a\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b21a385147a34571aa736282f4b16b7f : SUCCESS in 44s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/62911aad51cc40ea94102dfb58d267fa : SUCCESS in 36s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/09d5f0e1b4f84e51a592ac3d8c2623e3 : SUCCESS in 39s (non-voting)","accounts_in_message":[],"_revision_number":26}],"current_revision_number":26,"current_revision":"252f0d36b7c97d1e3261fd4cf475b362ee985abe","revisions":{"f50923b491232613e1c3da2b43ee459d37d57352":{"kind":"REWORK","_number":1,"created":"2023-11-17 19:01:13.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/35/901335/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/35/901335/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/901335/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/901335/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/901335/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/35/901335/1"}}},"commit":{"parents":[{"commit":"7f54b63a1eb7ca7c5b85a62acf7701e18d29b65d","subject":"proxy: always use namespaces for updating and listing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7f54b63a1eb7ca7c5b85a62acf7701e18d29b65d"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-11-17 15:03:59.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-11-17 19:00:05.000000000","tz":0},"subject":"proxy: only use listing shards cache for \u0027auto\u0027 listings","message":"proxy: only use listing shards cache for \u0027auto\u0027 listings\n\nThe proxy should NOT read or write to memcache when handling a\ncontainer GET that explicitly requests \u0027shard\u0027 or \u0027object\u0027 record\ntype. A request for \u0027shard\u0027 record type may specify \u0027namespace\u0027\nformat, but this request is unrelated to container listings or object\nupdates and passes directly to the backend.\n\nThis patch also removes unnecessary JSON serialisation and\nde-serialisation of namespaces within the proxy GET path when a\nsharded object listing is being built. The final response body will\ncontainer a list of objects so there is no need to write intermediate\nresponse bodies with list of namespaces.\n\nRequests that explicitly specify record type of \u0027shard\u0027 will of\ncourse still have the response body with serialized shard dicts that\nis returned from the backend.\n\nChange-Id: Id79c156432350c11c52a4004d69b85e9eb904ca6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f50923b491232613e1c3da2b43ee459d37d57352"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f50923b491232613e1c3da2b43ee459d37d57352"}]},"branch":"refs/heads/master"},"c447e9adc7bcf2a56f34e61d3831f6f4acb51350":{"kind":"REWORK","_number":2,"created":"2023-11-17 19:14:46.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/35/901335/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/35/901335/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/901335/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/901335/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/901335/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/35/901335/2"}}},"commit":{"parents":[{"commit":"934fe827cb7120834769f2b623da9476d91138eb","subject":"proxy: always use namespaces for updating and listing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/934fe827cb7120834769f2b623da9476d91138eb"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-11-17 15:03:59.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-11-17 19:11:50.000000000","tz":0},"subject":"proxy: only use listing shards cache for \u0027auto\u0027 listings","message":"proxy: only use listing shards cache for \u0027auto\u0027 listings\n\nThe proxy should NOT read or write to memcache when handling a\ncontainer GET that explicitly requests \u0027shard\u0027 or \u0027object\u0027 record\ntype. A request for \u0027shard\u0027 record type may specify \u0027namespace\u0027\nformat, but this request is unrelated to container listings or object\nupdates and passes directly to the backend.\n\nThis patch also removes unnecessary JSON serialisation and\nde-serialisation of namespaces within the proxy GET path when a\nsharded object listing is being built. The final response body will\ncontainer a list of objects so there is no need to write intermediate\nresponse bodies with list of namespaces.\n\nRequests that explicitly specify record type of \u0027shard\u0027 will of\ncourse still have the response body with serialized shard dicts that\nis returned from the backend.\n\nChange-Id: Id79c156432350c11c52a4004d69b85e9eb904ca6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c447e9adc7bcf2a56f34e61d3831f6f4acb51350"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c447e9adc7bcf2a56f34e61d3831f6f4acb51350"}]},"branch":"refs/heads/master"},"756c72d43359b2a7911676892e40a008a6a7e0e6":{"kind":"REWORK","_number":3,"created":"2023-11-20 16:47:37.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/35/901335/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/35/901335/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/901335/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/901335/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/901335/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/35/901335/3"}}},"commit":{"parents":[{"commit":"38a21feddd570a1abb9c41119a76d7b02930245c","subject":"proxy: always use namespaces for updating and listing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/38a21feddd570a1abb9c41119a76d7b02930245c"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-11-17 15:03:59.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-11-20 16:41:58.000000000","tz":0},"subject":"proxy: only use listing shards cache for \u0027auto\u0027 listings","message":"proxy: only use listing shards cache for \u0027auto\u0027 listings\n\nThe proxy should NOT read or write to memcache when handling a\ncontainer GET that explicitly requests \u0027shard\u0027 or \u0027object\u0027 record\ntype. A request for \u0027shard\u0027 record type may specify \u0027namespace\u0027\nformat, but this request is unrelated to container listings or object\nupdates and passes directly to the backend.\n\nThis patch also removes unnecessary JSON serialisation and\nde-serialisation of namespaces within the proxy GET path when a\nsharded object listing is being built. The final response body will\ncontainer a list of objects so there is no need to write intermediate\nresponse bodies with list of namespaces.\n\nRequests that explicitly specify record type of \u0027shard\u0027 will of\ncourse still have the response body with serialized shard dicts that\nis returned from the backend.\n\nChange-Id: Id79c156432350c11c52a4004d69b85e9eb904ca6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/756c72d43359b2a7911676892e40a008a6a7e0e6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/756c72d43359b2a7911676892e40a008a6a7e0e6"}]},"branch":"refs/heads/master"},"7334c9c37d02adf5e37658af7f6bb9ccb49966c3":{"kind":"REWORK","_number":4,"created":"2023-11-20 18:37:10.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/35/901335/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/35/901335/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/901335/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/901335/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/901335/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/35/901335/4"}}},"commit":{"parents":[{"commit":"e4de470b690ca40491ae178764e2cc2d4e8e8018","subject":"proxy: always use namespaces for updating and listing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e4de470b690ca40491ae178764e2cc2d4e8e8018"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-11-17 15:03:59.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-11-20 18:36:42.000000000","tz":0},"subject":"proxy: only use listing shards cache for \u0027auto\u0027 listings","message":"proxy: only use listing shards cache for \u0027auto\u0027 listings\n\nThe proxy should NOT read or write to memcache when handling a\ncontainer GET that explicitly requests \u0027shard\u0027 or \u0027object\u0027 record\ntype. A request for \u0027shard\u0027 record type may specify \u0027namespace\u0027\nformat, but this request is unrelated to container listings or object\nupdates and passes directly to the backend.\n\nThis patch also removes unnecessary JSON serialisation and\nde-serialisation of namespaces within the proxy GET path when a\nsharded object listing is being built. The final response body will\ncontainer a list of objects so there is no need to write intermediate\nresponse bodies with list of namespaces.\n\nRequests that explicitly specify record type of \u0027shard\u0027 will of\ncourse still have the response body with serialized shard dicts that\nis returned from the backend.\n\nChange-Id: Id79c156432350c11c52a4004d69b85e9eb904ca6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7334c9c37d02adf5e37658af7f6bb9ccb49966c3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7334c9c37d02adf5e37658af7f6bb9ccb49966c3"}]},"branch":"refs/heads/master"},"765f1e6e88f551d8cbc11c9151726bd507c02ac4":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2023-11-21 10:54:11.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/35/901335/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/35/901335/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/901335/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/901335/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/901335/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/35/901335/5"}}},"commit":{"parents":[{"commit":"7cb6b2bb06f85f760e54c2ba45ada51a95690709","subject":"proxy: always use namespaces for updating and listing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7cb6b2bb06f85f760e54c2ba45ada51a95690709"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-11-17 15:03:59.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-11-21 10:30:32.000000000","tz":0},"subject":"proxy: only use listing shards cache for \u0027auto\u0027 listings","message":"proxy: only use listing shards cache for \u0027auto\u0027 listings\n\nThe proxy should NOT read or write to memcache when handling a\ncontainer GET that explicitly requests \u0027shard\u0027 or \u0027object\u0027 record\ntype. A request for \u0027shard\u0027 record type may specify \u0027namespace\u0027\nformat, but this request is unrelated to container listings or object\nupdates and passes directly to the backend.\n\nThis patch also removes unnecessary JSON serialisation and\nde-serialisation of namespaces within the proxy GET path when a\nsharded object listing is being built. The final response body will\ncontainer a list of objects so there is no need to write intermediate\nresponse bodies with list of namespaces.\n\nRequests that explicitly specify record type of \u0027shard\u0027 will of\ncourse still have the response body with serialized shard dicts that\nis returned from the backend.\n\nChange-Id: Id79c156432350c11c52a4004d69b85e9eb904ca6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/765f1e6e88f551d8cbc11c9151726bd507c02ac4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/765f1e6e88f551d8cbc11c9151726bd507c02ac4"}]},"branch":"refs/heads/master"},"e35ed2d9ef70cf1589829944355ff0f1af95ca66":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2023-11-21 18:37:29.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/35/901335/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/35/901335/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/901335/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/901335/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/901335/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/35/901335/6"}}},"commit":{"parents":[{"commit":"ee29faaa2d7dde2764a32938e904a0173dfcd25e","subject":"proxy: always use namespaces for updating and listing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ee29faaa2d7dde2764a32938e904a0173dfcd25e"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-11-17 15:03:59.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-11-21 18:35:47.000000000","tz":0},"subject":"proxy: only use listing shards cache for \u0027auto\u0027 listings","message":"proxy: only use listing shards cache for \u0027auto\u0027 listings\n\nThe proxy should NOT read or write to memcache when handling a\ncontainer GET that explicitly requests \u0027shard\u0027 or \u0027object\u0027 record\ntype. A request for \u0027shard\u0027 record type may specify \u0027namespace\u0027\nformat, but this request is unrelated to container listings or object\nupdates and passes directly to the backend.\n\nThis patch also removes unnecessary JSON serialisation and\nde-serialisation of namespaces within the proxy GET path when a\nsharded object listing is being built. The final response body will\ncontainer a list of objects so there is no need to write intermediate\nresponse bodies with list of namespaces.\n\nRequests that explicitly specify record type of \u0027shard\u0027 will of\ncourse still have the response body with serialized shard dicts that\nis returned from the backend.\n\nChange-Id: Id79c156432350c11c52a4004d69b85e9eb904ca6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e35ed2d9ef70cf1589829944355ff0f1af95ca66"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e35ed2d9ef70cf1589829944355ff0f1af95ca66"}]},"branch":"refs/heads/master"},"3c02958a69e3270894531b328bcfa741f431cb71":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2023-11-22 11:07:53.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/35/901335/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/35/901335/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/901335/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/901335/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/901335/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/35/901335/7"}}},"commit":{"parents":[{"commit":"139c51e53307a8a9cb6c15f6ac4a67df492bc68c","subject":"proxy: always use namespaces for updating and listing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/139c51e53307a8a9cb6c15f6ac4a67df492bc68c"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-11-17 15:03:59.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-11-22 11:07:31.000000000","tz":0},"subject":"proxy: only use listing shards cache for \u0027auto\u0027 listings","message":"proxy: only use listing shards cache for \u0027auto\u0027 listings\n\nThe proxy should NOT read or write to memcache when handling a\ncontainer GET that explicitly requests \u0027shard\u0027 or \u0027object\u0027 record\ntype. A request for \u0027shard\u0027 record type may specify \u0027namespace\u0027\nformat, but this request is unrelated to container listings or object\nupdates and passes directly to the backend.\n\nThis patch also removes unnecessary JSON serialisation and\nde-serialisation of namespaces within the proxy GET path when a\nsharded object listing is being built. The final response body will\ncontainer a list of objects so there is no need to write intermediate\nresponse bodies with list of namespaces.\n\nRequests that explicitly specify record type of \u0027shard\u0027 will of\ncourse still have the response body with serialized shard dicts that\nis returned from the backend.\n\nChange-Id: Id79c156432350c11c52a4004d69b85e9eb904ca6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3c02958a69e3270894531b328bcfa741f431cb71"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3c02958a69e3270894531b328bcfa741f431cb71"}]},"branch":"refs/heads/master"},"bbc91e1d086fdeddac8a332275e327fcd0bd364c":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2023-11-22 12:52:13.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/35/901335/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/35/901335/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/901335/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/901335/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/901335/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/35/901335/8"}}},"commit":{"parents":[{"commit":"6af59026dd9c2d0f126eaf07047a25a9d555d58b","subject":"proxy: always use namespaces for updating and listing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6af59026dd9c2d0f126eaf07047a25a9d555d58b"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-11-17 15:03:59.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-11-22 12:47:44.000000000","tz":0},"subject":"proxy: only use listing shards cache for \u0027auto\u0027 listings","message":"proxy: only use listing shards cache for \u0027auto\u0027 listings\n\nThe proxy should NOT read or write to memcache when handling a\ncontainer GET that explicitly requests \u0027shard\u0027 or \u0027object\u0027 record\ntype. A request for \u0027shard\u0027 record type may specify \u0027namespace\u0027\nformat, but this request is unrelated to container listings or object\nupdates and passes directly to the backend.\n\nThis patch also removes unnecessary JSON serialisation and\nde-serialisation of namespaces within the proxy GET path when a\nsharded object listing is being built. The final response body will\ncontainer a list of objects so there is no need to write intermediate\nresponse bodies with list of namespaces.\n\nRequests that explicitly specify record type of \u0027shard\u0027 will of\ncourse still have the response body with serialized shard dicts that\nis returned from the backend.\n\nChange-Id: Id79c156432350c11c52a4004d69b85e9eb904ca6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bbc91e1d086fdeddac8a332275e327fcd0bd364c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bbc91e1d086fdeddac8a332275e327fcd0bd364c"}]},"branch":"refs/heads/master"},"8e31247b13a644433fc0c31837302151dde7c253":{"kind":"REWORK","_number":9,"created":"2023-11-30 16:22:51.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/35/901335/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/35/901335/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/901335/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/901335/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/901335/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/35/901335/9"}}},"commit":{"parents":[{"commit":"b92d644e71309f57ff4028e82a07a923ad097bb5","subject":"sq: Namespace.name must be \u0027a/c\u0027 after all","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b92d644e71309f57ff4028e82a07a923ad097bb5"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-11-17 15:03:59.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-11-30 16:08:45.000000000","tz":0},"subject":"proxy: only use listing shards cache for \u0027auto\u0027 listings","message":"proxy: only use listing shards cache for \u0027auto\u0027 listings\n\nThe proxy should NOT read or write to memcache when handling a\ncontainer GET that explicitly requests \u0027shard\u0027 or \u0027object\u0027 record\ntype. A request for \u0027shard\u0027 record type may specify \u0027namespace\u0027\nformat, but this request is unrelated to container listings or object\nupdates and passes directly to the backend.\n\nThis patch also removes unnecessary JSON serialisation and\nde-serialisation of namespaces within the proxy GET path when a\nsharded object listing is being built. The final response body will\ncontainer a list of objects so there is no need to write intermediate\nresponse bodies with list of namespaces.\n\nRequests that explicitly specify record type of \u0027shard\u0027 will of\ncourse still have the response body with serialized shard dicts that\nis returned from the backend.\n\nChange-Id: Id79c156432350c11c52a4004d69b85e9eb904ca6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8e31247b13a644433fc0c31837302151dde7c253"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8e31247b13a644433fc0c31837302151dde7c253"}]},"branch":"refs/heads/master"},"9585159d16d47d5b6b493dadd1f66d18b8f0e904":{"kind":"TRIVIAL_REBASE","_number":10,"created":"2023-11-30 18:20:14.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/35/901335/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/35/901335/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/901335/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/901335/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/901335/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/35/901335/10"}}},"commit":{"parents":[{"commit":"1a532557e8b0b327381487e58ed7eb7627b2970e","subject":"sq: Namespace.name must be \u0027a/c\u0027 after all","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1a532557e8b0b327381487e58ed7eb7627b2970e"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-11-17 15:03:59.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-11-30 18:19:14.000000000","tz":0},"subject":"proxy: only use listing shards cache for \u0027auto\u0027 listings","message":"proxy: only use listing shards cache for \u0027auto\u0027 listings\n\nThe proxy should NOT read or write to memcache when handling a\ncontainer GET that explicitly requests \u0027shard\u0027 or \u0027object\u0027 record\ntype. A request for \u0027shard\u0027 record type may specify \u0027namespace\u0027\nformat, but this request is unrelated to container listings or object\nupdates and passes directly to the backend.\n\nThis patch also removes unnecessary JSON serialisation and\nde-serialisation of namespaces within the proxy GET path when a\nsharded object listing is being built. The final response body will\ncontainer a list of objects so there is no need to write intermediate\nresponse bodies with list of namespaces.\n\nRequests that explicitly specify record type of \u0027shard\u0027 will of\ncourse still have the response body with serialized shard dicts that\nis returned from the backend.\n\nChange-Id: Id79c156432350c11c52a4004d69b85e9eb904ca6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9585159d16d47d5b6b493dadd1f66d18b8f0e904"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9585159d16d47d5b6b493dadd1f66d18b8f0e904"}]},"branch":"refs/heads/master"},"d271079b917a510314c5fa206e2864b3c339e619":{"kind":"NO_CHANGE","_number":11,"created":"2023-12-04 09:30:09.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/35/901335/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/35/901335/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/901335/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/901335/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/901335/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/35/901335/11"}}},"commit":{"parents":[{"commit":"8212898c56041425d373364a36108cdd2adda343","subject":"proxy: always use namespaces for updating and listing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8212898c56041425d373364a36108cdd2adda343"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-11-17 15:03:59.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-12-04 09:29:51.000000000","tz":0},"subject":"proxy: only use listing shards cache for \u0027auto\u0027 listings","message":"proxy: only use listing shards cache for \u0027auto\u0027 listings\n\nThe proxy should NOT read or write to memcache when handling a\ncontainer GET that explicitly requests \u0027shard\u0027 or \u0027object\u0027 record\ntype. A request for \u0027shard\u0027 record type may specify \u0027namespace\u0027\nformat, but this request is unrelated to container listings or object\nupdates and passes directly to the backend.\n\nThis patch also removes unnecessary JSON serialisation and\nde-serialisation of namespaces within the proxy GET path when a\nsharded object listing is being built. The final response body will\ncontainer a list of objects so there is no need to write intermediate\nresponse bodies with list of namespaces.\n\nRequests that explicitly specify record type of \u0027shard\u0027 will of\ncourse still have the response body with serialized shard dicts that\nis returned from the backend.\n\nChange-Id: Id79c156432350c11c52a4004d69b85e9eb904ca6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d271079b917a510314c5fa206e2864b3c339e619"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d271079b917a510314c5fa206e2864b3c339e619"}]},"branch":"refs/heads/master"},"e66c13a4c4938f2acc9144f40524fe3b4309cc27":{"kind":"TRIVIAL_REBASE","_number":12,"created":"2023-12-05 13:58:14.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/35/901335/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/35/901335/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/901335/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/901335/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/901335/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/35/901335/12"}}},"commit":{"parents":[{"commit":"e3de45d302f9305c963df84c0a17f701ace7e8a8","subject":"proxy: always use namespaces for updating and listing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e3de45d302f9305c963df84c0a17f701ace7e8a8"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-11-17 15:03:59.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-12-05 13:52:17.000000000","tz":0},"subject":"proxy: only use listing shards cache for \u0027auto\u0027 listings","message":"proxy: only use listing shards cache for \u0027auto\u0027 listings\n\nThe proxy should NOT read or write to memcache when handling a\ncontainer GET that explicitly requests \u0027shard\u0027 or \u0027object\u0027 record\ntype. A request for \u0027shard\u0027 record type may specify \u0027namespace\u0027\nformat, but this request is unrelated to container listings or object\nupdates and passes directly to the backend.\n\nThis patch also removes unnecessary JSON serialisation and\nde-serialisation of namespaces within the proxy GET path when a\nsharded object listing is being built. The final response body will\ncontainer a list of objects so there is no need to write intermediate\nresponse bodies with list of namespaces.\n\nRequests that explicitly specify record type of \u0027shard\u0027 will of\ncourse still have the response body with serialized shard dicts that\nis returned from the backend.\n\nChange-Id: Id79c156432350c11c52a4004d69b85e9eb904ca6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e66c13a4c4938f2acc9144f40524fe3b4309cc27"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e66c13a4c4938f2acc9144f40524fe3b4309cc27"}]},"branch":"refs/heads/master"},"167554516e00d276a997593edf39396fcb592b76":{"kind":"NO_CHANGE","_number":13,"created":"2023-12-05 16:21:14.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/35/901335/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/35/901335/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/901335/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/901335/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/901335/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/35/901335/13"}}},"commit":{"parents":[{"commit":"559c3cc58514a3e3c367b8e51e419e7de0868395","subject":"proxy: always use namespaces for updating and listing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/559c3cc58514a3e3c367b8e51e419e7de0868395"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-11-17 15:03:59.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-12-05 16:14:11.000000000","tz":0},"subject":"proxy: only use listing shards cache for \u0027auto\u0027 listings","message":"proxy: only use listing shards cache for \u0027auto\u0027 listings\n\nThe proxy should NOT read or write to memcache when handling a\ncontainer GET that explicitly requests \u0027shard\u0027 or \u0027object\u0027 record\ntype. A request for \u0027shard\u0027 record type may specify \u0027namespace\u0027\nformat, but this request is unrelated to container listings or object\nupdates and passes directly to the backend.\n\nThis patch also removes unnecessary JSON serialisation and\nde-serialisation of namespaces within the proxy GET path when a\nsharded object listing is being built. The final response body will\ncontainer a list of objects so there is no need to write intermediate\nresponse bodies with list of namespaces.\n\nRequests that explicitly specify record type of \u0027shard\u0027 will of\ncourse still have the response body with serialized shard dicts that\nis returned from the backend.\n\nChange-Id: Id79c156432350c11c52a4004d69b85e9eb904ca6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/167554516e00d276a997593edf39396fcb592b76"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/167554516e00d276a997593edf39396fcb592b76"}]},"branch":"refs/heads/master"},"3e8864d9a71152e46934dd90c64458224bd53d7f":{"kind":"REWORK","_number":14,"created":"2023-12-05 18:38:59.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/35/901335/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/35/901335/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/901335/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/901335/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/901335/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/35/901335/14"}}},"commit":{"parents":[{"commit":"a6e9bb549c6acf98d4d4092565a1ac545d4f02e8","subject":"proxy: always use namespaces for updating and listing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a6e9bb549c6acf98d4d4092565a1ac545d4f02e8"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-11-17 15:03:59.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-12-05 18:23:41.000000000","tz":0},"subject":"proxy: only use listing shards cache for \u0027auto\u0027 listings","message":"proxy: only use listing shards cache for \u0027auto\u0027 listings\n\nThe proxy should NOT read or write to memcache when handling a\ncontainer GET that explicitly requests \u0027shard\u0027 or \u0027object\u0027 record\ntype. A request for \u0027shard\u0027 record type may specify \u0027namespace\u0027\nformat, but this request is unrelated to container listings or object\nupdates and passes directly to the backend.\n\nThis patch also removes unnecessary JSON serialisation and\nde-serialisation of namespaces within the proxy GET path when a\nsharded object listing is being built. The final response body will\ncontainer a list of objects so there is no need to write intermediate\nresponse bodies with list of namespaces.\n\nRequests that explicitly specify record type of \u0027shard\u0027 will of\ncourse still have the response body with serialized shard dicts that\nis returned from the backend.\n\nChange-Id: Id79c156432350c11c52a4004d69b85e9eb904ca6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3e8864d9a71152e46934dd90c64458224bd53d7f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3e8864d9a71152e46934dd90c64458224bd53d7f"}]},"branch":"refs/heads/master"},"19415232d6693577b0cabf0f9ce0fb6a62de12f6":{"kind":"REWORK","_number":15,"created":"2023-12-11 18:59:13.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/35/901335/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/35/901335/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/901335/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/901335/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/901335/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/35/901335/15"}}},"commit":{"parents":[{"commit":"9360a785c5936ea4af627cc3426cb3f79ca0577d","subject":"proxy: always use namespaces for updating and listing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9360a785c5936ea4af627cc3426cb3f79ca0577d"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-11-17 15:03:59.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-12-11 18:58:49.000000000","tz":0},"subject":"proxy: only use listing shards cache for \u0027auto\u0027 listings","message":"proxy: only use listing shards cache for \u0027auto\u0027 listings\n\nThe proxy should NOT read or write to memcache when handling a\ncontainer GET that explicitly requests \u0027shard\u0027 or \u0027object\u0027 record\ntype. A request for \u0027shard\u0027 record type may specify \u0027namespace\u0027\nformat, but this request is unrelated to container listings or object\nupdates and passes directly to the backend.\n\nThis patch also removes unnecessary JSON serialisation and\nde-serialisation of namespaces within the proxy GET path when a\nsharded object listing is being built. The final response body will\ncontainer a list of objects so there is no need to write intermediate\nresponse bodies with list of namespaces.\n\nRequests that explicitly specify record type of \u0027shard\u0027 will of\ncourse still have the response body with serialized shard dicts that\nis returned from the backend.\n\nChange-Id: Id79c156432350c11c52a4004d69b85e9eb904ca6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/19415232d6693577b0cabf0f9ce0fb6a62de12f6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/19415232d6693577b0cabf0f9ce0fb6a62de12f6"}]},"branch":"refs/heads/master"},"d399be872b7143dd4214246d3d053d116ea5849d":{"kind":"TRIVIAL_REBASE","_number":16,"created":"2024-01-08 15:01:05.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/35/901335/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/35/901335/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/901335/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/901335/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/901335/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/35/901335/16"}}},"commit":{"parents":[{"commit":"201a84c8984fd2ac54dc46f27b2c7be6154be14c","subject":"Proxy: Use namespaces when getting listing/updating shards","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/201a84c8984fd2ac54dc46f27b2c7be6154be14c"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-11-17 15:03:59.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-01-08 14:59:58.000000000","tz":0},"subject":"proxy: only use listing shards cache for \u0027auto\u0027 listings","message":"proxy: only use listing shards cache for \u0027auto\u0027 listings\n\nThe proxy should NOT read or write to memcache when handling a\ncontainer GET that explicitly requests \u0027shard\u0027 or \u0027object\u0027 record\ntype. A request for \u0027shard\u0027 record type may specify \u0027namespace\u0027\nformat, but this request is unrelated to container listings or object\nupdates and passes directly to the backend.\n\nThis patch also removes unnecessary JSON serialisation and\nde-serialisation of namespaces within the proxy GET path when a\nsharded object listing is being built. The final response body will\ncontainer a list of objects so there is no need to write intermediate\nresponse bodies with list of namespaces.\n\nRequests that explicitly specify record type of \u0027shard\u0027 will of\ncourse still have the response body with serialized shard dicts that\nis returned from the backend.\n\nChange-Id: Id79c156432350c11c52a4004d69b85e9eb904ca6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d399be872b7143dd4214246d3d053d116ea5849d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d399be872b7143dd4214246d3d053d116ea5849d"}]},"branch":"refs/heads/master"},"4c5f1189e3320f4d460db44f9c8a0b332feb9a94":{"kind":"TRIVIAL_REBASE","_number":17,"created":"2024-01-08 18:10:41.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/35/901335/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/35/901335/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/901335/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/901335/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/901335/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/35/901335/17"}}},"commit":{"parents":[{"commit":"0204e5cf3540c94fb49ef35b416a5f14dd5d1e84","subject":"Proxy: Use namespaces when getting listing/updating shards","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0204e5cf3540c94fb49ef35b416a5f14dd5d1e84"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-11-17 15:03:59.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-01-08 18:10:27.000000000","tz":0},"subject":"proxy: only use listing shards cache for \u0027auto\u0027 listings","message":"proxy: only use listing shards cache for \u0027auto\u0027 listings\n\nThe proxy should NOT read or write to memcache when handling a\ncontainer GET that explicitly requests \u0027shard\u0027 or \u0027object\u0027 record\ntype. A request for \u0027shard\u0027 record type may specify \u0027namespace\u0027\nformat, but this request is unrelated to container listings or object\nupdates and passes directly to the backend.\n\nThis patch also removes unnecessary JSON serialisation and\nde-serialisation of namespaces within the proxy GET path when a\nsharded object listing is being built. The final response body will\ncontainer a list of objects so there is no need to write intermediate\nresponse bodies with list of namespaces.\n\nRequests that explicitly specify record type of \u0027shard\u0027 will of\ncourse still have the response body with serialized shard dicts that\nis returned from the backend.\n\nChange-Id: Id79c156432350c11c52a4004d69b85e9eb904ca6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4c5f1189e3320f4d460db44f9c8a0b332feb9a94"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4c5f1189e3320f4d460db44f9c8a0b332feb9a94"}]},"branch":"refs/heads/master"},"21a956a5bed81b47b515fafe87f24e1735f1dd8a":{"kind":"TRIVIAL_REBASE","_number":18,"created":"2024-01-10 14:57:30.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/35/901335/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/35/901335/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/901335/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/901335/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/901335/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/35/901335/18"}}},"commit":{"parents":[{"commit":"e3bd775fafc62212750b2a366660f3055da50a2a","subject":"Proxy: Use namespaces when getting listing/updating shards","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e3bd775fafc62212750b2a366660f3055da50a2a"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-11-17 15:03:59.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-01-10 14:56:27.000000000","tz":0},"subject":"proxy: only use listing shards cache for \u0027auto\u0027 listings","message":"proxy: only use listing shards cache for \u0027auto\u0027 listings\n\nThe proxy should NOT read or write to memcache when handling a\ncontainer GET that explicitly requests \u0027shard\u0027 or \u0027object\u0027 record\ntype. A request for \u0027shard\u0027 record type may specify \u0027namespace\u0027\nformat, but this request is unrelated to container listings or object\nupdates and passes directly to the backend.\n\nThis patch also removes unnecessary JSON serialisation and\nde-serialisation of namespaces within the proxy GET path when a\nsharded object listing is being built. The final response body will\ncontainer a list of objects so there is no need to write intermediate\nresponse bodies with list of namespaces.\n\nRequests that explicitly specify record type of \u0027shard\u0027 will of\ncourse still have the response body with serialized shard dicts that\nis returned from the backend.\n\nChange-Id: Id79c156432350c11c52a4004d69b85e9eb904ca6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/21a956a5bed81b47b515fafe87f24e1735f1dd8a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/21a956a5bed81b47b515fafe87f24e1735f1dd8a"}]},"branch":"refs/heads/master"},"e1dc97831575a596a568d3fe26103a236c8857e8":{"kind":"TRIVIAL_REBASE","_number":19,"created":"2024-01-11 10:51:52.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/35/901335/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/35/901335/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/901335/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/901335/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/901335/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/35/901335/19"}}},"commit":{"parents":[{"commit":"03b66c94f423f06cd6c16742a8ae2ea4bbccbd70","subject":"Proxy: Use namespaces when getting listing/updating shards","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/03b66c94f423f06cd6c16742a8ae2ea4bbccbd70"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-11-17 15:03:59.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-01-11 10:50:32.000000000","tz":0},"subject":"proxy: only use listing shards cache for \u0027auto\u0027 listings","message":"proxy: only use listing shards cache for \u0027auto\u0027 listings\n\nThe proxy should NOT read or write to memcache when handling a\ncontainer GET that explicitly requests \u0027shard\u0027 or \u0027object\u0027 record\ntype. A request for \u0027shard\u0027 record type may specify \u0027namespace\u0027\nformat, but this request is unrelated to container listings or object\nupdates and passes directly to the backend.\n\nThis patch also removes unnecessary JSON serialisation and\nde-serialisation of namespaces within the proxy GET path when a\nsharded object listing is being built. The final response body will\ncontainer a list of objects so there is no need to write intermediate\nresponse bodies with list of namespaces.\n\nRequests that explicitly specify record type of \u0027shard\u0027 will of\ncourse still have the response body with serialized shard dicts that\nis returned from the backend.\n\nChange-Id: Id79c156432350c11c52a4004d69b85e9eb904ca6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e1dc97831575a596a568d3fe26103a236c8857e8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e1dc97831575a596a568d3fe26103a236c8857e8"}]},"branch":"refs/heads/master"},"0c6ccf2abe8c4dc3d98a210c20c647c750118f81":{"kind":"NO_CODE_CHANGE","_number":20,"created":"2024-01-16 10:20:31.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/35/901335/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/35/901335/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/901335/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/901335/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/901335/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/35/901335/20"}}},"commit":{"parents":[{"commit":"03b66c94f423f06cd6c16742a8ae2ea4bbccbd70","subject":"Proxy: Use namespaces when getting listing/updating shards","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/03b66c94f423f06cd6c16742a8ae2ea4bbccbd70"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-11-17 15:03:59.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-01-16 10:16:46.000000000","tz":0},"subject":"proxy: only use listing shards cache for \u0027auto\u0027 listings","message":"proxy: only use listing shards cache for \u0027auto\u0027 listings\n\nThe proxy should NOT read or write to memcache when handling a\ncontainer GET that explicitly requests \u0027shard\u0027 or \u0027object\u0027 record\ntype. A request for \u0027shard\u0027 record type may specify \u0027namespace\u0027\nformat, but this request is unrelated to container listings or object\nupdates and passes directly to the backend.\n\nThis patch also removes unnecessary JSON serialisation and\nde-serialisation of namespaces within the proxy GET path when a\nsharded object listing is being built. The final response body will\ncontain a list of objects so there is no need to write intermediate\nresponse bodies with a list of namespaces.\n\nRequests that explicitly specify record type of \u0027shard\u0027 will of\ncourse still have the response body with serialised shard dicts that\nis returned from the backend.\n\nChange-Id: Id79c156432350c11c52a4004d69b85e9eb904ca6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0c6ccf2abe8c4dc3d98a210c20c647c750118f81"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0c6ccf2abe8c4dc3d98a210c20c647c750118f81"}]},"branch":"refs/heads/master"},"b2a47afd4a788e499e533723a07e5755d16de0c7":{"kind":"REWORK","_number":21,"created":"2024-01-18 11:14:13.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/35/901335/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/35/901335/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/901335/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/901335/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/901335/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/35/901335/21"}}},"commit":{"parents":[{"commit":"03b66c94f423f06cd6c16742a8ae2ea4bbccbd70","subject":"Proxy: Use namespaces when getting listing/updating shards","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/03b66c94f423f06cd6c16742a8ae2ea4bbccbd70"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-11-17 15:03:59.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-01-18 11:04:07.000000000","tz":0},"subject":"proxy: only use listing shards cache for \u0027auto\u0027 listings","message":"proxy: only use listing shards cache for \u0027auto\u0027 listings\n\nThe proxy should NOT read or write to memcache when handling a\ncontainer GET that explicitly requests \u0027shard\u0027 or \u0027object\u0027 record\ntype. A request for \u0027shard\u0027 record type may specify \u0027namespace\u0027\nformat, but this request is unrelated to container listings or object\nupdates and passes directly to the backend.\n\nThis patch also removes unnecessary JSON serialisation and\nde-serialisation of namespaces within the proxy GET path when a\nsharded object listing is being built. The final response body will\ncontain a list of objects so there is no need to write intermediate\nresponse bodies with a list of namespaces.\n\nRequests that explicitly specify record type of \u0027shard\u0027 will of\ncourse still have the response body with serialised shard dicts that\nis returned from the backend.\n\nChange-Id: Id79c156432350c11c52a4004d69b85e9eb904ca6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b2a47afd4a788e499e533723a07e5755d16de0c7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b2a47afd4a788e499e533723a07e5755d16de0c7"}]},"branch":"refs/heads/master"},"731a7c36e35d41a3743f98249dec02b1014eb464":{"kind":"REWORK","_number":22,"created":"2024-01-23 13:02:06.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/35/901335/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/35/901335/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/901335/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/901335/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/901335/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/35/901335/22"}}},"commit":{"parents":[{"commit":"03b66c94f423f06cd6c16742a8ae2ea4bbccbd70","subject":"Proxy: Use namespaces when getting listing/updating shards","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/03b66c94f423f06cd6c16742a8ae2ea4bbccbd70"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-11-17 15:03:59.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-01-23 13:01:54.000000000","tz":0},"subject":"proxy: only use listing shards cache for \u0027auto\u0027 listings","message":"proxy: only use listing shards cache for \u0027auto\u0027 listings\n\nThe proxy should NOT read or write to memcache when handling a\ncontainer GET that explicitly requests \u0027shard\u0027 or \u0027object\u0027 record\ntype. A request for \u0027shard\u0027 record type may specify \u0027namespace\u0027\nformat, but this request is unrelated to container listings or object\nupdates and passes directly to the backend.\n\nThis patch also removes unnecessary JSON serialisation and\nde-serialisation of namespaces within the proxy GET path when a\nsharded object listing is being built. The final response body will\ncontain a list of objects so there is no need to write intermediate\nresponse bodies with a list of namespaces.\n\nRequests that explicitly specify record type of \u0027shard\u0027 will of\ncourse still have the response body with serialised shard dicts that\nis returned from the backend.\n\nChange-Id: Id79c156432350c11c52a4004d69b85e9eb904ca6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/731a7c36e35d41a3743f98249dec02b1014eb464"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/731a7c36e35d41a3743f98249dec02b1014eb464"}]},"branch":"refs/heads/master"},"57c4e5cd01b8d358714de66069bfa6e62b89aab5":{"kind":"REWORK","_number":23,"created":"2024-01-23 13:10:13.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/35/901335/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/35/901335/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/901335/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/901335/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/901335/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/35/901335/23"}}},"commit":{"parents":[{"commit":"03b66c94f423f06cd6c16742a8ae2ea4bbccbd70","subject":"Proxy: Use namespaces when getting listing/updating shards","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/03b66c94f423f06cd6c16742a8ae2ea4bbccbd70"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-11-17 15:03:59.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-01-23 13:10:05.000000000","tz":0},"subject":"proxy: only use listing shards cache for \u0027auto\u0027 listings","message":"proxy: only use listing shards cache for \u0027auto\u0027 listings\n\nThe proxy should NOT read or write to memcache when handling a\ncontainer GET that explicitly requests \u0027shard\u0027 or \u0027object\u0027 record\ntype. A request for \u0027shard\u0027 record type may specify \u0027namespace\u0027\nformat, but this request is unrelated to container listings or object\nupdates and passes directly to the backend.\n\nThis patch also removes unnecessary JSON serialisation and\nde-serialisation of namespaces within the proxy GET path when a\nsharded object listing is being built. The final response body will\ncontain a list of objects so there is no need to write intermediate\nresponse bodies with a list of namespaces.\n\nRequests that explicitly specify record type of \u0027shard\u0027 will of\ncourse still have the response body with serialised shard dicts that\nis returned from the backend.\n\nChange-Id: Id79c156432350c11c52a4004d69b85e9eb904ca6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/57c4e5cd01b8d358714de66069bfa6e62b89aab5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/57c4e5cd01b8d358714de66069bfa6e62b89aab5"}]},"branch":"refs/heads/master"},"e1b43037a365eb62c10bc37841cbd120bdf2dee4":{"kind":"REWORK","_number":24,"created":"2024-01-24 16:50:04.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/35/901335/24","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/35/901335/24","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/901335/24 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/901335/24 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/901335/24 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/35/901335/24"}}},"commit":{"parents":[{"commit":"03b66c94f423f06cd6c16742a8ae2ea4bbccbd70","subject":"Proxy: Use namespaces when getting listing/updating shards","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/03b66c94f423f06cd6c16742a8ae2ea4bbccbd70"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-11-17 15:03:59.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-01-24 16:49:55.000000000","tz":0},"subject":"proxy: only use listing shards cache for \u0027auto\u0027 listings","message":"proxy: only use listing shards cache for \u0027auto\u0027 listings\n\nThe proxy should NOT read or write to memcache when handling a\ncontainer GET that explicitly requests \u0027shard\u0027 or \u0027object\u0027 record\ntype. A request for \u0027shard\u0027 record type may specify \u0027namespace\u0027\nformat, but this request is unrelated to container listings or object\nupdates and passes directly to the backend.\n\nThis patch also removes unnecessary JSON serialisation and\nde-serialisation of namespaces within the proxy GET path when a\nsharded object listing is being built. The final response body will\ncontain a list of objects so there is no need to write intermediate\nresponse bodies with a list of namespaces.\n\nRequests that explicitly specify record type of \u0027shard\u0027 will of\ncourse still have the response body with serialised shard dicts that\nis returned from the backend.\n\nChange-Id: Id79c156432350c11c52a4004d69b85e9eb904ca6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e1b43037a365eb62c10bc37841cbd120bdf2dee4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e1b43037a365eb62c10bc37841cbd120bdf2dee4"}]},"branch":"refs/heads/master"},"b5869c93d360d5ff6109ab6a2048445758df7a65":{"kind":"REWORK","_number":25,"created":"2024-01-25 12:20:15.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/35/901335/25","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/35/901335/25","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/901335/25 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/901335/25 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/901335/25 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/35/901335/25"}}},"commit":{"parents":[{"commit":"03b66c94f423f06cd6c16742a8ae2ea4bbccbd70","subject":"Proxy: Use namespaces when getting listing/updating shards","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/03b66c94f423f06cd6c16742a8ae2ea4bbccbd70"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-11-17 15:03:59.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-01-25 12:17:50.000000000","tz":0},"subject":"proxy: only use listing shards cache for \u0027auto\u0027 listings","message":"proxy: only use listing shards cache for \u0027auto\u0027 listings\n\nThe proxy should NOT read or write to memcache when handling a\ncontainer GET that explicitly requests \u0027shard\u0027 or \u0027object\u0027 record\ntype. A request for \u0027shard\u0027 record type may specify \u0027namespace\u0027\nformat, but this request is unrelated to container listings or object\nupdates and passes directly to the backend.\n\nThis patch also removes unnecessary JSON serialisation and\nde-serialisation of namespaces within the proxy GET path when a\nsharded object listing is being built. The final response body will\ncontain a list of objects so there is no need to write intermediate\nresponse bodies with a list of namespaces.\n\nRequests that explicitly specify record type of \u0027shard\u0027 will of\ncourse still have the response body with serialised shard dicts that\nis returned from the backend.\n\nChange-Id: Id79c156432350c11c52a4004d69b85e9eb904ca6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b5869c93d360d5ff6109ab6a2048445758df7a65"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b5869c93d360d5ff6109ab6a2048445758df7a65"}]},"branch":"refs/heads/master"},"252f0d36b7c97d1e3261fd4cf475b362ee985abe":{"kind":"REWORK","_number":26,"created":"2024-01-31 11:03:11.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/35/901335/26","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/35/901335/26","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/901335/26 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/901335/26 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/901335/26 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/35/901335/26"}}},"commit":{"parents":[{"commit":"03b66c94f423f06cd6c16742a8ae2ea4bbccbd70","subject":"Proxy: Use namespaces when getting listing/updating shards","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/03b66c94f423f06cd6c16742a8ae2ea4bbccbd70"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-11-17 15:03:59.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-01-31 11:02:54.000000000","tz":0},"subject":"proxy: only use listing shards cache for \u0027auto\u0027 listings","message":"proxy: only use listing shards cache for \u0027auto\u0027 listings\n\nThe proxy should NOT read or write to memcache when handling a\ncontainer GET that explicitly requests \u0027shard\u0027 or \u0027object\u0027 record\ntype. A request for \u0027shard\u0027 record type may specify \u0027namespace\u0027\nformat, but this request is unrelated to container listings or object\nupdates and passes directly to the backend.\n\nThis patch also removes unnecessary JSON serialisation and\nde-serialisation of namespaces within the proxy GET path when a\nsharded object listing is being built. The final response body will\ncontain a list of objects so there is no need to write intermediate\nresponse bodies with a list of namespaces.\n\nRequests that explicitly specify record type of \u0027shard\u0027 will of\ncourse still have the response body with serialised shard dicts that\nis returned from the backend.\n\nChange-Id: Id79c156432350c11c52a4004d69b85e9eb904ca6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/252f0d36b7c97d1e3261fd4cf475b362ee985abe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/252f0d36b7c97d1e3261fd4cf475b362ee985abe"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
