)]}'
{"id":"openstack%2Fswift~901603","triplet_id":"openstack%2Fswift~master~I45f7678c9932afa038438ee841d1b262d53c9da8","project":"openstack/swift","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2023-12-01 01:48:57.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I45f7678c9932afa038438ee841d1b262d53c9da8","subject":"Remove per-service auto_create_account_prefix","status":"MERGED","created":"2023-11-21 16:29:07.000000000","updated":"2023-12-01 01:50:36.000000000","submitted":"2023-12-01 01:48:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"901603","meta_rev_id":"998a8020fd1ffda9b3babba4a85e5dc66708d70c","_number":901603,"virtual_id_number":901603,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-12-01 01:48:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"all":[{"value":2,"date":"2023-11-21 20:54:14.000000000","_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":1,"date":"2023-11-29 06:10:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-12-01 00:26:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-12-01 00:26:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-11-21 16:51:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-11-21 18:16:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-11-21 20:54:14.000000000","updated_by":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"reviewer":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"state":"REVIEWER"},{"updated":"2023-11-29 06:10:00.000000000","updated_by":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"reviewer":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"state":"REVIEWER"},{"updated":"2023-12-01 00:26:40.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"}],"messages":[{"id":"f68989db7274cb2b84b63287bbaa8bdde657980b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-11-21 16:29:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"71b15b1a933cd6ebd354ffee1630e60640e70507","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-21 16:51:18.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e511f522c87d42c2be0d3a4edea37fb9\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/2a9a3c5ab4a14ee288b7768047de92fa : FAILURE in 16m 34s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/3cf8b2fae8cb4f639b084b5cde10cb8e : FAILURE in 19m 26s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/e160bcf177a443478df9a375f5e56177 : SUCCESS in 21m 06s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/4c46c2879a4549be9ba4ec6bf6bbf02a : SUCCESS in 20m 25s","accounts_in_message":[],"_revision_number":1},{"id":"797b56e72c3b33bf57cbd9e7cb38936a195e7b16","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-11-21 16:58:20.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"14bf0ed9792d27bfa889e7cfc7ec09a428cbad04","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-21 18:16:00.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/99b4a8b1558d4340825305e60eebd21f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9be9a2544f4a41efae0e31ad61be49c7 : SUCCESS in 7m 55s\n- grenade https://zuul.opendev.org/t/openstack/build/7af042f875c1470c893ec46f87a9dcd4 : SUCCESS in 51m 19s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/15d3d309b7594403831154d2b69fb435 : SUCCESS in 59m 17s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/a45f9f2655e4497595893b3b59b2bad5 : SUCCESS in 40m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5e4ed6f1f95f433891d46f1314ef1390 : SUCCESS in 1h 10m 54s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e3ab81d29b6e45fa8dec78bfb349c3ac : SUCCESS in 25m 29s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/dae271bb1e33428582725caef1114bbe : SUCCESS in 18m 48s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f12521c96137412a9f4f762149cebb35 : SUCCESS in 28m 30s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/2c9b8df464cd469da1a18f251f88e9c0 : SUCCESS in 9m 47s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/a1412f6c1a8f4f4393498d9bca267f8a : SUCCESS in 5m 30s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/c4e5d6f9f7464b1db73d27d09fc4faff : SUCCESS in 17m 47s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/48e6e660dc4f406d8e03669d2519e6d1 : SUCCESS in 16m 48s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/46639a8b3b7e4946b8c91e04593923d7 : SUCCESS in 16m 14s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/5a306aa2e6c142199d148752118e05a2 : SUCCESS in 16m 12s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/f61605fc2013469f99923fe01df37503 : SUCCESS in 18m 02s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/62443f13e6164540a157987c387ad6a7 : SUCCESS in 19m 39s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/fc8bb5832d424568bcb74531a93bb5a0 : SUCCESS in 20m 37s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/cf7f36ad208548619d69497fd9aeb5c1 : SUCCESS in 16m 19s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/852692db51a4403ba78856d9d6094bc7 : SUCCESS in 13m 46s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/4e04e826a32f4b6a8b408a08bd93f6ab : SUCCESS in 15m 36s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/80c89ae1ec404eb1b7c8bf4804be09bc : SUCCESS in 15m 36s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/824d055b94d8400c9538205e53275e65 : SUCCESS in 7m 39s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/24b6290ed34e43088356fe92a66f71ab : SUCCESS in 18m 03s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/7989e76a270f48d3a272f9cc53c81d23 : SUCCESS in 7m 41s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/c2c15ae2dbb74ca5a2f446c67dcea806 : SUCCESS in 45m 55s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/7ed836630690444fabb4bc3735c97b87 : SUCCESS in 52m 50s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/d4adfa6f937048d0ad12bacca68888a7 : SUCCESS in 52m 22s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/239b7a649bdc4971b4b96364cc4d9f67 : SUCCESS in 59m 12s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4c1cab9cb5514492b13e7638f58b138c : SUCCESS in 17m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f99b327d8abb44eda32a3448fb8d6361 : SUCCESS in 4m 42s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/2885a352e1c04a4499c0453373c8371a : SUCCESS in 22m 05s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/979f11b311394abf80f9b88de19abb91 : SUCCESS in 1h 06m 08s","accounts_in_message":[],"_revision_number":2},{"id":"2d3299113ca185e06742e677595586e155746814","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-21 18:32:13.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b206634ed334499d914710700966cb27\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/6ee8d4ddcae3441ea4c2fd7a6e473c65 : TIMED_OUT in 1h 30m 58s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f311e3d3079b441d9f952a4b1f767a1c : SUCCESS in 52m 15s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/34dede879fb144009c4a32e4d5300ba0 : SUCCESS in 19m 18s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/68335e054e5e44179aa020f86ba295cd : SUCCESS in 1h 13m 54s","accounts_in_message":[],"_revision_number":2},{"id":"bfe89140128ab6bf63f3b97f6e12bcb0778a12a8","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2023-11-21 20:54:14.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"f30db09ef910747fc6d008e2c1bb3917abbfadab","author":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"date":"2023-11-29 06:10:00.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"37d51882f85d91eaa7e4f1558c3102aa466c3280","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-12-01 00:26:40.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"9d37f00696b2fcb5ddad793730e31a6bc1fb511a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-01 00:27:21.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"ac1376de31736c1e2f6a74d5188ddc9b0634370f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-01 01:48:56.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c166b1c426654c79a9b17f09efb7ae95\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aa8718fab8b84ef2bf8aaeab4b69277d : SUCCESS in 9m 51s\n- grenade https://zuul.opendev.org/t/openstack/build/7fe263f90969419eb17afce02e8cea6d : SUCCESS in 1h 05m 11s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/efaa073e05d4490b8111a5441ca7c59c : SUCCESS in 1h 02m 13s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/6ae0ee110d2a47bdb056053809d09b9e : SUCCESS in 33m 54s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/20a5c16ffcd4460d9d73de089cea1a91 : SUCCESS in 46m 39s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/5fc9d43dab1b4728ae658dd308429d17 : SUCCESS in 8m 20s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/48ad6bfe333445ccac59c9915b20c42e : SUCCESS in 5m 37s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/8a5a89c7e4ec4de19c6a3bb698dfd26f : SUCCESS in 16m 48s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/ae7e613cf67e4fbb93111ea04f383d00 : SUCCESS in 16m 08s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/903262d6ffb94ea3b24ef1c367110511 : SUCCESS in 15m 54s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/e90d92191aef4d398379a3b1b75dedb9 : SUCCESS in 15m 25s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/79ad486eb1cf4803aa529027d0aad41f : SUCCESS in 17m 33s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/cba33023cbdc4b5c9b51c3d1a971e956 : SUCCESS in 17m 37s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/58dd94dcead54054bddf1aa8fd6e8939 : SUCCESS in 19m 40s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/4f357a26883247509aea77759cd0853c : SUCCESS in 17m 56s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/81d9011d7c4b4693881139ae369c64dc : SUCCESS in 14m 34s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/e534dd53b8204724b9ea2b049b814e34 : SUCCESS in 15m 29s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/770f0dcce18542a798b7bcfcbefa7002 : SUCCESS in 17m 50s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/0901a8bc39aa4d479655cf0d5dafb55f : SUCCESS in 6m 47s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/64e9b5c9dbe344d0a082b83cffc18946 : SUCCESS in 8m 21s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/298c690d86ed4eea9bea2773170e32ff : SUCCESS in 1h 06m 22s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/f042b9126fa349e0bd4bf1aa4f83f4ee : SUCCESS in 1h 02m 28s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/0acc03021dff4e739c618a7c08fc2ef6 : SUCCESS in 58m 20s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/9ad598631ba84f4a87922164545f35af : SUCCESS in 57m 03s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a265612ae73a446d99fc9ff014a805d6 : SUCCESS in 16m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dd3918db4656435993a73f9c41dae8ec : SUCCESS in 6m 54s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ff6afa332d3e477d96b7d2f404e6d1f1 : SUCCESS in 1h 04m 04s","accounts_in_message":[],"_revision_number":2},{"id":"0426e18e4c2c5ea31a7f8c2fb4a799df660c2694","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-01 01:48:57.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"998a8020fd1ffda9b3babba4a85e5dc66708d70c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-01 01:50:36.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/70a397c118144500b325d42fb2ac925d\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7311b1cefc954d48ba460a1c9a2a376a : SUCCESS in 1m 11s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/a0b83c412f4f45c2b2e8051597e07de0 : SUCCESS in 33s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/ed1657509db84cca83c3c3ef36ec4f8e : SUCCESS in 32s (non-voting)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"49b19613d29bda5717ca5eec8c9f6bc9049f99c0","revisions":{"82a3369a6ff1868be0553b4693fb4242e907d000":{"kind":"REWORK","_number":1,"created":"2023-11-21 16:29:07.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/03/901603/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/03/901603/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/03/901603/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/03/901603/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/03/901603/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/03/901603/1"}}},"commit":{"parents":[{"commit":"5555980fb5fce79140efc0c8ab8a730bd552733c","subject":"Merge \"Swap find_executable for which\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5555980fb5fce79140efc0c8ab8a730bd552733c"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-21 16:16:39.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-21 16:25:30.000000000","tz":540},"subject":"Remove per-service auto_create_account_prefix","message":"Remove per-service auto_create_account_prefix\n\nThe per-service option was deprecated almost 4 years ago[1].\n\n[1] 4601548dabdec0a4dc89cefba11e963217255be3\n\nChange-Id: I45f7678c9932afa038438ee841d1b262d53c9da8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/82a3369a6ff1868be0553b4693fb4242e907d000"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/82a3369a6ff1868be0553b4693fb4242e907d000"}]},"branch":"refs/heads/master"},"49b19613d29bda5717ca5eec8c9f6bc9049f99c0":{"kind":"REWORK","_number":2,"created":"2023-11-21 16:58:20.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/03/901603/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/03/901603/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/03/901603/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/03/901603/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/03/901603/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/03/901603/2"}}},"commit":{"parents":[{"commit":"5555980fb5fce79140efc0c8ab8a730bd552733c","subject":"Merge \"Swap find_executable for which\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5555980fb5fce79140efc0c8ab8a730bd552733c"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-21 16:16:39.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-21 16:58:03.000000000","tz":540},"subject":"Remove per-service auto_create_account_prefix","message":"Remove per-service auto_create_account_prefix\n\nThe per-service option was deprecated almost 4 years ago[1].\n\n[1] 4601548dabdec0a4dc89cefba11e963217255be3\n\nChange-Id: I45f7678c9932afa038438ee841d1b262d53c9da8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/49b19613d29bda5717ca5eec8c9f6bc9049f99c0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/49b19613d29bda5717ca5eec8c9f6bc9049f99c0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
