)]}'
{"id":"openstack%2Fswift~901633","triplet_id":"openstack%2Fswift~master~Ibaeae082e0dcec1256177bc2d71d23990c647dca","project":"openstack/swift","branch":"master","topic":"p-proxy-namespaces","attention_set":{"1179":{"account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"last_update":"2023-11-21 23:36:56.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"Ibaeae082e0dcec1256177bc2d71d23990c647dca","subject":"tests: targeted tests for _get_updating_shard_ranges consumers","status":"NEW","created":"2023-11-21 22:25:23.000000000","updated":"2023-11-22 00:00:16.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"698f8e15752ebab541050d7df7b6e07b631a949d","_number":901633,"virtual_id_number":901633,"owner":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2023-11-21 23:36:56.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-11-21 23:36:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8fc280a97a8232ae1a218b0720309699b880365c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2023-11-21 22:25:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9d7fbe3005318090430753b0277ec08682c448ea","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2023-11-21 22:34:05.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"0e30aadeb0794f5d5839484be26e4a84c0c546bd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-21 23:36:56.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/55acffccffba4adc8c30a03adb8df644\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a07fc9e2dd644067ac1c46ca3f0cc033 : SUCCESS in 8m 46s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/2ce8377158dc446aa841778dcc56a4cf : SUCCESS in 59m 44s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/caa0a2984287444abb9574d1f0032bb7 : SUCCESS in 18m 26s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e276518ea2f947a39535de6bf1de53c4 : SUCCESS in 19m 25s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/434b03bdbc8b4797bc112a1cdd6e2883 : SUCCESS in 19m 05s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/54634cdf920540b88827d8f5ae28c690 : SUCCESS in 8m 18s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/2f385e2f38df4de295676560a23f696f : SUCCESS in 6m 42s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/fd4c93d184b249eabcc9665e71acd482 : FAILURE in 17m 57s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/f3a7ab49809040c08cc784d28e3a4702 : SUCCESS in 16m 39s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/8509cfe6cda749969b951212790c0ba0 : SUCCESS in 16m 20s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/c58ae783f0a64c8589aa5e1e9526966b : SUCCESS in 15m 27s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/93c043712ed74eb089343c31d87a1efb : SUCCESS in 15m 24s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/fdc51754272b42d6b6edf0fda6d1d41f : SUCCESS in 15m 12s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/645e95df813547d783ef1f01168ed6e8 : SUCCESS in 16m 22s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/970d0c585af3406493948b0c930adc2c : SUCCESS in 15m 39s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/d351fbd63331426a8a907eba444e44eb : SUCCESS in 19m 50s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/b37036be77ac48b7adf87781d1e73eec : SUCCESS in 21m 23s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/f6b6421069454fdda8abf16adade0d42 : SUCCESS in 25m 50s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/15e3f2e1f80a47deb9b97f3748055505 : SUCCESS in 51m 59s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/3c0a326e2e1b4daf906e0b716ebd40c6 : SUCCESS in 1h 02m 08s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3664f53ec2f74bc5a4a5b502834ede1b : FAILURE in 24m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a670e85a4693449d8353b46ac7ae48f6 : SUCCESS in 4m 41s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/ff9bcaa65310492582bb8031494d006b : SUCCESS in 23m 56s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"698f8e15752ebab541050d7df7b6e07b631a949d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-22 00:00:16.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ac6fe4f978484bc1bd9a627dabb12730\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ceeb817fa0f1470abb97d417932ca1f5 : SUCCESS in 16m 23s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ed57e441c44342739af5f71c74adf88f : SUCCESS in 16m 07s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/0e6b7108315e40029298c39a21fe9c18 : SUCCESS in 22m 26s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/a5659e35ff98445d8f470b0efce7f1a8 : SUCCESS in 1h 23m 38s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"4839793241ef6517d7b509396af14900275aed09","revisions":{"a7354cfe02186c8f3a42f8529002bb947e21dc77":{"kind":"REWORK","_number":1,"created":"2023-11-21 22:25:23.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/33/901633/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/33/901633/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/33/901633/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/33/901633/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/33/901633/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/33/901633/1"}}},"commit":{"parents":[{"commit":"6f890d2ba92a2b2a83db791e7af3e0aad661b01a","subject":"proxy: move _get_shard_ranges to ObjectController","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6f890d2ba92a2b2a83db791e7af3e0aad661b01a"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2023-11-21 22:19:42.000000000","tz":-360},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2023-11-21 22:20:47.000000000","tz":-360},"subject":"tests: targeted tests for _get_updating_shard_ranges consumers","message":"tests: targeted tests for _get_updating_shard_ranges consumers\n\nWe re-homed the interface the obj controller uses to fetch shard ranges\nin the related commit.\n\nThis interface is used either for includes\u003dobj lookups when shard range\ncachcing is disabled (non-default) or when getting the whole list of\nshard ranges for caching (default).  The existing tests also moved, but\ncovered mostly the non-default case, but created a new targeted test case\nthat we can expand to cover the primary consuming interface that hits\nboth paths into _get_updating_shard_ranges depending on the proxy\nconfiguration.  Some of the behaviors asserted in these new test methods\nare probably duplicated with other tests in test.unit.proxy.test_server\n\nRelated-Change-Id: I083e0c6898bf93d8a0dc593acd9723827e55508e\nChange-Id: Ibaeae082e0dcec1256177bc2d71d23990c647dca\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a7354cfe02186c8f3a42f8529002bb947e21dc77"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a7354cfe02186c8f3a42f8529002bb947e21dc77"}]},"branch":"refs/heads/master"},"4839793241ef6517d7b509396af14900275aed09":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2023-11-21 22:34:05.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/33/901633/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/33/901633/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/33/901633/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/33/901633/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/33/901633/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/33/901633/2"}}},"commit":{"parents":[{"commit":"6f890d2ba92a2b2a83db791e7af3e0aad661b01a","subject":"proxy: move _get_shard_ranges to ObjectController","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6f890d2ba92a2b2a83db791e7af3e0aad661b01a"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2023-11-21 22:19:42.000000000","tz":-360},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2023-11-21 22:33:40.000000000","tz":-360},"subject":"tests: targeted tests for _get_updating_shard_ranges consumers","message":"tests: targeted tests for _get_updating_shard_ranges consumers\n\nWe re-homed the interface the obj controller uses to fetch shard ranges\nin the related commit.\n\nThis interface is used either for includes\u003dobj lookups when shard range\ncachcing is disabled (non-default) or when getting the whole list of\nshard ranges for caching (default).  The existing tests also moved, but\ncovered mostly the non-default case, but created a new targeted test\ncase that we can expand to cover the primary consuming interface that\nhits both paths into _get_updating_shard_ranges depending on the proxy\nconfiguration.  Some of the behaviors asserted in these new test methods\nare probably duplicated with other tests in test.unit.proxy.test_server\n\nRelated-Change-Id: I083e0c6898bf93d8a0dc593acd9723827e55508e\nChange-Id: Ibaeae082e0dcec1256177bc2d71d23990c647dca\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4839793241ef6517d7b509396af14900275aed09"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4839793241ef6517d7b509396af14900275aed09"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
