)]}'
{"id":"openstack%2Fswift~902000","triplet_id":"openstack%2Fswift~master~Id2b3c6efeb8afec786b2fe9d07ab835626a21891","project":"openstack/swift","branch":"master","topic":"extend-slo-attrs-patch52","hashtags":[],"change_id":"Id2b3c6efeb8afec786b2fe9d07ab835626a21891","subject":"implement delete handling for part-num queries","status":"ABANDONED","created":"2023-11-27 20:13:29.000000000","updated":"2023-11-27 23:22:43.000000000","total_comment_count":0,"unresolved_comment_count":0,"work_in_progress":true,"has_review_started":false,"meta_rev_id":"dc6dd3073cf5b8d9772da7c914bb4af3c7680f22","_number":902000,"virtual_id_number":902000,"owner":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2023-11-27 22:01:49.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"reviewer_updates":[{"updated":"2023-11-27 22:01:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9fe31e99587c00bc9a73dc27319f124e1f199c87","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2023-11-27 20:13:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a215182c246559ac8a7b8c5a9da001094ed708fd","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2023-11-27 20:50:29.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"c0572d9b294735150a6740dd270fc2e80a84625b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-27 22:01:49.000000000","message":"Patch Set 2: Verified-1\n\n(3 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9780cc3823dd4691ae15f97ac0999bb2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8beeecff60054bee99ceec55a0866002 : SUCCESS in 8m 54s\n- grenade https://zuul.opendev.org/t/openstack/build/ab50a0d951b84d1d994fc723e7e40ebe : SUCCESS in 1h 03m 16s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/bad0d71bd824411e9c4b66deeba6d4c7 : SUCCESS in 46m 21s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/a006f3812e934e28a1e0e2537b27e75a : SUCCESS in 52m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/921b6afda3194a5a8033c05f9213cfef : SUCCESS in 1h 02m 34s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/28ee921adb724cc291aaebc86b52f418 : FAILURE in 14m 05s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/416597690d8a49dda01e8d5e286c5c94 : FAILURE in 19m 31s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7c3da08a10424941b3ebaba2c12d0558 : FAILURE in 19m 42s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/f03453eac5fd471bb942cdcecc2952e9 : SUCCESS in 9m 19s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/96b6a360f75e40dcb7d9381d08ba121e : SUCCESS in 6m 18s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/b823bff02c8342a099fdd52bbf988690 : FAILURE in 14m 01s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/3e15b7898d2848e5916845a3e01d9857 : FAILURE in 17m 48s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/f267de3cd8a841d996f394a8f6921ca0 : FAILURE in 16m 17s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/edde5d7ee67c407f976ac794dece83c8 : FAILURE in 15m 44s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/b41b836af07e459db7084f612baec675 : FAILURE in 13m 05s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/7627658683f0460e921acea2454f369d : FAILURE in 18m 14s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/f03af08dd78c4c7e8580d8b456279c08 : FAILURE in 18m 57s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/74c05d535f364e14bb20232c68c31822 : FAILURE in 18m 09s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/6ea7d5e54bcd4011b194ff41c82511e9 : FAILURE in 20m 42s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/c085143c408f4189b360c6b16ba2a11f : FAILURE in 17m 52s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/b92761bb15a34fde9275a8f00e46dd39 : FAILURE in 17m 09s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/3bdf9f7c837d4fb58f48c004789c7f9f : SUCCESS in 6m 26s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/9eb032d61ce7452cafe7e695cb20b1f7 : SUCCESS in 19m 15s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/705395f0a8d1470c8bee64d4cfcf3b03 : SUCCESS in 7m 29s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/3c589805f8934072854b4cba4f0d9e35 : SUCCESS in 58m 24s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/30e37ce9e4f4494ba0abf7d583166772 : SUCCESS in 59m 57s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/e23021822af54de69f6f08f43beeff60 : FAILURE in 41m 31s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/ebf4addbbe1543d4a75f1f15f855adaa : FAILURE in 27m 39s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2d3dbd8043ae413f992316b806a6b144 : FAILURE in 17m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f312aab0d8a241ada56ee961890ae7dc : FAILURE in 4m 22s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/2e30308644c1406cab3718138160fe53 : SUCCESS in 25m 14s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8a69ce73c0dd4ac6ad535648385977e6 : SUCCESS in 1h 03m 06s","accounts_in_message":[],"_revision_number":2},{"id":"853937c0ecf3cfaa6573f764e41ee664bf7c0b04","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-27 22:02:08.000000000","message":"Patch Set 2:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/swift for 894570,65","accounts_in_message":[],"_revision_number":2},{"id":"dc6dd3073cf5b8d9772da7c914bb4af3c7680f22","tag":"autogenerated:gerrit:abandon","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2023-11-27 23:22:43.000000000","message":"Abandoned\n\nsquashed into https://review.opendev.org/c/openstack/swift/+/894570","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"d14fee637918e9196c02e2306334f29cc125cc55","revisions":{"460fdfdfda04738f92cea62b0b0f707404c038b8":{"kind":"REWORK","_number":1,"created":"2023-11-27 20:13:29.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/00/902000/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/00/902000/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/00/902000/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/00/902000/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/00/902000/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/00/902000/1"}}},"commit":{"parents":[{"commit":"f975440442770a690052cb5033b5b601874a2e33","subject":"slo: part-number\u003dN query parameter support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f975440442770a690052cb5033b5b601874a2e33"}]}],"author":{"name":"indianwhocodes","email":"nairashwin952013@gmail.com","date":"2023-11-27 20:13:18.000000000","tz":-480},"committer":{"name":"indianwhocodes","email":"nairashwin952013@gmail.com","date":"2023-11-27 20:13:18.000000000","tz":-480},"subject":"implement delete handling for part-num queries","message":"implement delete handling for part-num queries\n\nChange-Id: Id2b3c6efeb8afec786b2fe9d07ab835626a21891\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/460fdfdfda04738f92cea62b0b0f707404c038b8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/460fdfdfda04738f92cea62b0b0f707404c038b8"}]},"branch":"refs/heads/master"},"d14fee637918e9196c02e2306334f29cc125cc55":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2023-11-27 20:50:29.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/00/902000/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/00/902000/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/00/902000/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/00/902000/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/00/902000/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/00/902000/2"}}},"commit":{"parents":[{"commit":"707e5f30714181e966fbbb8e2c53229e1d0aff38","subject":"slo: part-number\u003dN query parameter support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/707e5f30714181e966fbbb8e2c53229e1d0aff38"}]}],"author":{"name":"indianwhocodes","email":"nairashwin952013@gmail.com","date":"2023-11-27 20:13:18.000000000","tz":-480},"committer":{"name":"ASHWIN A NAIR","email":"nairashwin952013@gmail.com","date":"2023-11-27 20:50:29.000000000","tz":0},"subject":"implement delete handling for part-num queries","message":"implement delete handling for part-num queries\n\nChange-Id: Id2b3c6efeb8afec786b2fe9d07ab835626a21891\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d14fee637918e9196c02e2306334f29cc125cc55"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d14fee637918e9196c02e2306334f29cc125cc55"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
