)]}'
{"id":"openstack%2Fswift~902114","triplet_id":"openstack%2Fswift~master~Ifb6b2cad7e4db124d30991788efb07d3de29700d","project":"openstack/swift","branch":"master","topic":"extend-slo-attrs-patch52","attention_set":{},"removed_from_attention_set":{"1179":{"account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"last_update":"2024-03-04 20:04:45.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Ifb6b2cad7e4db124d30991788efb07d3de29700d","subject":"sq: more slo part-number unittests","status":"ABANDONED","created":"2023-11-28 22:00:25.000000000","updated":"2024-03-04 20:04:45.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"95439ccba966bb8a24779a7d3e588a5080cf6440","_number":902114,"virtual_id_number":902114,"owner":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},{"tag":"autogenerated:zuul:check","value":1,"date":"2023-11-28 23:24:10.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"all":[{"value":1,"date":"2023-11-30 20:30:31.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-11-28 22:00:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-11-30 20:30:31.000000000","updated_by":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"reviewer":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"state":"REVIEWER"}],"messages":[{"id":"dc02c04cdeecedae3741642e78e2804abb1fdebe","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2023-11-28 22:00:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c32d5711f7e3cda2c9588fae556f0363ebe2d123","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-28 22:00:44.000000000","message":"Patch Set 1: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/swift for 894570,70","accounts_in_message":[],"_revision_number":1},{"id":"e8bea1a0dcd5541336dc6470ab042bf5c7fe811c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-28 22:00:46.000000000","message":"Patch Set 1:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/swift for 894570,70","accounts_in_message":[],"_revision_number":1},{"id":"5971bb49852fb7fcd540ff522c8348cdd4f423ac","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2023-11-28 22:06:19.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"cc26bb7df947a696b52f792a5b0d11fb6882b081","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-28 23:24:10.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/388e39351e1c4d8480af7af63e553b3e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c86922ad232c40cf83e706a1430222a8 : SUCCESS in 9m 16s\n- grenade https://zuul.opendev.org/t/openstack/build/bd052e7a12bb46b6998a6b8bc001610e : SUCCESS in 1h 09m 29s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/34dc17a3ec4247b8b874b343173c8d01 : SUCCESS in 1h 03m 01s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/5fb5e19774924642a101ed86451e7926 : SUCCESS in 50m 54s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/74f437d14f6c4195a98314321d753b36 : SUCCESS in 48m 35s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b75de23d7fc74cb486f42af1ad32b71e : SUCCESS in 19m 15s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/71b5efeaf03e4e5e856a15fd6fcaba7d : SUCCESS in 16m 12s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/873b2f3da5834e038285841f028b928b : SUCCESS in 18m 42s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/36dff92aeb0f4b14b70937e30f6079c6 : SUCCESS in 8m 08s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/c3a5af8112b24a329346bea24fb16448 : SUCCESS in 5m 11s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/33df63193d40417d9c88e9d5044410d4 : SUCCESS in 18m 47s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/687969aadc52410683f1b926a8eed045 : SUCCESS in 16m 03s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/2e323c9d0a494bf68b66a79b8e102803 : SUCCESS in 13m 30s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/6e88e9f6074d4e5c94152bc5d8518f5c : SUCCESS in 16m 09s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/35f59007f8b84fc59a8a74c178734224 : SUCCESS in 16m 31s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/a45d5497ff734b72b4dc3656859ed0a0 : SUCCESS in 21m 45s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/0179979b35f945bb9861297e3a8f4524 : SUCCESS in 19m 10s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/dd3c3ad5a31b4692bc8009c4d9dc04ec : SUCCESS in 16m 42s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/6b8ea042eb8a485a980156c1dceb1919 : SUCCESS in 15m 38s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/ccd45747f8304e9b9b60ef73716a6f11 : SUCCESS in 16m 14s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/390ea4dc0a764d52ab51dc45a1a3d1ea : SUCCESS in 18m 30s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/90b47f298d81435fbedc5f02d5621f3f : SUCCESS in 7m 19s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/76c185632ef5431085b26898a81f3e08 : SUCCESS in 16m 07s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/a6bb626a924b46d184d01a3331053cca : SUCCESS in 8m 21s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/6b5e208e572a411f96f0690366a212a6 : SUCCESS in 56m 48s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/0d1f28060ac441bdb3d0eb3c4af0dcf7 : SUCCESS in 1h 07m 34s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/b0bd994dd9724a91beb43ec9eee62bbf : SUCCESS in 1h 05m 14s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/7e2ce51bffe649f4afad1305408c49ca : SUCCESS in 48m 17s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/374540d0b89242f58966ff82f657dbe2 : SUCCESS in 19m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dbef20e78a944d13b3f6f768496a6a92 : SUCCESS in 5m 05s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/34f660094e7840e0b784bb245eecbec1 : SUCCESS in 25m 50s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/12e0b0550fcd4ad69f03a9bb73ec1421 : SUCCESS in 57m 21s","accounts_in_message":[],"_revision_number":2},{"id":"d89c2b0c7603dad87162fdbb8330772fe0c60e41","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-28 23:39:48.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/96fc66bd3b3444c19f071534e4af920c\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/cc1745faab1c469aa740c8333c08c618 : SUCCESS in 1h 25m 53s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/57562192d93e4c0d9057d5abd8a62b64 : TIMED_OUT in 1h 30m 50s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/aaae432ece9f41ecbef606adf21b25a1 : SUCCESS in 22m 12s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/0898d420adeb4e53aa990594c461c861 : TIMED_OUT in 1h 30m 58s","accounts_in_message":[],"_revision_number":2},{"id":"788d735748152230e47f62b78f389efa67f8d0e0","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2023-11-30 20:30:31.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"95439ccba966bb8a24779a7d3e588a5080cf6440","tag":"autogenerated:gerrit:abandon","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-03-04 20:04:45.000000000","message":"Abandoned\n\nSquashed into parent","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"9f6b8aa1e27e3fde141fbfcff675ecf9f77abccd","revisions":{"176c88c12669d917b88c5c9ec6888b60e8d75a99":{"kind":"REWORK","_number":1,"created":"2023-11-28 22:00:25.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/14/902114/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/14/902114/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/14/902114/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/14/902114/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/14/902114/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/14/902114/1"}}},"commit":{"parents":[{"commit":"9f87d716ce45ce5f1a75054b4b52493bbda952be","subject":"slo: part-number\u003dN query parameter support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9f87d716ce45ce5f1a75054b4b52493bbda952be"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2023-11-28 21:58:37.000000000","tz":-360},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2023-11-28 21:58:56.000000000","tz":-360},"subject":"sq: more slo part-number unittests","message":"sq: more slo part-number unittests\n\nChange-Id: Ifb6b2cad7e4db124d30991788efb07d3de29700d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/176c88c12669d917b88c5c9ec6888b60e8d75a99"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/176c88c12669d917b88c5c9ec6888b60e8d75a99"}]},"branch":"refs/heads/master"},"9f6b8aa1e27e3fde141fbfcff675ecf9f77abccd":{"kind":"REWORK","_number":2,"created":"2023-11-28 22:06:19.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/14/902114/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/14/902114/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/14/902114/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/14/902114/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/14/902114/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/14/902114/2"}}},"commit":{"parents":[{"commit":"9c57584525b1d5f75ff2ebc98a2c7ba59568496a","subject":"slo: part-number\u003dN query parameter support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9c57584525b1d5f75ff2ebc98a2c7ba59568496a"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2023-11-28 21:58:37.000000000","tz":-360},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2023-11-28 22:05:26.000000000","tz":-360},"subject":"sq: more slo part-number unittests","message":"sq: more slo part-number unittests\n\nChange-Id: Ifb6b2cad7e4db124d30991788efb07d3de29700d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9f6b8aa1e27e3fde141fbfcff675ecf9f77abccd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9f6b8aa1e27e3fde141fbfcff675ecf9f77abccd"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
