)]}'
{"id":"openstack%2Fswift~902434","triplet_id":"openstack%2Fswift~master~If6b9c0e289f2039f8a93c07c23627adea1dd7522","project":"openstack/swift","branch":"master","topic":"fix-s3acl2","attention_set":{"1179":{"account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"last_update":"2023-12-01 19:10:53.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"If6b9c0e289f2039f8a93c07c23627adea1dd7522","subject":"wip: update headers in s3acl","status":"NEW","created":"2023-12-01 18:05:22.000000000","updated":"2023-12-01 19:10:53.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"c3e4492eaf242852adc54b2c5d7b7c1ccdbf7cfb","_number":902434,"virtual_id_number":902434,"owner":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2023-12-01 19:10:53.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-12-01 18:30:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-12-01 19:10:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"dfd7c2b0f23d5e12da251730901cf229972e6ade","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2023-12-01 18:05:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"44f8337fa78f9cf3f3b4f00e3d955b678075813c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-01 18:30:06.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/93115cf451d2403aacbca84478d6aa4a\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/84f8318c887048eeb30c35bf55a9d082 : FAILURE in 14m 28s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/77c1ae8024df4be881e7b143a3af5e69 : FAILURE in 16m 21s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/5005ff6552a54f99942f73ad33287956 : SUCCESS in 23m 38s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/2703d48752864605b9b874a9a2246b43 : SUCCESS in 22m 15s","accounts_in_message":[],"_revision_number":1},{"id":"c3e4492eaf242852adc54b2c5d7b7c1ccdbf7cfb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-01 19:10:53.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a7b1325f3bd5483f9047a86d6c78d67e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/570cc4d7dc154a139f9833c1c4f3505b : SUCCESS in 9m 39s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/4c483c64455b4a5a9881ad1cc6109563 : SUCCESS in 59m 33s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c865866da29144efb13df1eff28e39d8 : SUCCESS in 18m 30s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5592f4ec75d24a3997da39aac5d1499e : SUCCESS in 19m 05s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2e6d2c98361d49ae9096825dd3b34e83 : SUCCESS in 15m 05s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/95f9e83190d245a28c6bebd1d7db5467 : SUCCESS in 8m 05s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/64067572b69843c490b403fdb2e603a2 : SUCCESS in 6m 10s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/c5ceb0dba935496e98887bd3736e18db : FAILURE in 25m 02s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/9d4f54ca23674f548eddbfacddf34cca : FAILURE in 17m 22s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/0596308245a946478b755b65735a1220 : FAILURE in 17m 56s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/ef052d23bdd140b68df424c4f5479e60 : FAILURE in 17m 08s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/8a87f41d6eec46398081fd39db34c410 : FAILURE in 15m 15s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/2dd2d7b527324e1f9b19a5ad1f9038f0 : SUCCESS in 16m 52s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/73215a2947754e2e8bf73e5d17aeced9 : SUCCESS in 18m 04s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/05a87dd8bfaa40aea00ae77975a8e743 : SUCCESS in 19m 11s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/b81b842afe2d4dac90f91800b5f34a09 : SUCCESS in 15m 50s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/e58c1304aae6493c9a3b50030fa68de1 : SUCCESS in 16m 20s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/915edbc7723f46e9ae019f84253cab01 : SUCCESS in 15m 56s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/5c4d0eaffe1d404bb94055820f808e4d : SUCCESS in 54m 32s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/648498a9b20a4f08b053bd851ad56875 : SUCCESS in 42m 54s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c9b0d5afc17843ed943041d0b688457f : FAILURE in 17m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6bddd72b8e384fc9985dca62b85d2515 : SUCCESS in 5m 07s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/b76725663644476aace196a42148de76 : SUCCESS in 20m 01s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b61e55b95024974543a3fca475b20d57c4d17341","revisions":{"b61e55b95024974543a3fca475b20d57c4d17341":{"kind":"REWORK","_number":1,"created":"2023-12-01 18:05:22.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/34/902434/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/34/902434/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/34/902434/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/34/902434/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/34/902434/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/34/902434/1"}}},"commit":{"parents":[{"commit":"190b8576ef89094a42b0852a91adfe06f7805daa","subject":"Merge \"replication network aware helper for container-updater in proxy\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/190b8576ef89094a42b0852a91adfe06f7805daa"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2023-12-01 18:04:42.000000000","tz":-360},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2023-12-01 18:04:42.000000000","tz":-360},"subject":"wip: update headers in s3acl","message":"wip: update headers in s3acl\n\nChange-Id: If6b9c0e289f2039f8a93c07c23627adea1dd7522\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b61e55b95024974543a3fca475b20d57c4d17341"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b61e55b95024974543a3fca475b20d57c4d17341"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
