)]}'
{"id":"openstack%2Fswift~902435","triplet_id":"openstack%2Fswift~master~Iaea2b7d765a2cef229d9f1bbb9f299eec5276f3e","project":"openstack/swift","branch":"master","topic":"fix-s3acl","attention_set":{"1179":{"account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"last_update":"2023-12-01 19:20:08.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"Iaea2b7d765a2cef229d9f1bbb9f299eec5276f3e","subject":"wip: fix s3acl header setup","status":"NEW","created":"2023-12-01 18:05:34.000000000","updated":"2023-12-04 11:49:48.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":2,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"1a84c622fac71843aa3e1e6c2d5fb35a59552976","_number":902435,"virtual_id_number":902435,"owner":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2023-12-01 19:20:08.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-12-01 19:20:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-12-04 11:49:48.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"CC"}],"messages":[{"id":"4ed101c7805e8efb3b1b09bf722553d7d6d007bc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2023-12-01 18:05:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5883d79ddf4a58f26fd72ca3bca64a06f6d2a603","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-01 19:20:08.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/cf3e21f4a2334c78bddc13428bd51ebd\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4845c852f23a481fae4d34b121e87ae6 : SUCCESS in 9m 06s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/7c9b6e6a4d154bd5906d7d806358b2b6 : SUCCESS in 50m 53s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c993dd73b3a14af3b987ddaca90380f1 : SUCCESS in 19m 40s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/546ec15323c1421c9365f5672c09866e : SUCCESS in 19m 16s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3d0ddcd3f0d44cfe9dbd864dd9130b9f : SUCCESS in 24m 07s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/66a4cc2c24fb4dcfa62292fa4ec1e368 : SUCCESS in 8m 47s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/867ed018b0684386808287bd163e4f97 : SUCCESS in 5m 01s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/d949142a1c78471fa939b858535316c8 : FAILURE in 18m 18s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/1e0ae305014b4a12bab0bec7667752c2 : FAILURE in 13m 24s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/31007c1794164406a57fa7b7db73b839 : FAILURE in 14m 21s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/83ccc6f2f5ab4295afba0b8bc16100c6 : FAILURE in 16m 23s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/f8e523116103499db0af1b854d9fe790 : FAILURE in 15m 58s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/70dbab2458cf4164a9541459ace818ed : SUCCESS in 20m 11s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/bd72e0eb3dbf4113be025e59b8e14cd8 : SUCCESS in 17m 38s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/d7de6a840728437a97c9c6843b4d09c3 : SUCCESS in 17m 35s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/2eff7735fc104e1d8ff8304d4762343d : SUCCESS in 24m 25s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/83513c1458b94014aa95272529978ff1 : SUCCESS in 17m 27s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/0ba11c1e96734789a19d61df09c9e2b6 : SUCCESS in 28m 22s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/f7b54683f84f43a18d320e34e0b402ce : SUCCESS in 1h 08m 12s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/1e562382deb04fb5a4ea664d0f5c9564 : SUCCESS in 47m 55s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0600f680c34048f3a5064596571131a4 : FAILURE in 19m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c664fc11dd0f4b988325d94612a8af61 : SUCCESS in 4m 57s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/14079f87b4fe46b888b67d37f548de6b : SUCCESS in 20m 38s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e7a2ce7265f9c862c26b0ad567639c25f462e1d7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-01 19:38:41.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4fa056fa7ca5430b98428f5a7f1475a8\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/5d42abdb3218438b87a4a57d2a1c7d2c : FAILURE in 1h 05m 40s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7d9ca2984c124082a6417dc03eb72e0d : FAILURE in 14m 21s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/f185e298089748c18ab84db4b6b82f92 : TIMED_OUT in 1h 30m 54s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/d95fa86d176f4171b2f072de223aa122 : SUCCESS in 20m 15s","accounts_in_message":[],"_revision_number":1},{"id":"1a84c622fac71843aa3e1e6c2d5fb35a59552976","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-12-04 11:49:48.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"18b7f09a66703203e1f9c7612f20198944772620","revisions":{"18b7f09a66703203e1f9c7612f20198944772620":{"kind":"REWORK","_number":1,"created":"2023-12-01 18:05:34.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/35/902435/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/35/902435/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/902435/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/902435/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/902435/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/35/902435/1"}}},"commit":{"parents":[{"commit":"190b8576ef89094a42b0852a91adfe06f7805daa","subject":"Merge \"replication network aware helper for container-updater in proxy\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/190b8576ef89094a42b0852a91adfe06f7805daa"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2023-11-30 18:33:31.000000000","tz":-360},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2023-12-01 16:27:20.000000000","tz":-360},"subject":"wip: fix s3acl header setup","message":"wip: fix s3acl header setup\n\nChange-Id: Iaea2b7d765a2cef229d9f1bbb9f299eec5276f3e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/18b7f09a66703203e1f9c7612f20198944772620"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/18b7f09a66703203e1f9c7612f20198944772620"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
