)]}'
{"id":"openstack%2Fswift~902724","triplet_id":"openstack%2Fswift~master~I0197795a93d791cc230cf61f695dbec48dda4836","project":"openstack/swift","branch":"master","topic":"p-get-path-fixes","attention_set":{},"removed_from_attention_set":{"1179":{"account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"last_update":"2024-01-25 14:18:27.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I0197795a93d791cc230cf61f695dbec48dda4836","subject":"tests: add more assertions on requests","status":"ABANDONED","created":"2023-12-05 20:03:52.000000000","updated":"2024-01-25 14:18:27.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"8b66cb5a73ac126807e05af35e44e33f0e01fa1d","_number":902724,"virtual_id_number":902724,"owner":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2023-12-05 20:55:30.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-12-05 20:55:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-01-25 14:18:20.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"CC"}],"messages":[{"id":"667136b4b0f703faf2408847fc7c7292a18f70be","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2023-12-05 20:03:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"48eb1196eb603dfeb80c2197bd28094b63d1fa97","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-05 20:55:30.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2226935885b548d4832edb0c5320397c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9db1088fa7824212b7b8d44371513b33 : SUCCESS in 8m 27s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/c7fcd7b2599e4fed8e4b35139a92b082 : SUCCESS in 46m 04s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/76926cdab33340299fd44ad5270a2bdd : SUCCESS in 14m 48s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/aae8048e84634a6a8949fcfa6e38a4cc : SUCCESS in 15m 36s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5ba3448d0c594da78e7af2b0e2e0be5e : SUCCESS in 17m 19s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/e6d8f859465648aabc1ffb2a2775c6ef : SUCCESS in 8m 03s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/4130251377564183bcb1ad3b84362ba9 : SUCCESS in 7m 52s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/8d8650c0912f40cf9b667a0b7bdbf1a2 : SUCCESS in 17m 35s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/ad2d501af9824751a384a3f18aaa7f08 : SUCCESS in 15m 39s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/9159cea1e4564ba58482767ff9cf2ae7 : SUCCESS in 15m 02s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/69793af0fa8342fb8810b67acc4cd6fc : SUCCESS in 16m 23s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/7b49a970a30143a9bf0319245ed6568c : SUCCESS in 21m 14s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/a541c438c0d24f579bd9720c176f66ed : SUCCESS in 16m 52s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/7f6378178f454b3c9e5fb7b226f83ad5 : SUCCESS in 18m 13s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/b716f91e20a240be8fadea347cab21af : SUCCESS in 18m 41s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/fe6d89c0f9e44ec1b25d182032cec944 : SUCCESS in 15m 31s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/ef59c7bc882c460c82ed789e84b57520 : SUCCESS in 15m 40s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/11e07af518cf4481b6aa0d79700090ca : SUCCESS in 19m 41s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/68011278a6704159aebdcb5028f6cbb4 : SUCCESS in 49m 27s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/cdfd21bcae0d4933aa4746769c1c8438 : SUCCESS in 42m 57s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/13336f0b7cf94e439c66697b6d46220e : SUCCESS in 19m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d5dee62f1fc4419ca54c50d3bfc6bf2f : SUCCESS in 5m 45s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/eeea888b0c2843dbb6600064b2e716e1 : SUCCESS in 18m 38s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"fe058d23a5cfaf0ac33c6884db6bb673e9c1c3dd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-05 21:09:42.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/029307a094ff48e28b9a1a0a76f4a0b5\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b01f08c9af2042569262c507222ac67a : SUCCESS in 15m 04s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/fbc8a87d0eab46f580ccb65ee1840def : SUCCESS in 1h 03m 21s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/ace0b2ae8e1a4e389a583a9a19666faf : SUCCESS in 50m 53s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/4a0db17e344242a7a2d475e1d1b3cb2d : SUCCESS in 17m 32s","accounts_in_message":[],"_revision_number":1},{"id":"9093c7e2cc77851882a305c1c3459ea88903a9db","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-01-25 14:18:20.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8b66cb5a73ac126807e05af35e44e33f0e01fa1d","tag":"autogenerated:gerrit:abandon","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-01-25 14:18:27.000000000","message":"Abandoned\n\nsquashed this patch into the parent","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5010b373551f29154a09bd4a37548fe29356d414","revisions":{"5010b373551f29154a09bd4a37548fe29356d414":{"kind":"REWORK","_number":1,"created":"2023-12-05 20:03:52.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/24/902724/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/24/902724/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/24/902724/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/24/902724/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/24/902724/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/24/902724/1"}}},"commit":{"parents":[{"commit":"a00132523c175e7a57f168dbfb98342f42d00783","subject":"Improve unit tests for proxy GET ChunkReadTimeouts","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a00132523c175e7a57f168dbfb98342f42d00783"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2023-12-05 19:52:43.000000000","tz":-360},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2023-12-05 20:02:25.000000000","tz":-360},"subject":"tests: add more assertions on requests","message":"tests: add more assertions on requests\n\nWe updated some timeout tests and I needed to convince myself we\u0027re\nstubbing the responses correctly for the requests being made.\n\nDrive-by: remove some ring device manipulation setup that\u0027s not needed\n\nRelated-Change-Id: I38c7fa648492c9bd2173ecf92f89e423bee4abf3\nChange-Id: I0197795a93d791cc230cf61f695dbec48dda4836\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5010b373551f29154a09bd4a37548fe29356d414"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5010b373551f29154a09bd4a37548fe29356d414"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
