)]}'
{"id":"openstack%2Fswift~902726","triplet_id":"openstack%2Fswift~master~Ia9eeae6ed231a9be504967208efa3a0124f8cf09","project":"openstack/swift","branch":"master","topic":"p-get-path-fixes","attention_set":{},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2023-12-06 15:27:08.000000000","reason":"\u003cGERRIT_ACCOUNT_7847\u003e replied on the change","reason_account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}},"1179":{"account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"last_update":"2023-12-06 17:50:35.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ia9eeae6ed231a9be504967208efa3a0124f8cf09","subject":"test: use debug_logger instead of Mock logger","status":"MERGED","created":"2023-12-05 20:33:18.000000000","updated":"2023-12-06 17:51:59.000000000","submitted":"2023-12-06 17:50:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"902726-p-get-path-fixes","meta_rev_id":"b97d082570c4e56f9adc9d27633ea540f6f9aac2","_number":902726,"virtual_id_number":902726,"owner":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-12-06 17:50:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":2,"date":"2023-12-06 10:36:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":1,"date":"2023-12-06 15:27:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-12-05 21:27:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-12-05 21:56:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-12-06 10:35:04.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"CC"},{"updated":"2023-12-06 10:36:11.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"REVIEWER"}],"messages":[{"id":"09d492835927c9a9d493f448e62ebcdd19e66dcf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2023-12-05 20:33:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7f7d4beec865032f0745816161be7a3bd26309d0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-05 21:27:03.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cb1d822d5591403697d710b28472b908\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/2b28d276b8ce464ab9e14b952960c2cf : FAILURE in 14m 55s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9fbd71dccd284eec8c94f247771e288c : FAILURE in 15m 57s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/a381ff71a2794c088e17842df6f6e69a : SUCCESS in 22m 50s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/2949b753d9ca4035b2b10e560329a772 : SUCCESS in 51m 47s","accounts_in_message":[],"_revision_number":1},{"id":"2d2c1f89a022775dfecc1692541f1eb82801c1f0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-05 21:56:58.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0bfd3c70061a4f87924a0db55aa4cb29\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/227af2c540fa4c70a232808e68d870a5 : SUCCESS in 7m 12s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/2413186efb984f08bd88dd1f4dc0c828 : SUCCESS in 1h 14m 40s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/faded538c9cd4985b08f4bda6306ff9c : SUCCESS in 19m 45s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9ec7538ea3eb466abcca273fe5fb5161 : SUCCESS in 19m 12s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/634d0befa0384286877a1a97f0eacf08 : SUCCESS in 17m 34s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/c53e6e076f234a199dcfedfcf1296fa6 : SUCCESS in 8m 46s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/6540d615341e42418663043e2626fccd : SUCCESS in 5m 41s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/949b6d403a674276842cc3f5dfcd10fa : FAILURE in 15m 02s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/a0b57d69aff04560873d72de8b9afcef : FAILURE in 13m 36s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/bc2390c0e8b54c0ca07fd1719d8c6922 : FAILURE in 13m 32s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/69102f4bef22491db3579ec005652ebf : FAILURE in 13m 03s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/fc7c47c69c064c08a1a411dd8154add7 : FAILURE in 13m 05s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/6f9537c20ecf4e3f9edaf13c198ba033 : SUCCESS in 18m 00s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/da40d4fab73b432cae2a000bb55f735d : SUCCESS in 18m 05s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/856e5ed1eda84d8c9d5f66bff4642872 : SUCCESS in 17m 43s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/5b67f7813b8c44e2bfc5c8eb61f267c8 : SUCCESS in 15m 27s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/0b9f6777ac4f4b80a4bdfbf0ef447810 : SUCCESS in 15m 52s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/414f56d2d51849ec8c281a99813ab9b1 : SUCCESS in 17m 07s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/2a1c750f2abc48a58978e21c4e011891 : SUCCESS in 38m 04s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/d99b041e9e934d37a6fb1c444a8ea517 : SUCCESS in 56m 00s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6f61afa64e2942d5be87b57a02ff0f1f : FAILURE in 16m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/44ecbf96ca29494990d9c954908ff7d1 : SUCCESS in 4m 14s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/e0fc1e82e44e4e37b8c6af84b913c72e : SUCCESS in 22m 01s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"afc6b0f00727c764f90ab14040d94e55eedb18c8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-12-06 10:31:31.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"31c33b97eef235bf034f5568e297c3dd1eebcbc6","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-12-06 10:35:04.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"69a2e97206397ba020d11ae563997c362e8d50fb","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-12-06 10:36:11.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"34b7511f32f4193450c162f838751b5ae913db96","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-06 11:24:45.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/020091473d704e7fa2024688b3d6ff5f\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/eb65128099fd471a9256d62e235929cb : SUCCESS in 16m 40s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/81bb7b18e1914d4b92b0170f4e6d0b4a : SUCCESS in 20m 24s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/6a50dfe48b994196a031b76c4c93c83e : SUCCESS in 19m 41s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/1c773c52ee37495185b16050bbd8708e : SUCCESS in 50m 45s","accounts_in_message":[],"_revision_number":2},{"id":"59a2c3f5d0743fe516b9a4d2ee75dfd9df875db4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-06 11:37:37.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/db4859b33fcc4a09b3ab3f8fe6668b1e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d873ac79cfd248e5979b5d7c9ffb8884 : SUCCESS in 8m 44s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/19f3dd0ebf7e482faab77e6a9fbcd693 : SUCCESS in 1h 00m 47s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/af853e12a04948748b9acf66817b6c2c : SUCCESS in 21m 02s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/37ef037f35264d10b91663b7662d2e84 : SUCCESS in 21m 13s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/25f601a237dd402f814c79857ee5ae54 : SUCCESS in 18m 48s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/ef918e003e394bb687113eb816a2ffd2 : SUCCESS in 8m 40s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/4b8ae09702c24573b794beecc493dac9 : SUCCESS in 5m 30s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/fb5f7c66c5af497783bb488757af991d : SUCCESS in 18m 00s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/29e63ab4a75e48b88f7b7ac3c349dc55 : SUCCESS in 13m 39s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/7f666c0b5e7d4c8d946e017d7d73a1fb : SUCCESS in 18m 31s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/099c74cc2820422dad715073a95fb82d : SUCCESS in 16m 35s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/d676249c76914f458cf0d36182b93b2c : SUCCESS in 16m 59s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/1ecfc8d0a39b4d55ac96d2a2bd388e53 : SUCCESS in 15m 46s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/8328c8fd9e934649a1fe77229a0a3789 : SUCCESS in 18m 16s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/a056b9acddb54e54a6278b757f62578a : SUCCESS in 21m 03s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/65fade6df6f046569a4174c0e718dc14 : SUCCESS in 13m 44s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/c930e6c6157f4473820eba6bd134c9b6 : SUCCESS in 16m 04s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/eba03c860e534b88b03e0e37ba1fd73f : SUCCESS in 19m 20s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/8c4eb902544f4172a55015a0f68ebedc : SUCCESS in 55m 55s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/0d02ddb170d944f1b9cd4850cab62f61 : SUCCESS in 56m 57s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/75ca4c4f72d747179177f66f0695e982 : SUCCESS in 19m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/04f2c0c5526940feb35e31ee24ce0c5a : SUCCESS in 4m 42s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/3dc2fcaf8df54b2ea2fd960554f4db85 : SUCCESS in 21m 25s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"7388628a704e41f70fe2c84d72a94107a9af8c4f","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2023-12-06 15:02:02.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"c7b009cc2c8b878d87321d38ca624692413f45f8","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-12-06 15:27:08.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"453a7cb953345d31ea625a756d4f2b26cbd61911","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-06 15:27:42.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"f408e9226e5aeeb02988bb31c4db84ecf7c454fa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-06 17:50:35.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/94295a5a3da34fffb92597bcc87453d4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1af9a0caf2554f7eab3f512c518cff1c : SUCCESS in 8m 00s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/cf0f5260aaed459bb89379384cf7c1a6 : SUCCESS in 46m 38s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/ce3fd05850a349978d8c74d1ced93e12 : SUCCESS in 9m 47s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/b4b9519ac3b3494ca32a064c9bce0108 : SUCCESS in 6m 00s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/f626c44d772741ac822caf48ac73e010 : SUCCESS in 15m 22s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/27c1cf7ae8124464a5d4f9f2aa4cc6bf : SUCCESS in 13m 13s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/cabc389494c149c082bc94401ffde265 : SUCCESS in 21m 16s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/58bcf327aa514530bd96b12a8c474015 : SUCCESS in 21m 55s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/b36f30ddcf4f4080b0973c2808d765cc : SUCCESS in 13m 06s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/a926a6824188419388eb970d96942c8a : SUCCESS in 21m 47s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/765f254e090149f1acc7d8c817775791 : SUCCESS in 18m 38s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/885c201c9ab24390ba125b2f6d8b88af : SUCCESS in 20m 13s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/e5b0ac0eb6374c51a9c6df5fd27902b6 : SUCCESS in 21m 37s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/3fcbc291af4740769437227a401343e4 : SUCCESS in 16m 32s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/85e47dc94f31485c84c002e77253408c : SUCCESS in 14m 34s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/d1fe468921fe49c1ad889824c80c9ed1 : SUCCESS in 7m 54s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/d62c163ca9874327871a1119b7757c30 : SUCCESS in 8m 11s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/00a79698dbed43d3b03d5dd17b830cb4 : SUCCESS in 53m 34s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/e26f1830de1a4eb4871a61a513f5ab1d : SUCCESS in 1h 04m 04s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3e7312b89f0e46c08649a395656b28a1 : SUCCESS in 14m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/211e5d4a126144bd94785b808e9a11b8 : SUCCESS in 4m 53s","accounts_in_message":[],"_revision_number":2},{"id":"af9953aff824c8956347ab88dcc8837bbd81b85c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-06 17:50:35.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"b97d082570c4e56f9adc9d27633ea540f6f9aac2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-06 17:51:59.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6c5ea720f1864e8a85fd934d43c91559\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a3840984c69b419cbbf23f39d220303b : SUCCESS in 52s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/b9997b73dd4c4814abd1ea136872711f : SUCCESS in 35s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/b2a043bc42e74982a92fa21cd013c045 : SUCCESS in 34s (non-voting)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"7dadc5cac1e47605def54c78ebb10744e0239620","revisions":{"a3e983482c769e7eda6e050dd9dd32f7c1e7b7c7":{"kind":"REWORK","_number":1,"created":"2023-12-05 20:33:18.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/26/902726/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/26/902726/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/26/902726/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/26/902726/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/26/902726/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/26/902726/1"}}},"commit":{"parents":[{"commit":"36ce42f54e48cdbf6edfc63dcb7a20a08b4e89a4","subject":"Relocate GetOrHeadHandler unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/36ce42f54e48cdbf6edfc63dcb7a20a08b4e89a4"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2023-12-05 20:30:40.000000000","tz":-360},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2023-12-05 20:32:57.000000000","tz":-360},"subject":"test: use debug_logger instead of Mock logger","message":"test: use debug_logger instead of Mock logger\n\nModern tests mostly consistently use a debug_logger when available and I\nfind the assertions a little more intuative to read than Mock\u0027s\nassert_calls helpers.\n\nDrive-by: remove un-needed patch_policies on TestCase\n\nRelated-Change-Id: I13a24c4a0cd4045ab2fffb18067674b3e3bf742b\nChange-Id: Ia9eeae6ed231a9be504967208efa3a0124f8cf09\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a3e983482c769e7eda6e050dd9dd32f7c1e7b7c7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a3e983482c769e7eda6e050dd9dd32f7c1e7b7c7"}]},"branch":"refs/heads/master"},"7dadc5cac1e47605def54c78ebb10744e0239620":{"kind":"REWORK","_number":2,"created":"2023-12-06 10:31:31.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/26/902726/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/26/902726/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/26/902726/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/26/902726/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/26/902726/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/26/902726/2"}}},"commit":{"parents":[{"commit":"36ce42f54e48cdbf6edfc63dcb7a20a08b4e89a4","subject":"Relocate GetOrHeadHandler unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/36ce42f54e48cdbf6edfc63dcb7a20a08b4e89a4"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2023-12-05 20:30:40.000000000","tz":-360},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-12-06 10:31:17.000000000","tz":0},"subject":"test: use debug_logger instead of Mock logger","message":"test: use debug_logger instead of Mock logger\n\nModern tests mostly consistently use a debug_logger when available and I\nfind the assertions a little more intuative to read than Mock\u0027s\nassert_calls helpers.\n\nRelated-Change-Id: I13a24c4a0cd4045ab2fffb18067674b3e3bf742b\nChange-Id: Ia9eeae6ed231a9be504967208efa3a0124f8cf09\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7dadc5cac1e47605def54c78ebb10744e0239620"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7dadc5cac1e47605def54c78ebb10744e0239620"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
