)]}'
{"id":"openstack%2Fswift~903101","triplet_id":"openstack%2Fswift~master~I62f97d814e1e226dcd88f2c04a19041afa58379a","project":"openstack/swift","branch":"master","topic":"bugfix/2045912","attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2023-12-11 03:31:13.000000000","reason":"\u003cGERRIT_ACCOUNT_36602\u003e replied on the change","reason_account":{"_account_id":36602,"name":"chenmingjian","display_name":"chenmingjian","email":"starvingnow@live.cn","username":"chenmingjian"}},"1179":{"account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"last_update":"2023-12-09 07:05:04.000000000","reason":"Change was marked ready for review"},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2023-12-09 07:05:04.000000000","reason":"Change was marked ready for review"}},"removed_from_attention_set":{"36602":{"account":{"_account_id":36602,"name":"chenmingjian","display_name":"chenmingjian","email":"starvingnow@live.cn","username":"chenmingjian"},"last_update":"2023-12-11 03:31:13.000000000","reason":"\u003cGERRIT_ACCOUNT_36602\u003e replied on the change","reason_account":{"_account_id":36602,"name":"chenmingjian","display_name":"chenmingjian","email":"starvingnow@live.cn","username":"chenmingjian"}}},"hashtags":[],"change_id":"I62f97d814e1e226dcd88f2c04a19041afa58379a","subject":"s3api: When no Delimiter return NextMarker","status":"NEW","created":"2023-12-07 16:12:34.000000000","updated":"2023-12-15 05:49:29.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"f60f2ea7d324b1977930d810325c5f515e2ad755","_number":903101,"virtual_id_number":903101,"owner":{"_account_id":36602,"name":"chenmingjian","display_name":"chenmingjian","email":"starvingnow@live.cn","username":"chenmingjian"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":36602,"name":"chenmingjian","display_name":"chenmingjian","email":"starvingnow@live.cn","username":"chenmingjian"},{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:check","value":1,"date":"2023-12-09 08:25:07.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":36602,"name":"chenmingjian","display_name":"chenmingjian","email":"starvingnow@live.cn","username":"chenmingjian"},"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":1,"date":"2023-12-09 06:59:02.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":36602,"name":"chenmingjian","display_name":"chenmingjian","email":"starvingnow@live.cn","username":"chenmingjian"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":36602,"name":"chenmingjian","display_name":"chenmingjian","email":"starvingnow@live.cn","username":"chenmingjian"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":36602,"name":"chenmingjian","display_name":"chenmingjian","email":"starvingnow@live.cn","username":"chenmingjian"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-12-07 16:15:03.000000000","updated_by":{"_account_id":36602,"name":"chenmingjian","display_name":"chenmingjian","email":"starvingnow@live.cn","username":"chenmingjian"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"},{"updated":"2023-12-07 16:24:06.000000000","updated_by":{"_account_id":36602,"name":"chenmingjian","display_name":"chenmingjian","email":"starvingnow@live.cn","username":"chenmingjian"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"},{"updated":"2023-12-07 17:55:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-12-07 18:10:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-12-11 03:31:13.000000000","updated_by":{"_account_id":36602,"name":"chenmingjian","display_name":"chenmingjian","email":"starvingnow@live.cn","username":"chenmingjian"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"REVIEWER"}],"messages":[{"id":"d9e858f84f12ab61e30b5c8221f96ea19e587556","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36602,"name":"chenmingjian","display_name":"chenmingjian","email":"starvingnow@live.cn","username":"chenmingjian"},"date":"2023-12-07 16:12:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"84a366180df4220d089251987dcf1fc5ddcce3a5","author":{"_account_id":36602,"name":"chenmingjian","display_name":"chenmingjian","email":"starvingnow@live.cn","username":"chenmingjian"},"date":"2023-12-07 16:13:16.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"56cca40fce2cf0354c6973192cbd7fdedda6cde9","tag":"autogenerated:gerrit:deleteVote","author":{"_account_id":36602,"name":"chenmingjian","display_name":"chenmingjian","email":"starvingnow@live.cn","username":"chenmingjian"},"date":"2023-12-07 16:13:22.000000000","message":"Removed Code-Review+1 by \u003cGERRIT_ACCOUNT_36602\u003e\n","accounts_in_message":[{"_account_id":36602,"name":"chenmingjian","display_name":"chenmingjian","email":"starvingnow@live.cn","username":"chenmingjian"}],"_revision_number":1},{"id":"dde1d0a02296f227716db6ca0e889748fa06d88b","author":{"_account_id":36602,"name":"chenmingjian","display_name":"chenmingjian","email":"starvingnow@live.cn","username":"chenmingjian"},"date":"2023-12-07 16:15:03.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"88665dfb4e24c04f77504ed2e688dc850567ce38","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-07 17:55:53.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/06e64692dc0a4f6f80bd55ca259e65b0\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1da3acae445942daac33041df6ff69ea : SUCCESS in 17m 13s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/75067a24057344d89b01de9141456448 : SUCCESS in 17m 22s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/21e5dc549bbb46308dab7106ee8f1dcd : SUCCESS in 22m 35s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/993da97b32c2483c8dcea41a1bd2bba5 : SUCCESS in 1h 20m 36s","accounts_in_message":[],"_revision_number":1},{"id":"cdc43ffc9f776133f1acce7f78af1903438ca3cd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-07 18:10:21.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e5313cd8d83b4e24a84d96c60b1d615a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8d643e3fe53641198fd91aa8929f7e17 : SUCCESS in 8m 38s\n- grenade https://zuul.opendev.org/t/openstack/build/81bc8ed4a06645e3b8aadc83802f0fea : SUCCESS in 49m 13s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/fd4ce461cfcb43aaafb1c81b060e3bf9 : SUCCESS in 48m 58s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/2b51c62b994545c48339788ad8dad634 : SUCCESS in 42m 01s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e26e4091f3c24f1da2cd724a465e4f6f : SUCCESS in 52m 20s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/13387b855fb64d47aa56f05476d1264a : SUCCESS in 26m 10s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c03a7c24a46f41bfaf867517e62839c2 : SUCCESS in 26m 36s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/150a23611323450cb10604a40fc6c7d6 : SUCCESS in 27m 48s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/2474a6a74f444c25a745c7f467415b63 : SUCCESS in 8m 52s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/cace2f4f5e4a418795ceca7264e9a9a7 : SUCCESS in 7m 00s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/f83aaed4c3404f5b960344c198ceaf30 : SUCCESS in 17m 01s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/86c4b16a88154f83821c30a30feef061 : SUCCESS in 14m 24s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/af8aab8ebbbf40b09e6ee1526eab4fdb : SUCCESS in 14m 21s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/b29c12a410894983a4c233d2f69d9cfd : SUCCESS in 13m 09s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/6210d3ab5b9a44fa9f143088b258d5bb : SUCCESS in 15m 38s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/1ce5cf87c25b4730ab06b3bb0b90620a : SUCCESS in 17m 47s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/9d2f542b94354757b1bffc2f2a45c293 : SUCCESS in 18m 16s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/a4d0ecbced0941ffa18086ed4517d3ec : SUCCESS in 20m 00s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/703f68dd875641faae5721237c35fb43 : SUCCESS in 13m 42s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/aa57c12dd1654fb08acabacca559faea : SUCCESS in 17m 45s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/6bbedcead6dc473ead3a06f8beb3fd80 : SUCCESS in 15m 36s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/dc1e68992a04490c895f0edc4a74f16e : SUCCESS in 7m 24s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/0e4d5633a25e428c9dd9aa9cccd246ee : SUCCESS in 17m 53s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/448cea51a87c47c1ab1557198ebf8d56 : SUCCESS in 8m 08s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/2194eee252f846ef9965994b58f1514e : SUCCESS in 1h 00m 34s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/5406d412878c465fb7ce73a4bbd46648 : SUCCESS in 1h 03m 19s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/44601d87e76f4f28a69dae46bcbf75ec : SUCCESS in 50m 01s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/538b145a2935441080f524f5c6274875 : SUCCESS in 59m 15s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/13fad1eb2554452996a497bcbe81c3eb : SUCCESS in 18m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8fd75f43945642159d5e523f550d40ff : SUCCESS in 4m 47s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/60df6b9a962d4a378416a46e3e78591d : SUCCESS in 20m 45s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5ff4979c4c0c48b8af2c4306d0a506f8 : SUCCESS in 47m 38s","accounts_in_message":[],"_revision_number":1},{"id":"0b8d79f7c714eaf0491f21234f4d4cd234e0d564","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36602,"name":"chenmingjian","display_name":"chenmingjian","email":"starvingnow@live.cn","username":"chenmingjian"},"date":"2023-12-09 06:56:36.000000000","message":"Patch Set 2: Commit message was updated.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"2b5121f28d6753f41cf4100a0d9c7543e2bfdc3d","author":{"_account_id":36602,"name":"chenmingjian","display_name":"chenmingjian","email":"starvingnow@live.cn","username":"chenmingjian"},"date":"2023-12-09 06:59:02.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"d5667bae5bff9dec2611ad7adb80447814b5de71","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":36602,"name":"chenmingjian","display_name":"chenmingjian","email":"starvingnow@live.cn","username":"chenmingjian"},"date":"2023-12-09 07:04:58.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":2},{"id":"84cf2ed1261afcd2ab9295d4565d1f3fa1a2f5c8","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":36602,"name":"chenmingjian","display_name":"chenmingjian","email":"starvingnow@live.cn","username":"chenmingjian"},"date":"2023-12-09 07:05:04.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":2},{"id":"3350d7556e9f2f156becf7b0f6062e7e8cf02709","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-09 08:06:54.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d7d93597f1754262a91a77437aa29bb7\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/87ee6e122ae54ba5b11be12987456cbb : SUCCESS in 17m 14s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/981444b7e07d48b89cc36a8db1c3131c : SUCCESS in 1h 07m 52s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/eb56f8a299a9492a86a68fc856c907de : SUCCESS in 24m 29s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/c32d39598ccb42be9ee6df9de118325a : SUCCESS in 57m 49s","accounts_in_message":[],"_revision_number":2},{"id":"dd1f541a744796504196c3aab2ead551e687097a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-09 08:25:07.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b677dc0039e3484ca5fec5813590b57a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/737173b0d3ab4dafbffaa988ec7de67c : SUCCESS in 8m 18s\n- grenade https://zuul.opendev.org/t/openstack/build/9b3ebd4f2e3549ad903fa28e9c05590d : SUCCESS in 59m 48s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/d511e0480f454472bd6cfe2cd8a641bd : SUCCESS in 43m 28s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/fb7c25838c1d48db8a66a44ce2eee34f : SUCCESS in 49m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/18ecf653dc4847c7abd5a077805f2fb1 : SUCCESS in 51m 09s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0756a7946c914b4589b7424776724707 : SUCCESS in 19m 07s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/fd94917baaad4d45870c242939eeb2e4 : SUCCESS in 25m 46s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b5b9018cf71248918f6e86465853eccb : SUCCESS in 19m 07s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/9056287dddc847cc929e108db43024be : SUCCESS in 8m 51s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/7c8969eff9fd4d34b42929d700d02b43 : SUCCESS in 6m 15s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/f63820dde4b74e658435f8672f1c0e1c : SUCCESS in 16m 53s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/d46da7650c474684aead5e628ab8632f : SUCCESS in 21m 58s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/231369eab08442ce8dd0f7467b11278e : SUCCESS in 17m 50s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/84158e72a9024a2ba533fd789d719120 : SUCCESS in 22m 27s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/32dc1fd285c5469db0af099ed1e2f74e : SUCCESS in 17m 07s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/362a35c7d5f54899bcf28aebc46a5572 : SUCCESS in 19m 03s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/78e0961e95a4489e9b774a34e3a36b83 : SUCCESS in 20m 16s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/f4d1a7733fdc4222b1d9209e71bea07f : SUCCESS in 18m 09s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/054eaea73f9f42b5803acc33ebd37a1e : SUCCESS in 16m 47s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/7cdbf2cd720143a29a5af0b3478f9340 : SUCCESS in 17m 27s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/0c0181ebedce45f7b6ee89ffe021bf9f : SUCCESS in 24m 48s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/de17e3b17e9249bda1f5cd41f60f25cc : SUCCESS in 7m 27s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/c8f19fe118aa49be9e08ee9776cffd85 : SUCCESS in 17m 28s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/26e95ea8572e4c6dafa18f96da1012fb : SUCCESS in 8m 44s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/1844cf2a7b994025aff0e50df9fb4cc8 : SUCCESS in 53m 48s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/7ce879bd66c4417fa198e83eeb1ed6e2 : SUCCESS in 1h 15m 24s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/6c76ddc69b894bf9bc81790d7ee18adb : SUCCESS in 50m 53s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/34ab96554f3c4472b73b39eed16e2aa8 : SUCCESS in 50m 08s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e66c16933a6846ad931f0f6798ce6c2f : SUCCESS in 17m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/037815348ad349159f8776b711a182f7 : SUCCESS in 5m 38s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/452f54966ac3423ba195e41012593dc2 : SUCCESS in 22m 37s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c43d8bd7f5f8466bb4978068733d36d6 : SUCCESS in 1h 07m 15s","accounts_in_message":[],"_revision_number":2},{"id":"f60f2ea7d324b1977930d810325c5f515e2ad755","author":{"_account_id":36602,"name":"chenmingjian","display_name":"chenmingjian","email":"starvingnow@live.cn","username":"chenmingjian"},"date":"2023-12-15 05:49:29.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"8395e4d6a4aa4f71ce0f6ae8ac5f284297c07c26","revisions":{"3d2b6624ae14e3846caeb610bb171db0ee4e8174":{"kind":"REWORK","_number":1,"created":"2023-12-07 16:12:34.000000000","uploader":{"_account_id":36602,"name":"chenmingjian","display_name":"chenmingjian","email":"starvingnow@live.cn","username":"chenmingjian"},"ref":"refs/changes/01/903101/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/01/903101/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/903101/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/903101/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/903101/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/01/903101/1"}}},"commit":{"parents":[{"commit":"7dadc5cac1e47605def54c78ebb10744e0239620","subject":"test: use debug_logger instead of Mock logger","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7dadc5cac1e47605def54c78ebb10744e0239620"}]}],"author":{"name":"mingjian chen","email":"starvingnow@live.cn","date":"2023-12-07 15:56:22.000000000","tz":480},"committer":{"name":"mingjian chen","email":"starvingnow@live.cn","date":"2023-12-07 15:56:22.000000000","tz":480},"subject":"when no Delimiter return NextMarker","message":"when no Delimiter return NextMarker\n\nChange-Id: I62f97d814e1e226dcd88f2c04a19041afa58379a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3d2b6624ae14e3846caeb610bb171db0ee4e8174"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3d2b6624ae14e3846caeb610bb171db0ee4e8174"}]},"branch":"refs/heads/master"},"8395e4d6a4aa4f71ce0f6ae8ac5f284297c07c26":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2023-12-09 06:56:36.000000000","uploader":{"_account_id":36602,"name":"chenmingjian","display_name":"chenmingjian","email":"starvingnow@live.cn","username":"chenmingjian"},"ref":"refs/changes/01/903101/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/01/903101/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/903101/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/903101/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/903101/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/01/903101/2"}}},"commit":{"parents":[{"commit":"7dadc5cac1e47605def54c78ebb10744e0239620","subject":"test: use debug_logger instead of Mock logger","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7dadc5cac1e47605def54c78ebb10744e0239620"}]}],"author":{"name":"mingjian chen","email":"starvingnow@live.cn","date":"2023-12-07 15:56:22.000000000","tz":480},"committer":{"name":"chenmingjian","email":"starvingnow@live.cn","date":"2023-12-09 06:56:36.000000000","tz":0},"subject":"s3api: When no Delimiter return NextMarker","message":"s3api: When no Delimiter return NextMarker\n\nbugfix: https://bugs.launchpad.net/swift/+bug/2045912\n\nChange-Id: I62f97d814e1e226dcd88f2c04a19041afa58379a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8395e4d6a4aa4f71ce0f6ae8ac5f284297c07c26"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8395e4d6a4aa4f71ce0f6ae8ac5f284297c07c26"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":36602,"name":"chenmingjian","display_name":"chenmingjian","email":"starvingnow@live.cn","username":"chenmingjian"}},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
