)]}'
{"id":"openstack%2Fswift~903266","triplet_id":"openstack%2Fswift~master~I01f9d2c58e265041c2adbac7dd2c5a5f90c58d25","project":"openstack/swift","branch":"master","topic":"extend-slo-attrs","attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2024-03-04 20:14:52.000000000","reason":"\u003cGERRIT_ACCOUNT_15343\u003e replied on the change","reason_account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},"1179":{"account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"last_update":"2023-12-10 23:08:28.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I01f9d2c58e265041c2adbac7dd2c5a5f90c58d25","subject":"wip: cross-compat test for non-mpu","status":"NEW","created":"2023-12-10 22:00:46.000000000","updated":"2024-03-04 20:14:52.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":4,"unresolved_comment_count":2,"has_review_started":true,"meta_rev_id":"d321bcbb096541a3c18004360ad162bdd398583a","_number":903266,"virtual_id_number":903266,"owner":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2024-02-05 16:04:03.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-12-10 22:45:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-12-10 23:08:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-02-05 15:46:51.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"CC"},{"updated":"2024-03-04 20:14:52.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"CC"}],"messages":[{"id":"1647d45a5c79b4b94344fd274585acfa7dda1472","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2023-12-10 22:00:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"693cfb977f31556f6f92cc4a592cf1a562b246d3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-10 22:45:58.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/04143226f1984609894b1805daffaf76\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c6ca1331a69f45d584c995bdc671c48e : FAILURE in 17m 14s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1989e99c644a418b9d12906d95d305d0 : FAILURE in 15m 17s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/162b4671cec4482e885d297275d357b5 : SUCCESS in 20m 38s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/ca0d8b72719947299c26f28346006e6e : SUCCESS in 42m 23s","accounts_in_message":[],"_revision_number":1},{"id":"c6a1ac18eb5272744fd4d55ef170a967230f62cf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-10 23:08:28.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/aa46ac25bfb24e25a91a094add6b1ab7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c10df46e2b004a41b66098c470a099ed : SUCCESS in 8m 57s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/aa02c21139e14b449054a6013c6778ea : SUCCESS in 1h 01m 51s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/a73f7d92e95845d3a9c5ce61a53bdbf4 : SUCCESS in 9m 52s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/29aa985c727d45eebe4f59a847a89421 : SUCCESS in 5m 55s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/0f4a9b5be337486da1b698f0b51b96f9 : FAILURE in 8m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c262530d51fc4d2783f7f152445caa6e : SUCCESS in 4m 53s","accounts_in_message":[],"_revision_number":1},{"id":"6d2e383ef7ea532f0e5b2f79d380ce35864c2971","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-02-05 15:46:07.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"122452cbc1cb928496c49d386e4b2a23d79c1c54","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-02-05 15:46:51.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"7e8f5e6e4a2dff53119eed13a65e25f2bc1748da","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-05 16:04:03.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/77acc860007d45248375d9b3aa2d5553\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c237ee5ee73f48f9871497d6675680c9 : SUCCESS in 9m 31s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/8ebdceb848574309a274ae9b5804dec6 : SUCCESS in 12m 07s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/02018f41ecdb4176b46b06f6d8cddce8 : SUCCESS in 5m 32s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/d4eeaa79a841490abe5eace95917998a : SUCCESS in 8m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8ec5fa0e7803489caef3c0998db29dc9 : SUCCESS in 4m 52s","accounts_in_message":[],"_revision_number":2},{"id":"ac5ea49cf10c88d004222458d5dbc01ff99b6503","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-05 17:05:41.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/07cdc916ceba40ed8816cdcc1861f807\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/773b9dfaf99944b486dbd5583c2f4420 : SUCCESS in 15m 20s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e84abca3b4e3423399745bc186b60e3a : SUCCESS in 57m 51s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/aa400600dfe444d995e87d5c6a9c0e46 : SUCCESS in 22m 29s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/4a063169d15546ae89b511c98c9bd3c3 : SUCCESS in 1h 17m 26s","accounts_in_message":[],"_revision_number":2},{"id":"d321bcbb096541a3c18004360ad162bdd398583a","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-03-04 20:14:52.000000000","message":"Patch Set 2:\n\n(3 comments)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"47198cf36e888b877274c0f6126719da29d70da0","revisions":{"3fec40237ace87d49bdb94814ed6c469a4ba811c":{"kind":"REWORK","_number":1,"created":"2023-12-10 22:00:46.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/66/903266/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/66/903266/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/66/903266/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/66/903266/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/66/903266/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/66/903266/1"}}},"commit":{"parents":[{"commit":"7269d7461403367a6389789ecb5b2bf3b2a66e82","subject":"s3api: Support GET/HEAD request with PartNumber","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7269d7461403367a6389789ecb5b2bf3b2a66e82"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2023-12-10 22:00:34.000000000","tz":-360},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2023-12-10 22:00:34.000000000","tz":-360},"subject":"wip: cross-compat test for non-mpu","message":"wip: cross-compat test for non-mpu\n\nChange-Id: I01f9d2c58e265041c2adbac7dd2c5a5f90c58d25\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3fec40237ace87d49bdb94814ed6c469a4ba811c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3fec40237ace87d49bdb94814ed6c469a4ba811c"}]},"branch":"refs/heads/master"},"47198cf36e888b877274c0f6126719da29d70da0":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-02-05 15:46:07.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/66/903266/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/66/903266/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/66/903266/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/66/903266/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/66/903266/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/66/903266/2"}}},"commit":{"parents":[{"commit":"212d5d438ac99288b32e68a03e288924f59cb262","subject":"s3api: Support GET/HEAD request with ?partNumber","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/212d5d438ac99288b32e68a03e288924f59cb262"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2023-12-10 22:00:34.000000000","tz":-360},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-02-05 15:42:54.000000000","tz":0},"subject":"wip: cross-compat test for non-mpu","message":"wip: cross-compat test for non-mpu\n\nChange-Id: I01f9d2c58e265041c2adbac7dd2c5a5f90c58d25\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/47198cf36e888b877274c0f6126719da29d70da0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/47198cf36e888b877274c0f6126719da29d70da0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
