)]}'
{"id":"openstack%2Fswift~903537","triplet_id":"openstack%2Fswift~master~I9639940f9a09a6707954db2e0606c37bf9e4c7d3","project":"openstack/swift","branch":"master","topic":"extend-slo-attrs-patch52","hashtags":[],"change_id":"I9639940f9a09a6707954db2e0606c37bf9e4c7d3","subject":"sq: include backend headers with 416 response","status":"ABANDONED","created":"2023-12-12 20:01:00.000000000","updated":"2023-12-14 10:03:34.000000000","total_comment_count":1,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"ab1c27a5e80765dc5ed821bfd8a8066ac2c9e966","_number":903537,"virtual_id_number":903537,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2023-12-12 21:33:01.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-12-12 20:24:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-12-12 21:33:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9946854c93722008c7fa4290221de570f4f7da55","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-12-12 20:01:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c82251ea839f1e142ad96ad06ddf5bec3d16a66c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-12-12 20:02:05.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"125cf2375daa1e69610af30c7c218a3b6b3cefc9","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-12-12 20:13:27.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"78b4eafd321c91c55e6f9cdb013b946cabda3b2f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-12 20:24:53.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/025f8fbb71264c2aa90177574db24732\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e99ef4a0230949478ec766f7c98b8fa6 : SUCCESS in 15m 35s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8fe3faac97874cd9a2ecc2c4fa91092c : SUCCESS in 18m 20s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/328ba0353ede4dd28b4fdec36594a88f : SUCCESS in 22m 06s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/34823238a3e74bdc8a27cf647d883ea8 : SUCCESS in 21m 21s","accounts_in_message":[],"_revision_number":2},{"id":"42a25292f75fe0bc9de80fd46f00bc5e5ca8c194","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-12 21:33:01.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fe9ff675ed624a5ca498bc881c8680aa\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/984490ad92fe456ebbdf07103fd5b993 : SUCCESS in 9m 06s\n- grenade https://zuul.opendev.org/t/openstack/build/2344b0631dea45b79dfa2715ff6d4fb8 : SUCCESS in 1h 08m 46s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/6215f55535fd4511a4a7b211684fb990 : SUCCESS in 1h 24m 18s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/15e9d225451346629cb4ec684c436ea4 : SUCCESS in 52m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2f9e63d883bd42d28d02fbe23a32ab78 : SUCCESS in 1h 11m 42s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b137918c24fc4975ac0d09e045774055 : SUCCESS in 19m 56s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3818f6b3a2da4f41b2dc47a8096ce661 : SUCCESS in 19m 46s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/49ce08eeb4b84a05816cdb1500a74d93 : SUCCESS in 20m 04s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/0e155e43f0c34f66be7c142347d0f0f8 : SUCCESS in 8m 48s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/1e4c12795af14742a3d2f90833a244bf : SUCCESS in 5m 56s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/8ff484884f15440f9a6199dcb9f7fd8d : SUCCESS in 23m 30s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/2c468b8e69904387907bd526553d4b1d : SUCCESS in 23m 55s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/a9d41c7b762948edb15c846deb9c0556 : SUCCESS in 16m 16s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/292f784420594550b8d888d0af353c84 : SUCCESS in 14m 37s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/18792c76d00f426da199087473d7c35a : SUCCESS in 25m 01s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/17113b24966845f0a5162e7872711034 : SUCCESS in 23m 45s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/53b64a7088ff4833acdf7f452e70c847 : SUCCESS in 17m 54s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/3ab4467e19a144ff92b434b9099faded : SUCCESS in 20m 12s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/c744140defb34f228c66922c95fa23b4 : SUCCESS in 14m 08s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/938d9b36d2f74071baf56ba449c791c1 : SUCCESS in 16m 51s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/d7350444e10a47d989333b8a81d2e92c : SUCCESS in 15m 21s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/ea00ec3379434b09bd65ca6d80460a7f : SUCCESS in 18m 50s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/36648604cedc495f83a99c61e9852bb2 : SUCCESS in 16m 46s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/5eba35737d654b7f9eb0b91a0dbfa97f : SUCCESS in 8m 44s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/f3cbe9c8fad946fa9e03368534addc94 : SUCCESS in 59m 28s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/b1dab20846b64de4b3cf436734e80f2c : SUCCESS in 1h 10m 49s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/5d7029e271f349758cef6234247bcae4 : SUCCESS in 56m 12s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/00b377befd4d4bc8ac501745eec2aa92 : SUCCESS in 56m 53s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fafbc55718644a77bd9bc4faa52ed0a8 : SUCCESS in 24m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/34ad66d486c94ca6969627aa16ae9757 : SUCCESS in 5m 21s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/c2fc602abe15408396cd035f6ebacc51 : SUCCESS in 20m 34s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/788d9394c10945aa805bca0cbb15b3cb : SUCCESS in 1h 01m 35s","accounts_in_message":[],"_revision_number":2},{"id":"ab1c27a5e80765dc5ed821bfd8a8066ac2c9e966","tag":"autogenerated:gerrit:abandon","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-12-14 10:03:34.000000000","message":"Abandoned\n\nsquashed into parent","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"7bca44b0993e78a93af31f5c02436be88d16df5e","revisions":{"5c14a5bde8d3eb7174437122532de04c607c45b3":{"kind":"REWORK","_number":1,"created":"2023-12-12 20:01:00.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/37/903537/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/37/903537/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/37/903537/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/37/903537/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/37/903537/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/37/903537/1"}}},"commit":{"parents":[{"commit":"7ab63dc431d00b8ca0a62d49d89ba529eaa8960b","subject":"slo: part-number\u003dN query parameter support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7ab63dc431d00b8ca0a62d49d89ba529eaa8960b"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-12-12 19:58:46.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-12-12 19:58:46.000000000","tz":0},"subject":"sq: include backend headers with 416 response","message":"sq: include backend headers with 416 response\n\nChange-Id: I9639940f9a09a6707954db2e0606c37bf9e4c7d3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5c14a5bde8d3eb7174437122532de04c607c45b3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5c14a5bde8d3eb7174437122532de04c607c45b3"}]},"branch":"refs/heads/master"},"7bca44b0993e78a93af31f5c02436be88d16df5e":{"kind":"REWORK","_number":2,"created":"2023-12-12 20:02:05.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/37/903537/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/37/903537/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/37/903537/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/37/903537/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/37/903537/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/37/903537/2"}}},"commit":{"parents":[{"commit":"7ab63dc431d00b8ca0a62d49d89ba529eaa8960b","subject":"slo: part-number\u003dN query parameter support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7ab63dc431d00b8ca0a62d49d89ba529eaa8960b"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-12-12 19:58:46.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-12-12 20:01:34.000000000","tz":0},"subject":"sq: include backend headers with 416 response","message":"sq: include backend headers with 416 response\n\nChange-Id: I9639940f9a09a6707954db2e0606c37bf9e4c7d3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7bca44b0993e78a93af31f5c02436be88d16df5e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7bca44b0993e78a93af31f5c02436be88d16df5e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
