)]}'
{"id":"openstack%2Fswift~903814","triplet_id":"openstack%2Fswift~master~I739c9f51cf7f656a7fbdfeb7575ef6ed4642fa11","project":"openstack/swift","branch":"master","topic":"p-proxy-namespaces","attention_set":{"34930":{"account":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"last_update":"2023-12-17 07:14:54.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I739c9f51cf7f656a7fbdfeb7575ef6ed4642fa11","subject":"DNM: test get_namespaces to reuse _get_shard_range_rows.","status":"NEW","created":"2023-12-17 05:42:01.000000000","updated":"2023-12-17 07:17:53.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"9978a88fa7d8e5633c4d235b1acda787561864c9","_number":903814,"virtual_id_number":903814,"owner":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2023-12-17 07:14:54.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-12-17 07:14:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"142743cddae9e6d9d19084600d85028931282e62","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2023-12-17 05:42:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3ee5d3b7da8ec3a6e804d0a25adb750f482952a7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-17 07:14:54.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/af8965e7ea6243aaa1b17e178e0d9b20\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/68c3cae1ee604005a1e0c57a204f45cf : SUCCESS in 10m 11s\n- grenade https://zuul.opendev.org/t/openstack/build/a26731d1dfa04d7694e59c19a85f69aa : SUCCESS in 1h 03m 55s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/333ac0e9bd934dd794020f807306e508 : SUCCESS in 1h 01m 33s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/8426420213aa4217ae864134a9f530b6 : SUCCESS in 52m 54s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5f9cd3c082464ababf3291cf93f2820e : SUCCESS in 1h 01m 39s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a92a510e7d67431f99328edb3712eb33 : SUCCESS in 19m 17s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/72351805324a4ba286edfc18e3eec142 : SUCCESS in 20m 06s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/716627e3b6d34fab8aea72286c047d71 : SUCCESS in 23m 50s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/40a406a4dbb14a4999c8ba3c15ddec68 : SUCCESS in 10m 13s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/db59b044a02940389ea8a29c8048e821 : SUCCESS in 8m 36s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/b821f33d0c0b40f499ab315147a050f9 : FAILURE in 22m 25s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/ef4935953b034bf8b08027d58c0e2ef2 : FAILURE in 22m 07s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/744e986f447e461f89d6cd8045d79860 : FAILURE in 18m 56s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/1e0619ed864c4ca6928251fc0261c458 : FAILURE in 19m 54s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/bfc798e717364fedba9e789f2a9d912d : FAILURE in 18m 38s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/958f8731fda94f42a0cf6c1a2958c81f : SUCCESS in 17m 41s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/b9434a4f57be4bc1897d76321d4fab46 : SUCCESS in 18m 59s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/17e227c888d64b8b83f539153a4a6fa4 : SUCCESS in 23m 26s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/f8fafb032ddd413389e2f9e114795c57 : SUCCESS in 17m 28s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/15632f310b8645d99d91664c1108040b : SUCCESS in 16m 02s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/8160e494631c467ebf205334a44f3403 : SUCCESS in 16m 32s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/a33c6e3352154c4bb9e6881ff5a5f9a3 : SUCCESS in 7m 49s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/2e6af41aa9754b9a9fcf4abd2655a318 : SUCCESS in 17m 32s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/a3eb3a7ed91a41f5ab4ff25959f952fe : SUCCESS in 10m 49s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/52d92fbfe2504bc5b98c2c169b2d068f : FAILURE in 1h 07m 16s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/47b4b58d07f64c0ba2c714197346332c : FAILURE in 1h 18m 51s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/10d616a174b945dea2a88261afa648b5 : SUCCESS in 1h 01m 43s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/56cb83f72f904e1c87311977875314a2 : SUCCESS in 1h 04m 03s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/83dd760cf0ef487aa7b0c0559819f2d1 : FAILURE in 22m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c7de148563aa4799bb2304f98b63a230 : SUCCESS in 6m 15s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/55a11890296948d7b4ca74a88ecbf09d : SUCCESS in 20m 23s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7803ca015856462f9a71f60630f25a6d : SUCCESS in 55m 07s","accounts_in_message":[],"_revision_number":1},{"id":"9978a88fa7d8e5633c4d235b1acda787561864c9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-17 07:17:53.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6091aa7f6686455a9e968fca4e6514a7\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/cf8744df193e44e4be726232b4656461 : TIMED_OUT in 1h 31m 15s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/612ba205ed9d4d3e831f68f64772fe92 : FAILURE in 1h 20m 04s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/29c727ed1e4644698e736973142e2ddd : SUCCESS in 23m 24s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/3fc5545bf58642549c00fc34f7d6a412 : SUCCESS in 1h 25m 50s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4ab015e31ed8af559a409f4861ce811eae79cc3e","revisions":{"4ab015e31ed8af559a409f4861ce811eae79cc3e":{"kind":"REWORK","_number":1,"created":"2023-12-17 05:42:01.000000000","uploader":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"ref":"refs/changes/14/903814/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/14/903814/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/14/903814/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/14/903814/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/14/903814/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/14/903814/1"}}},"commit":{"parents":[{"commit":"c5ea203be8f91f270cb1b82404c43316d15360e8","subject":"container: add Namespace marker/end_marker/include/reverse support.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c5ea203be8f91f270cb1b82404c43316d15360e8"}]}],"author":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2023-12-17 05:40:23.000000000","tz":-480},"committer":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2023-12-17 05:41:16.000000000","tz":-480},"subject":"DNM: test get_namespaces to reuse _get_shard_range_rows.","message":"DNM: test get_namespaces to reuse _get_shard_range_rows.\n\nChange-Id: I739c9f51cf7f656a7fbdfeb7575ef6ed4642fa11\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4ab015e31ed8af559a409f4861ce811eae79cc3e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4ab015e31ed8af559a409f4861ce811eae79cc3e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
