)]}'
{"id":"openstack%2Fswift~903847","triplet_id":"openstack%2Fswift~master~Ic907321a53e518ccf6a4acaf82f82c6698a40b90","project":"openstack/swift","branch":"master","topic":"extend-slo-attrs","attention_set":{},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2023-12-20 14:31:41.000000000","reason":"\u003cGERRIT_ACCOUNT_7847\u003e replied on the change","reason_account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}},"1179":{"account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"last_update":"2024-02-05 17:50:41.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Ic907321a53e518ccf6a4acaf82f82c6698a40b90","subject":"more s3api part-num tests","status":"ABANDONED","created":"2023-12-17 22:27:33.000000000","updated":"2024-02-05 17:50:41.000000000","total_comment_count":43,"unresolved_comment_count":4,"has_review_started":true,"meta_rev_id":"846b68d42e721a59b2e2ffc71b1e7ebc2a34ab01","_number":903847,"virtual_id_number":903847,"owner":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2023-12-19 18:03:33.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":-1,"date":"2024-02-05 17:50:21.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-12-17 23:34:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-12-18 00:05:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-12-18 16:36:43.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"CC"},{"updated":"2024-02-05 17:50:21.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"REVIEWER"}],"messages":[{"id":"8c3a50add1028c5cf4f699def55a47a738b94667","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2023-12-17 22:27:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3672a6100e30b1336147a45bb4bde18c3f419ad1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-17 23:34:25.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/112b6b8cec61475e9a7831ac44984074\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/2ed3470a88514a69920555fa0ab559f4 : SUCCESS in 49m 10s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/bcd0e43133e249efbe0957f2a9ea4484 : SUCCESS in 51m 00s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/1a6162dcc22d465c843de62730025efe : SUCCESS in 1h 03m 05s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/f1355b9ed2314342b20b6586797b6f00 : SUCCESS in 22m 58s","accounts_in_message":[],"_revision_number":1},{"id":"392456b8b4ac467435c0a901e37558a9bda1d7ab","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-18 00:05:50.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/53918854fe274d718687b297a919741c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0474feda41784bbc9859717081010839 : SUCCESS in 9m 19s\n- grenade https://zuul.opendev.org/t/openstack/build/9195071e83ef43c9940c3f1b32df8bc7 : FAILURE in 40m 04s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/d5d2e76ec65644f99db9c82b5c8a7cc0 : SUCCESS in 1h 02m 08s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/85ecc5ec4abb4b5b8f5695854bdd4379 : SUCCESS in 39m 14s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/74c61076a2ad4f95b9779bec75b13873 : SUCCESS in 1h 03m 26s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e1a5f75a0e0a45878d160104a7ee6fc5 : SUCCESS in 19m 26s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5c5a8eadc63847baa475ee33b65fec3d : SUCCESS in 20m 37s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/23d5d1e23b6d495dbfd75c5232c11b03 : SUCCESS in 19m 58s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/ef666de75826432e85a0b5530b937139 : SUCCESS in 10m 19s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/2a3f95eea6a348c692e866e0762cd8a6 : SUCCESS in 6m 22s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/6447dd38fe9c4cdbb740bb87dcdc4126 : SUCCESS in 18m 10s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/067ed3115c1041b098d4ac5fe9a4e7a7 : SUCCESS in 18m 00s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/435c1eb4ac6b4d8db4ef5fa1d304a020 : SUCCESS in 17m 22s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/e543736e1765469388b6e26d7afa4054 : SUCCESS in 17m 21s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/51668a6b4ed245289c67a2284837e2f1 : SUCCESS in 18m 58s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/244021a1a29b44388f4606293c54797d : SUCCESS in 28m 44s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/214a8dde25c34f8aad551b6865051100 : SUCCESS in 17m 44s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/0b2eeba330ff47e49f62017aaed01d2b : SUCCESS in 19m 53s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/326772ea89324d63b24b24252b6579b7 : SUCCESS in 14m 22s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/dc9dd570d7ff4709a001be3bcb7326d7 : SUCCESS in 14m 14s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/b67365f21e3b48f8b1d3b52dcaf31815 : SUCCESS in 14m 48s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/c02fecc807f1408a89a482b6d4b47710 : SUCCESS in 7m 48s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/5fca56317b6c4ef2aabb9d4fb904c219 : SUCCESS in 18m 50s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/943a6dba53764971959619d6147580a9 : FAILURE in 9m 00s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/93b114fab0bd4d429ff57ca5961b5843 : SUCCESS in 49m 29s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/09398affd2984e70a61983e60341d051 : SUCCESS in 57m 45s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/5e633ef16e044191a92c46885535510d : SUCCESS in 48m 25s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/5095a952948b4229a18094ec0b54c7a4 : SUCCESS in 46m 16s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6d86343b172a4787a0a6f3772d3df5a8 : SUCCESS in 18m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2a73a3936b3b420d8b69732d741e62a9 : FAILURE in 4m 30s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/ec42370b465a43c293093db81795d066 : SUCCESS in 21m 05s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cfd132aaed0044dc870cc135e6e003e0 : SUCCESS in 1h 06m 19s","accounts_in_message":[],"_revision_number":1},{"id":"2677e9df1dbef52235fa2d93ca72a8511b04727e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-12-18 16:24:22.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"62db7cad9c372d87701458abb7c69d73ff4307a5","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-12-18 16:36:43.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"387a5f6c675fc6a46bd1e9391520699ac73bb894","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-18 17:16:10.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/08706f56df264f32979f1e568d15ff32\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9eb1f7ccf64849e9a1055c237dfc25e9 : SUCCESS in 17m 53s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/08c776bcd046456db933fb63f1a27050 : SUCCESS in 48m 54s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/1345ccb5c16a451bac07030c0f90d22e : SUCCESS in 20m 53s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/fb931e1403eb43f9a8dd83bb98bb342c : SUCCESS in 44m 22s","accounts_in_message":[],"_revision_number":2},{"id":"de921b422488464f815fc90396d842b47585ed00","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-18 17:51:07.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ec0b91449f0e4b5b9e8390073734ee2d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/449e5284d53a482086e65583066ec99c : SUCCESS in 9m 05s\n- grenade https://zuul.opendev.org/t/openstack/build/565502f4c9334f4d8df25424873f5931 : SUCCESS in 1h 03m 35s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/6664626f918a4fb5aef26979c6128b8c : SUCCESS in 1h 16m 12s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/1f798ac203794f66a6c940ea82b7907c : SUCCESS in 55m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0a0836ce9c78461fb47a52a19495c90c : SUCCESS in 1h 04m 10s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e726d2ac354549c487c500db588cdc9b : SUCCESS in 14m 58s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/11fb9c07d85a43b88fc1fd8e994f4a1d : SUCCESS in 21m 15s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d134103472534fb88b49c1fe63a62e3c : SUCCESS in 23m 33s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/11777752bab24a3e8f60fa85e0b9d9d7 : SUCCESS in 9m 15s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/2486e84133b141cb9828dbabfc316304 : SUCCESS in 5m 30s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/d51b24d992254266b2b04627bbb1b6c8 : SUCCESS in 17m 21s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/1d2daed427cd431d80ad3e5239495423 : SUCCESS in 19m 04s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/baff9dc5dba149ad9360e3444aa4cb63 : SUCCESS in 15m 47s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/e513c67295c84ec6836dbaf9343941c1 : SUCCESS in 14m 14s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/870b5dcd8849407e99757696078f0a9b : SUCCESS in 19m 56s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/1ab971ca150443c48b2069c75748fa2c : FAILURE in 20m 47s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/31c5925affc24698afa7e10768da7e40 : FAILURE in 19m 55s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/c6283ca9b90d43c2ba713c217307a5b1 : FAILURE in 20m 39s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/2fa1c68ea2104ea58c7d4f448fd1d9ec : SUCCESS in 13m 35s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/390bb0159e3741488a2739c75b34de20 : SUCCESS in 14m 30s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/f233ee9077224ffb8bb6afeebe41224c : SUCCESS in 15m 23s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/532fec72f6a346caa6118c462ffd8584 : SUCCESS in 7m 56s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/4b3220aeee194369b6f8e8947f8da8b4 : SUCCESS in 16m 48s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/c08ac0a5adde4ee3bb0b9dc74251a6bd : FAILURE in 7m 15s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/c1a908200b9c4360b9487d6da159c172 : SUCCESS in 51m 20s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/bb12f68a2d3f44368a09c344872ebd6e : SUCCESS in 57m 01s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/5e1347b328ec4e6ba7329a53af2b58b5 : SUCCESS in 1h 03m 53s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/45e3b63cc1eb40268cc2a03555b105f7 : SUCCESS in 42m 36s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/13e1b701f7fd4568a0e5c8a9c6462c5d : SUCCESS in 16m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d8ad252a5bf941d68dd5250e1b74546e : SUCCESS in 5m 20s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/9aaaab575f854b99b5bf22fc41ae8863 : SUCCESS in 22m 39s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/47e81ed16dc5447bb3ef5f4dcac6bc50 : SUCCESS in 50m 09s","accounts_in_message":[],"_revision_number":2},{"id":"38c88feaef1c2971aafc1292de54d67bd6ef211b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2023-12-19 01:24:45.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"f5a83409e5a7f375fd8391f649fed358f61bd888","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2023-12-19 01:42:02.000000000","message":"Patch Set 3: Code-Review-1 Workflow-1\n\n(7 comments)","accounts_in_message":[],"_revision_number":3},{"id":"3ef8c93be08565f4ab4f282740cf7e327377039c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-19 02:42:52.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ac15155a401048e58c9fdc927fb0f71c\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/de5382149ce14bb7a581c07316be8362 : SUCCESS in 17m 30s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/891214dce6794470891dfd5db00f6f09 : SUCCESS in 1h 14m 15s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/363cd6b8ee174cee937b1ce5ef744029 : SUCCESS in 23m 14s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/2b3c2e312dba4434a707d00b181cb3b9 : SUCCESS in 1h 11m 04s","accounts_in_message":[],"_revision_number":3},{"id":"8acf22ef7b3e32b09120ec13d4031150ecfa84c8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-19 03:06:27.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/dd76323f6ab34679bb7303ad37fc03f7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7a341a3d1b10441e9f29d71bf1ed15a7 : SUCCESS in 10m 03s\n- grenade https://zuul.opendev.org/t/openstack/build/1a7f98a232224c859d4903adc60f9ee2 : SUCCESS in 1h 31m 29s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/eaeda2f05a8546f3a1736fdb8368360c : SUCCESS in 52m 54s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/f00263bcca374a10ad2e9631262d0100 : SUCCESS in 57m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/18d1898036714f8c84efc7cfdd6cd603 : SUCCESS in 1h 08m 59s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c0ca5dc489cd4dc99a823edcf3158389 : SUCCESS in 19m 15s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c5fb19336f284800b925fd5e24defb47 : SUCCESS in 27m 01s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/335ef13f9b344779ab026011d5732701 : SUCCESS in 23m 23s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/9ea68b8b31ba41b78053d10cc01fb1eb : SUCCESS in 8m 39s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/1d0ce915af0f4101b757ccc266f1124b : SUCCESS in 5m 35s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/79859b37bbea4983a511e83d30c7a8a7 : SUCCESS in 18m 00s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/b2e29758bd9747fbbb2ec9c3e38b6c7e : SUCCESS in 19m 51s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/21da3f112e8e4398b78e1c0f60736a6b : SUCCESS in 14m 10s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/8dcbc3d80ce54090bb4061017dcb12fc : SUCCESS in 16m 06s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/e99b786daf744090a742983c3a73a6a4 : SUCCESS in 17m 39s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/dda38c757f9843099a45c3db9034ee92 : FAILURE in 19m 07s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/a87002acfb5b46c780c571b41a857738 : FAILURE in 19m 49s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/89675c1720e24bc783bacf9a240bcf6c : FAILURE in 25m 51s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/c7b8dcd18656405fab9a04860984dce3 : SUCCESS in 16m 11s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/6b2c8dea77734f5b93f1dbb9573f1c62 : SUCCESS in 16m 50s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/8d38b7da5a674cbf90b1694c19ca57ad : SUCCESS in 16m 04s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/d88d444c6a084c688d3acd6c8f47c25a : SUCCESS in 7m 14s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/5e53c112c9c54688a8eaf7340693723f : SUCCESS in 18m 19s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/10d883f3edd94921a628478341f16ea0 : FAILURE in 7m 57s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/5d2ad95d3f81426d9f9e2536c3f08b10 : SUCCESS in 53m 49s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/04b023adc5da400b950a7f820344c5c9 : FAILURE in 1h 03m 11s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/256b8bfda289402494daa481dadc601e : SUCCESS in 46m 45s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/f0abda7c33f4499996d8c0e2fe7a4af4 : SUCCESS in 42m 52s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2d655aedbdb843bcb54844b677cf15ba : SUCCESS in 17m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b0f88014f76941d4aea22ac48c6f9a4d : SUCCESS in 4m 57s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/a002a3adcf074a3c8b22b597997644ef : SUCCESS in 23m 42s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/52756fef0dc744d3afd82e9cf2aa6b7f : SUCCESS in 1h 03m 19s","accounts_in_message":[],"_revision_number":3},{"id":"0fb0a555bc0926efeb329c2ba8363b3bb18e6d67","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-12-19 15:40:41.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":4},{"id":"0c6904f5ee9536248d66896b83f5a66fa0d638cd","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2023-12-19 15:54:52.000000000","message":"Patch Set 4: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":4},{"id":"0940b617ca37dacec2fefe3f7f0ec6e737f2fd27","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-12-19 16:13:29.000000000","message":"Patch Set 4:\n\n(3 comments)","accounts_in_message":[],"_revision_number":4},{"id":"e53eae0378ff6beb538cf77f9645afddaf65fa92","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-19 16:26:38.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/382f30bcd8af49b0861857b294491e10\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/3d618fc10bcd40e9b8ea3502a8287d4c : SUCCESS in 17m 22s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f198de27deda4fce810d24ae116d06a6 : SUCCESS in 18m 05s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/f10ea6e96d0d4abab291390aa1db0d11 : SUCCESS in 23m 26s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/f703afe2549f470388aeff2061aca1e6 : SUCCESS in 22m 49s","accounts_in_message":[],"_revision_number":4},{"id":"012c12848dfdb781d8193736a3a995acfcd3d4d7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-12-19 16:34:16.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.\n\nCopied Votes:\n* Code-Review-1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n","accounts_in_message":[],"_revision_number":5},{"id":"654e39841d944f70799c1cbbde2f5b75a8f5b6ad","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-12-19 16:46:20.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":6},{"id":"5d0c5eafc5f1d0487106fd29708726b716e12272","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-12-19 16:47:44.000000000","message":"Patch Set 6:\n\n(3 comments)","accounts_in_message":[],"_revision_number":6},{"id":"2d5ff70609b14502581b42063e42c6ea58fc084c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-19 18:03:31.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f982a2eabd514a868f9d0e1da0f3b907\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/faf7b83a16964b99b202a4208a30ef4b : SUCCESS in 53m 49s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/44e348240f2c46b4b216456b6129cf01 : SUCCESS in 1h 14m 56s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/9b4070df36274fc69b5ea15417e0788f : SUCCESS in 31m 13s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/a587dee4ab32468ab9db8bec4504402d : SUCCESS in 1h 09m 28s","accounts_in_message":[],"_revision_number":6},{"id":"f9862f0ffe9b0f297f10f3a678080e1b081b202b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-19 18:03:33.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/71dbdea93b7d46e09afdc66c6091bd31\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/179ca76ae9f740309f251c8549f699e4 : SUCCESS in 9m 25s\n- grenade https://zuul.opendev.org/t/openstack/build/3055ca4b42324ab798c511efe33b2785 : SUCCESS in 1h 10m 06s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/1042a2cb0b5a4928bebd867b5b3f44d0 : SUCCESS in 56m 28s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/a5745bbf0c87426d820b80ae086173b7 : SUCCESS in 1h 00m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/790b74eef0ed44fab19432c6782433be : SUCCESS in 1h 09m 55s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a81f554a785145958089bf036221c26b : SUCCESS in 19m 46s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d10d0da8d9d84051810c3fc5ef901b6a : SUCCESS in 27m 35s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f94bc3f3d20742c28dee0c4f0250c3a5 : SUCCESS in 18m 11s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/b0fdcb12ae584fdbad1b224ccad95bda : SUCCESS in 7m 37s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/8c8072fc12b442c5ae68623292882fe6 : SUCCESS in 5m 22s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/7e6e9046173e4733bb60430e7300be6b : SUCCESS in 28m 17s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/4c97dfcb52c24a08a6aa918d25bf9409 : SUCCESS in 18m 36s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/48b7cc65fbd644518a8018677d278564 : SUCCESS in 19m 26s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/15e3230394104a61a10717ea8fe2920f : SUCCESS in 17m 55s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/461936059a6f4751817e0fbb61fa02b4 : SUCCESS in 17m 50s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/aea242687735442cb9aebcb38440115c : SUCCESS in 18m 31s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/b2231a5cf759402ca38c318b7adb8daf : SUCCESS in 18m 50s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/fa5679806dd74445ab31abba72be1149 : SUCCESS in 18m 39s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/fdda3ca075964e82a8a9f9c022ed00df : SUCCESS in 18m 38s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/9d38518e0a9f4ef4b6e413b9d915d435 : SUCCESS in 17m 17s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/425f06ec7e4e4a6fb7660dc9ec6fc8fb : SUCCESS in 18m 44s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/a4210638686e43da8d568b03e92e08a1 : SUCCESS in 7m 18s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/a51be0d7f3434a4c804562d631af597b : SUCCESS in 20m 54s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/07bdb0469b944910b7efe584279c3635 : FAILURE in 8m 29s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/a929a64032eb4105a499dbd981c12dc7 : SUCCESS in 55m 07s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/cd5329a09cd14d18bf8b6f3b97419f6c : SUCCESS in 1h 05m 28s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/dac96436813f4a11800121241c003bfd : SUCCESS in 55m 11s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/c88faa31861347aa9fc5380ba3a89344 : SUCCESS in 51m 08s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/60c6425b5bc24cd4b8e1b6f97a0849b2 : SUCCESS in 15m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0fad66e92ce347279e8f5460aad4945f : SUCCESS in 5m 40s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/f2598490d3f74b6b804d238b246584d7 : SUCCESS in 18m 19s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bf4dfc016b7648b8b47435bdf431fbf8 : SUCCESS in 48m 06s","accounts_in_message":[],"_revision_number":6},{"id":"074172b4e6cadaea6dc1d7518a9099652533a8f1","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-12-19 19:03:54.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"4a74980339ac615aea9f738e0f0a99067850bf1d","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-12-19 19:29:16.000000000","message":"Patch Set 6:\n\n(3 comments)","accounts_in_message":[],"_revision_number":6},{"id":"31bc4863f943d9522f477e64bc8b58bdfb5b3a5f","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2023-12-20 01:27:35.000000000","message":"Patch Set 6:\n\n(11 comments)","accounts_in_message":[],"_revision_number":6},{"id":"78215065cde1342b4d9b850f52c8ff164d7bf46c","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-12-20 14:31:41.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"bfef8c6908ec40b2b643753937b242f92eb43523","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-12-20 14:37:22.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"5269fae4a61bbf209d99fa44ec468f37017eaa38","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-02-05 17:50:21.000000000","message":"Patch Set 6: Code-Review-1\n\n(8 comments)","accounts_in_message":[],"_revision_number":6},{"id":"846b68d42e721a59b2e2ffc71b1e7ebc2a34ab01","tag":"autogenerated:gerrit:abandon","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-02-05 17:50:41.000000000","message":"Abandoned\n\nchanges made it into https://review.opendev.org/c/openstack/swift/+/894580","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"b41f59bbb2b3d84ff19ca889cbc140b73fa06564","revisions":{"52e3577979e9f6909282cd92fc0be2ced6d2358c":{"kind":"REWORK","_number":1,"created":"2023-12-17 22:27:33.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/47/903847/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/47/903847/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/903847/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/903847/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/903847/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/47/903847/1"}}},"commit":{"parents":[{"commit":"61d73adc3539703fe29164cd92e5b14bd537ad27","subject":"s3api: Support GET/HEAD request with PartNumber","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/61d73adc3539703fe29164cd92e5b14bd537ad27"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2023-12-17 22:27:26.000000000","tz":-360},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2023-12-17 22:27:26.000000000","tz":-360},"subject":"more s3api part-num tests","message":"more s3api part-num tests\n\nChange-Id: Ic907321a53e518ccf6a4acaf82f82c6698a40b90\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/52e3577979e9f6909282cd92fc0be2ced6d2358c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/52e3577979e9f6909282cd92fc0be2ced6d2358c"}]},"branch":"refs/heads/master"},"788a4e5156fb6671e1f0e4170316b821992a4a36":{"kind":"REWORK","_number":2,"created":"2023-12-18 16:24:22.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/47/903847/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/47/903847/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/903847/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/903847/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/903847/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/47/903847/2"}}},"commit":{"parents":[{"commit":"c09525899875d2b8c866ba63564c50aa4fe2872e","subject":"s3api: Support GET/HEAD request with PartNumber","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c09525899875d2b8c866ba63564c50aa4fe2872e"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-12-18 15:29:12.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-12-18 16:23:48.000000000","tz":0},"subject":"more s3api part-num tests","message":"more s3api part-num tests\n\nChange-Id: Ic907321a53e518ccf6a4acaf82f82c6698a40b90\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/788a4e5156fb6671e1f0e4170316b821992a4a36"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/788a4e5156fb6671e1f0e4170316b821992a4a36"}]},"branch":"refs/heads/master"},"3269fb664d6cb54abd46bdfe615dd887d6d512d7":{"kind":"REWORK","_number":3,"created":"2023-12-19 01:24:45.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/47/903847/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/47/903847/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/903847/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/903847/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/903847/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/47/903847/3"}}},"commit":{"parents":[{"commit":"059517f49abbd7534e00a427d30472c361f6056d","subject":"s3api: Support GET/HEAD request with PartNumber","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/059517f49abbd7534e00a427d30472c361f6056d"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-12-18 15:29:12.000000000","tz":0},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2023-12-18 23:51:22.000000000","tz":-360},"subject":"more s3api part-num tests","message":"more s3api part-num tests\n\nChange-Id: Ic907321a53e518ccf6a4acaf82f82c6698a40b90\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3269fb664d6cb54abd46bdfe615dd887d6d512d7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3269fb664d6cb54abd46bdfe615dd887d6d512d7"}]},"branch":"refs/heads/master"},"6a6293b359b26a8b8dd7943ecdd2320abe88c248":{"kind":"REWORK","_number":4,"created":"2023-12-19 15:40:41.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/47/903847/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/47/903847/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/903847/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/903847/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/903847/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/47/903847/4"}}},"commit":{"parents":[{"commit":"8060edfe7e3dc12d47cdc3ebdb8f930c007dc1f2","subject":"s3api: Support GET/HEAD request with PartNumber","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8060edfe7e3dc12d47cdc3ebdb8f930c007dc1f2"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-12-18 15:29:12.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-12-19 11:27:43.000000000","tz":0},"subject":"more s3api part-num tests","message":"more s3api part-num tests\n\nChange-Id: Ic907321a53e518ccf6a4acaf82f82c6698a40b90\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6a6293b359b26a8b8dd7943ecdd2320abe88c248"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6a6293b359b26a8b8dd7943ecdd2320abe88c248"}]},"branch":"refs/heads/master"},"25ca6f970bb1084ab027d24f5644e3c2383fa1e9":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2023-12-19 16:34:16.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/47/903847/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/47/903847/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/903847/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/903847/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/903847/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/47/903847/5"}}},"commit":{"parents":[{"commit":"172fb5f7d242ba93208647c01d348268702df585","subject":"s3api: Support GET/HEAD request with PartNumber","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/172fb5f7d242ba93208647c01d348268702df585"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-12-18 15:29:12.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-12-19 16:33:12.000000000","tz":0},"subject":"more s3api part-num tests","message":"more s3api part-num tests\n\nChange-Id: Ic907321a53e518ccf6a4acaf82f82c6698a40b90\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/25ca6f970bb1084ab027d24f5644e3c2383fa1e9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/25ca6f970bb1084ab027d24f5644e3c2383fa1e9"}]},"branch":"refs/heads/master"},"b41f59bbb2b3d84ff19ca889cbc140b73fa06564":{"kind":"REWORK","_number":6,"created":"2023-12-19 16:46:20.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/47/903847/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/47/903847/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/903847/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/903847/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/903847/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/47/903847/6"}}},"commit":{"parents":[{"commit":"172fb5f7d242ba93208647c01d348268702df585","subject":"s3api: Support GET/HEAD request with PartNumber","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/172fb5f7d242ba93208647c01d348268702df585"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-12-18 15:29:12.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-12-19 16:45:53.000000000","tz":0},"subject":"more s3api part-num tests","message":"more s3api part-num tests\n\nChange-Id: Ic907321a53e518ccf6a4acaf82f82c6698a40b90\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b41f59bbb2b3d84ff19ca889cbc140b73fa06564"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b41f59bbb2b3d84ff19ca889cbc140b73fa06564"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
