)]}'
{"id":"openstack%2Fswift~904131","triplet_id":"openstack%2Fswift~master~Ie835e6e6bb384d6ab97b806fd34aa79bf6042a94","project":"openstack/swift","branch":"master","topic":"extend-slo-attrs","attention_set":{},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2024-01-02 15:44:03.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Ie835e6e6bb384d6ab97b806fd34aa79bf6042a94","subject":"sq: cleanup test_obj.py","status":"ABANDONED","created":"2023-12-20 15:01:17.000000000","updated":"2024-01-02 15:44:03.000000000","total_comment_count":11,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"b9488cb21245174a96e9ed297432c9a436f1c04e","_number":904131,"virtual_id_number":904131,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:check","value":1,"date":"2023-12-20 16:17:03.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"recommended":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"all":[{"value":1,"date":"2023-12-28 09:21:14.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":2,"date":"2023-12-20 17:44:02.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-12-20 16:17:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-12-20 17:44:02.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"},{"updated":"2023-12-28 09:21:14.000000000","updated_by":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"reviewer":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"state":"REVIEWER"}],"messages":[{"id":"99d8c6c71f5406a66bd7e5e9180fc1a0bb9b7826","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-12-20 15:01:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"327079e1c6b773f89afd38fe9fed5af65d5ec0f0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-20 16:17:03.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5311f0d308d74a14bc3b3460b44406d5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b00e1463abf44f0b8d02991d3f71b949 : SUCCESS in 9m 03s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/1dcb7a1f778b4394b7920827206614fd : SUCCESS in 57m 03s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/624d95d965ac4f9f8cb49dd6687ae64d : SUCCESS in 18m 48s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/33ceda36899c4c2cb5fc60528a41ddab : SUCCESS in 18m 49s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e88c0a6b6ae44b9d9eccb73b7440fa25 : SUCCESS in 23m 24s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/91e3d5d93ee34896b723e203c344d5fe : SUCCESS in 9m 27s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/1525f0384fc647eda231610c8e5b394c : SUCCESS in 6m 41s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/4167f19a23a04cefbd94f630f56449ad : SUCCESS in 19m 28s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/df014618dd0e44a1b3c56faca6dee013 : SUCCESS in 16m 15s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/6d405aa93de64d07a4a9ac206d84e873 : SUCCESS in 16m 23s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/a9edf0cafd504142b639663efffb1bca : SUCCESS in 20m 44s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/f883e882d8a844a78d6d43bf7262a9ca : SUCCESS in 16m 34s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/9291293602604ee2ac3f0f847ddb9bb7 : SUCCESS in 20m 40s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/11b3d404f7434394a9129184b9e68aa3 : SUCCESS in 18m 50s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/e9f5bca47ffa43d1a91567407284b698 : SUCCESS in 17m 39s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/02d1954cbc334b739b5df57903e203ea : SUCCESS in 18m 57s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/9219ab23d3474d8b8e59e6bbf52b6b04 : SUCCESS in 15m 44s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/7b4d736e87c94ef6afa7c51a6ae041b8 : SUCCESS in 16m 39s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/6fa61cf24f034362a4215561029bf4a3 : SUCCESS in 1h 08m 36s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/31ba2aa54f1d4ad18cbb32e40d04cd71 : SUCCESS in 58m 12s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7c9b38462c464fcaa0c42db781334bb0 : SUCCESS in 22m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b3341b8e0dbb4326b980fc2d3063fdb9 : SUCCESS in 5m 30s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/fb2f4adcc33941de86e3a18d4f1f7efe : SUCCESS in 21m 40s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"80b44eb6e9448348108502c3161472f07aeb256b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-20 16:46:46.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5dc9c5f33c0a4c05b0be8e4b2f0fb95d\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b5fc59cd920541dd8205f7acd0e2ae98 : SUCCESS in 1h 27m 22s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/075ed0adfefd436790605a7b091b3294 : SUCCESS in 17m 44s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/3ced2527acc14e298a10abedc440a7a9 : TIMED_OUT in 1h 31m 20s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/626aa5622c164172b793d0b2b881e531 : SUCCESS in 23m 06s","accounts_in_message":[],"_revision_number":1},{"id":"d5554dd8e0c28cbc0a45a98afd01b99f3c25a94c","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2023-12-20 17:44:02.000000000","message":"Patch Set 1: Code-Review+2\n\n(6 comments)","accounts_in_message":[],"_revision_number":1},{"id":"6910183746b0e531ff1aa0de5932c4bed3712130","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2023-12-21 17:09:43.000000000","message":"Patch Set 1:\n\n(5 comments)","accounts_in_message":[],"_revision_number":1},{"id":"aa683a79150cc3208bf96a46f8c0651534a4f9bf","author":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"date":"2023-12-28 09:21:14.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"b9488cb21245174a96e9ed297432c9a436f1c04e","tag":"autogenerated:gerrit:abandon","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-01-02 15:44:03.000000000","message":"Abandoned\n\nnow part of https://review.opendev.org/c/openstack/swift/+/904234","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"eb05494b7c0964798dd9e9918f5367784a8a5a16","revisions":{"eb05494b7c0964798dd9e9918f5367784a8a5a16":{"kind":"REWORK","_number":1,"created":"2023-12-20 15:01:17.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/31/904131/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/31/904131/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/904131/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/904131/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/904131/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/31/904131/1"}}},"commit":{"parents":[{"commit":"2e9d68a0fd67db02696806b56fce1825041640a3","subject":"s3api: Support GET/HEAD request with ?partNumber","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2e9d68a0fd67db02696806b56fce1825041640a3"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-12-20 15:00:33.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-12-20 15:00:33.000000000","tz":0},"subject":"sq: cleanup test_obj.py","message":"sq: cleanup test_obj.py\n\nChange-Id: Ie835e6e6bb384d6ab97b806fd34aa79bf6042a94\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/eb05494b7c0964798dd9e9918f5367784a8a5a16"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/eb05494b7c0964798dd9e9918f5367784a8a5a16"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
