)]}'
{"id":"openstack%2Fswift~904137","triplet_id":"openstack%2Fswift~master~I4305e5a1853ce96dd99d10061f551a5c073387cc","project":"openstack/swift","branch":"master","topic":"extend-slo-attrs","attention_set":{},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2024-01-02 11:40:32.000000000","reason":"\u003cGERRIT_ACCOUNT_7847\u003e replied on the change","reason_account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}}},"hashtags":[],"change_id":"I4305e5a1853ce96dd99d10061f551a5c073387cc","subject":"WIP s3api: enforce configured max part num","status":"ABANDONED","created":"2023-12-20 16:16:05.000000000","updated":"2024-01-02 15:43:37.000000000","total_comment_count":21,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"f66fb06749d0c5c3d276627a857b77e2262180f8","_number":904137,"virtual_id_number":904137,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:check","value":1,"date":"2023-12-20 17:32:48.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":2,"date":"2023-12-21 17:06:18.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":-1,"date":"2023-12-21 15:46:07.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-12-20 17:23:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-12-20 17:32:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-12-21 17:06:18.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"}],"messages":[{"id":"025dde50ab0aa83144c51e7f9d462def48bb496b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-12-20 16:16:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8534d0c76bcf565df5bea8b903d4b5d91fd4f016","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-20 17:23:26.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/04dfaee45a6f44cb81716e99dc8021f6\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/776d6d87e9f04e1ebb7a1d93f7518aa5 : SUCCESS in 16m 57s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/cafe1cdecf424c4392dd668036e8f187 : SUCCESS in 1h 04m 30s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/667a66c543a4417485ce7a4f81b35d9e : SUCCESS in 22m 15s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/507e914c129c4bf1871958a60f868610 : SUCCESS in 22m 28s","accounts_in_message":[],"_revision_number":1},{"id":"84d499f0894513c45265b6b3dce72d776372b30b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-20 17:32:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/89561da6e35b40079bab86b27a23fe57\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c6f6a186c611401aaedda2adfdade556 : SUCCESS in 8m 48s\n- grenade https://zuul.opendev.org/t/openstack/build/665e49df82774dcaa797e6dcea7a6fd0 : SUCCESS in 46m 24s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/d61464eee01c438eb5ae359f3aa5dcec : SUCCESS in 59m 35s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/bd54646fd5de4490a91aef78ae1940a4 : SUCCESS in 51m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e78ad3b638cc4224a03330964e52e289 : SUCCESS in 1h 03m 10s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/029b6f828eb348768a9307c190917d58 : SUCCESS in 20m 00s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3948b6f535e64031aa010354308c40c0 : SUCCESS in 20m 01s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/212f835868bb492b8c5a6a8f416b6d46 : SUCCESS in 20m 18s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/54bc93bfd6134f34a5ccfcf89f7fb129 : SUCCESS in 9m 02s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/42b4bec9ecca4c7f868140bfdbb90ab2 : SUCCESS in 6m 27s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/83225000e2ff49098e3f24767d92d394 : SUCCESS in 18m 28s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/8de8f53496934e8983a4a733346cb8f0 : SUCCESS in 19m 03s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/690a0da8a56841e89c7a82ee3f493a09 : SUCCESS in 16m 40s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/b31a3ba0b96e4846863a168f8935b344 : SUCCESS in 16m 33s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/fbae008f0bce495c8394c1a2e2c369c2 : SUCCESS in 18m 20s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/a9a7caff5a7e43038778b2d49d4e202a : SUCCESS in 14m 55s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/cb4e1f795d0e4279b26cd6bd0aba371b : SUCCESS in 17m 31s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/67a030eb6b484ae384ced8f025f8c1e6 : SUCCESS in 18m 41s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/8a2460bd070249f8860907543e17e99c : SUCCESS in 18m 38s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/8611ec06e47e4e79b10eb82884cc42ff : SUCCESS in 13m 56s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/1e9bba2bc2924449acd26755ebaf993d : SUCCESS in 18m 15s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/5cc9748316834e928c2cacc57a6f73d0 : SUCCESS in 7m 33s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/e3f39975012e49c6aa7fab8ffa9f810a : SUCCESS in 19m 39s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/6697878e902f4a5f9f564304abf76512 : SUCCESS in 8m 20s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/5b61b71f6cc745e592598aedf76c4d69 : SUCCESS in 1h 07m 59s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/a8f78653c82842859640998032ec92cf : SUCCESS in 1h 06m 08s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/5a0367c43ca44b0abb8f1ae6304aec52 : SUCCESS in 40m 54s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/19afe28d8b694aa69eb8edc82cf05830 : SUCCESS in 36m 07s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8465b1177d234208ac70e8ce907a8e28 : SUCCESS in 14m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/699bf8fd44f44bb99635890a5ae94047 : SUCCESS in 4m 52s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/7ea28d8f4daf4c3a935e92b9cf5f0594 : SUCCESS in 21m 36s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/dcc12e854cbc4cb49686fbad8383bd9c : SUCCESS in 39m 44s","accounts_in_message":[],"_revision_number":1},{"id":"eb53eb9aef8a9778273fdd4f93b429f9a0dca8a7","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-12-21 15:46:07.000000000","message":"Patch Set 1: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ea4dad3fd7894df372874dc79bf4ea5c25176610","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2023-12-21 17:06:18.000000000","message":"Patch Set 1: Code-Review+2\n\n(10 comments)","accounts_in_message":[],"_revision_number":1},{"id":"09f600dac71d8e9c3a985c158df8d2bc5e075011","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2023-12-21 17:45:34.000000000","message":"Patch Set 1:\n\n(9 comments)","accounts_in_message":[],"_revision_number":1},{"id":"86741300c86790ad7e79d8b559861c16455056b7","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-01-02 11:40:32.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f66fb06749d0c5c3d276627a857b77e2262180f8","tag":"autogenerated:gerrit:abandon","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-01-02 15:43:37.000000000","message":"Abandoned\n\nnow part of https://review.opendev.org/c/openstack/swift/+/904234","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"162224ed1e424e05dfa913756e9497d94fcf2102","revisions":{"162224ed1e424e05dfa913756e9497d94fcf2102":{"kind":"REWORK","_number":1,"created":"2023-12-20 16:16:05.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/37/904137/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/37/904137/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/37/904137/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/37/904137/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/37/904137/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/37/904137/1"}}},"commit":{"parents":[{"commit":"5232ee2e04e86fc6afef403cb6d813f62cc9aae1","subject":"s3api: unify partNumber validation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5232ee2e04e86fc6afef403cb6d813f62cc9aae1"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-12-20 16:11:06.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-12-20 16:11:06.000000000","tz":0},"subject":"WIP s3api: enforce configured max part num","message":"WIP s3api: enforce configured max part num\n\nAWS has a limit of 10000 parts but swift\u0027s limit is configurable and\ndefaults to 1000.\n\nChange-Id: I4305e5a1853ce96dd99d10061f551a5c073387cc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/162224ed1e424e05dfa913756e9497d94fcf2102"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/162224ed1e424e05dfa913756e9497d94fcf2102"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":["label:Workflow\u003dMIN"],"failing_atoms":["label:Workflow\u003dMAX"],"atom_explanations":{}}}]}
