)]}'
{"id":"openstack%2Fswift~904234","triplet_id":"openstack%2Fswift~master~Ic5f54da3a4a0dba9acec73f7f46beac7ce7aac24","project":"openstack/swift","branch":"master","topic":"extend-slo-attrs","attention_set":{},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2024-01-05 15:58:11.000000000","reason":"Change was abandoned"},"32761":{"account":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"last_update":"2024-01-05 15:58:11.000000000","reason":"Change was abandoned"},"1179":{"account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"last_update":"2024-01-05 15:58:11.000000000","reason":"Change was abandoned"},"32238":{"account":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"last_update":"2024-01-05 15:58:11.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Ic5f54da3a4a0dba9acec73f7f46beac7ce7aac24","subject":"s3api: unify enforcement of max part number and handle exceptions","status":"ABANDONED","created":"2023-12-21 17:56:21.000000000","updated":"2024-01-05 15:58:11.000000000","total_comment_count":12,"unresolved_comment_count":7,"has_review_started":true,"meta_rev_id":"e4b95f47cee1a90debc6ff3419972bcdcc005917","_number":904234,"virtual_id_number":904234,"owner":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:check","value":1,"date":"2024-01-04 00:10:45.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}],"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-12-21 17:56:21.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"CC"},{"updated":"2023-12-21 18:19:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-12-21 19:17:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-12-28 09:21:09.000000000","updated_by":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"reviewer":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"state":"REVIEWER"},{"updated":"2023-12-29 07:08:29.000000000","updated_by":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"reviewer":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"state":"REVIEWER"},{"updated":"2024-01-02 15:42:22.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"REVIEWER"},{"updated":"2024-01-03 22:31:00.000000000","updated_by":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"CC"}],"messages":[{"id":"6d829c8ddcefc0fbcb8a6eb631cd8d09e1d8f7ea","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2023-12-21 17:56:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a3671221782f6105c8c249f9972068ba27b18d54","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2023-12-21 18:09:35.000000000","message":"Patch Set 1: Code-Review+2\n\n(6 comments)","accounts_in_message":[],"_revision_number":1},{"id":"8cd8f7a183481745d9b2afe429a15078c4622c6a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-21 18:19:29.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/de3116f761f14ef38ab49cefb58d3b72\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/6b58984356b342ecb03ffbb05140c50a : SUCCESS in 14m 46s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/72f7a864e63f492d964bf841257997c3 : SUCCESS in 14m 43s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/c03cb90de2c54ee2bfd9c01c590f3be3 : SUCCESS in 19m 50s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/00018548b4cf42b4b3c629312c305ff1 : SUCCESS in 22m 05s","accounts_in_message":[],"_revision_number":1},{"id":"97ef8c7bdd8d1b847620bdf86d02b75d51c3811f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-21 19:17:27.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ad9ff750867b436b815de2e58c24c450\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e26a034ae2c04b26ba09d8c6c0ef82c3 : SUCCESS in 8m 18s\n- grenade https://zuul.opendev.org/t/openstack/build/e86e1e5ddc354064b44e6fa17e597fa9 : SUCCESS in 1h 13m 55s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/fb4b9a48d15542a8950d519866e657aa : SUCCESS in 56m 21s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/67540d09ae184f90ac7f580ab66f2899 : SUCCESS in 51m 45s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/47f92d9b0f5d4111a91601a3fa3a2fdc : SUCCESS in 1h 09m 36s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/bbe4f089414f42b3b1ca81f8b7c3f2ab : SUCCESS in 19m 51s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4cab81611d0445cdb58aed7beece2ab1 : SUCCESS in 18m 08s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d1baf62a55724519afe05376b8d0c4c0 : SUCCESS in 19m 49s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/8976dc1b407a4f228f606ae9b8e79b23 : SUCCESS in 9m 47s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/2fa664aa8bb249dc99129faf7a46d115 : SUCCESS in 7m 10s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/a7b8cab3bb04461fbf173b370c8359f3 : SUCCESS in 15m 55s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/369c69ca470f494b9968bcb7716161a1 : SUCCESS in 18m 01s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/7cba617e55c74eafa2c3cc494ec6ffc0 : SUCCESS in 16m 14s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/ba1069d35a0442359a7399212866a3d8 : SUCCESS in 13m 56s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/dd5d5ad914bb4d4788df2671a500e6c6 : SUCCESS in 16m 48s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/1d74dd3e49ad49fea68363bfab073311 : SUCCESS in 20m 48s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/e1f19dc8007d430382365fbd65e6004c : SUCCESS in 18m 33s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/1740907b1b974a24905b97beeb844f81 : SUCCESS in 23m 10s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/a77f5d73eb844e3e8bb54de8911618d3 : SUCCESS in 20m 39s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/959635b972264fb5a22d890665353e1c : SUCCESS in 14m 41s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/29f63ba1b77f46ca91a05750abb1a3f8 : SUCCESS in 21m 12s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/f7ac4a6c84b54de88e1b568e4f904198 : SUCCESS in 9m 54s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/99573e8b86d64a90a15ac0830af07033 : SUCCESS in 18m 25s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/2162492217de480dbe580e8a88e72b3c : SUCCESS in 8m 34s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/4c4a55ef3fc444d1bf0f98206582f764 : SUCCESS in 1h 12m 58s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/3d6580ba895d467ba682d41004dc5e15 : SUCCESS in 47m 41s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/d468b4d2b5f14993b48853bdbba2f750 : SUCCESS in 44m 22s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/329da0b1264148089cd4602b7180e9aa : SUCCESS in 48m 36s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7430d7d8b26a4e589aefde0768525ad9 : SUCCESS in 15m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b9923e29f79a43059ec438e43b196d6b : SUCCESS in 4m 40s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/9eec7d47fb7246e9b1d168860f060a51 : SUCCESS in 20m 40s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/729c93f2851c405789e396d70384b668 : SUCCESS in 51m 52s","accounts_in_message":[],"_revision_number":1},{"id":"59dab8112921ef80c98c3fb1c82cd412b333d4fc","author":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"date":"2023-12-28 09:21:09.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"b625103c22644dbab4e32e6ba6435ae20985fab3","author":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"date":"2023-12-29 07:08:29.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"2954672d3374f6ca35bfd0e6423caba79a92e3b0","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-01-02 15:42:22.000000000","message":"Patch Set 1: Code-Review+1\n\n(6 comments)","accounts_in_message":[],"_revision_number":1},{"id":"435e1a670ad00e91eb682b5f9ed60bdba25df261","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"real_author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-01-03 19:19:49.000000000","message":"Patch Set 2: Patch Set 1 was rebased on behalf of \u003cGERRIT_ACCOUNT_1179\u003e\n\nCopied Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}],"_revision_number":2},{"id":"f078d4472126f7476eaa563e05b012d7b7c61cf4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-03 20:19:25.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4ec5d7fa7da04eb895a1993f60aa8622\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0f9b0af135284c6ba57089a16aaab8f7 : SUCCESS in 16m 44s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/3a1e52b3bff1447a989b5bc1f5de9e53 : SUCCESS in 50m 00s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/98407be6ca5b4f73bb4876bf3df08a1a : SUCCESS in 24m 09s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/55d2d2f0f1b54d09a351f07b1b1bc9ad : SUCCESS in 57m 43s","accounts_in_message":[],"_revision_number":2},{"id":"da78873f8d6a19a6f4057b1339847cfbcdb6ebe4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-03 20:30:14.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ca0123a1fdf5416081ec13fc881c2342\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/00c23b16cb87428eae378426283e56ae : SUCCESS in 10m 28s\n- grenade https://zuul.opendev.org/t/openstack/build/5a4c779b84894b9e8947d6f3d6e70828 : SUCCESS in 1h 00m 33s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/64deef542bb04c51a1b0f9768c2a3d2f : SUCCESS in 59m 11s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/b8012860ced94281ae448652dc2ba316 : SUCCESS in 44m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e39f67e92f804987bb13af39bb2d2536 : SUCCESS in 1h 04m 33s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7c8a481b7eff4c62ad1393bacbfc573a : SUCCESS in 19m 49s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b9216a8373664494ad61887469d2dd0b : SUCCESS in 20m 34s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a07eb9ca3fbd49f58c6fb9fca52bd02a : SUCCESS in 20m 29s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/8126a827722d4b359be28a13bda340c0 : SUCCESS in 11m 04s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/75872a4d2e414208801a6b465adb627c : SUCCESS in 6m 52s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/3a042fe2e0a443f08fd91ea4f8552a40 : SUCCESS in 16m 46s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/f539ea135d2d41f9b6a60e56e8449a88 : SUCCESS in 15m 59s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/05cb5230aacf4b4a8c5f100f3d0e6fa1 : SUCCESS in 13m 03s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/bafe44a5171341d5901ec7c6964b19ac : SUCCESS in 18m 00s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/2665e88af61a4ad2981fda5d769ab007 : SUCCESS in 15m 19s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/9b07fb44a3a942168f9d6227a86b76cf : SUCCESS in 21m 06s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/ffea849b3e024973a6ea2a10bf7de7c2 : SUCCESS in 20m 06s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/e441ed37e7894707a2ae2f47fa5c0689 : SUCCESS in 23m 29s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/c2e7c092652145a2b91f1fe8e870d962 : SUCCESS in 19m 46s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/7c92709fb38f4ed58efc590b4cde0d9f : SUCCESS in 17m 04s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/ff90bdee764a445890abade868be2468 : SUCCESS in 18m 40s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/c0ea91b18b3846bfbd2bed262c7f9861 : SUCCESS in 6m 59s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/690c888589f447b9821785c6e4405abe : SUCCESS in 17m 25s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/e6a181d2c2174310a621b4075e5d62fb : SUCCESS in 9m 05s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/7c8858c752e34d749c4cde41527279ac : SUCCESS in 56m 45s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/f296fa56755d42d583de33dd572fbbc8 : SUCCESS in 1h 03m 01s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/480904df1b36472d98b0b0ce1b6d1da4 : SUCCESS in 57m 18s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/7bc3c750b0694b909425c78b5b8eb28c : SUCCESS in 51m 40s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/34454995f98944b5b006799dfe2eae6b : SUCCESS in 16m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7d9289af2ff04a949e0a98a46873dc68 : SUCCESS in 5m 05s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/a119eced141140469e2e7289905fee38 : SUCCESS in 21m 00s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/83169b381acf4d82945f0e18e6dd8fe4 : SUCCESS in 49m 46s","accounts_in_message":[],"_revision_number":2},{"id":"c9fcb7279fbd10c4a089928698e51a75812d09c1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-01-03 22:31:00.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"61350708b41d0a90142e426497cd5348727ad95f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-01-03 22:31:38.000000000","message":"Patch Set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"97fada7962c148dfbcb5d6e7c1adcf33c5498b8c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-03 22:53:32.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/551648fe4ba0413490ee9b0dba7c8ad4\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a822eae5a2fb4331b777cfebe32fb077 : SUCCESS in 14m 49s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/3d241ead85f04ae093c444c1aaa5e813 : SUCCESS in 14m 40s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/ab418129ac5b4694a299f78366f7eed8 : SUCCESS in 19m 56s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/4790517fe342405e818b7c51ad6c73b9 : SUCCESS in 21m 26s","accounts_in_message":[],"_revision_number":4},{"id":"04df214be370d48f645a2cad09a981ecca3989b0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-04 00:10:45.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/67a74063b4e240928377651efc4691db\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/14606e9ee6964d3298f151b9379ea226 : SUCCESS in 9m 07s\n- grenade https://zuul.opendev.org/t/openstack/build/b6167bcb028b4f3eb223429726759a42 : SUCCESS in 1h 02m 55s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/62d8f7dabdad43528d988aa962e380f2 : SUCCESS in 46m 17s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/b34888fb73624a129ba68120749e25da : SUCCESS in 33m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0263bb4b4ddc4d6db3ca4a2db506bdba : SUCCESS in 1h 32m 08s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/092bf7844ac9487c93bda164debe3d8f : SUCCESS in 18m 06s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7d58f78091014906a7405f130fda758a : SUCCESS in 20m 21s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/fae694f408824c6d985dadb56235cff8 : SUCCESS in 16m 08s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/3cbc40549af64224bf02e12b88035a67 : SUCCESS in 11m 45s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/ab7aeef6b64d4185bfb000c2cd4ce5d4 : SUCCESS in 5m 40s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/089aa9d4380348dcbb06d6fe2ab5928d : SUCCESS in 18m 19s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/94b8b122edc840f58053cd0e5a394716 : SUCCESS in 17m 31s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/fc315e7bf4474fcb85611d1f861847ed : SUCCESS in 16m 28s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/1065776aca7b4eadb9b103844370fba9 : SUCCESS in 15m 29s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/b839bbe6eff743d1be80fcb0ec061408 : SUCCESS in 17m 16s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/12b854f417cd4c1fbb6f4086e3011c03 : SUCCESS in 21m 25s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/578741dd2d5a4cdca8897bafd7c08faf : SUCCESS in 19m 16s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/bce316b29d1b4ddbbd07e763273bc2d6 : SUCCESS in 18m 18s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/fb2094f5111f4a99b555ecdc11083b3c : SUCCESS in 16m 32s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/c6815e6fc3d54239b2475873b512ad5a : SUCCESS in 15m 57s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/67fbdd8b9bc84d8cafe6f14ca946821c : SUCCESS in 16m 08s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/258f3fe6330d410caaa4752d9d4c5a1e : SUCCESS in 7m 38s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/90dcff5192614c44a22df4e4c3ecdc51 : SUCCESS in 16m 33s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/4998ea4921774d69825c12abc2007cba : SUCCESS in 8m 24s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/d6bb4da6592c4a08a69328f380cfe271 : SUCCESS in 57m 04s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/59838d29716642088726cf9a79cace90 : SUCCESS in 51m 29s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/82b26caeafba4b488cd6a719c9334d83 : SUCCESS in 54m 37s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/43475719246f4623918da9b7ed2633c8 : SUCCESS in 38m 56s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/814a4964fead4696bcda2f23ba06c802 : SUCCESS in 16m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e4e6a9e3c8674412b06dfd641999ccff : SUCCESS in 4m 19s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/f985f5efdd3d4bc4ad2301bc341a2954 : SUCCESS in 20m 06s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bf3e714b03d74f21ae01867a7238c3cd : SUCCESS in 40m 06s","accounts_in_message":[],"_revision_number":4},{"id":"e4b95f47cee1a90debc6ff3419972bcdcc005917","tag":"autogenerated:gerrit:abandon","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-01-05 15:58:11.000000000","message":"Abandoned\n\nsquashed into parent","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"57645005a6d7834c9d8eee0fa2dd8ece759b9da8","revisions":{"47ba17fc7836ea5193ea7bef05e442833ea1490e":{"kind":"REWORK","_number":1,"created":"2023-12-21 17:56:21.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/34/904234/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/34/904234/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/34/904234/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/34/904234/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/34/904234/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/34/904234/1"}}},"commit":{"parents":[{"commit":"2e9d68a0fd67db02696806b56fce1825041640a3","subject":"s3api: Support GET/HEAD request with ?partNumber","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2e9d68a0fd67db02696806b56fce1825041640a3"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-12-20 15:00:33.000000000","tz":0},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2023-12-21 17:44:34.000000000","tz":-360},"subject":"s3api: unify enforcement of configured max part num","message":"s3api: unify enforcement of configured max part num\n\nUse a single method to validate partNumber.\n\nAWS has a limit of 10000 parts but swift\u0027s limit is configurable and\ndefaults to 1000.\n\nChange-Id: Ic5f54da3a4a0dba9acec73f7f46beac7ce7aac24\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/47ba17fc7836ea5193ea7bef05e442833ea1490e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/47ba17fc7836ea5193ea7bef05e442833ea1490e"}]},"branch":"refs/heads/master"},"01842843d1a9cb1d5a2d3584ad4c8c0e973519a0":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-01-03 19:19:49.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"real_uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/34/904234/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/34/904234/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/34/904234/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/34/904234/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/34/904234/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/34/904234/2"}}},"commit":{"parents":[{"commit":"67cdbb04d9bfa55f5218b92c8cde431152798298","subject":"s3api: Support GET/HEAD request with ?partNumber","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/67cdbb04d9bfa55f5218b92c8cde431152798298"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-12-20 15:00:33.000000000","tz":0},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-01-03 19:19:49.000000000","tz":0},"subject":"s3api: unify enforcement of configured max part num","message":"s3api: unify enforcement of configured max part num\n\nUse a single method to validate partNumber.\n\nAWS has a limit of 10000 parts but swift\u0027s limit is configurable and\ndefaults to 1000.\n\nChange-Id: Ic5f54da3a4a0dba9acec73f7f46beac7ce7aac24\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/01842843d1a9cb1d5a2d3584ad4c8c0e973519a0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/01842843d1a9cb1d5a2d3584ad4c8c0e973519a0"}]},"branch":"refs/heads/master","description":"Rebase"},"fb47bcf4d53f32cf0eba8559a08aacbb8f0ea6dd":{"kind":"REWORK","_number":3,"created":"2024-01-03 22:31:00.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/34/904234/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/34/904234/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/34/904234/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/34/904234/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/34/904234/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/34/904234/3"}}},"commit":{"parents":[{"commit":"67cdbb04d9bfa55f5218b92c8cde431152798298","subject":"s3api: Support GET/HEAD request with ?partNumber","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/67cdbb04d9bfa55f5218b92c8cde431152798298"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-12-20 15:00:33.000000000","tz":0},"committer":{"name":"indianwhocodes","email":"nairashwin952013@gmail.com","date":"2024-01-03 22:27:29.000000000","tz":-480},"subject":"This change unifies the enforcement of the configured maximum part number using a single method to validate partNumber. While AWS has a limit of 10000 parts, Swift\u0027s limit is configurable and defaults to 1000.","message":"This change unifies the enforcement of the configured maximum\npart number using a single method to validate partNumber. While\nAWS has a limit of 10000 parts, Swift\u0027s limit is configurable\nand defaults to 1000.\n\nAdditionally, if a Multi-Part Upload (MPU) exists with\na number of parts (N) greater than the configured\nmax_upload_part_number (M), we allow an object GET or HEAD\nto have N \u003e\u003d partNumber \u003e M. If a GET or HEAD to the MPU has\npartNumber \u003e N, then the error response will indicate that the\nmax is N, not M.\n\nChange-Id: Ic5f54da3a4a0dba9acec73f7f46beac7ce7aac24\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fb47bcf4d53f32cf0eba8559a08aacbb8f0ea6dd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fb47bcf4d53f32cf0eba8559a08aacbb8f0ea6dd"}]},"branch":"refs/heads/master"},"57645005a6d7834c9d8eee0fa2dd8ece759b9da8":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2024-01-03 22:31:38.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/34/904234/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/34/904234/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/34/904234/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/34/904234/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/34/904234/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/34/904234/4"}}},"commit":{"parents":[{"commit":"67cdbb04d9bfa55f5218b92c8cde431152798298","subject":"s3api: Support GET/HEAD request with ?partNumber","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/67cdbb04d9bfa55f5218b92c8cde431152798298"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-12-20 15:00:33.000000000","tz":0},"committer":{"name":"ASHWIN A NAIR","email":"nairashwin952013@gmail.com","date":"2024-01-03 22:31:38.000000000","tz":0},"subject":"s3api: unify enforcement of max part number and handle exceptions","message":"s3api: unify enforcement of max part number and handle exceptions\n\nThis change unifies the enforcement of the configured maximum\npart number using a single method to validate partNumber. While\nAWS has a limit of 10000 parts, Swift\u0027s limit is configurable\nand defaults to 1000.\n\nAdditionally, if a Multi-Part Upload (MPU) exists with\na number of parts (N) greater than the configured\nmax_upload_part_number (M), we allow an object GET or HEAD\nto have N \u003e\u003d partNumber \u003e M. If a GET or HEAD to the MPU has\npartNumber \u003e N, then the error response will indicate that the\nmax is N, not M.\n\nChange-Id: Ic5f54da3a4a0dba9acec73f7f46beac7ce7aac24\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/57645005a6d7834c9d8eee0fa2dd8ece759b9da8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/57645005a6d7834c9d8eee0fa2dd8ece759b9da8"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
