)]}'
{"id":"openstack%2Fswift~904325","triplet_id":"openstack%2Fswift~master~Ic885e8bde7c1bbb3d93d032080b591db1de80970","project":"openstack/swift","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-01-03 20:18:58.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ic885e8bde7c1bbb3d93d032080b591db1de80970","subject":"Document allowed_digests for formpost middleware","status":"MERGED","created":"2023-12-25 08:19:07.000000000","updated":"2024-01-03 20:20:06.000000000","submitted":"2024-01-03 20:18:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"904325","meta_rev_id":"658fdf0ed97bbff36060291bfee75c3e071aa618","_number":904325,"virtual_id_number":904325,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-01-03 20:18:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"recommended":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-12-26 01:52:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"value":2,"date":"2024-01-03 18:53:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"value":1,"date":"2024-01-03 18:53:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-12-25 09:01:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-12-25 09:33:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-12-26 01:52:03.000000000","updated_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"reviewer":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"state":"REVIEWER"},{"updated":"2024-01-03 18:53:05.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"}],"messages":[{"id":"169821b443b5763849a12ea95657a06238c925f7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-12-25 08:19:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"811429786180d7e36230b17e0384f8c3dd4052bf","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-25 09:01:11.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1313933a109a460b8869209a3fd1471f\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d19656edd93d4b78a9644c5b017df72a : SUCCESS in 39m 39s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7ea53355e5774fb9a0ded9cec6a57f77 : SUCCESS in 40m 11s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/4bffb36109304adeb1a85a8508ea3279 : SUCCESS in 22m 35s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/d72f69962b194c80b95eb2b0fd9943eb : SUCCESS in 19m 28s","accounts_in_message":[],"_revision_number":1},{"id":"cda0a040fa24aff7a144b5981bb237ba1750245d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-25 09:33:41.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d23ac9bd9a9c4971a981017d99be11bf\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1f3014e81b9344faacb1c86d6848168a : SUCCESS in 9m 00s\n- grenade https://zuul.opendev.org/t/openstack/build/2fbd3d38b0e6498a976b28bbc599a7e3 : SUCCESS in 1h 05m 52s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/36c0ce23396c4472916a36a37ff96541 : SUCCESS in 40m 20s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/d6c17ab05e03469da214cfc800823640 : SUCCESS in 52m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7ad27728162c421fb1a372f890c17e71 : SUCCESS in 1h 03m 42s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9d99d849771e4c0586682c0243a8040c : SUCCESS in 14m 58s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b03c8204611344569361350384ba7819 : SUCCESS in 19m 51s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/caf77669404d47a69a81249848daf140 : SUCCESS in 19m 36s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/042d4c03805c4ede9fc97a099c1f2815 : SUCCESS in 9m 53s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/3788c13a4de94114b915a50b4966953a : SUCCESS in 5m 04s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/4cdbfa728a4e427988f4250378d46c4b : SUCCESS in 18m 06s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/709ef32ce8bf4cb0b8005503bbcb683c : SUCCESS in 17m 25s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/baa619e1a86d491da93b4a27b9679cb8 : SUCCESS in 13m 41s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/06116c12533c45abbf476c9a624fbab6 : SUCCESS in 15m 38s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/7fef938a9f55419bb382719c31373dc0 : SUCCESS in 14m 54s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/fdccc2fe258c49bd8c44875f84a292f8 : SUCCESS in 18m 19s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/8e09b8108f57427d960dd1e7b06d09db : SUCCESS in 17m 02s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/25b99f2db7734dbd9682f528bd71e0d7 : SUCCESS in 18m 49s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/960cdf3a4e6b4df0b0911a186994d291 : SUCCESS in 14m 37s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/27455370da974d28bbc7c5e9b1de9b31 : SUCCESS in 15m 36s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/6150c1bf6a2e4fe3885219479fff2320 : SUCCESS in 15m 08s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/0b47988dbe2c4354bb47852b952045c5 : SUCCESS in 5m 59s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/8a4263e4cbb549999d5873cd008d3713 : SUCCESS in 19m 39s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/60f2313ab14b4dae9f790128da8b8ee2 : SUCCESS in 8m 30s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/11a413b70b24427b85338e78c074138d : SUCCESS in 1h 08m 35s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/fac8077f058a425c9d0f8f189b2976d1 : SUCCESS in 1h 02m 59s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/7cdd76e396d347f389cdcf5e1515a9a5 : SUCCESS in 33m 33s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/14dfed5915004a3e8a7ed2c1066936cd : SUCCESS in 59m 37s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/928b4abd809e407fbd1bdbe01214fa6d : SUCCESS in 14m 08s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/5cd0fbdcaac04a8680346c2c77998173 : SUCCESS in 20m 08s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2e9683ad5e164c379c9a16ba0f0c74d2 : SUCCESS in 59m 44s","accounts_in_message":[],"_revision_number":1},{"id":"098d2df186f0ce352f713196a4d1526b1b24531e","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-12-26 01:52:03.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"0729de69f0d00409b1c2ce7cc6eb6f2ce870db24","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-01-03 18:53:05.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"195cff85a7bf01fe4e29c84ef45336e4dc84278d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-03 18:53:37.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"268df2460edb932af3cd79319baadd2c073d8704","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-03 20:18:57.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/67e214f5b823425fb97e8b957ed7f564\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1ec14d6d13f641678e1779267b5d9abe : SUCCESS in 7m 50s\n- grenade https://zuul.opendev.org/t/openstack/build/1323146af8094e6bacf25d13d47da854 : SUCCESS in 1h 16m 34s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/f3a33c8de5984320968a5e30f3744ca8 : SUCCESS in 1h 08m 55s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/a79957722f9948c49b852b5f4808ce4e : SUCCESS in 57m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/63ed17c5494c4deaa88d48f22003d219 : SUCCESS in 1h 04m 27s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/70b6b282174f40aab57fb12992b24eac : SUCCESS in 10m 46s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/a6461f669493470ea005772daad4bd4a : SUCCESS in 5m 51s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/212fd84956334d5fa54e4e3f828b7887 : SUCCESS in 18m 45s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/7ca44237761046b5a9c8df3cec84cfdb : SUCCESS in 15m 38s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/c63bc6966fa24b25a8ba36549d90add4 : SUCCESS in 16m 35s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/7a90741053244a9cafc25807b015205a : SUCCESS in 15m 03s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/c1e165570d3d45f885ccf6e399a02f2d : SUCCESS in 15m 21s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/8e7eacf9e171470f8b3dbd83b4830a88 : SUCCESS in 20m 22s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/ba79051d0e064eecaf96378059c95772 : SUCCESS in 17m 58s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/99eb1013436348fbbee0d09af4f3665b : SUCCESS in 19m 11s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/1b4b07e0e45b471596eddb34a7045ef6 : SUCCESS in 16m 48s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/57527b05d8db4c4bb6f0a2fcd700a360 : SUCCESS in 16m 27s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/96fcdf83369d441ab1db7a60a6c96d8c : SUCCESS in 17m 45s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/1593053b2941499bad940e25ae27a622 : SUCCESS in 7m 37s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/1986e601a70a40c4b890a3df7b5354e0 : SUCCESS in 7m 56s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/ea97ba9597b8448caf01ca9cbf565a51 : SUCCESS in 44m 01s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/f534e1fb4fb44a408a9eec6737843ed4 : SUCCESS in 54m 41s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/08ee20ddf2604f2c84ce2f6717c6067c : SUCCESS in 34m 50s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/6be5e0a527c94974bf0cf7798d3bbaff : SUCCESS in 40m 59s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/31b1a672c7c24ad78a612a7ee0fd2fff : SUCCESS in 13m 28s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c543a5711e4a4fa6acfc32e675f4bf53 : SUCCESS in 43m 20s","accounts_in_message":[],"_revision_number":1},{"id":"a489d9aaa5c5a089508ebcea13e563a09513bf9c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-03 20:18:58.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"658fdf0ed97bbff36060291bfee75c3e071aa618","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-03 20:20:06.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e6e4c4d79ead4f6796516f41a8eec467\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/311456b420f24052a91a77c6d3f2643f : SUCCESS in 52s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/376ef35f305b489897c9a4239030e6c3 : SUCCESS in 29s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/8398ff9d9ecf40cfb2b5177b79423107 : SUCCESS in 35s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"bd64748a030809553022234849786b003de29bfa","revisions":{"bd64748a030809553022234849786b003de29bfa":{"kind":"REWORK","_number":1,"created":"2023-12-25 08:19:07.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/25/904325/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/25/904325/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/25/904325/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/25/904325/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/25/904325/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/25/904325/1"}}},"commit":{"parents":[{"commit":"6f890d2ba92a2b2a83db791e7af3e0aad661b01a","subject":"proxy: move _get_shard_ranges to ObjectController","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6f890d2ba92a2b2a83db791e7af3e0aad661b01a"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-12-25 08:17:39.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-12-25 08:17:39.000000000","tz":540},"subject":"Document allowed_digests for formpost middleware","message":"Document allowed_digests for formpost middleware\n\nThe allowed_digests option were added to the formpost middleware in\naddition to the tempurl middleware[1], but the option was not added to\nthe formpost section in the example proxy config file.\n\n[1] 2d063cd61f6915579840a41ac0248a26085e0245\n\nChange-Id: Ic885e8bde7c1bbb3d93d032080b591db1de80970\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bd64748a030809553022234849786b003de29bfa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bd64748a030809553022234849786b003de29bfa"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
