)]}'
{"id":"openstack%2Fswift~904601","triplet_id":"openstack%2Fswift~master~Icae2f1e3e95f216d214636bd5a6d1f40aacab20d","project":"openstack/swift","branch":"master","topic":"py312","attention_set":{},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2024-02-06 09:24:49.000000000","reason":"\u003cGERRIT_ACCOUNT_7847\u003e replied on the change","reason_account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2024-02-06 10:16:53.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Icae2f1e3e95f216d214636bd5a6d1f40aacab20d","subject":"tests: Fix float expectations for py312","status":"MERGED","created":"2024-01-04 01:33:13.000000000","updated":"2024-02-06 10:18:08.000000000","submitted":"2024-02-06 10:16:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"904601-py312","meta_rev_id":"036dfe6315e0b6f581c11fac1d0967988bffe081","_number":904601,"virtual_id_number":904601,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-02-06 10:16:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":2,"date":"2024-02-06 09:24:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":1,"date":"2024-02-06 09:24:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-01-04 02:54:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-02-05 10:56:34.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"REVIEWER"}],"messages":[{"id":"545bccbff23941aad84897366206e517d13d0854","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-01-04 01:33:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3322ddef9a5bb8a4df6cd29855b5248384f9e014","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-01-04 01:35:12.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"00ce1b783eb7b9f8c72a1db32665e729e1d79f41","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-04 02:54:15.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d63b10f7a7544db482c1c8ece6069c3a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b5f88293a3f342598596d13d80f2c80e : SUCCESS in 8m 58s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/05503902e1f943d69b5a474bdf125799 : SUCCESS in 1h 11m 09s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b897163d6d4f4c8a86b871eb2170b1c0 : SUCCESS in 21m 16s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ebc19e8ef34644b5bc88bdf70eed8dd4 : SUCCESS in 15m 06s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/346cc9558e4e4784a78b983a51b49e54 : SUCCESS in 15m 51s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/391c9c08c0e6434e9b369da8219aa181 : SUCCESS in 9m 16s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/167686a712474e0b82ba4c277c89390f : SUCCESS in 5m 31s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/ac1d3231c19d4b3697cad10d71aec675 : SUCCESS in 19m 46s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/64a918b9534c4fff935a81a067e6899a : SUCCESS in 17m 10s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/f7b328791f2342a286e578dff18e7e4d : SUCCESS in 16m 06s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/1d6b431730e846af8119757c72f95b1c : SUCCESS in 17m 50s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/a562b3420e514d70910c1d018d451333 : SUCCESS in 16m 19s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/5ce65c2bd2794f83a30b474dd441c17b : SUCCESS in 20m 31s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/674abd402f524eec85a9ed4c540f1ff3 : SUCCESS in 16m 51s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/24d571b8c6c84ac2838074b06ee9ff1f : SUCCESS in 17m 14s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/51acba1dfedc4cdba3a09b507d70fac6 : SUCCESS in 15m 20s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/3612a0870a6a412da3a00385247e9996 : SUCCESS in 16m 34s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/d6ae7bc866fc4ec89eabb0aa9feef70a : SUCCESS in 15m 52s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/7bcb85a71f294d80aa143c1c6e98e0de : SUCCESS in 53m 55s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/eea4f124ff664b1489a5eda10662b305 : SUCCESS in 59m 02s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fec54208196245dd88ce2e7bf2e5f496 : SUCCESS in 15m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/30061743b5184a7a9dd3c1b59e7050b1 : SUCCESS in 5m 21s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/060ce355233f492289305dfc44a3b5ea : SUCCESS in 21m 35s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"ed44d7501f1c181dde34431306fc3f566ef59558","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-04 03:08:05.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d9ce9aa3501f4098be855201c7ac98d8\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/fa9bc77cfa3d43f2903ef4f1cc2c7328 : SUCCESS in 14m 17s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e79473ce41414eefbdf807693197b8ab : SUCCESS in 57m 33s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/1fd3fd97923d499388bf9106798b229e : SUCCESS in 22m 08s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/fded84a5dc9e48eaaa39fe62dd690a3c : TIMED_OUT in 1h 30m 58s","accounts_in_message":[],"_revision_number":2},{"id":"9ec72d3923d066dc07060d8d6126d6eae997b1d1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-01-24 20:40:30.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"a4a92f89bd089ff63317d1aa1d12365409e78d2a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-24 21:51:56.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/251ba24d74db4ad090676f20395c7728\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/71f283c280ac44149f8bfaa08cc28306 : SUCCESS in 9m 59s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d4e339ec05994ae5ad494a22a3c34dc0 : SUCCESS in 16m 44s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/aeaf99ce9f744eb8ba17a0262f911967 : SUCCESS in 24m 02s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/12fe6cab39b84fc3bae1497bb14f5019 : SUCCESS in 20m 09s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/0340f957cc5f47228f7d35107042748a : SUCCESS in 13m 08s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/37a44f5b150046abb9502346403f7ced : SUCCESS in 6m 18s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/039afd5378d0449d917afaffbcaf8eb1 : SUCCESS in 22m 00s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/1f990d13a66d4fc283f68844c07a0b24 : SUCCESS in 20m 52s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/ce41aaa4e9ac4e29a36d6976c56c5f7d : SUCCESS in 12m 54s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/de1b498b1f7940349a9affcaa3e6932b : SUCCESS in 17m 35s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/18f6c16d423d4690bf46a72040fad35e : SUCCESS in 18m 44s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/0a2d0e39e8cf47dcb915c7bf06123294 : SUCCESS in 19m 06s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/33f1f3a5baad4d6d833a25e8e76dbd1a : SUCCESS in 20m 01s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/969ea80e31ec4203a1fa0c163d12395f : SUCCESS in 20m 48s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/88bbf1b0a97545a788d9f8328dffbb1f : SUCCESS in 17m 21s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/1f0a12d32cbd4128a23aba75e04db9a0 : SUCCESS in 17m 07s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/ed421249a7f14082a3e2a08951f5ea7a : SUCCESS in 16m 54s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/772ccefc207b44cc8aa199bce961203a : SUCCESS in 59m 46s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/b48c0a08710c4ee299ecf3e9f7e57ac6 : SUCCESS in 1h 05m 19s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c42daa5fb93c489db2179c846e3d6be5 : SUCCESS in 17m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/edb940ef071a4f05a98072a6cb2ed5bf : SUCCESS in 5m 01s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/9b2042e1a49f496d934223e9a51920af : SUCCESS in 21m 30s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"7dced207540432a784f86180cef45b806694b129","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-24 22:13:24.000000000","message":"Patch Set 3:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4f5aea57ab894d29a6cef11dd699d6ce\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b89ca9f5506d4c7cafe3c56c1ae70d07 : SUCCESS in 16m 32s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b1f360dcc6ed43bca6b28bf9c9793b0c : SUCCESS in 17m 27s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/9357a99d24514aec9777bc58682ca1e0 : TIMED_OUT in 1h 31m 09s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/0e6fc5db3c134c249f803e28f5d6da7f : SUCCESS in 21m 03s","accounts_in_message":[],"_revision_number":3},{"id":"dd40ddf06af6c269e476bc20e51ea20f1f6f29be","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-01-25 17:39:03.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"2e9cae13dbbcf73b881619ae0300199be1d205dc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-25 18:44:06.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a75a82dec3d94f9dab96c6cc670cc800\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/fc9cd19f7d6d468a9884719f6f4056a8 : SUCCESS in 1h 03m 16s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/98506e125b3b4be8b7bd19836f251164 : SUCCESS in 19m 12s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/f2b97e8d325642648186976449b39f7a : SUCCESS in 22m 35s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/2e9fa1ad64a747fc9d73ed7cc5c17cdf : SUCCESS in 59m 03s","accounts_in_message":[],"_revision_number":4},{"id":"7cd51d5ba8bdba85cbc461f286f911e8735ba62d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-25 18:49:59.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cbd4ca54f57e48a8ad4f28e001ad24ff\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/50ac909b484e45a38d3b760034896814 : SUCCESS in 8m 32s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/44e32be577334da88c47d5d851432537 : SUCCESS in 19m 38s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4512c7d1e5554d9d88e75740b920e86a : SUCCESS in 17m 53s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2902c359a77f4020936f10add342e50c : SUCCESS in 20m 43s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/41ddab1dcbba42d4bf404b203b45e5bc : SUCCESS in 12m 09s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/b7822a81e99749afbaa40e914f82e9b9 : SUCCESS in 6m 05s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/8cc5e20305254766a298636880b034d4 : SUCCESS in 18m 51s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/67a45545a14648c58ed23d146bceeb7e : SUCCESS in 19m 56s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/3cd0eef116244ffeb40e128873950a5f : SUCCESS in 14m 29s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/3d176eb31ea441c792138d3f50c6ae50 : SUCCESS in 17m 18s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/aa5ca9717cd341379bb5665bce85a066 : SUCCESS in 18m 15s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/99797d4c953f49bc9367faf7637081c4 : SUCCESS in 18m 19s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/aa54fe73c959401a946481d01361aaf5 : SUCCESS in 18m 50s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/ab0d08a585f448e4b9ae0e9546a65ace : SUCCESS in 18m 17s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/7077e07e5d0844e18d0023e968e4e746 : SUCCESS in 15m 34s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/f28c203a93c740738e7c80086769fc8e : SUCCESS in 18m 42s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/e40e7fce7fda4ebc85e8520d8536707a : SUCCESS in 17m 33s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/95f6c5fd7eb04ec38ca155fd4afaebbb : SUCCESS in 57m 02s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/25c57c040d1f4a439d7b3db6a915c77f : SUCCESS in 42m 50s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b63936a09dc64f549596f5e60b949757 : SUCCESS in 14m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/122cd91cbf4a48f4a21d449a86401ccb : SUCCESS in 5m 19s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/178ceda38c43423ea543b54f3e878c49 : SUCCESS in 24m 35s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"63ca57d1dd916d24d06309b7bd3f1e9130cd894a","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-02-05 10:56:34.000000000","message":"Patch Set 4: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":4},{"id":"f02cef76ad4600d4eb92ad958e245a7774e800e4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-02-05 18:30:05.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"3464bad95ca482c6c341b537e299c9178069c3d1","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-02-05 18:30:12.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"bbc90bfa7858fc34a11e9d650149abd7f7f6ad7c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-05 19:16:24.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3798e07b59c64f4d8473729f7d93304c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aa621cef5b924a27bb3250c22001d425 : SUCCESS in 9m 08s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/90d84b6b783a4e5989af2a7e0e2288eb : SUCCESS in 25m 21s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ead4c03cdd8545408edb999c5d4c5677 : SUCCESS in 18m 02s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/cbe3bc85f5f94e96936554e09020e05c : SUCCESS in 19m 56s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/32e5a62cef7648fcb4dd57b9ab88b0ce : SUCCESS in 11m 19s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/4f9350fd66214922a758e40fab5432be : SUCCESS in 5m 26s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/f50da16aad764b34b116fa2455561af6 : SUCCESS in 17m 57s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/112c450ce932495b9e92d47407fe3913 : SUCCESS in 16m 47s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/9229d6535fa04cd0b9cae3a2c85159b1 : SUCCESS in 25m 18s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/3c11198a76d84aff84094b5a3e0e8c52 : SUCCESS in 17m 20s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/4cf263fd8ac54c3d9a5c58ec95768a80 : SUCCESS in 24m 58s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/57cfc2bc78674b429ac8c00733925d5d : SUCCESS in 18m 12s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/376b8a1006664d98a9f827d70e59d4fe : SUCCESS in 19m 06s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/2d574a1f15c14aa4bead78e14c20b5ed : SUCCESS in 20m 05s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/cf3d422abc544b2bb0b9b98e26788a16 : SUCCESS in 23m 11s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/ed4a4985f97c492692ab508d6f62b293 : SUCCESS in 19m 14s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/165f08f8a0444ae195aed9b32af620f7 : SUCCESS in 17m 22s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/725979fde1364cfba03dbe38d80f2209 : SUCCESS in 38m 55s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/2404a6c8b5904caa8655285b031d18c2 : SUCCESS in 45m 13s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f9fcbd720e234113b88b33c558f0bdf2 : SUCCESS in 21m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/80c6e445f88c48428bda8ee7922ab8c8 : SUCCESS in 5m 34s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/0f44c246c22e4a389b0b15768a7a4ed1 : SUCCESS in 21m 03s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"3d0331074147786203b8a927d9981713067f4f5d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-05 20:02:45.000000000","message":"Patch Set 5:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8d36d69cf2894bc7926fd8781129ae33\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9d32ac3dcb3847baa18ae8309263d37f : SUCCESS in 15m 12s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0ec6e9f77e0f4002a604301b366f3346 : TIMED_OUT in 1h 30m 54s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/57f43305febb41a1ac506eaa3921a7ef : SUCCESS in 17m 56s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/3bd3891562a24e0b944307156d12fca0 : SUCCESS in 1h 09m 24s","accounts_in_message":[],"_revision_number":5},{"id":"fdb6fc4fb6e5d829288801d583023b428980bfb0","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-02-06 09:24:49.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"82bd68b3e589dd067bf31da81f2e99d2d40bd528","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-06 09:25:21.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"1c8aa54f3b63dd19cb62ab3b69274d342cd72630","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-06 10:16:53.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/87dc59ee17364ba0b097cbb801670673\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/857994192b3e4cd4bac7a750959580db : SUCCESS in 8m 55s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/68b5e77ff7504cff942a3fa133f733db : SUCCESS in 10m 27s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/4c0d544e300e4dbd920ea415aaa71993 : SUCCESS in 6m 05s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/b85f4f354fd74b4eb5dd44adef50d58d : SUCCESS in 20m 45s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/7351c18685ec48eca16fc61881d703de : SUCCESS in 14m 15s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/43eb1839195042f18582324da16870e5 : SUCCESS in 17m 18s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/2095d3b4f0644dabaa2081c3b17a4270 : SUCCESS in 16m 57s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/c087d0b00c4749a6982a5d3621123d99 : SUCCESS in 17m 25s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/f9985bdf0f5f46f581c9586ffd53313c : SUCCESS in 19m 42s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/6488848bcfcd435381f4f6830c6673e3 : SUCCESS in 19m 35s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/e1156ff8c7044515a18eeaf47e8f38d8 : SUCCESS in 18m 56s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/3f9069542cbd4379a664f06eb4df9a01 : SUCCESS in 16m 40s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/38e104c1343b4345a352815e734e9d1f : SUCCESS in 17m 31s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/15ff22624aa74c50a810c8fed8c4f72f : SUCCESS in 16m 29s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/ca8ee1d36e7b4c59a7353869a525cfc6 : SUCCESS in 6m 35s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/ef79118e44a54591a9491a4176a1aa58 : SUCCESS in 7m 02s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/440e2d26e94644a994ffbac9e3ca6cf2 : SUCCESS in 47m 52s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/ba469a27e2224b128ca2bdb897e07e21 : SUCCESS in 34m 57s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f4119bc2bb4343489a6a2797f98328ea : SUCCESS in 19m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/67b59410b74a4985a5319c9935783aef : SUCCESS in 5m 01s","accounts_in_message":[],"_revision_number":5},{"id":"efa461bd8245bbbbff7ddd32a397a19372fbde74","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-06 10:16:53.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"036dfe6315e0b6f581c11fac1d0967988bffe081","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-06 10:18:08.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/57d1b11daa914aacb5671dba0a09f256\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6b5500c7a00f46a998e3138ce18da4ba : SUCCESS in 53s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/6f0efb16228c46e2ae32bb5628ab74b9 : SUCCESS in 38s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/28f0ac316e574688a7ad8748fe323654 : SUCCESS in 33s (non-voting)","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"e96a0810249bfcbb2c069b9b159c318ca2d22fe3","revisions":{"8e02174235d1cf66894c3120c22b1a79ca5ee9f5":{"kind":"REWORK","_number":1,"created":"2024-01-04 01:33:13.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/01/904601/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/01/904601/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/904601/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/904601/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/904601/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/01/904601/1"}}},"commit":{"parents":[{"commit":"8bdd8f206ab89c1da918c5dbc5a270ce44f5130b","subject":"Merge \"Document allowed_digests for formpost middleware\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8bdd8f206ab89c1da918c5dbc5a270ce44f5130b"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-01-04 01:31:15.000000000","tz":0},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-01-04 01:31:21.000000000","tz":0},"subject":"tests: Fix float expectations for py312","message":"tests: Fix float expectations for py312\n\nFrom https://docs.python.org/3/whatsnew/3.12.html :\n\n   sum() now uses Neumaier summation to improve accuracy and\n   commutativity when summing floats or mixed ints and floats.\n\nAt least, I *think* that\u0027s what was causing the ring builder failures.\n\nChange-Id: Icae2f1e3e95f216d214636bd5a6d1f40aacab20d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8e02174235d1cf66894c3120c22b1a79ca5ee9f5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8e02174235d1cf66894c3120c22b1a79ca5ee9f5"}]},"branch":"refs/heads/master"},"3122a8ba23c41c1305780455606295e4af14d9c0":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2024-01-04 01:35:12.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/01/904601/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/01/904601/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/904601/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/904601/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/904601/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/01/904601/2"}}},"commit":{"parents":[{"commit":"8bdd8f206ab89c1da918c5dbc5a270ce44f5130b","subject":"Merge \"Document allowed_digests for formpost middleware\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8bdd8f206ab89c1da918c5dbc5a270ce44f5130b"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-01-04 01:31:15.000000000","tz":0},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-01-04 01:33:53.000000000","tz":0},"subject":"tests: Fix float expectations for py312","message":"tests: Fix float expectations for py312\n\nFrom https://docs.python.org/3/whatsnew/3.12.html :\n\n   sum() now uses Neumaier summation to improve accuracy and\n   commutativity when summing floats or mixed ints and floats.\n\nAt least, I *think* that\u0027s what was causing the ring builder failures.\n\nPartial-Bug: #2046352\nChange-Id: Icae2f1e3e95f216d214636bd5a6d1f40aacab20d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3122a8ba23c41c1305780455606295e4af14d9c0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3122a8ba23c41c1305780455606295e4af14d9c0"}]},"branch":"refs/heads/master"},"e74c37feb39b269ca66e91fca769330c54203454":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2024-01-24 20:40:30.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/01/904601/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/01/904601/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/904601/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/904601/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/904601/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/01/904601/3"}}},"commit":{"parents":[{"commit":"1d9ab5d0a8372e907ae0f1029e743b7a97554aaa","subject":"Merge \"Only try to install py2 dev libraries for py2 jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1d9ab5d0a8372e907ae0f1029e743b7a97554aaa"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-01-04 01:31:15.000000000","tz":0},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-01-24 19:17:37.000000000","tz":-480},"subject":"tests: Fix float expectations for py312","message":"tests: Fix float expectations for py312\n\nFrom https://docs.python.org/3/whatsnew/3.12.html :\n\n   sum() now uses Neumaier summation to improve accuracy and\n   commutativity when summing floats or mixed ints and floats.\n\nAt least, I *think* that\u0027s what was causing the ring builder failures.\n\nPartial-Bug: #2046352\nChange-Id: Icae2f1e3e95f216d214636bd5a6d1f40aacab20d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e74c37feb39b269ca66e91fca769330c54203454"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e74c37feb39b269ca66e91fca769330c54203454"}]},"branch":"refs/heads/master"},"a94343863fa8e578d6555324074caddf50d7d5ee":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2024-01-25 17:39:03.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/01/904601/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/01/904601/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/904601/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/904601/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/904601/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/01/904601/4"}}},"commit":{"parents":[{"commit":"a16e1f55a742f33bdfda55d38d6962a785fe5c97","subject":"Improve unit tests for proxy GET ChunkReadTimeouts","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a16e1f55a742f33bdfda55d38d6962a785fe5c97"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-01-04 01:31:15.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-01-25 17:00:24.000000000","tz":0},"subject":"tests: Fix float expectations for py312","message":"tests: Fix float expectations for py312\n\nFrom https://docs.python.org/3/whatsnew/3.12.html :\n\n   sum() now uses Neumaier summation to improve accuracy and\n   commutativity when summing floats or mixed ints and floats.\n\nAt least, I *think* that\u0027s what was causing the ring builder failures.\n\nPartial-Bug: #2046352\nChange-Id: Icae2f1e3e95f216d214636bd5a6d1f40aacab20d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a94343863fa8e578d6555324074caddf50d7d5ee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a94343863fa8e578d6555324074caddf50d7d5ee"}]},"branch":"refs/heads/master"},"e96a0810249bfcbb2c069b9b159c318ca2d22fe3":{"kind":"REWORK","_number":5,"created":"2024-02-05 18:30:05.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/01/904601/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/01/904601/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/904601/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/904601/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/904601/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/01/904601/5"}}},"commit":{"parents":[{"commit":"a16e1f55a742f33bdfda55d38d6962a785fe5c97","subject":"Improve unit tests for proxy GET ChunkReadTimeouts","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a16e1f55a742f33bdfda55d38d6962a785fe5c97"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-01-04 01:31:15.000000000","tz":0},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-02-05 18:29:32.000000000","tz":-480},"subject":"tests: Fix float expectations for py312","message":"tests: Fix float expectations for py312\n\nFrom https://docs.python.org/3/whatsnew/3.12.html :\n\n   sum() now uses Neumaier summation to improve accuracy and\n   commutativity when summing floats or mixed ints and floats.\n\nAt least, I *think* that\u0027s what was causing the ring builder failures.\n\nPartial-Bug: #2046352\nChange-Id: Icae2f1e3e95f216d214636bd5a6d1f40aacab20d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e96a0810249bfcbb2c069b9b159c318ca2d22fe3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e96a0810249bfcbb2c069b9b159c318ca2d22fe3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
