)]}'
{"id":"openstack%2Fswift~904815","triplet_id":"openstack%2Fswift~master~Ic5a68998222709ab088caafe9dbea2b726db38c6","project":"openstack/swift","branch":"master","hashtags":[],"change_id":"Ic5a68998222709ab088caafe9dbea2b726db38c6","subject":"staticweb: Generate HTML5 pages","status":"MERGED","created":"2024-01-05 05:57:47.000000000","updated":"2024-01-18 21:55:07.000000000","submitted":"2024-01-18 21:52:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"904815","meta_rev_id":"3203560bc26037ef9869e430aaabe6bebfd8f512","_number":904815,"virtual_id_number":904815,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-01-18 21:52:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-01-18 19:54:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-01-18 19:54:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-01-05 06:33:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-01-05 07:13:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8bc45d7b20784ebe9115997ba1121b8572c226ae","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-01-05 05:57:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"dee149a078b5cdbb6f95424500ce081d85337444","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-01-05 06:09:44.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"e14e56ed48fc66df668760e4aa7b35df3c803821","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-01-05 06:19:32.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"2dddb203e3e60b8072843f11b5950d71de693e1b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-05 06:33:13.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/82f62838514a423e8c13516f55ddf84e\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/2359fe4c66b5428f82607cb5c79a4da2 : SUCCESS in 20m 28s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e92d433601b6412cb635b85e1faf2e59 : SUCCESS in 16m 46s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/530e1e0e0ef34ed1bf301ea2f61c9d10 : SUCCESS in 22m 30s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/2b5c1658d6ba44c2a88510104b720959 : SUCCESS in 20m 16s","accounts_in_message":[],"_revision_number":2},{"id":"1e0758b13c0efe368dae9bf197e3010862479666","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-05 07:13:14.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ee5bbad900304c29879942cec6c3522a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/97053177b72a40b890a7f99809ddcb9c : SUCCESS in 7m 25s\n- grenade https://zuul.opendev.org/t/openstack/build/e70fcf907333442cb09fd8211455d70d : SUCCESS in 40m 04s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/b391331944a54127a7ca9f3fd7b3500e : SUCCESS in 42m 00s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/5bd9501127834ccfa94291f6a45e3b78 : SUCCESS in 50m 04s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/940eb6191e324cf3ae75d986b526c9bc : SUCCESS in 46m 57s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0fca122cce7646258ef3ccbc28cf814b : SUCCESS in 18m 35s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/48e65fc660e74bdfa710e1f7671bb5fc : SUCCESS in 17m 50s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/58598d4aded14631bae823e3eeabc6af : SUCCESS in 19m 23s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/0ec1560c02ec4043870254060407cc3c : SUCCESS in 10m 20s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/cf0073ed92c54919932997cf7676e463 : SUCCESS in 6m 20s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/e3ddeedf0c074779990c18e1714fcf02 : SUCCESS in 19m 14s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/a56abbe19cab42438e074bf7ce7d8236 : SUCCESS in 17m 35s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/8de14a84b19f4dcf84a38d9151336792 : SUCCESS in 15m 46s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/aa399a6117264c808ddb9f62281a0607 : SUCCESS in 17m 20s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/3e40a9017add4f98888469c275f44022 : SUCCESS in 14m 51s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/a48f9492c0ae44418b03b473517656a8 : SUCCESS in 18m 16s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/3befd07231b64464809eabf844a3c352 : SUCCESS in 19m 20s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/8cb8f4f6b8574f42bfb972bedb56a9ff : SUCCESS in 19m 08s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/cfbac14e14a84caa9dcae0c74ace8d18 : SUCCESS in 16m 27s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/cda2fb5bc23847088ed8c00ab4e67cb4 : SUCCESS in 16m 56s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/f69e007786aa462581131e3373ea8633 : SUCCESS in 16m 50s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/d352aa897cb54b069324063fb9c62fd4 : SUCCESS in 6m 11s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/cc07c6f6d71a4d469b1276456531ba9a : SUCCESS in 16m 28s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/12389b51d7f54fefbc74ffd80bcced2b : SUCCESS in 7m 55s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/b4d8cb92e6ec43fe8f72ccb55557e7d7 : SUCCESS in 48m 00s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/21f5b57a401e405a8988d193ed20bcad : SUCCESS in 50m 02s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/6bb25d34290446bc8c0d537eac471e05 : SUCCESS in 49m 09s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/196cb236669d40228b8650a734b46c52 : SUCCESS in 54m 14s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7d5780e7e4a84a4b978f59e76d0ab525 : SUCCESS in 18m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b5a0730ebdb84ec9b1a4aae0e1fe6a89 : SUCCESS in 5m 05s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/e02b83d4c10f4f5d9863109ced8c86f6 : SUCCESS in 27m 00s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7ebe963cc36442d0857e56676b7d5c68 : SUCCESS in 44m 18s","accounts_in_message":[],"_revision_number":2},{"id":"89c4e0e4a3c9541135c0473e0cb32ea7b6cbb5aa","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-01-11 17:12:38.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"e4b33feeddbea5d05d432c3897e2d086814f581d","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-01-18 19:54:41.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"1def6a83068b64b0bfc7c78cdb44585523e48da1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-18 19:55:18.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"324cea4b19a69b6080a6f3722f41ccd5498c2592","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-18 21:52:03.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1ae4e573b0664414b3ec9e34736cde62\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1e97746728bd4d4786b60d2bc2f2a6ba : SUCCESS in 9m 14s\n- grenade https://zuul.opendev.org/t/openstack/build/32c926e8700a43d0b9cd66c3fd66e1ed : SUCCESS in 48m 07s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/2ed67e3bbaf44010bac83de55541af8c : SUCCESS in 53m 53s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/2c56c4490fdb47aa9c9df997834d277a : SUCCESS in 32m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d5871e57722d4f318346de7a96e50965 : SUCCESS in 28m 13s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/a9c90ad73cac42aeb1b626789e49558f : SUCCESS in 10m 32s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/08134997a6ed432ea93c3994a8fbdabc : SUCCESS in 5m 37s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/f14ca00a14c04ac7ae93c45e6de523e0 : SUCCESS in 17m 31s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/717f543baae240309992caddc285ad92 : SUCCESS in 20m 46s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/e2a295698e2e4bc3b1b025517205b7ca : SUCCESS in 17m 13s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/c3f4243ebe96435a93d3b97651fcb04d : SUCCESS in 14m 10s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/208e9f62d6bf4ffa9e712e4d44e3f899 : SUCCESS in 17m 25s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/05e7f9b672c04ad2a1cf2a704220120a : SUCCESS in 18m 20s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/36d08fdbca6347829db882827bf44cda : SUCCESS in 19m 14s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/14a2947445754f439cd841befe2d591e : SUCCESS in 18m 13s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/7f9df09647514c6ca77873a582c968e1 : SUCCESS in 15m 15s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/2a8a32dfef9944aca3e302a175d6a8ba : SUCCESS in 16m 33s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/3345d030aed24235b30fe6b79465d397 : SUCCESS in 16m 41s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/e44b8e57d3c44759b0f2d0520babde27 : SUCCESS in 6m 33s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/94443fbe41eb4784a97a8b38614b1716 : SUCCESS in 7m 29s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/8cde4b1e984e401f8b9b0365bc5ac072 : SUCCESS in 57m 18s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/a39920a609254f0ca56f4cb2def285d2 : SUCCESS in 53m 38s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/95d7a821cdb44368921494676a7b9e40 : SUCCESS in 42m 17s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/7a68e9f518d148f3b55e96efcbf28b53 : SUCCESS in 56m 04s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9d5d59af790e49f69e219658f11b5478 : SUCCESS in 17m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/efbb5fe19f87430195eafb1e14893aeb : SUCCESS in 5m 04s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fd123358aa8f474286ce3855a36e44de : SUCCESS in 44m 31s","accounts_in_message":[],"_revision_number":2},{"id":"69bbf67b9c1ea6c79c38e057400ecab357f38a99","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-18 21:52:03.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"3203560bc26037ef9869e430aaabe6bebfd8f512","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-18 21:55:07.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/749b323ff5644c1ba52d64baf2aa8fd5\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/61690ec32b0149e185bf2870c78c0f0d : SUCCESS in 44s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/1795cc1111944846af77b64a40467797 : SUCCESS in 34s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/1fc385ab75db46f183b3846bc20f5245 : SUCCESS in 35s (non-voting)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"ac7eb8ac9d9d03e584a570ac1aac5ff8e015914d","revisions":{"0100bff85503763f27b166568f177802f8f431bd":{"kind":"REWORK","_number":1,"created":"2024-01-05 05:57:47.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/15/904815/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/15/904815/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/15/904815/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/15/904815/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/15/904815/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/15/904815/1"}}},"commit":{"parents":[{"commit":"365db20275d7798c30f10ee03221561a28bb7a79","subject":"FakeSwift: use HTTPMethodNotAllowed not HTTPNotImplemented","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/365db20275d7798c30f10ee03221561a28bb7a79"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-01-05 05:55:03.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-01-05 05:56:35.000000000","tz":-480},"subject":"staticweb: Generate HTML5 pages","message":"staticweb: Generate HTML5 pages\n\nHTML 4.01 is so 1999.\n\nCloses-Bug: #2047679\nChange-Id: Ic5a68998222709ab088caafe9dbea2b726db38c6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0100bff85503763f27b166568f177802f8f431bd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0100bff85503763f27b166568f177802f8f431bd"}]},"branch":"refs/heads/master"},"ac7eb8ac9d9d03e584a570ac1aac5ff8e015914d":{"kind":"REWORK","_number":2,"created":"2024-01-05 06:09:44.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/15/904815/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/15/904815/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/15/904815/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/15/904815/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/15/904815/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/15/904815/2"}}},"commit":{"parents":[{"commit":"365db20275d7798c30f10ee03221561a28bb7a79","subject":"FakeSwift: use HTTPMethodNotAllowed not HTTPNotImplemented","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/365db20275d7798c30f10ee03221561a28bb7a79"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-01-05 05:55:03.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-01-05 06:09:17.000000000","tz":-480},"subject":"staticweb: Generate HTML5 pages","message":"staticweb: Generate HTML5 pages\n\nHTML 4.01 is so 1999.\n\nCloses-Bug: #2047679\nChange-Id: Ic5a68998222709ab088caafe9dbea2b726db38c6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ac7eb8ac9d9d03e584a570ac1aac5ff8e015914d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ac7eb8ac9d9d03e584a570ac1aac5ff8e015914d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
