)]}'
{"id":"openstack%2Fswift~904870","triplet_id":"openstack%2Fswift~master~I57dffe74555b87124dda6c5542dba6b637784094","project":"openstack/swift","branch":"master","attention_set":{"1179":{"account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"last_update":"2024-01-09 00:03:48.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I57dffe74555b87124dda6c5542dba6b637784094","subject":"Include % free when FALLOCATE_RESERVE fails","status":"NEW","created":"2024-01-05 17:33:16.000000000","updated":"2024-01-09 19:14:25.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":4,"unresolved_comment_count":2,"has_review_started":true,"meta_rev_id":"745ef523752cf03c9d1e16fa97f2a4e481284b7c","_number":904870,"virtual_id_number":904870,"owner":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2024-01-09 05:33:28.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-01-05 18:19:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-01-05 18:44:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-01-09 03:11:39.000000000","updated_by":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"reviewer":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"state":"CC"},{"updated":"2024-01-09 19:14:25.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"CC"}],"messages":[{"id":"4170626796b3b2355dbf9949405666fb6173caab","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-01-05 17:33:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"58c0d4bd354b103dc2368dcdde5808eccd8d2286","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-05 18:19:49.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/453efc5b23154b72bbe677ecacca3626\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/62973e9ff1d94ab68d74ee48b77a288a : SUCCESS in 14m 54s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0a1d95dc118c403b8cbc687c5c6c12a6 : SUCCESS in 15m 58s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/12d01546a5394bc7a7b9a7a59a80d3cf : SUCCESS in 22m 20s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/d0ea5eca2541415db66ce6a885ab92ba : SUCCESS in 44m 58s","accounts_in_message":[],"_revision_number":1},{"id":"d88227f53cfe193e55aa0130b3cd72f954739be8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-05 18:44:43.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/3ad1412d8cb3407b93436ea85499c271\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ed167ba247d24b0da3bbf80b9e047ec2 : SUCCESS in 9m 02s\n- grenade https://zuul.opendev.org/t/openstack/build/59869710163f4674a2eb155af2da8df9 : SUCCESS in 1h 02m 31s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/1f05aa85faef441eb5f7053c70857e23 : SUCCESS in 59m 22s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/908262c754074bd88b9df83f502639c7 : SUCCESS in 49m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/00f7a950de3742fc9d196251dfa711c5 : SUCCESS in 1h 05m 46s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8b867f3b8c324c40bea0b82f4f7092bc : SUCCESS in 21m 01s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f80e9a48ee4c4deabf04c458472035d1 : SUCCESS in 19m 08s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/cf922b4bdf954279906871d2f556d6bd : SUCCESS in 21m 17s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/c0303a93da0244d3a10ed8cc83c08e9e : SUCCESS in 9m 10s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/13c814926b554a7c8d915780d26bc2d8 : SUCCESS in 6m 59s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/f65b80d3b4ed447bab401ffa10ac0b6e : FAILURE in 13m 07s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/fea85aec3f584cfc84e5d02d9641148d : SUCCESS in 17m 05s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/fb6b1c06a58e455ea400a43e8e3e0299 : SUCCESS in 14m 57s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/60c71d7fd5a344e5825b0c75434170aa : SUCCESS in 11m 55s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/c9466063fbe14f7181eef24ca93143ce : SUCCESS in 17m 04s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/cd57df5f06714225a5365f50f821a6cf : SUCCESS in 19m 51s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/ed35ba9cdd39490491fa616e52a6f5ea : SUCCESS in 19m 33s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/8e6370bab6d84b4ea4664eb4be183830 : SUCCESS in 20m 19s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/c8175966a7eb4124af3b3aed0c3a315b : SUCCESS in 15m 18s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/084ce140bd0a47bea33c301ced622f26 : SUCCESS in 14m 56s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/6589e058717840818988da11b8295be3 : SUCCESS in 16m 41s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/487c4a0eb60b46219decc1114677885d : SUCCESS in 7m 35s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/9922f03e721f4da389bc0fb0cf8d118f : SUCCESS in 16m 57s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/49c0c61c04bd4bc0990cfeeae0c2dce1 : SUCCESS in 7m 53s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/c473240530674e819e07e6b41d34c10d : SUCCESS in 51m 42s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/c23bd87fc2e1484481cca3f3a910e4b1 : SUCCESS in 58m 24s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/8c427f9dea8a45e19e3b459c3ad547f6 : SUCCESS in 56m 11s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/36ca8871821d4830baebe8f72ac13bf7 : SUCCESS in 38m 45s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3714ae978abc426cb1ee9c7b042133db : FAILURE in 19m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6aea31896dfa4f44b4ff2e8a89f59724 : SUCCESS in 4m 59s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/d55fcf72570846059c9b832fb2be90cf : SUCCESS in 17m 51s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/33b8ca80b7fc43008483c6175b6b1e25 : SUCCESS in 40m 17s","accounts_in_message":[],"_revision_number":1},{"id":"3691a5509c00a32a808798e2554ed2c53c3ad2b3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-01-08 21:08:29.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"1b11df3c1b5847e37a45dddb5a2293c2218c5152","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-01-08 22:12:45.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"cbf0f9eb61a94781134153b0d26d0b235ac379e7","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-01-08 22:12:47.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"3472c557bf93df504ee4e6527a13795c38654ebb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-08 23:45:58.000000000","message":"Patch Set 3:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/580cd24752ab4f488b84535836ffff1a\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/8f17bf2e0a904efb97825161e04e034a : SUCCESS in 1h 11m 49s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/34b67fec1d344750afd9a8d3989ffdfd : SUCCESS in 1h 23m 25s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/0af46f6247324c698ffa7abc26e9e679 : TIMED_OUT in 1h 31m 19s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/77b746e164dd4712bd3c53dddef25c7e : SUCCESS in 1h 20m 04s","accounts_in_message":[],"_revision_number":3},{"id":"266929452a760aa72d899769535543ef7cdba6fd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-09 00:03:48.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/34d73851e7bd4866856b1ecb9dbd8b95\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6cd34b2d41d54e28ab8d4352d7ab9535 : SUCCESS in 9m 23s\n- grenade https://zuul.opendev.org/t/openstack/build/5c4e4887c32f4692ba1254b4f101bb8d : SUCCESS in 1h 03m 17s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/cbcdf7942f2c4c229c0dd8c3e5f39376 : SUCCESS in 1h 01m 21s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/3ee2bb4335d840a49f7bd193512ad0be : SUCCESS in 36m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cdbace165fdd469db667a4e6af11d604 : SUCCESS in 40m 37s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/88abd5c97d2e40c0855a5093d8a1dc52 : SUCCESS in 20m 13s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3264a6769436480eb1e7e248b2cb35cf : SUCCESS in 21m 07s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f25dbc2ab978443298b28ed8e5f08ea3 : SUCCESS in 17m 55s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/56ac389eff744467bd8d50b6d865b2b9 : SUCCESS in 10m 48s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/4bb0b4054a724b6799d469a5e37bc978 : SUCCESS in 6m 07s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/7654dd87e89147cc97b5f52e433cd19d : FAILURE in 16m 07s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/9856109067aa4cbaa8974b6f88e13825 : SUCCESS in 18m 13s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/f4c365bc7c814ebe93ddbbd4d2271fc4 : SUCCESS in 17m 04s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/56aa28caba334d26a1cc0fdeb19f2571 : SUCCESS in 15m 39s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/6a12b56a5bcc4946ab3a3937407f6c3f : SUCCESS in 16m 11s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/a54c4de72a794f27980fefc5c48e976f : SUCCESS in 18m 23s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/f64d59734b424a62a72c9989872e5d2e : SUCCESS in 19m 25s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/47020c44ae4547ce8e531dfa04f39d5a : SUCCESS in 19m 30s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/4ee97a8f7d4242509b4b4b4fa7a1802b : SUCCESS in 15m 00s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/159645672a5447ab88f9ceb2313886da : SUCCESS in 16m 54s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/0b39cca52691404fb53991117049104c : SUCCESS in 18m 01s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/422f09e9799b498894995b0d5a69ef4e : SUCCESS in 7m 23s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/d0ac6b3f2f024d43aeb8f8ec3a4391d7 : SUCCESS in 17m 46s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/f1295ed994b247d29c738e9425fab442 : SUCCESS in 8m 08s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/10bab6fcd423407a9d23f8494996d728 : SUCCESS in 58m 18s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/ed7c21e237a34969a0b2e0cdc346e3f4 : SUCCESS in 1h 01m 06s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/cee81f19e6d847389ed380083a1ed9d4 : SUCCESS in 1h 06m 51s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/b2389091aded4427b2e4905219ac5f02 : SUCCESS in 59m 47s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0d9c5852d2cc473b8d9f4a26ab982e7d : FAILURE in 17m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9a64052d42b642aeba920cd454b6636f : SUCCESS in 5m 08s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/9344e5ae21fa4f6786b7a1647f2c7c50 : SUCCESS in 20m 34s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/166c87b86e4d4038aaf3a94947fb4470 : SUCCESS in 42m 53s","accounts_in_message":[],"_revision_number":3},{"id":"8a75b7d1be6b1aaae6c07f5c156009130de4789b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-01-09 02:21:09.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"f0d9cfd38f95d4c4e4098f5f080c687127566696","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2024-01-09 03:11:39.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"ae3d099561f42f35db2beaa7ce388c2b4b4981fc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-09 03:51:07.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b14d524e3ac0425d99ab93255e87c105\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0b5318ce6aaa4b689a81c7aae3cd6b0c : SUCCESS in 22m 07s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/02c9793ea2b14c1790801e4e565b4ca6 : SUCCESS in 1h 27m 30s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/dd0eb363941e4a41a49d58dc52aaa5cc : SUCCESS in 28m 16s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/57e722d10ccf4ecfb57d97ebdfd82e77 : SUCCESS in 26m 18s","accounts_in_message":[],"_revision_number":4},{"id":"491b937d28e7ad0e28eebe6ebe462e5c888af855","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-09 05:33:28.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/143b37d9967d493f85341fc808717406\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/db070712b12b46abbf7d675f431c5c94 : SUCCESS in 8m 41s\n- grenade https://zuul.opendev.org/t/openstack/build/0d5e07bf27114642848e54c6ebfe1eec : SUCCESS in 59m 56s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/9fb9bdfebe944c0293f79f71e36fdb50 : SUCCESS in 2h 05m 54s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/21660e32e735400690a605b2d6639d58 : SUCCESS in 53m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2c91bd016f944e3fb5557ae731ee29da : SUCCESS in 31m 59s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2cf340db89f243079dab6ac11ac32ee5 : SUCCESS in 22m 35s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9222972b8a3546b7b77c6eb6031604c4 : SUCCESS in 21m 19s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f0e42e26d2054d3d882d47678f4ece91 : SUCCESS in 18m 54s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/2bcba8eeceaf494cb3c64e2ae8c5596b : SUCCESS in 11m 37s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/5a00e9ce9bb34cf982c6f5a05597bb62 : SUCCESS in 8m 24s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/fb502a8bd42943839703ba2ed7ca8833 : SUCCESS in 17m 44s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/8d83a4fdffee495dad5c6b2dd25d4295 : SUCCESS in 17m 14s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/92fd55042d0c43a8ad821b056f456b17 : SUCCESS in 15m 59s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/5c489195bcee4078a7f911b2541e018a : SUCCESS in 17m 47s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/bd0da0aa0b7b4528b0a69530f0471ff7 : SUCCESS in 18m 13s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/571a30cc79a14f25ba54f3ae5407f41b : SUCCESS in 20m 15s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/2df85cf9e34b4befafaf1cd22979c87e : SUCCESS in 18m 49s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/068fef9d1b68475d8ebc6ab313a30b70 : SUCCESS in 23m 12s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/1e8e259e02064d588daef4b3f2d6becf : SUCCESS in 17m 11s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/a76aa5ac70b747a091a0bff182aee106 : SUCCESS in 17m 27s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/d3402144b55e49e7ae3fa9a7961d3e24 : SUCCESS in 16m 36s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/1e9565bc5e634856bbc780a95d5e425d : SUCCESS in 8m 53s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/379bed9d5d414b518436802f09c8d71b : SUCCESS in 18m 43s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/db6e45f3569a4d10873b9ce494bb542c : SUCCESS in 9m 16s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/570ae9290f544266b500f0abb461e8a2 : SUCCESS in 56m 18s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/b02e7b0b7c5741018fafd275a62734ce : SUCCESS in 1h 04m 01s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/4722441bbaa84a3f96f53a98773b45c1 : SUCCESS in 1h 04m 15s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/3c82dfcd059f4bf8bd6373050eb87b6d : SUCCESS in 51m 43s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4e693e2407904169879ebf7a33ba43fc : SUCCESS in 23m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d5447db6c1c84657aea7257099197704 : SUCCESS in 5m 06s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/b33fdb8fd6bc4493b72b77be4e3355d5 : SUCCESS in 21m 00s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e8f96d4c6f2a482bb9d7b34a222c71ec : SUCCESS in 57m 31s","accounts_in_message":[],"_revision_number":4},{"id":"745ef523752cf03c9d1e16fa97f2a4e481284b7c","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-01-09 19:14:25.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"22993ab7ae57c59c16cb20708630e57df0e4978f","revisions":{"1ef90c378d36d5d17d66396802d5f27373cf8904":{"kind":"REWORK","_number":1,"created":"2024-01-05 17:33:16.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/70/904870/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/70/904870/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/70/904870/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/70/904870/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/70/904870/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/70/904870/1"}}},"commit":{"parents":[{"commit":"0f5cccfe0952c30a4203608391b347537b392950","subject":"Split reserve functionality from C calls in fallocate","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0f5cccfe0952c30a4203608391b347537b392950"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-01-05 17:33:07.000000000","tz":-360},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-01-05 17:33:07.000000000","tz":-360},"subject":"wip: include % free when FALLOCATE_RESERVE fails","message":"wip: include % free when FALLOCATE_RESERVE fails\n\nChange-Id: I57dffe74555b87124dda6c5542dba6b637784094\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1ef90c378d36d5d17d66396802d5f27373cf8904"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1ef90c378d36d5d17d66396802d5f27373cf8904"}]},"branch":"refs/heads/master"},"afb1d1770d5606e3e8b8b69cca3bc11e09658e88":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-01-08 21:08:29.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/70/904870/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/70/904870/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/70/904870/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/70/904870/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/70/904870/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/70/904870/2"}}},"commit":{"parents":[{"commit":"86212fd89a986c9d14f92bbf11303f4597412368","subject":"Split reserve functionality from C calls in fallocate","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/86212fd89a986c9d14f92bbf11303f4597412368"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-01-05 17:33:07.000000000","tz":-360},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-01-08 21:06:33.000000000","tz":-360},"subject":"wip: include % free when FALLOCATE_RESERVE fails","message":"wip: include % free when FALLOCATE_RESERVE fails\n\nChange-Id: I57dffe74555b87124dda6c5542dba6b637784094\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/afb1d1770d5606e3e8b8b69cca3bc11e09658e88"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/afb1d1770d5606e3e8b8b69cca3bc11e09658e88"}]},"branch":"refs/heads/master"},"7cb583a1ffb86e2c52b7de658a7880d7e276113b":{"kind":"REWORK","_number":3,"created":"2024-01-08 22:12:45.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/70/904870/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/70/904870/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/70/904870/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/70/904870/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/70/904870/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/70/904870/3"}}},"commit":{"parents":[{"commit":"86212fd89a986c9d14f92bbf11303f4597412368","subject":"Split reserve functionality from C calls in fallocate","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/86212fd89a986c9d14f92bbf11303f4597412368"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-01-05 17:33:07.000000000","tz":-360},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-01-08 22:12:21.000000000","tz":-360},"subject":"Include % free when FALLOCATE_RESERVE fails","message":"Include % free when FALLOCATE_RESERVE fails\n\nChange-Id: I57dffe74555b87124dda6c5542dba6b637784094\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7cb583a1ffb86e2c52b7de658a7880d7e276113b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7cb583a1ffb86e2c52b7de658a7880d7e276113b"}]},"branch":"refs/heads/master"},"22993ab7ae57c59c16cb20708630e57df0e4978f":{"kind":"REWORK","_number":4,"created":"2024-01-09 02:21:09.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/70/904870/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/70/904870/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/70/904870/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/70/904870/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/70/904870/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/70/904870/4"}}},"commit":{"parents":[{"commit":"86212fd89a986c9d14f92bbf11303f4597412368","subject":"Split reserve functionality from C calls in fallocate","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/86212fd89a986c9d14f92bbf11303f4597412368"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-01-05 17:33:07.000000000","tz":-360},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-01-09 02:21:00.000000000","tz":-360},"subject":"Include % free when FALLOCATE_RESERVE fails","message":"Include % free when FALLOCATE_RESERVE fails\n\nChange-Id: I57dffe74555b87124dda6c5542dba6b637784094\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/22993ab7ae57c59c16cb20708630e57df0e4978f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/22993ab7ae57c59c16cb20708630e57df0e4978f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
