)]}'
{"id":"openstack%2Fswift~905011","triplet_id":"openstack%2Fswift~master~Ia4ebe4869b45ab00d8d956ed327789e9e7117aed","project":"openstack/swift","branch":"master","topic":"py312","attention_set":{},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2024-01-24 20:57:17.000000000","reason":"Change was abandoned"},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2024-01-13 06:17:47.000000000","reason":"\u003cGERRIT_ACCOUNT_15343\u003e replied on the change","reason_account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}},"hashtags":[],"change_id":"Ia4ebe4869b45ab00d8d956ed327789e9e7117aed","subject":"wip: ClosingIterator class","status":"ABANDONED","created":"2024-01-08 16:56:00.000000000","updated":"2024-01-24 20:57:17.000000000","total_comment_count":33,"unresolved_comment_count":6,"has_review_started":true,"meta_rev_id":"f791176fbe568ee4f9a8f0278655d6c7fc6b40c7","_number":905011,"virtual_id_number":905011,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2024-01-12 15:53:41.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-01-08 18:15:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-01-10 21:01:57.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"CC"}],"messages":[{"id":"e48e9a66f7edafb65e89a513a2048318470302ae","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-01-08 16:56:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4a3c158e19aaa609f3b6f93efc53930d77f840b7","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-01-08 17:40:13.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ba7ce30d9a0ead5e164520b199001709f1ec5ebe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-08 18:15:19.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d5aca2d41def4556bc4f52549b521757\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0c03fd8b9f2f49d68ff0e7716cb95dce : SUCCESS in 9m 30s\n- grenade https://zuul.opendev.org/t/openstack/build/2e5ac51aa075408cbc8a767f41534392 : SUCCESS in 1h 06m 56s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/af31082c7aca4bd3a492705ccee42918 : SUCCESS in 1h 10m 18s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/488795ad335d499faa6a9ce3a1693129 : SUCCESS in 50m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/acc877b8579d4ed5a6f111508cd5db21 : SUCCESS in 54m 17s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/47fe6febe51b48d68cdb29971d36bc70 : SUCCESS in 20m 59s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2bb6b7ce6f794cf68d8e7c37632cd3f3 : SUCCESS in 21m 41s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/65dda1df5db347a8aa87cb247d7e2f31 : SUCCESS in 19m 30s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/27a94d21407c4b40aa885972c13064f8 : SUCCESS in 12m 18s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/0463cc7649bc4779b332bcee53887b2a : SUCCESS in 5m 45s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/37083bf49f064213ab1b968531aac229 : FAILURE in 16m 24s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/bb723ea49fc24f3fa4026e018f2613d0 : FAILURE in 19m 35s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/d87e64bb77e54de1b1a3918af25ac4b5 : FAILURE in 15m 21s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/9472cd0dc4b5409eaa209ac6f5bc9843 : FAILURE in 16m 52s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/e28b094791d84e7fb4edbb71598e0bb5 : FAILURE in 16m 04s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/ff6e69b40ac34c0ebd93a80257e9f2ae : SUCCESS in 18m 08s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/ebf38e7de68f4312b7575e5f6ad8cea5 : SUCCESS in 18m 04s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/792c9a08f31746bca1c7fa6b8942d3c9 : SUCCESS in 20m 11s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/411d87a113df4142a9c3f5d754952623 : SUCCESS in 16m 09s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/1ab88b9c83e3414192d088851eb6ef3a : SUCCESS in 18m 22s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/8c0c07129f944d7888661caaec68cd98 : SUCCESS in 17m 32s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/893294becc2c4d83b836a4965bf39501 : SUCCESS in 7m 31s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/604f3132690e45a59c5b6454e2c0a056 : SUCCESS in 16m 11s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/ec3dce31e593469e93e1a489145925c1 : SUCCESS in 9m 57s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/6fb58527fca04537b6841426639b137e : SUCCESS in 1h 10m 23s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/23661a2212a4422c84fba9f324802666 : SUCCESS in 1h 02m 07s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/4f375fe05eb14d6ea7b97eafafc051c7 : SUCCESS in 52m 12s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/2375ead2215c498e91a1ec785807a8c2 : SUCCESS in 44m 01s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/551148b590e74f2eb465babe466b2e42 : FAILURE in 18m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2f857e442c284f61b51deed2c7b47165 : SUCCESS in 5m 00s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/5d693eb131b24e33be1c596141a8cf34 : SUCCESS in 23m 12s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0da861fed88c4715b48d85194bc40338 : SUCCESS in 56m 25s","accounts_in_message":[],"_revision_number":1},{"id":"7a26d9e309711939a8061a381bbe4670440f03a0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-08 18:30:32.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cb719be54516434a943256565a1d3ca7\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/2c7e680fd0374106a8c0005a35f3fe2d : FAILURE in 58m 20s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8a171cabd9d54264a30994e2bb459098 : TIMED_OUT in 1h 31m 02s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/b096b060c17a4ce99b1ea432818d9539 : SUCCESS in 1h 21m 03s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/78a8a1fedb844690b964a957727497f7 : SUCCESS in 1h 06m 46s","accounts_in_message":[],"_revision_number":1},{"id":"dac06c96b1a3160d30b4e55354717a73d57c2ac7","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-01-10 21:01:57.000000000","message":"Patch Set 1:\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"b3e926728dc99858d98d95b0c3e8177cb3186ad5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-01-11 19:14:19.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"67854287a4e45bedfc7d7eee0a87a03bfc144943","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-01-11 19:23:39.000000000","message":"Patch Set 2:\n\n(6 comments)","accounts_in_message":[],"_revision_number":2},{"id":"2952902b05f52b827a56272252fcc08d2addc449","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-11 20:42:51.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eb3b94799613412c9327779e00df3c46\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/18de4c56d217409a853bc58b83a96118 : SUCCESS in 16m 32s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8ad16836d8224f7484a9cab63064f220 : SUCCESS in 17m 15s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/7090c06bcd0c40749c9af82da1794088 : SUCCESS in 22m 21s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/00b32dfdcde7481b95138902067eae7a : SUCCESS in 1h 26m 35s","accounts_in_message":[],"_revision_number":2},{"id":"e3c0266af3569f8881934420e8de510c3b726999","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-11 20:43:49.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ec70079248fd4c228fec35da53c81586\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0cdfb8a8fbc24e29abe41bc2fe2ab9e2 : SUCCESS in 8m 51s\n- grenade https://zuul.opendev.org/t/openstack/build/e264710e653e4a2a9b820d77fd58017f : SUCCESS in 1h 06m 20s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/a610cca70546413c90f9f9c7a89dbd5f : SUCCESS in 1h 07m 31s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/b8ece9905a104859b38db18683c46a71 : SUCCESS in 54m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/459eaa320b6a4707a62be0819c40feff : SUCCESS in 38m 26s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/82a292e6d34144afaec9fc8820f8adf4 : SUCCESS in 15m 01s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e2f36214c1694061831ffc43ee35e74e : SUCCESS in 19m 20s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5dc433aef961478286f52f5edcff4ad5 : SUCCESS in 19m 51s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/b0d426d513554c66b256d7f67567e52f : SUCCESS in 9m 33s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/0b9f7d0bc775460c9e2210a871e964ae : SUCCESS in 5m 27s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/ac957867c5ec4ba48a62fd0147511637 : FAILURE in 16m 48s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/a7a0862dcfdf4abf8e1eedcd843a018e : SUCCESS in 16m 48s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/1d2f63f9bde14a38bea437d53c631f18 : SUCCESS in 15m 41s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/aca5756aa9f644d889fe431f66e7f64a : SUCCESS in 20m 18s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/fbe5aa552e394c8dae8d9160f3ac69b5 : SUCCESS in 19m 43s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/611ac75d61cd4b04ad06f84cd86ed37a : SUCCESS in 19m 43s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/b839c11b565146e58f9db7438653ede7 : SUCCESS in 18m 29s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/16473065573d4a539c2ae001f0b16c83 : SUCCESS in 19m 35s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/cac35f2b914c44a4895d7803e5b200b6 : SUCCESS in 16m 00s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/55709fe1c4304549b7af93505ec99a86 : SUCCESS in 14m 59s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/c29c5a3d66ef427fbbe68e7237b3dd7b : SUCCESS in 14m 52s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/9e6b0b2f1a4a45eca9068d8f0eed543f : SUCCESS in 9m 20s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/8f4cda2345454b919c1065a227fd9d31 : SUCCESS in 17m 08s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/74cfed2b3e5744bdbf842717c749de18 : SUCCESS in 10m 39s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/5715f2330b514656b00f5006b100075b : SUCCESS in 1h 00m 07s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/da3cf000a1bb47129f3ffb2e47a8a90b : SUCCESS in 1h 01m 36s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/3145676d2d8b4938b39f0a1ca3d1336b : SUCCESS in 54m 59s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/9e623557bb304829afa8af2f0efb2651 : SUCCESS in 1h 02m 16s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/316ba834d7234c4ea7d582ac6db7fbbf : FAILURE in 18m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2ba545196ad7407abfd15ee1e2637b86 : SUCCESS in 4m 57s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/cd50cdc54499466585cf598730e3902e : SUCCESS in 23m 59s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/20d3882c3290466faa3c0b0e408b21ab : SUCCESS in 1h 00m 49s","accounts_in_message":[],"_revision_number":2},{"id":"53d5d9b4d043ed9b4840416aec0d88a35b348614","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-01-11 20:54:52.000000000","message":"Patch Set 2:\n\n(10 comments)","accounts_in_message":[],"_revision_number":2},{"id":"6b78072bbca403675d18df8cabe60ff133365ad3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-01-12 14:31:10.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"fa9a08671d31a66d7dbfbfd4ffa621436f7dec00","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-01-12 14:32:51.000000000","message":"Patch Set 3:\n\n(8 comments)","accounts_in_message":[],"_revision_number":3},{"id":"c170da99fa901251264782bc41380ace818350e2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-12 15:31:28.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8f95b9847a6d4ba8b750cc9c25a6bfab\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1ecefd67779547918eb6ea28dfa84f45 : SUCCESS in 15m 31s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d44a743f708e45c3ad5b48441d98d135 : SUCCESS in 17m 31s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/5233fdefeed440808ae1e025c197fbab : SUCCESS in 20m 41s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/ca3ef1a1c05941f4b869b1b782adf9f6 : SUCCESS in 58m 27s","accounts_in_message":[],"_revision_number":3},{"id":"67f9898f9f421b9cbcd5df35f49db24b03d605d0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-12 15:53:41.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f4aeb92b389e4227b290d182e4c96d78\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ee20ab979ebd4ff98490ec45fec0f7d5 : SUCCESS in 9m 26s\n- grenade https://zuul.opendev.org/t/openstack/build/127ad160b7ee4e0595117825be24b704 : SUCCESS in 1h 02m 56s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/0152a8f6d7514c059cdc314c1c80272f : SUCCESS in 1h 06m 29s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/0c286ee619e24e01b78c3e1348097197 : SUCCESS in 48m 55s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3752967cb9014664be6d0ba70db8aa6e : SUCCESS in 36m 35s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6cfa8406e91e40b8b26114d0b6275c2e : SUCCESS in 20m 57s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4a8929d505454010ab080280ddfa56fc : SUCCESS in 19m 24s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/717ebfad9fb5461bbc87b87e13700a77 : SUCCESS in 21m 30s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/47a30a76d0ba4716b02e3395ddc0afed : SUCCESS in 10m 59s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/2192b694d0e84728be4b5d658e3d3bb3 : SUCCESS in 5m 50s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/bbcb0c0046d94492bc85671ffa627334 : SUCCESS in 17m 06s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/8f3502f7adab4897a505ca717d62af0c : SUCCESS in 18m 02s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/70b1fba3552f41afb6ea9efbc202bbf4 : SUCCESS in 19m 09s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/f3763d9dce594160b76efd66a4ac3862 : SUCCESS in 14m 29s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/b8019f04b2174f9391eb2c6d729e01d5 : SUCCESS in 17m 10s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/c6394cb42e48400999a26393d3a10d58 : SUCCESS in 17m 42s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/63cda2c4a1ea4f67a61dfe22c2937c90 : SUCCESS in 17m 51s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/dadff0b9180d482f8e8244f0006e2d11 : SUCCESS in 18m 45s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/9676bf425911410a9d2293e3339c833b : SUCCESS in 14m 09s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/72cdcd5a028842e587ea246e2f99e2ed : SUCCESS in 16m 08s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/f157a087efbc4e53b0ec602ff844492f : SUCCESS in 16m 46s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/75dbf766f4d34782b5269b15c852f694 : SUCCESS in 8m 00s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/91068605402a494bae96b7089780274e : SUCCESS in 15m 46s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/98e3da9d201140b788903057f5851ef3 : SUCCESS in 8m 43s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/bc39ad0e5aaa4239b487e80008bdb971 : SUCCESS in 1h 07m 03s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/5cbb4918788c4e39a9eca05a67efb0c9 : SUCCESS in 1h 05m 52s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/edf10d6c819a4177bea00b2bd96c8553 : SUCCESS in 58m 48s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/9b32fc0af5384b95aa521879d2abde69 : SUCCESS in 1h 03m 35s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d4890c9ba32246b0994cababe4616bfa : SUCCESS in 17m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/10628bb4d97f4564bd7ddc34cae4940f : SUCCESS in 4m 57s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/98024aa93ebb4be88059a101754eb995 : SUCCESS in 19m 19s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/14a1b46008664aaf94b3ae2649203aa2 : SUCCESS in 1h 07m 47s","accounts_in_message":[],"_revision_number":3},{"id":"05e5016c3949217a8a0ba4fe8c5896c448c9efb6","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-01-13 06:17:47.000000000","message":"Patch Set 3:\n\n(3 comments)","accounts_in_message":[],"_revision_number":3},{"id":"7e1f7f498c9a1104f9f16092caff81f8740d5080","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-01-24 19:33:51.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"f791176fbe568ee4f9a8f0278655d6c7fc6b40c7","tag":"autogenerated:gerrit:abandon","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-01-24 20:57:17.000000000","message":"Abandoned\n\nSquashed into https://review.opendev.org/c/openstack/swift/+/904652","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"154c3e1e1cc0e9f765fca55cd8726d5130a39784","revisions":{"4139b81ace834cb64291e4d1a2de4850cc22e677":{"kind":"REWORK","_number":1,"created":"2024-01-08 16:56:00.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/11/905011/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/11/905011/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/905011/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/905011/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/905011/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/11/905011/1"}}},"commit":{"parents":[{"commit":"78a40666153e641bbf613998841ec3f7d7d0c78c","subject":"Be more explicit about closes in string_along","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/78a40666153e641bbf613998841ec3f7d7d0c78c"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-01-08 16:55:47.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-01-08 16:55:47.000000000","tz":0},"subject":"wip: ClosingIterator class","message":"wip: ClosingIterator class\n\nChange-Id: Ia4ebe4869b45ab00d8d956ed327789e9e7117aed\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4139b81ace834cb64291e4d1a2de4850cc22e677"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4139b81ace834cb64291e4d1a2de4850cc22e677"}]},"branch":"refs/heads/master"},"a1d9cf95888e901b7c8043ce0a0162a55b9ce93a":{"kind":"REWORK","_number":2,"created":"2024-01-11 19:14:19.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/11/905011/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/11/905011/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/905011/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/905011/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/905011/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/11/905011/2"}}},"commit":{"parents":[{"commit":"78a40666153e641bbf613998841ec3f7d7d0c78c","subject":"Be more explicit about closes in string_along","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/78a40666153e641bbf613998841ec3f7d7d0c78c"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-01-08 16:55:47.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-01-11 19:12:35.000000000","tz":0},"subject":"wip: ClosingIterator class","message":"wip: ClosingIterator class\n\nChange-Id: Ia4ebe4869b45ab00d8d956ed327789e9e7117aed\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a1d9cf95888e901b7c8043ce0a0162a55b9ce93a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a1d9cf95888e901b7c8043ce0a0162a55b9ce93a"}]},"branch":"refs/heads/master"},"154c3e1e1cc0e9f765fca55cd8726d5130a39784":{"kind":"REWORK","_number":3,"created":"2024-01-12 14:31:10.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/11/905011/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/11/905011/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/905011/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/905011/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/905011/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/11/905011/3"}}},"commit":{"parents":[{"commit":"78a40666153e641bbf613998841ec3f7d7d0c78c","subject":"Be more explicit about closes in string_along","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/78a40666153e641bbf613998841ec3f7d7d0c78c"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-01-08 16:55:47.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-01-12 14:29:29.000000000","tz":0},"subject":"wip: ClosingIterator class","message":"wip: ClosingIterator class\n\nChange-Id: Ia4ebe4869b45ab00d8d956ed327789e9e7117aed\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/154c3e1e1cc0e9f765fca55cd8726d5130a39784"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/154c3e1e1cc0e9f765fca55cd8726d5130a39784"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
