)]}'
{"id":"openstack%2Fswift~905748","triplet_id":"openstack%2Fswift~master~Id2ced8c6d8e817bdf3cfe9cdd57af8a156f37adc","project":"openstack/swift","branch":"master","topic":"fix-bug-2040332","attention_set":{},"removed_from_attention_set":{"36697":{"account":{"_account_id":36697,"name":"Yihong Jin","display_name":"Yihong Jin","email":"yihongjin404@gmail.com","username":"yihongjin"},"last_update":"2024-01-24 16:53:20.000000000","reason":"Change was abandoned"},"1179":{"account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"last_update":"2024-01-24 16:53:13.000000000","reason":"\u003cGERRIT_ACCOUNT_1179\u003e replied on the change","reason_account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2024-01-24 16:53:20.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Id2ced8c6d8e817bdf3cfe9cdd57af8a156f37adc","subject":"sq: some tests for account suppression","status":"ABANDONED","created":"2024-01-16 17:51:05.000000000","updated":"2024-01-25 00:00:06.000000000","total_comment_count":12,"unresolved_comment_count":2,"has_review_started":true,"meta_rev_id":"94ca886bc5188b08967c2c19e2d0fb1e09da6737","_number":905748,"virtual_id_number":905748,"owner":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"_account_id":36697,"name":"Yihong Jin","display_name":"Yihong Jin","email":"yihongjin404@gmail.com","username":"yihongjin"},{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2024-01-16 19:04:06.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"all":[{"value":2,"date":"2024-01-16 19:33:46.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":36697,"name":"Yihong Jin","display_name":"Yihong Jin","email":"yihongjin404@gmail.com","username":"yihongjin"},{"value":-1,"date":"2024-01-24 16:53:13.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2024-01-22 20:47:29.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"_account_id":36697,"name":"Yihong Jin","display_name":"Yihong Jin","email":"yihongjin404@gmail.com","username":"yihongjin"},{"value":-1,"date":"2024-01-24 16:53:13.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":36697,"name":"Yihong Jin","display_name":"Yihong Jin","email":"yihongjin404@gmail.com","username":"yihongjin"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-01-16 19:04:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-01-16 19:33:46.000000000","updated_by":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"reviewer":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"state":"REVIEWER"},{"updated":"2024-01-22 20:47:29.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"},{"updated":"2024-01-24 06:35:03.000000000","updated_by":{"_account_id":36697,"name":"Yihong Jin","display_name":"Yihong Jin","email":"yihongjin404@gmail.com","username":"yihongjin"},"reviewer":{"_account_id":36697,"name":"Yihong Jin","display_name":"Yihong Jin","email":"yihongjin404@gmail.com","username":"yihongjin"},"state":"REVIEWER"}],"messages":[{"id":"41883f28b250c46d0109a7da0d1a193cb785061e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-01-16 17:51:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"738c3706f64d5f13b1339cee58c34255318af575","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-16 19:04:06.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/2f57f919c340442f9ca92f18f33a8b37\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ebcd26329f9c416ba3c212388118d0a3 : SUCCESS in 8m 41s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/f585695431df485a96b0625239367140 : SUCCESS in 1h 05m 07s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1574835306a74576ba6665e5c56f43ce : SUCCESS in 18m 45s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1bf181ed3d774b7ebed8c4af11fe4ed8 : SUCCESS in 19m 01s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/58ab3e3203534ab38dec3754c0288109 : SUCCESS in 21m 55s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/6f5b9f5791b748ddb151825c854438f3 : SUCCESS in 9m 35s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/49cf36d994e34faa843dbeafad4bce48 : SUCCESS in 7m 13s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/e2c234b268ac40d989524e34ab63de17 : FAILURE in 21m 01s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/a87fa1d5369c42248b9098a95f6c5510 : SUCCESS in 18m 15s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/7ba4bcdc37594b2ea6b6b09516e712a0 : FAILURE in 15m 26s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/dc85b67be10947d5b56efb4535627465 : FAILURE in 14m 10s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/8c72e0135cf24d33a06b7c7a03c87f33 : FAILURE in 17m 16s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/84499cb6890b4bd9964c1fcd8844268a : SUCCESS in 18m 51s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/74d8ad30c2c64cf884be52d13f76a53e : SUCCESS in 20m 16s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/327ebf22b2124fb09426d895c030624f : SUCCESS in 20m 11s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/fb592383c56442d994a3d36cfc29b50e : SUCCESS in 18m 17s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/ee9c2f7e367a41aa9973f20a5bd00599 : SUCCESS in 16m 24s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/b4a1a8473dd847718b5a7544d6b85027 : SUCCESS in 16m 28s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/b84cf072a545401e992c36bea01df9f8 : SUCCESS in 50m 51s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/ccb1c1c737f045389894d92c744cee4f : SUCCESS in 34m 15s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/71e828d9011849b8a2f83507a2bcedc2 : FAILURE in 18m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ad2cd8e58efa44acbc51da185d01f962 : SUCCESS in 5m 55s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/9fe29f9cb2ce4669b940d5647e32f2d1 : SUCCESS in 20m 52s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f6c5062ce6e54457f4c9312a2faa883f938a0d13","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-16 19:22:29.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/61026d51ac3d424ea44858a4b9d0b665\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/4f3a5b4b428c4f9686e9bae4c28e992d : FAILURE in 13m 55s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/660bd26a2f8846f886e36c8150e1071f : FAILURE in 1h 05m 19s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/722f3753dfda442792e6a54bba111fb4 : SUCCESS in 23m 20s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/76b71a75a51f43869a360cc9a942b407 : SUCCESS in 1h 29m 39s","accounts_in_message":[],"_revision_number":1},{"id":"835b52cbb2508f18082e9380a567de0540204561","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2024-01-16 19:33:46.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5d44fd54d252a1eeb2a3448579fa9033ba06b656","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-01-22 20:47:29.000000000","message":"Patch Set 1: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"b3ef597776063dd20717882049d3861f8ee83c58","author":{"_account_id":36697,"name":"Yihong Jin","display_name":"Yihong Jin","email":"yihongjin404@gmail.com","username":"yihongjin"},"date":"2024-01-24 07:50:31.000000000","message":"Patch Set 1:\n\n(4 comments)","accounts_in_message":[],"_revision_number":1},{"id":"6c4924e1f4c377238462782e3cb7a8b4a2f8de37","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-01-24 16:53:13.000000000","message":"Patch Set 1: Code-Review-1 Workflow-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":1},{"id":"492fc437fc9005f71d1d56625ac7bfd2b3045bd0","tag":"autogenerated:gerrit:abandon","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-01-24 16:53:20.000000000","message":"Abandoned\n\nsquashed 905554: Container-updater: fix account-suppresion issue | https://review.opendev.org/c/openstack/swift/+/905554","accounts_in_message":[],"_revision_number":1},{"id":"94ca886bc5188b08967c2c19e2d0fb1e09da6737","author":{"_account_id":36697,"name":"Yihong Jin","display_name":"Yihong Jin","email":"yihongjin404@gmail.com","username":"yihongjin"},"date":"2024-01-25 00:00:06.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"418012f1ad3e54abc69c0488bd17b8e3d590677f","revisions":{"418012f1ad3e54abc69c0488bd17b8e3d590677f":{"kind":"REWORK","_number":1,"created":"2024-01-16 17:51:05.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/48/905748/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/48/905748/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/905748/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/905748/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/905748/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/48/905748/1"}}},"commit":{"parents":[{"commit":"c551a16db38e52d10e235b1acc7a668c19c822f3","subject":"Container-updater: fix account-suppresion issue","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c551a16db38e52d10e235b1acc7a668c19c822f3"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-01-16 17:50:56.000000000","tz":-360},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-01-16 17:50:56.000000000","tz":-360},"subject":"sq: some tests for account suppression","message":"sq: some tests for account suppression\n\nChange-Id: Id2ced8c6d8e817bdf3cfe9cdd57af8a156f37adc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/418012f1ad3e54abc69c0488bd17b8e3d590677f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/418012f1ad3e54abc69c0488bd17b8e3d590677f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":["label:Workflow\u003dMIN"],"failing_atoms":["label:Workflow\u003dMAX"],"atom_explanations":{}}}]}
