)]}'
{"id":"openstack%2Fswift~905945","triplet_id":"openstack%2Fswift~master~I524fe7164ce5106f3adcba4f479472baac9faa45","project":"openstack/swift","branch":"master","topic":"bump-eventlet","attention_set":{},"removed_from_attention_set":{"28522":{"account":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"last_update":"2024-01-18 10:39:37.000000000","reason":"\u003cGERRIT_ACCOUNT_28522\u003e replied on the change","reason_account":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2024-01-19 12:44:44.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I524fe7164ce5106f3adcba4f479472baac9faa45","subject":"bump eventlet to latest version that support python 3.12","status":"ABANDONED","created":"2024-01-17 15:30:40.000000000","updated":"2024-01-19 12:44:44.000000000","total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"de1ea4f828bbc055ccec45f387d5d20f3116f037","_number":905945,"virtual_id_number":905945,"owner":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2024-01-17 18:03:21.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"rejected":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-2,"date":"2024-01-17 16:17:00.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-01-17 16:17:00.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"},{"updated":"2024-01-17 16:45:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-01-17 18:03:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"54def6a0e4bb971627445bfb8c0cae6a14d9a8f0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2024-01-17 15:30:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1c0659ec77f63e0726361b91daa23e42c99b0c8b","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-01-17 16:17:00.000000000","message":"Patch Set 1: Code-Review-2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"87ca97cdc1ba636a86b250298ec34e9e72fd9181","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2024-01-17 16:43:24.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f66e5b083bd7969e49f4b87c857755cba6415f06","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-17 16:45:38.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b6c4566c0cf04ab3ae6f50a2ba24a3d7\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9a9e1a4114df4ead9a7b815f242a3a5c : FAILURE in 23m 16s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b39029d69caa4fedb3004b88c5e521e9 : FAILURE in 1h 12m 53s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/64b7798b4e6d42c88f3794798ff62826 : FAILURE in 6m 26s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/435e39c6bdd24f7c80f2e976fa222b2f : FAILURE in 6m 24s","accounts_in_message":[],"_revision_number":1},{"id":"bcfa6bb3ab32dbbb4912851aee1d679cb3c38145","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-17 18:03:21.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/147489e2104149728218d8c5b727de4f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5ab86fdfc30b43389c6720bc668ad6e5 : SUCCESS in 9m 29s\n- requirements-check https://zuul.opendev.org/t/openstack/build/cb1e57603f864d93a544595ffec6b00b : FAILURE in 3m 54s\n- grenade https://zuul.opendev.org/t/openstack/build/8c1fd80d6a644a7299b006ba98bdeb74 : SUCCESS in 1h 04m 41s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/85fdec2c460f4bc49b8d816b6530fd45 : SUCCESS in 1h 02m 47s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/a241d180acbc46f9aa0dce4835c57b2c : SUCCESS in 50m 27s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0aedb363ab7c435abb9ad51c3107c02a : SUCCESS in 39m 31s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/12ba265fd73040b99bfe3e4898dd6297 : FAILURE in 6m 23s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ee215873435a4236ac35c0165c3ef079 : FAILURE in 5m 36s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ad9b06da1fda4f90b8df5acd9065f264 : FAILURE in 6m 43s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/3efed54bd1cc446ab8d02de77278ae78 : SUCCESS in 7m 18s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/1201a16c916a4465bdad17257da16e2e : SUCCESS in 5m 55s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/590295ce46594afdb3431d99af0213a5 : FAILURE in 3m 41s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/6aa41e521d084c24a6741dfe44859f36 : FAILURE in 2m 59s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/a046bbd764fd41fabf4e41f7f991445e : FAILURE in 18m 10s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/83bedf38e28b469c92852afb289a1b2d : FAILURE in 16m 01s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/80a1a6a3c6b54669a6ef8443f5629144 : FAILURE in 16m 42s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/288cf7bfa5054d93921ecc0ec5af25a5 : FAILURE in 3m 57s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/c5a7af40858e4bfc876f2bfe05c5775c : FAILURE in 3m 23s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/201c38561331495c8aa84bbeffbe0c81 : FAILURE in 5m 20s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/dd4d7db4446f4bf3aa1b48831945dc69 : FAILURE in 4m 02s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/9db6e6ed7856437b9becdb00b7f163d4 : FAILURE in 2m 41s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/e2ff3f943e0041b395b3ac5c484c4857 : FAILURE in 3m 26s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/5877ce6398814124b5c116cda1579649 : RETRY_LIMIT in 8m 17s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/664e558eaf79435b8455a468b53c4fcb : RETRY_LIMIT in 6m 02s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/6ff0089c8d944e8680572b939869d6e5 : RETRY_LIMIT in 5m 22s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/0e94cd7508ee49a2a2ff5426569fee5c : RETRY_LIMIT in 6m 48s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/eaecb13e99c84bab9fc55d3d993612fd : RETRY_LIMIT in 5m 46s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/526b610c603e49eea8133ef60615b5fb : FAILURE in 24m 21s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/85d73326969c4ca9b355317124bf8da5 : FAILURE in 24m 20s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ef30fab554aa4521a552f7ef5e321eb2 : FAILURE in 4m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f2eae2f6544643ed9c294c9dc07825d1 : SUCCESS in 6m 27s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/ed81fdcf98a44c07afc6b4830abfd177 : TIMED_OUT in 1h 31m 03s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3e15da270bd34170bf1c2131f44e12c3 : SUCCESS in 1h 08m 04s","accounts_in_message":[],"_revision_number":1},{"id":"a710878d7ac3ec62fce4f213a72e18d6cccb6b5c","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-01-17 22:54:41.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"12df2f07323994322abd49f3e25bd0cfaa8dc06e","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2024-01-18 10:39:37.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"de1ea4f828bbc055ccec45f387d5d20f3116f037","tag":"autogenerated:gerrit:abandon","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2024-01-19 12:44:44.000000000","message":"Abandoned\n\nThis version have been reverted from Upper Constraints. https://review.opendev.org/c/openstack/requirements/+/905905","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b988aa526fa71c3e567e5a4d4036d26579b754ab","revisions":{"b988aa526fa71c3e567e5a4d4036d26579b754ab":{"kind":"REWORK","_number":1,"created":"2024-01-17 15:30:40.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/45/905945/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/45/905945/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/45/905945/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/45/905945/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/45/905945/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/45/905945/1"}}},"commit":{"parents":[{"commit":"2331c9abf2c4472d7274874cf9edc37cc764fa81","subject":"Merge \"tests: Switch get_v4_amz_date_header to take timedeltas\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2331c9abf2c4472d7274874cf9edc37cc764fa81"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2024-01-17 15:30:30.000000000","tz":60},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2024-01-17 15:30:38.000000000","tz":60},"subject":"bump eventlet to latest version that support python 3.12","message":"bump eventlet to latest version that support python 3.12\n\nChange-Id: I524fe7164ce5106f3adcba4f479472baac9faa45\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b988aa526fa71c3e567e5a4d4036d26579b754ab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b988aa526fa71c3e567e5a4d4036d26579b754ab"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":["label:Code-Review\u003dMIN"],"failing_atoms":["label:Code-Review\u003dMAX"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
