)]}'
{"id":"openstack%2Fswift~906228","triplet_id":"openstack%2Fswift~master~I39c9c268df80de97744cc0fbf578b5eaafbd2cc4","project":"openstack/swift","branch":"master","topic":"redundant_authorize","attention_set":{},"removed_from_attention_set":{"34930":{"account":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"last_update":"2024-01-22 06:45:38.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I39c9c268df80de97744cc0fbf578b5eaafbd2cc4","subject":"proxy: remove redundant authorization in container GETorHEAD.","status":"ABANDONED","created":"2024-01-21 20:03:58.000000000","updated":"2024-01-22 06:45:38.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"b413a0a9ac33a53975f27b1043a7c44b4cb0c20d","_number":906228,"virtual_id_number":906228,"owner":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2024-01-21 21:52:40.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-01-21 20:27:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-01-21 21:52:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9b3706c744480df8335083a7fe8990ed111b174a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2024-01-21 20:03:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6543b859d9ba13044d49aac535c9bce61d5f2002","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-21 20:27:05.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ed7597c15e144dd1bd0714389a26f221\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c4d422904f8d479c98340f91a75b8fc9 : FAILURE in 15m 29s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ebfec895a38c4d8f9f534958e5542c1c : FAILURE in 15m 14s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/78b032f2560b4b688ac704725bb85339 : FAILURE in 22m 10s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/14526bf818a14a1bb7f9c70f61a8b549 : FAILURE in 20m 06s","accounts_in_message":[],"_revision_number":1},{"id":"02666b879feb7b5c54e788d1e6339e45d515a837","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-21 21:52:40.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4c96c90e48e84051bd8f03eb361e4c8c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6623b1cbe10649a8b855f5d3a9463b56 : SUCCESS in 8m 35s\n- grenade https://zuul.opendev.org/t/openstack/build/20e411f6720945de9170f8cf4e440e9e : SUCCESS in 1h 07m 28s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/2b43d2d9947d4fbabc95ffdbc4f01bd0 : SUCCESS in 40m 43s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/b8543f26520844c4a6794b36506ec8c0 : SUCCESS in 59m 33s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e1874e49f2794ab2a3d06a55fd5679ac : SUCCESS in 23m 25s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b84745b0456b4519bfe6c1f4a39238cc : FAILURE in 24m 40s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/72c41fda781e407583a7c5b6e99586db : FAILURE in 19m 24s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8466f38a00a844048ca9da4168dcb538 : FAILURE in 20m 06s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/c9e3a2c31421440bb0c892547413c169 : SUCCESS in 9m 07s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/3164774432004f96a868b41b9a227742 : SUCCESS in 5m 12s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/1f9cdadd24bf4eb8979ad31f57fd03d7 : FAILURE in 12m 24s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/23f25ec3c8494d838be7bb16a0352c17 : FAILURE in 11m 38s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/522b5a40eda8435790fb8252cbc82998 : FAILURE in 13m 26s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/b0af62e34837494499c5cb1be79606a2 : FAILURE in 17m 40s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/ab7a1b4c714b4a908f9afdca62dce9c1 : FAILURE in 17m 34s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/4506a6d4ea0f49aa8c88f57a265cb3a4 : FAILURE in 18m 17s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/49eab1b619774a05bc5cb085ee888b90 : FAILURE in 21m 22s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/5f6986a56ede454e9e5bac875e839853 : FAILURE in 18m 18s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/a52d4e843c2f4e8ba2e8aeafaafb1457 : FAILURE in 16m 25s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/03c7e98ce62f45ea940a2dfd5929d580 : FAILURE in 14m 50s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/d4c60912cdfb4fda94371a7e5873ce13 : FAILURE in 17m 05s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/55f9f81c4e62453b92ce86681fc1a706 : FAILURE in 9m 15s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/7bbb703ec2ae4ba2a8a94952ad14fc16 : SUCCESS in 17m 30s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/0dd84a7013634119994e2fe5ad885033 : SUCCESS in 10m 50s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/d45bce2c44d54699ac78579c6e760e4f : SUCCESS in 58m 55s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/cd8a112c2e2645798fad762cc22de6e1 : SUCCESS in 1h 16m 15s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/52ced6723017470ab25e6d9cd6027c7d : FAILURE in 35m 27s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/b5b60f519d044a0ebb8f0f6273861fc4 : FAILURE in 26m 20s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/238004763f2c41aa9f1d9a2d28bb5419 : FAILURE in 18m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dee4ded86194416f9229c84d09c76a1b : SUCCESS in 5m 41s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/a775a65948294eb49751aa66465261f4 : SUCCESS in 21m 21s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7acc4600dd9e46f3b2d466e520541b59 : SUCCESS in 1h 06m 47s","accounts_in_message":[],"_revision_number":1},{"id":"b413a0a9ac33a53975f27b1043a7c44b4cb0c20d","tag":"autogenerated:gerrit:abandon","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2024-01-22 06:45:38.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"07fa6be97c36c02290eefe114453734df4559f30","revisions":{"07fa6be97c36c02290eefe114453734df4559f30":{"kind":"REWORK","_number":1,"created":"2024-01-21 20:03:58.000000000","uploader":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"ref":"refs/changes/28/906228/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/28/906228/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/28/906228/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/28/906228/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/28/906228/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/28/906228/1"}}},"commit":{"parents":[{"commit":"ca00e6b853458c11caed6d8cc83340ad56fd37a6","subject":"Merge \"CI: Treat grenade-skip-level jobs like grenade jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ca00e6b853458c11caed6d8cc83340ad56fd37a6"}]}],"author":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2024-01-21 20:02:33.000000000","tz":-480},"committer":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2024-01-21 20:02:33.000000000","tz":-480},"subject":"proxy: remove redundant authorization in container GETorHEAD.","message":"proxy: remove redundant authorization in container GETorHEAD.\n\nChange-Id: I39c9c268df80de97744cc0fbf578b5eaafbd2cc4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/07fa6be97c36c02290eefe114453734df4559f30"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/07fa6be97c36c02290eefe114453734df4559f30"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
