)]}'
{"id":"openstack%2Fswift~906338","triplet_id":"openstack%2Fswift~master~Ida70824287129fa3823b575c0bdffaa353c7bd95","project":"openstack/swift","branch":"master","topic":"p-proxy-namespaces","attention_set":{},"removed_from_attention_set":{"34930":{"account":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"last_update":"2024-01-23 23:54:27.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Ida70824287129fa3823b575c0bdffaa353c7bd95","subject":"sq: use guard return in _get_listing_namespaces_from_cache()","status":"ABANDONED","created":"2024-01-23 05:37:33.000000000","updated":"2024-01-23 23:54:27.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"dc645d947aca7962b54591540cf9840369d6f9b4","_number":906338,"virtual_id_number":906338,"owner":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2024-01-23 06:54:15.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-01-23 06:54:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"366b48b602c4650dbf563c5c7de6720ffe0ef10e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2024-01-23 05:37:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8050a5b6b2f39a54177ee8de2d8a60be3d8eccb6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-23 06:54:15.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f37473ebfdae4a33a72278a112ea797c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/904c3b3c385a45ec9b75a74261c6a541 : SUCCESS in 10m 10s\n- grenade https://zuul.opendev.org/t/openstack/build/b449800cdd6d4eee8bb7669c1f8b63af : FAILURE in 34m 58s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/a45d9404a80b4bf6aaf3275b60ffaeff : SUCCESS in 50m 50s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/706c14440b494c6ebc3a81e5cc3b7e39 : SUCCESS in 43m 00s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3cd6da632ef84c54814e611ce4d5a684 : SUCCESS in 32m 18s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5b8bde9e7afb455592c9f04ddd253a6a : SUCCESS in 22m 33s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1406b64bb301443886368cebfb34ab5c : SUCCESS in 20m 18s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7825088aa0314384a27d937dab0f2f3b : SUCCESS in 20m 41s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/a64a7b8eff314a848ac96fedd381bb34 : SUCCESS in 10m 51s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/bc1fbd8563954ea4a6d244c373088d3b : SUCCESS in 5m 38s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/4504b4e03ced45e1a458c1873a6e3e22 : SUCCESS in 19m 27s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/59f01e18266e4a19bb170d934e392449 : SUCCESS in 16m 29s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/f9945e0387404f589ab745390c350edf : SUCCESS in 17m 15s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/ec3cd9b1e9174accbe18af652e9d4277 : SUCCESS in 17m 33s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/e71c17f3bcde4d92a4ea4f9e92235a19 : SUCCESS in 16m 49s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/3ec0c08fd06e46e7905230cf7d112841 : SUCCESS in 18m 58s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/0b4880e36eb44effa92a32d16f7a3e8c : SUCCESS in 22m 54s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/994cb9144df64a50ad741cb2185d9c92 : SUCCESS in 20m 40s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/51fb82eff0174693b7c38f6e6fa355f6 : SUCCESS in 14m 35s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/78b92269bbd4415082664df66b310dfe : SUCCESS in 16m 41s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/5fcbdf7141c94a118da418c1f0141447 : SUCCESS in 16m 44s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/4224c00768994f52938b2893c1737f8b : SUCCESS in 6m 57s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/169e7b6c0a3641b9872eb4ac0e7cd128 : SUCCESS in 16m 42s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/3800396d1f98431cbd6e97658feec6f5 : SUCCESS in 8m 06s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/adb17e7ff2cf44d5b39bb3e8d06b718d : SUCCESS in 1h 08m 05s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/8689fe8f1d2143d0a03f7418ece053b3 : SUCCESS in 1h 03m 59s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/a936e59c34504afb974e26f6540ec0c8 : SUCCESS in 1h 07m 24s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/79df5e45f3ef462c93f7fbd513a43d9d : SUCCESS in 57m 58s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/780126c4335c44a19eada9fdcb552578 : SUCCESS in 14m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ed99bcc6991546fb805588f78312bbba : SUCCESS in 5m 45s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/e10e72d9345f4f409cc36aeab665590e : SUCCESS in 21m 30s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/759f1da93e2c4f2288d8a6b8ccdf6936 : SUCCESS in 54m 06s","accounts_in_message":[],"_revision_number":1},{"id":"3d1c3c97aebc2ad232bd47b8d2ce4063b6783178","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-23 07:10:41.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8b1557b39aed48bc80280b00e663a5eb\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/17f2b64dab1f480087f0b842d9e6e1aa : TIMED_OUT in 1h 30m 57s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d3f812cd50bd4bf790f8059511a7b4a0 : SUCCESS in 15m 54s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/533ce6f18d4348edb1a3cb7a66bbe7a4 : TIMED_OUT in 1h 30m 56s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/1f97099ee78f4a60854004c3fd7e3d28 : SUCCESS in 18m 47s","accounts_in_message":[],"_revision_number":1},{"id":"dc645d947aca7962b54591540cf9840369d6f9b4","tag":"autogenerated:gerrit:abandon","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2024-01-23 23:54:27.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e61d87871eb27b7aa1d210190f66cfa906e65907","revisions":{"e61d87871eb27b7aa1d210190f66cfa906e65907":{"kind":"REWORK","_number":1,"created":"2024-01-23 05:37:33.000000000","uploader":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"ref":"refs/changes/38/906338/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/38/906338/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/38/906338/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/38/906338/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/38/906338/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/38/906338/1"}}},"commit":{"parents":[{"commit":"b2a47afd4a788e499e533723a07e5755d16de0c7","subject":"proxy: only use listing shards cache for \u0027auto\u0027 listings","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b2a47afd4a788e499e533723a07e5755d16de0c7"}]}],"author":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2024-01-23 05:35:37.000000000","tz":-480},"committer":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2024-01-23 05:35:37.000000000","tz":-480},"subject":"sq: use guard return in _get_listing_namespaces_from_cache()","message":"sq: use guard return in _get_listing_namespaces_from_cache()\n\nChange-Id: Ida70824287129fa3823b575c0bdffaa353c7bd95\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e61d87871eb27b7aa1d210190f66cfa906e65907"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e61d87871eb27b7aa1d210190f66cfa906e65907"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
