)]}'
{"id":"openstack%2Fswift~906391","triplet_id":"openstack%2Fswift~master~I8fa3588d0ce92914bf27df5e58c81d6b5a07bdd5","project":"openstack/swift","branch":"master","attention_set":{"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2024-09-24 16:34:56.000000000","reason":"\u003cGERRIT_ACCOUNT_34892\u003e replied on the change","reason_account":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"}}},"removed_from_attention_set":{"34892":{"account":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"last_update":"2024-09-24 16:34:56.000000000","reason":"\u003cGERRIT_ACCOUNT_34892\u003e replied on the change","reason_account":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"}}},"hashtags":[],"change_id":"I8fa3588d0ce92914bf27df5e58c81d6b5a07bdd5","subject":"slo: Support range requests for part-number queries","status":"NEW","created":"2024-01-23 17:25:27.000000000","updated":"2024-11-04 20:46:03.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":6,"unresolved_comment_count":2,"has_review_started":true,"meta_rev_id":"c5c52c3af90981de9bf137886a8897fd97867be0","_number":906391,"virtual_id_number":906391,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2024-11-04 20:30:55.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-01-23 18:15:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-01-23 18:50:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-01-26 17:50:47.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"CC"},{"updated":"2024-03-15 23:37:04.000000000","updated_by":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"reviewer":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"state":"REVIEWER"}],"messages":[{"id":"4cf21d4078072d0019747c86c0cd31725c82c43e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-01-23 17:25:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7718e4ec11186c7c452491af949cc7c66c31b5b4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-23 18:15:24.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4ad178eaf9fc4375aae2243e10b46474\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/26fe197a1dd54c55b8c262e15940936c : SUCCESS in 13m 39s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/44f6e707288c4009b8c55b32f5cfe0d3 : SUCCESS in 15m 32s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/c1bb913dbb484dfb94f14a63dd892be0 : SUCCESS in 48m 04s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/647085a8f45f409491f8e2f3a1297a27 : SUCCESS in 23m 52s","accounts_in_message":[],"_revision_number":1},{"id":"2c45f65c4ec12a072c91ac5f9da1f21d63512cb6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-23 18:50:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/41c5d6e0043e4057aeb447a1cf318a30\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/079e96fe4a8443deaa3793c8a12fa5ef : SUCCESS in 7m 26s\n- grenade https://zuul.opendev.org/t/openstack/build/8068d31cc3e7485ca80cb31ca777a995 : SUCCESS in 1h 03m 54s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/fd09481af15147d98a37dfc109ad181f : SUCCESS in 41m 30s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/8f4e671923844508a3225604a7e4be67 : SUCCESS in 51m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/22251a46351341ac8ee1f9bc54df233b : SUCCESS in 24m 27s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d21482740d1e4a1bacc76edc995d700c : SUCCESS in 18m 56s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6654b1fc4ea9400b865c6bdd2214d306 : SUCCESS in 19m 12s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/cc54a8c8b8c94f969d8fe73c40969601 : SUCCESS in 21m 02s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/98b071f57802441e9c42642097d1b0cc : SUCCESS in 9m 59s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/0b0662a93ea1458eb66687382c81074c : SUCCESS in 5m 41s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/c175d169960c41d490056d6b66239073 : SUCCESS in 19m 22s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/c3e58372e0af4fa2844ebe09ce96dacc : SUCCESS in 17m 37s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/d85a780b7d1848f5bed1147761ecaae8 : SUCCESS in 18m 22s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/5d5ede1d66454dc89ff9ee537b8ff9cf : SUCCESS in 15m 55s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/235da25e540748b2b9e4b374b10565a9 : SUCCESS in 17m 28s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/a81ef9a340204b54ad193d557a4cf230 : SUCCESS in 19m 04s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/315f9aa09197431089aa947f5326ad38 : SUCCESS in 18m 32s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/c3519e232c7b422dbdc790305f235b04 : SUCCESS in 20m 38s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/7e4cb18529ff4003ae1dddf2ef094a42 : SUCCESS in 21m 25s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/601529fd4e3e4928b4b91ab824581382 : SUCCESS in 16m 39s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/591d1fb67a654397b82b5c66acfac971 : SUCCESS in 17m 50s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/696c1e8900064120a5ba3bc595f4b12c : SUCCESS in 7m 13s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/0f215a25c75e42b8960e8ee88f0f0215 : SUCCESS in 15m 34s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/17402cee535e4997b76b421560fc3d37 : SUCCESS in 7m 41s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/ab59c566df924dd782157b73a89c8ebb : SUCCESS in 50m 43s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/ca615b0c24274e97b1b66e71cfd7c34e : SUCCESS in 58m 17s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/62426d31ec29426ab5b8624c00eb19ad : SUCCESS in 59m 49s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/4cf97531f4bb4d53b2e03b9bbb1c5fdf : SUCCESS in 56m 33s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/040289aca3ee405daf009029deb0a2c0 : SUCCESS in 20m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fa72b5d7ad164fbe89f7554b4b9f69f3 : SUCCESS in 4m 18s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/7a515556d2e14e0b9261a5cf46899abf : SUCCESS in 21m 23s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/043b90d679dd4eb98b19c1074e3a93bb : SUCCESS in 1h 04m 50s","accounts_in_message":[],"_revision_number":1},{"id":"051a1ff1684fc419e47344cfedc1f32590664500","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-01-26 17:50:47.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"fd7c1fdc635975c8cd4bf7b18a5c5b138ecc2a20","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-02-01 00:01:08.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"f434f36dd74bdfad10432d80116641adebfea8a9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-01 01:07:13.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0e33c935009c4c43a2b1b71d9436cd5f\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0cc654f3584b4605b331c8e2af2580e3 : SUCCESS in 16m 57s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ab29bf074bf740f782f0583c4306297e : SUCCESS in 1h 03m 33s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/6ee51036fe1c43768d3248432ea1deab : SUCCESS in 21m 39s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/4ec25ddd4d5944579f9a36455fd69668 : SUCCESS in 57m 04s","accounts_in_message":[],"_revision_number":2},{"id":"5296ab90a842338d15274ce6f5b33aa2f0bfc802","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-01 01:30:15.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cd6db0865b874a9e81675ba4dcbf7046\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8b8065f490bf4e16a8c2289c3a3cddc8 : SUCCESS in 8m 55s\n- grenade https://zuul.opendev.org/t/openstack/build/380ea92a7efe4da29e9c0809d771ebb3 : SUCCESS in 43m 37s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/d916d586b3b8456e953a67327bb46dc9 : SUCCESS in 1h 20m 32s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/d89be8613bdd49ef9de8f5cecd4a41c4 : SUCCESS in 41m 33s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6baeb385406c4819b1ed453e26033be7 : SUCCESS in 37m 07s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/bd273aed24b9496a985ae41ece23aec0 : SUCCESS in 22m 19s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/73315eaebb6f42619233ac0f67cbbb9e : SUCCESS in 17m 18s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1543eb0002364065ad7b20e5e7eb1c94 : SUCCESS in 25m 29s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/a63eb95d3b5a4c4683dd92bd1e47c5cf : SUCCESS in 10m 56s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/a0366dd5dedc4d1d907a883f29170d4d : SUCCESS in 5m 52s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/3b63762c815346a4ba4834701cb8d6be : SUCCESS in 17m 50s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/3a5db38162fa4373945c740c7c93714d : SUCCESS in 17m 26s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/96b04e8ecd1f4de3a241ce49dc5b6493 : SUCCESS in 17m 05s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/aabb0e2a00494c4293ce14705bea61c4 : SUCCESS in 14m 00s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/e133e44142304b5ba49320db7c280be6 : SUCCESS in 16m 13s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/ce93ee33244c4057bf9d3bd64f6318f1 : SUCCESS in 15m 17s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/b1bbcddbe3674a52b9b0eca78bfb2bb1 : SUCCESS in 19m 11s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/c63fcbc4807042deb7799c65bb6c61ab : SUCCESS in 19m 48s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/d7936030ab8e4ee7af7cd83db0092527 : SUCCESS in 14m 00s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/76f5110e93d34ec3a493e5bd5010a974 : SUCCESS in 16m 36s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/e1e1be9d4e3e4344af1c1df9374ac6bd : SUCCESS in 14m 48s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/707e6d2e374c4982be7a879f5f2c0b48 : SUCCESS in 9m 03s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/f4a7c6cf983942a093aad48669618b0e : SUCCESS in 14m 35s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/0121b0583ab6465f87c1ff4fe249fd70 : SUCCESS in 8m 01s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/a860da2592f64c79a5f0efaf534816b8 : SUCCESS in 53m 51s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/4ceb6e51c75c40a89db96cf0610abdd4 : SUCCESS in 1h 01m 08s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/6a74044d33c5418db993ab2fe3355c15 : SUCCESS in 55m 37s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/0bd18493e6694bc792d0e2b383ab14dc : SUCCESS in 54m 58s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2fc6125143fb49ccaf491d7f9ab554ae : SUCCESS in 24m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/855fefb809df4224b604274796ba706c : SUCCESS in 4m 59s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/55e833597c53436d9dfe8d9876909808 : SUCCESS in 24m 41s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9e9ed074bd064194a1ac2f347f41267f : SUCCESS in 1h 01m 11s","accounts_in_message":[],"_revision_number":2},{"id":"ece68b6200d12ea16bd1406bf7082cba27c8703e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"real_author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-02-20 21:16:24.000000000","message":"Patch Set 3: Patch Set 2 was rebased on behalf of \u003cGERRIT_ACCOUNT_15343\u003e\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"_revision_number":3},{"id":"4736a7a5c0b252b4e66e7e294ec6ae7263b8fd77","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-20 22:13:33.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1773a2f7bdb94210a226248c983aba3d\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d32eaab466324deb94fd1395e30d3ad2 : SUCCESS in 16m 05s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9419a6c97d5f4ed0965213b40905ab50 : SUCCESS in 55m 29s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/5fe87e67423e4059b4d395a2adcca64b : SUCCESS in 20m 04s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/ec0ef4f2105b4629b01d8b180f617c9e : SUCCESS in 52m 45s","accounts_in_message":[],"_revision_number":3},{"id":"2021e10c7398ecda248124e7697993f76256839a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-20 22:30:25.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/30784b0cf1ba4a28879b7d683ea8e17a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/17bde557e9b646c3b5289659bbf603df : SUCCESS in 9m 16s\n- grenade https://zuul.opendev.org/t/openstack/build/1f945154ac314faf8ea5687695f816a8 : SUCCESS in 52m 32s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/9023a311236c4076be135ac680462913 : SUCCESS in 54m 37s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/d06d73671b7642b18e502543c2b92c17 : SUCCESS in 1h 06m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bf42d778e17d476698e5289abeb025d6 : SUCCESS in 44m 36s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d494fff07a0c4e5699d14b8bed1985bc : SUCCESS in 22m 25s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f23953a91f694f33a19fa22d54e9570d : SUCCESS in 19m 17s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/91724cd5c9a844a1b6e3d6431de540bf : SUCCESS in 16m 41s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/cfdeac6b1f284f5e824e5be65b376917 : SUCCESS in 11m 05s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/9b8f395da508433280750d4880527bbd : SUCCESS in 5m 20s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/3ab25f5611b74dbfb07ab41dd15b4293 : SUCCESS in 26m 48s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/006f09255e6643f0b1b646384183f754 : SUCCESS in 18m 50s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/26da73952604407199763fb94d4bb55c : SUCCESS in 12m 30s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/1a729cf65bf74738920b190d4bb30df8 : SUCCESS in 21m 07s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/2ef1eee51a33451188f5e8411493f6d1 : SUCCESS in 12m 42s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/81f38b1b5f3d443eae67f244a4ba7d72 : SUCCESS in 23m 06s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/6a1cc1733af946f8ba681bbb0a9e581e : SUCCESS in 18m 25s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/172d701ca6d641d5a20deeb892d40163 : SUCCESS in 19m 06s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/8cd0106135b24be38f554b5f18d48780 : SUCCESS in 16m 53s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/5e152a684e514f70be1f5532be3aed65 : SUCCESS in 14m 15s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/112223d742b748e0846333f57c759d1c : SUCCESS in 18m 40s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/2c3d67b76dfb4d31964302ebd53ad3b8 : SUCCESS in 8m 06s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/3e3202c619ae4914a7663a50e91c15c0 : SUCCESS in 15m 16s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/143370f8f1924458a8864d91e72acf5d : SUCCESS in 7m 20s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/da3543e7310e4a05a6a9fd063230d9ba : SUCCESS in 54m 22s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/63fc6494244d4214a88fb6f95d777de8 : SUCCESS in 51m 28s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/7a36bc34fdc740eca3be79818d2155fb : SUCCESS in 53m 30s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/ea4d7d896c8a41469bffc7c9ab5fbc4a : SUCCESS in 53m 54s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a4786766682449e481b15630e503c6f4 : SUCCESS in 20m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c41702ecada84ea98a78145739510e5a : SUCCESS in 4m 21s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/d5bce4bdd4494549b348487e41443130 : SUCCESS in 17m 22s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e848c679f4f7459e936395d866a918ad : SUCCESS in 1h 04m 25s","accounts_in_message":[],"_revision_number":3},{"id":"6883568888f2f0aeaead851e4227c52be1f86a99","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"real_author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-02-27 18:44:37.000000000","message":"Patch Set 4: Patch Set 3 was rebased on behalf of \u003cGERRIT_ACCOUNT_15343\u003e\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"_revision_number":4},{"id":"adb275bdfed4496c78ba1d672e56d76672fb9128","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-27 19:52:54.000000000","message":"Patch Set 4:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7fbef3d04a7c434ba2a73c4f60a370f3\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/fe1b3c96f06d494fb023457735f12f8b : FAILURE in 18m 25s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2fe486a0b42e459195984284058cbd81 : FAILURE in 19m 12s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/517da1c20a5247d3b04d6e9252337b27 : SUCCESS in 1h 07m 45s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/5442733e99dd4cc58a57492ed4d8a479 : SUCCESS in 55m 19s","accounts_in_message":[],"_revision_number":4},{"id":"2630dd8d452ccfec1361e7761453e826d186a6a5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-27 20:00:11.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9f0582d198c94490b73c446bcc96e044\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d2c02515f6994a538be74b5a76090f74 : SUCCESS in 11m 20s\n- grenade https://zuul.opendev.org/t/openstack/build/30c35019612141c78bd8d6e0adebe1e8 : SUCCESS in 57m 49s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/d382a6d92abb44cb91b2f1e00a5677ab : SUCCESS in 58m 34s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/c06d99b50b994be580ff2d50e8d29715 : SUCCESS in 46m 08s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e5e04f20319a4a33aa27e340e288395e : SUCCESS in 56m 59s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a40b1059864e412b808d1de1c4693b2f : SUCCESS in 22m 30s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/badcc8470d994f1b8a2a9f6210ecb14b : SUCCESS in 21m 43s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/86a9b58ed30a40fca2e3a32992ca8331 : SUCCESS in 19m 54s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/96a19b99c35e43fd9d7f47e9f8f43f5b : SUCCESS in 12m 26s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/ddc2d82f42ef400d8f91db27b313567a : SUCCESS in 5m 49s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/ed48bc48b2c84c1390225a84343037ca : FAILURE in 19m 39s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/75ab0b1246aa43738ccb0a8aca58f891 : FAILURE in 23m 08s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/2f9e8eaa3f8f4fc1b97c28fc606a8651 : FAILURE in 19m 35s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/01f392d1ea7342a5bba359a61ac18f70 : FAILURE in 21m 35s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/9264de864b8444328b91cc4fa57d5940 : FAILURE in 19m 23s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/43f81f11a3894fc48449af87abccda62 : SUCCESS in 22m 40s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/a5cb3832e0da4667ad57aee623f44490 : SUCCESS in 17m 13s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/20e56b7ad26542599f635933510059ba : SUCCESS in 19m 54s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/e5f4c5ae33524a90bf1bdc4f0a1b78f8 : SUCCESS in 14m 19s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/df426949b07944c5a811d16e063cbb14 : SUCCESS in 33m 09s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/98fb9421a9e24023a29c614568b6d4b2 : SUCCESS in 34m 53s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/2b071bc74f8442e694980a344a49247d : SUCCESS in 9m 15s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/782e71ff446b43eb8894e2e0223699f0 : SUCCESS in 18m 54s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/9ab731856b13401da72aa4f93bec6ddb : SUCCESS in 10m 41s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/c1c3ae7df27c4ea0a87af49e30a540a8 : SUCCESS in 1h 09m 52s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/69df038a8c194d5aa6cee24aa7baa4e4 : SUCCESS in 1h 04m 49s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/a5f4f251e3b04535926542082dc62e64 : SUCCESS in 52m 50s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/bbbe87ab3a3d4d019f95cb1f0e2f698f : SUCCESS in 59m 50s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/75473dee822c402cafc3ff5505bbe481 : FAILURE in 16m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/78e32ec4a48c4bb5bff0f5b7459d6e79 : SUCCESS in 4m 43s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/9a8b4c9fab3f4b3586747778e43de89b : SUCCESS in 21m 00s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/109c21a34f064b2ea832dded61bd70c9 : SUCCESS in 1h 02m 21s","accounts_in_message":[],"_revision_number":4},{"id":"2477b10ee379302d20a7568f2f15d2608ddcb6c8","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-03-13 16:11:36.000000000","message":"Patch Set 5: Patch Set 4 was rebased\n\nThe following files contain Git conflicts:\n* swift/common/middleware/slo.py\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"a30e4d244cc9cd901b65f513df778e225a80d225","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-13 16:19:43.000000000","message":"Patch Set 5:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2fd40986614646039a4fdc615845f109\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1c4099f6bb6d4e42a38416e14b8983ee : FAILURE in 6m 16s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d8f291329cd849b389fde8200dce613b : FAILURE in 5m 01s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/239551890deb4fc2a0de1ad89592f552 : FAILURE in 4m 34s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/3eb9ee56863b44f9a6028e9b584b231a : FAILURE in 4m 23s","accounts_in_message":[],"_revision_number":5},{"id":"51fdec2ae8aa46b1a43c9de667ee0a30e424ee1f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-13 18:17:57.000000000","message":"Patch Set 5: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/fca3e6c8e8bc43e19d03d8a0e431e617\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7d2b13fb67a34478aeeff567993783de : FAILURE in 6m 10s\n- grenade https://zuul.opendev.org/t/openstack/build/feeac3f494fc4bd9a1a304779fb74818 : FAILURE in 37m 24s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/9da4f438ce34440fb34e3cac7a8d0cfc : FAILURE in 39m 35s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/1c6f7d725c4140c082ba4b38baf8c99e : FAILURE in 18m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4a4c3474c3cf4a52aba3da4764b3078f : POST_FAILURE in 19m 09s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d3bbcd749a594bc99c39b10b275b06ec : FAILURE in 8m 09s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c8625d17db7d48b2af0a008324409e46 : FAILURE in 8m 20s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/638588a16bfe423d87fc9be14b58d71c : FAILURE in 6m 35s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/16e93aa6e1d640a78485237a8110dfd2 : SUCCESS in 11m 49s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/db11bbf2888c4802a51a99f1c25d5233 : SUCCESS in 7m 02s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/1bfd6f7f6afa4ade8e7528d9a2bb9ef1 : FAILURE in 4m 34s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/6c183168be37475fa569368fff93b803 : FAILURE in 5m 11s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/5acb44bacd2e4901b4b0f237650a0470 : FAILURE in 5m 37s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/b041fe163fbf45fb93241c93ef4f0f87 : FAILURE in 4m 49s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/e1a6572d649446709b2a1e89633c515d : FAILURE in 3m 34s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/fa0cf9ca141447d2be4fd15f58c56b9f : FAILURE in 7m 29s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/612db37b800148b4b9471aa07c6492f1 : FAILURE in 4m 34s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/eb977b19b5334b5a932bea4de8fc258a : FAILURE in 4m 07s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/8a80196effb441d4b91445bc4bb86746 : FAILURE in 3m 49s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/37b48358f946497f81af9ddd6f039ce3 : FAILURE in 3m 54s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/46b18e3a435d4c969ca07fba96c9a53d : FAILURE in 6m 19s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/1c1abaf8d4614584825c275deb3b09c2 : RETRY_LIMIT in 5m 15s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/6835fe1e14a94ccc94e1c5b983330552 : RETRY_LIMIT in 5m 40s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/a6d5bd58d6b94ee285235ad8bc93fe2b : TIMED_OUT in 2h 00m 51s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/4de1f6f476054e29b65fe32f92af3e3d : RETRY_LIMIT in 10m 51s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/22abcfd0655e41a8ac66bb0eb43587c0 : RETRY_LIMIT in 22m 26s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/abafe6560fe44f418543bb5b375d44dd : FAILURE in 5m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/10c57204dffd4531b8c25da3f47e51eb : FAILURE in 4m 42s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/33c9cc0c6bd84708b0d35ac8257ee737 : SUCCESS in 18m 23s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/50fb5e32464a4716ae34d27f90978a3b : FAILURE in 18m 25s","accounts_in_message":[],"_revision_number":5},{"id":"48b3415c6732b9628e7cef49eed36d33dd7f1286","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-03-14 20:52:57.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"bc10959120528445ca58d3fb004c0a1c3bfcdcda","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-14 21:12:51.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/28ecbdbd46ce4d5196211c7a024901e5\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/254a85765394449186e6445606cffb28 : SUCCESS in 16m 33s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/27670b03b90b45729d320e6f04dd8c37 : SUCCESS in 15m 59s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/28c6d511db104338ad615ea52789b32c : SUCCESS in 18m 38s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/1e35688f2b254f60b4fddb6b8c44ac90 : SUCCESS in 18m 49s","accounts_in_message":[],"_revision_number":6},{"id":"92954926727506266b216b0b90aeecfec09329dc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-14 22:03:27.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d5ec2521dccf4dc899fb165754980fdd\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/167c833bd6c04a4c949cb78af0b2df45 : SUCCESS in 8m 31s\n- grenade https://zuul.opendev.org/t/openstack/build/a616d7525be84846961103ebc1a306c7 : SUCCESS in 50m 39s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/b58d379cebb64724a003b6e3398f3a5c : SUCCESS in 57m 17s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/64a26f7fa4ec4188b1a0211ef08c31f2 : SUCCESS in 50m 34s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/117e05f366ce4115b3879f49363ac98e : SUCCESS in 57m 48s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/fa9efffb152940ffb3b62665486fd1d0 : SUCCESS in 16m 05s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a4aa7346ee244b15856e7d130a116e65 : SUCCESS in 15m 44s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/014477a8ccbf4c459b8121977c7aa898 : SUCCESS in 20m 40s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/1a52880aa2dd46bb897766e7b3dd76a0 : SUCCESS in 9m 00s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/1c25ab71f6254ffc8bcb62d9f55b6af3 : SUCCESS in 5m 17s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/9c21039357f64ff49812925845645d63 : SUCCESS in 14m 18s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/9fe59684f4c54ef78511d9eed7291505 : SUCCESS in 13m 16s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/bc2d630107274dfe97715e6246c88263 : SUCCESS in 13m 20s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/baede66df4804f0181004f67411c1b00 : SUCCESS in 15m 09s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/b833f2ec5c9442f8b794e469be0abdf7 : SUCCESS in 17m 21s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/a9e92091b1ad4e5f9bbfc22a4e702d16 : SUCCESS in 20m 28s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/8369768171b74f5aa897dc56ea4e1d9b : SUCCESS in 20m 08s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/a1d8c2cc2c5648329bc3a0a54709dc52 : SUCCESS in 16m 33s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/a336a8c15ebb45d1bf77ae37018a8cbe : SUCCESS in 15m 23s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/ed18a1b4fdbe48449934a72f73a4a05c : SUCCESS in 19m 17s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/971774f22cc34c58af5730073bc5e63f : SUCCESS in 6m 53s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/36a0d6bc1cb546c8b6ce85cb1955f8c2 : SUCCESS in 19m 22s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/1ad9e51bfacc4fca8b0cfa90a932dfb7 : SUCCESS in 6m 17s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/f0f16564e5d24222b7cb2434c62ce560 : SUCCESS in 1h 05m 31s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/39adffad43944eafbb7c6a2070a1deed : SUCCESS in 43m 25s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/d4b04d9b44ab4e6ea86de8431e37a46d : SUCCESS in 53m 00s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fe0c22d0053244e897065d158b960cb6 : SUCCESS in 23m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7ae3b878f12c47879bb933d94b224dc7 : SUCCESS in 4m 35s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/1d64375422ee4f44975bfe93d99e367c : SUCCESS in 17m 39s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b52279cf477d414898b804a2f6c653f3 : SUCCESS in 59m 45s","accounts_in_message":[],"_revision_number":6},{"id":"fedebba75f6b4f44f6d4f4ee30182c92734245b8","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-03-15 16:22:00.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":6},{"id":"f4fff9ebfe621b84a1607222789046b3525dee90","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-03-15 16:22:05.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"fe491e74973759bf44190802c7a0423fdb959eeb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"real_author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-03-15 23:36:52.000000000","message":"Patch Set 7: Patch Set 6 was rebased on behalf of \u003cGERRIT_ACCOUNT_15343\u003e\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"_revision_number":7},{"id":"2dac6f987fd45f26f717a9a771c11737e2357f50","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-16 00:18:22.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1f889a952e7a4c2c9571eddd48cc4ec9\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/8581c44565be4a978d15711f58eaa85e : SUCCESS in 15m 14s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/cda94d522ea14e74bfe6e8cc7048577d : SUCCESS in 17m 31s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/ade4f95ee4624f08bf79b584545e18e2 : SUCCESS in 20m 55s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/aeadba8882104e9280fc76ae98620c1d : SUCCESS in 40m 04s","accounts_in_message":[],"_revision_number":7},{"id":"2b7ca46985291b6b24267dee1123db824feaabba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-16 00:49:38.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a7f9129cee09464ab9acce2cf17e532a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/11b4e152e3da4f4ca5f7242ded812a08 : SUCCESS in 8m 25s\n- grenade https://zuul.opendev.org/t/openstack/build/028cc39bc84146eaa98477481fcbb65e : SUCCESS in 45m 00s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/561dfbcfc53342ec947ec831418ab04e : SUCCESS in 44m 09s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/3888d5fb9f8b4d41b1ddbecd004cf93f : SUCCESS in 35m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/edbd547fb15749e18861c45bdbe0d2cb : SUCCESS in 1h 02m 08s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/267150bed095474aacd023a9d5ab5bac : SUCCESS in 18m 33s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/cde5a558313a412aa720a57d7160f60e : SUCCESS in 21m 31s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c012843274944e17bc6fa7da5b3880b4 : SUCCESS in 18m 44s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/372a4a0703514b958e2497167f993fb5 : SUCCESS in 10m 15s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/e02b6dad3fdd425785e1e58da3bbdff6 : SUCCESS in 5m 03s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/ce60dedaba954c809084d1af31ee640c : SUCCESS in 19m 49s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/969f8ad342da4e00a7ec5679b9fcae3d : SUCCESS in 16m 24s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/a7762f268f7d4b59ad387e9b55ef2b0d : SUCCESS in 17m 38s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/6903e3e3bc87478c9ab11359d8a6d6ec : SUCCESS in 15m 26s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/e790f5b6af7e445c81696a996ed81eaf : SUCCESS in 18m 19s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/7f82be31e28b44a584bcd26fd44b3f07 : SUCCESS in 18m 58s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/3166e3b48780430c93f67feb2bac420d : SUCCESS in 20m 10s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/395cd94e780943c2970d70a1e444b81f : SUCCESS in 15m 30s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/b6f0cf042df0449e93856f91507f72fb : SUCCESS in 15m 45s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/5e375b28fe0546c9983d3253d8df175d : SUCCESS in 16m 27s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/e95730a36f794406a9d0f26923a1e7c5 : SUCCESS in 6m 13s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/c8100a627a37486f9aff8bea90c3bd23 : SUCCESS in 15m 43s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/d45965539beb4ad3a4fc6fa1e3534462 : SUCCESS in 6m 56s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/c054c97fc2a340618ea7143490540ce4 : SUCCESS in 50m 43s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/96404e676bb0447ab475c3e3898f2a28 : SUCCESS in 32m 52s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/5793736283644654885a3527d76a6a49 : SUCCESS in 55m 03s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/06dc2256d99c43fd9aabd50e5d2f9e60 : SUCCESS in 19m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5ac54e8fd1b1484a8cfa060293c17c19 : SUCCESS in 4m 28s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/e91fb66325f74caabae28ae8228569e9 : SUCCESS in 16m 58s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c00c5d2baebc4892aabd5b58f2e1ff70 : SUCCESS in 39m 24s","accounts_in_message":[],"_revision_number":7},{"id":"693d1a8a8663ebc6d96d165ff163eab1c382ea06","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"real_author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-04-16 14:52:56.000000000","message":"Patch Set 8: Patch Set 7 was rebased on behalf of \u003cGERRIT_ACCOUNT_15343\u003e\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"_revision_number":8},{"id":"6be8f2cdac772218c87f26705ef33e74868a7979","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-16 15:44:29.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/914f9c6da1e445acb16dfd3291a2dc93\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/edb5ffa91aa64aa9aab90b9d5d550d39 : SUCCESS in 15m 22s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7a596f03635d4a11ac3ab1c9a2b0c086 : SUCCESS in 15m 35s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/e35386cbb80a482b85c1d4542e7f77cf : SUCCESS in 42m 27s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/d66988fdcfcb464d86f78814fb59bc09 : SUCCESS in 49m 50s","accounts_in_message":[],"_revision_number":8},{"id":"5c815300de0716dddf037568ae2074221b666319","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-16 16:01:05.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/66fa7f4bbe744577ba8f1377ffeb2419\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e7379379f0294f3f83c40bc91f4ed456 : SUCCESS in 8m 00s\n- grenade https://zuul.opendev.org/t/openstack/build/339338fb65564f2182060521698140da : SUCCESS in 57m 45s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/2d9ea1eca4e6473da38c74a114bdaf99 : SUCCESS in 42m 51s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/063b9406f39c4e439b4eaef483f82930 : SUCCESS in 1h 02m 28s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7b0c763d1dd64920bd90292a5db6fbdc : SUCCESS in 18m 53s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/234c09675e5d42bbae4a761b04f00a48 : SUCCESS in 15m 28s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/44ff8b4fd14545168b5fd1abc935beea : SUCCESS in 20m 00s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/dcb8851a22f1406c984bf1c26bdb5fbc : SUCCESS in 9m 17s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/f5d7a36466d741aeaaf0ed7af822fc2c : SUCCESS in 4m 56s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/6a444137ff42417fbbc18795b5f6b486 : SUCCESS in 16m 57s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/37dda17f515545a8997fecc5c8d94a2f : SUCCESS in 22m 21s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/855c7c0ed9ca4451ae79026c68359d5c : SUCCESS in 14m 05s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/999f2074968c449bbcabe4606092bb8d : SUCCESS in 23m 46s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/e5e940a1d7a44020a4a1dc2b39c26aa5 : SUCCESS in 16m 35s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/1484560bda27405ab88b87e261d71140 : SUCCESS in 23m 40s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/1f1c76c9e46b4b79a52aaeb3d26935c9 : SUCCESS in 14m 56s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/4971c1e7d3d54b689c6c1329a948c7c7 : SUCCESS in 16m 51s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/58e0cda3b251457ba9b758877fa7ddc2 : SUCCESS in 13m 33s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/b5ca0806b2394a02ba8a0088123f0248 : SUCCESS in 23m 05s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/a3654b9bd99e4873afde5b6dd9fca27e : SUCCESS in 5m 23s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/d03b14e62a5249d192bd74827fd4d7e1 : SUCCESS in 18m 48s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/b5ae48741baf40578d2b8c3757122107 : SUCCESS in 7m 24s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/c4ed4e10d45a45b3959c2cf9787b346d : SUCCESS in 52m 38s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/58ac840c4f3540dc9a526915b41c30b5 : SUCCESS in 49m 59s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/60fb4ac1936e48f08196f0b51013d9b1 : SUCCESS in 48m 40s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e22d091dd8114004a6511548f8a00e24 : SUCCESS in 20m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/81e04f53d36044fc9d53cdbd7ef3810a : SUCCESS in 3m 49s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/93b881ba036e488fae158fe5cddcc613 : SUCCESS in 20m 03s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/03a1f50d4bfd4a9e8ea45890d5368790 : SUCCESS in 56m 43s","accounts_in_message":[],"_revision_number":8},{"id":"0cbe0529eaaab00d56c57602fb71eb94fa0189b9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"real_author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-05-16 22:24:50.000000000","message":"Patch Set 9: Patch Set 8 was rebased on behalf of \u003cGERRIT_ACCOUNT_15343\u003e\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"_revision_number":9},{"id":"ca14d48564f32924f82bb1e42d2d37278884d885","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-16 23:07:54.000000000","message":"Patch Set 9:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7caf0b6fe2b649d5935bd5b006a6be89\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/4364837083ba4c51866195703d99ef21 : FAILURE in 16m 57s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0d2bac65109542ea9861898ce724b15e : SUCCESS in 17m 42s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/fa02071318ce4a63a1100c92893e517c : SUCCESS in 41m 32s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/566352276c574ff094ff27141256df58 : SUCCESS in 20m 15s","accounts_in_message":[],"_revision_number":9},{"id":"04422f211ef89b3ea367b080ed45172f2f66cd92","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-16 23:45:46.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/63cef4e0fbd64fafa0a23ed76a51375a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c6d8d57bf49d4881a1264e6f4b200119 : SUCCESS in 8m 29s\n- grenade https://zuul.opendev.org/t/openstack/build/ad4d79a9795b42ff826158190f4d4689 : SUCCESS in 1h 00m 52s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/f81bd1bea0954512b303ac3aea52f177 : SUCCESS in 57m 27s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ff51899b5d2d462d893a5b6b6e1711e2 : SUCCESS in 1h 03m 10s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/331ce549b5d34f08b3ef44ef19a468bc : SUCCESS in 22m 10s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1e0b8b0aacdf4542a762c29b75be75dc : SUCCESS in 19m 06s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/558712475af548e8b222e343d794282e : SUCCESS in 25m 09s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/efc52042dae44fe6bd25e63e4a36abe5 : SUCCESS in 10m 29s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/ceb4ff692fdd42ce8d696a1720e7bc79 : SUCCESS in 5m 38s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/911b1b7da2444e3ab220c41c8f439552 : SUCCESS in 18m 55s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/ddd6d0676a7b40259dde6a3aa8675ade : SUCCESS in 18m 35s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/4332694ffab141edbc75d4bad5cece5e : SUCCESS in 19m 47s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/947540782e4f4b5bbfc0d0e912149f4c : SUCCESS in 18m 13s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/5dfa1a50b0d14e43a83c0ba9360e53ff : SUCCESS in 23m 03s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/9429991106cd47d3b944285bc24c5747 : SUCCESS in 17m 21s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/224c63dd0687419283019cc91034052d : SUCCESS in 19m 36s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/8391fddd4df8422ab539f12dffd1a8b1 : SUCCESS in 18m 44s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/637ad9eb64af417ca31c773976164d63 : SUCCESS in 14m 11s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/bf291e1e841c476f93d375ec1c0f0cf7 : SUCCESS in 17m 23s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/d36b844f8de940aab43afd80c02140b9 : SUCCESS in 17m 35s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/ff0816c75b81450eb4535378a8b5a21a : SUCCESS in 6m 49s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/9d17848188e34748807f6995318940d6 : SUCCESS in 15m 46s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/3cfaf3b123e544b4aa2c16f9cef2338a : SUCCESS in 7m 00s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/da424eeb98ef4015ba8f59fa83043af0 : SUCCESS in 55m 37s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/3a31951b209842749aeb8dab678629ac : SUCCESS in 55m 23s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/477d9608f26245759b1e530a96e80b66 : SUCCESS in 55m 55s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b8a201209e064381b6951a38e1c2b43b : SUCCESS in 14m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d2ba35277abe4d589ad9982d25f96a83 : SUCCESS in 3m 55s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/bec869eb2dcf4240bdb724be35b00447 : SUCCESS in 20m 16s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/aa32ca8d9873403d91dc33d7696375f3 : SUCCESS in 1h 02m 21s","accounts_in_message":[],"_revision_number":9},{"id":"05fe5435231ca6b10a352e74c0843a75ac934df7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"real_author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-09-23 18:59:53.000000000","message":"Patch Set 10: Patch Set 9 was rebased on behalf of \u003cGERRIT_ACCOUNT_15343\u003e\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"_revision_number":10},{"id":"2b4f1930f81ed9c55204b41d8830f98bdebd934b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-23 19:53:28.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3fd1d8fc6bd340cfb644aa52f66d0f5b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/523406cba0314cbd8852d179c3e03ecb : SUCCESS in 9m 19s\n- grenade https://zuul.opendev.org/t/openstack/build/1e2f9d9bd4764339bd94078403a1c9f6 : SUCCESS in 52m 51s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/3c29c88702004e79ace9bb00cbf14f42 : SUCCESS in 48m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/09b20e5aa856432aa409d208a77e1a8f : SUCCESS in 25m 55s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ed467aee751e49f4a3c228b9628770df : SUCCESS in 17m 38s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a22e56bf468f42bca133505dcbbf2bf7 : SUCCESS in 18m 00s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/38ab99599bf1472f8ef9fd97089588b8 : SUCCESS in 11m 00s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/008924e015834c758f750da6bd8a59da : SUCCESS in 6m 50s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/42f3c84f08ed4339b31c1b4fec86c277 : SUCCESS in 17m 53s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/6ec44442ad034e7eafd2810586878cd0 : SUCCESS in 18m 57s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/1210f0846478475facfc401f022ab71d : SUCCESS in 16m 01s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/f47b65fa85a44cc7992d3e63d579d782 : SUCCESS in 24m 41s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/827c1683f16d439a8231de9c7a0fd1a6 : SUCCESS in 19m 22s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/a2861fde930746e1b0f46c265d26031e : SUCCESS in 14m 42s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/d335e6aceb3449eb9c9ec2311038f214 : SUCCESS in 14m 59s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/92c1915c4fcb48e8bf2dd2df2de38d79 : SUCCESS in 17m 37s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/ccb3128099c74190a3a6bf0f408d7216 : SUCCESS in 12m 37s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/002a08fcba244f8d857c90958bdbc5ff : SUCCESS in 16m 52s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/a3893356c6f44b0abef9ba1a9a56026a : SUCCESS in 13m 18s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/6672819d41634df5b903e40ca0f11427 : SUCCESS in 4m 57s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/0f8f5c9d547046349a99682bfcd96f14 : SUCCESS in 14m 52s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/61c7bfbca2404a6fb45d13f0ef65f372 : SUCCESS in 6m 15s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/e976fe678f6a4f479c306f73af274f8c : SUCCESS in 45m 14s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/0e323eb3e07348c191a21d18a1d838f6 : SUCCESS in 36m 47s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/415da8e398dc4a2eb951020d1b1ba0e7 : SUCCESS in 29m 59s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5c0b32ad59bf45b6b11dd89dc2ee3c3a : SUCCESS in 19m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7fbdfa49216a442780d3dd627ff6bbd6 : SUCCESS in 4m 03s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/65e9008f882f4b94a283c98eaebe4528 : SUCCESS in 18m 02s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bbcc7e05aa7349d4af6b58fb82caef39 : SUCCESS in 48m 24s","accounts_in_message":[],"_revision_number":10},{"id":"80f2882ddd01cd021243f466a13e2c4370734ba8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-23 21:23:46.000000000","message":"Patch Set 10:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9c07ed7cf2cf4aff815c4e2b6616f1e1\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c2b0f14882c742a290c2573162efcb58 : TIMED_OUT in 2h 00m 58s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/13a82187bdaa4a17a10699d2d918d915 : TIMED_OUT in 2h 01m 06s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/10bbb4d1f3d041dbb201f2800cdbadc9 : SUCCESS in 1h 44m 38s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/56eb098230e44aa19370abad30ec6547 : TIMED_OUT in 2h 00m 50s","accounts_in_message":[],"_revision_number":10},{"id":"931f369c387366c685da2fffe1596e72c300d83b","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-09-24 16:34:56.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"bb508d758d23836b1172216a39097122238a2490","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-24 17:39:18.000000000","message":"Patch Set 10:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/01e1e2475d3a4602ae1fea7bab54d996\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/09b944a5b51d4238a32f565b99b6cabf : SUCCESS in 10m 16s\n- grenade https://zuul.opendev.org/t/openstack/build/e646ae0e564c481e83a912fa1ab8cdfc : SUCCESS in 56m 08s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/79f4a37393b941d7b4b1d2b5271b8083 : SUCCESS in 41m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a3595c69416b4e1cb57418097c7452a5 : SUCCESS in 21m 37s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/defa7c11a12d48c09ab0e89d4b125be8 : SUCCESS in 17m 25s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b0cfed9fe639446f9e353ccaafeb15f2 : SUCCESS in 17m 11s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0254f66d264d42999f3419ae3ba6627f : SUCCESS in 18m 03s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/7f3c64007c00443e9573cf780628307a : SUCCESS in 6m 45s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/d311a777a63746369b598df4951711a4 : SUCCESS in 17m 51s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/e0cf25ed7f0e4174bac30f2ddd9e83a5 : SUCCESS in 20m 34s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/e3f0c3912dd447b18d7c88c2a22fffd6 : SUCCESS in 18m 40s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/618fdb549fb240439f4eefb87fe48fed : SUCCESS in 18m 05s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/0805cb53e9d6460b93bcbf8863c944da : SUCCESS in 19m 35s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/4c656316f0fd4f5bbaf0b4837a4e36b4 : SUCCESS in 15m 39s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/375eef138dda42249c05d78bf3338feb : SUCCESS in 11m 58s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/7da05b76a0f94bb5b7cd6e52ac6d3e93 : SUCCESS in 17m 59s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/f2a70cf2da1d4520b431b20027bf4dff : SUCCESS in 15m 16s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/bca0f05b7f3e48efb42c67b6ddf7e729 : SUCCESS in 15m 34s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/c3581bb554b4477689979a288ebf66b4 : SUCCESS in 9m 11s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/6e58ca1d2484401184b23019bb134941 : SUCCESS in 6m 42s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/69cc7d0b09ab4f8ab34aa8224a1ea6c6 : SUCCESS in 10m 49s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/d5b34ff4f1ab4153ae2e921d51d8e228 : SUCCESS in 6m 05s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/67ad53afe3e54f2eb5b90fd7389461f9 : SUCCESS in 51m 28s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/af90050b334949db9cb81ebbb6306191 : SUCCESS in 58m 03s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/e93b495aafbd4cf18af5316fbe81fee6 : SUCCESS in 43m 49s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c79be96e8ac34d789a1cbb621e8b764e : SUCCESS in 20m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ea602cd8e9f04b8fbd66569e313c1821 : SUCCESS in 4m 32s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/8be8e37522224fdfa55aa253053514bb : SUCCESS in 24m 11s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9faa46ac780b48709b2a593c2c91d386 : SUCCESS in 42m 54s","accounts_in_message":[],"_revision_number":10},{"id":"40325cf66a53c699de383fc030a74eb7b5946e4e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-24 18:36:05.000000000","message":"Patch Set 10:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/941c7d338e264a5e939ec95e4af62ed4\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/59f36b6ca4c64b04ac90444f75429a7a : SUCCESS in 1h 14m 21s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/de745763dafb4e399f2644dc2b662a81 : SUCCESS in 1h 51m 16s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/9340f184ebf64bedb94a1aa69461dcef : SUCCESS in 1h 59m 07s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/f37674f3b7124b43b58bc5ced3a512ef : FAILURE in 1h 34m 35s","accounts_in_message":[],"_revision_number":10},{"id":"7531646cbeb08d162a3c23cf9689203b083879a9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"real_author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-10-02 16:10:24.000000000","message":"Patch Set 11: Patch Set 10 was rebased on behalf of \u003cGERRIT_ACCOUNT_15343\u003e\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"_revision_number":11},{"id":"14abca96e140766bf940863b724bc2e3036fd86d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-02 17:18:50.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f7da7fdbf63c41499e22826d1d239bd9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/338233578ea14362bc2bb26b9dfe4cf2 : SUCCESS in 9m 23s\n- grenade https://zuul.opendev.org/t/openstack/build/55d3548ab1bf49889426b5346617a5a7 : SUCCESS in 57m 25s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/8f1b81db8f2c433db9ec7f9ab8698472 : SUCCESS in 1h 00m 48s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/8a1b5d03c9ec40a7b6c01c2a79118dc6 : SUCCESS in 50m 14s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/52c50ff1c8e446c1a86cac73ea1a00b4 : FAILURE in 29m 10s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7c76e362c812419bacde44f643e2514f : SUCCESS in 17m 31s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/35676eb0681b4c13a5ab3c66de9dbfdc : SUCCESS in 17m 30s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/bcb5b141baba4c1c9b993a4938c148a4 : SUCCESS in 18m 24s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/910537555f4041218430fa4252129b4d : SUCCESS in 6m 57s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/17a2f3b3f1764e28b67e1d6a2e3a3371 : SUCCESS in 19m 20s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/511d42c3354a4906a66d6202ffa50a9e : SUCCESS in 17m 08s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/a949a56ed8154241a68b1799efaa6f19 : SUCCESS in 19m 35s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/0eda1a204bc2496fbc4d0f33d0b4fd00 : SUCCESS in 21m 32s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/8291d45dae724c969512bce24ba7e6e4 : SUCCESS in 25m 59s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/8b9da2612a014d64b5e002bcf66c7a9a : SUCCESS in 18m 44s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/0ca504c5f2d144f7b43903ba6bbe1a33 : SUCCESS in 10m 47s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/fa62b9bdad7a40c5bc5b2be246c18196 : SUCCESS in 10m 51s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/8cdef7454e6346389a0a8f9130580b67 : SUCCESS in 14m 33s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/d3cce6e501d2483c87f67f237b266067 : SUCCESS in 15m 05s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/e3df71aa5bcb4d4e85d5a11bae0858be : SUCCESS in 8m 57s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/57b3b094e54344df9bd9f3ff8a21260c : SUCCESS in 6m 49s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/6a44d119e3a542c392be0c693e93616b : SUCCESS in 17m 06s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/1dcc4d6661ef41feb4de58211db1decb : SUCCESS in 6m 21s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/2ae49024ee2144f1a8e56fbe726f4ac4 : SUCCESS in 55m 28s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/4b6f378cbab34d04a25658b3d060a4da : SUCCESS in 45m 33s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/d993daf6a95f4ff5b4eeef55f59ad5c5 : SUCCESS in 45m 37s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8eabcb690d7a494dae0bf1849d406c41 : SUCCESS in 23m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bda6a5b82c3a46a29ca6fb8edbd21044 : SUCCESS in 4m 25s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/9fe99745a1ea4ebfad1cc6808a4326a5 : SUCCESS in 22m 27s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c252faa2fc7b4a389f16cc4d28f7fed1 : SUCCESS in 44m 34s","accounts_in_message":[],"_revision_number":11},{"id":"cf65b70df2b196545850f8aded8155d4bb1a0b7e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-02 22:14:07.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ba78083f54714ac193ea29494d717f31\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d0c37847598243ac883d1463e890d1fa : SUCCESS in 59m 19s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0194d44b464c44c5a4acd926affffa8c : SUCCESS in 1h 10m 14s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/6c2b1a727786410ebdfc172840d1f578 : SUCCESS in 1h 42m 25s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/4c84df9b1d624d66bffa9f7b6545ebbd : SUCCESS in 55m 45s","accounts_in_message":[],"_revision_number":11},{"id":"03ca902c08386b5bc0a7de8f081cc38491067d91","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"real_author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-10-09 23:33:59.000000000","message":"Patch Set 12: Patch Set 11 was rebased on behalf of \u003cGERRIT_ACCOUNT_15343\u003e\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"_revision_number":12},{"id":"eac1a3e7502b30cbcab7f6f7c741417398eb48b2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-10 00:48:26.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/11c62e764550447e9523d88b4b467d56\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/df32692d32f5456b94d31454394f8182 : SUCCESS in 6m 37s\n- grenade https://zuul.opendev.org/t/openstack/build/60864357aa754c3087393b5b7d7c1b2d : SUCCESS in 1h 08m 47s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/87a437a6fa744997b9991f33b19a0696 : SUCCESS in 1h 00m 44s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/960842dc818449b39befa2b5113f7ff6 : SUCCESS in 57m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f8edefbb926f42669507c674463891b7 : SUCCESS in 52m 51s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a44c0f2da48545d4b40bfb77719b54be : SUCCESS in 13m 20s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/356824d8169e4b48b6bac0823b4cec3c : SUCCESS in 13m 25s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/723f3cb567ee4b53bd70884924d4276b : SUCCESS in 19m 38s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/50923a3d725146fb9134b88e5f3739d9 : SUCCESS in 4m 59s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/ed21ecdfddcb45668ddd34d94be0fea4 : SUCCESS in 18m 27s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/c18ce00f85ed41d6917b47cd6b2b1698 : SUCCESS in 13m 49s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/553c6e61ccf8443b8893bfaa55a436bb : SUCCESS in 20m 20s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/fb26c4d229c8498d8428e49990dea81d : SUCCESS in 21m 53s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/23d676742970461b98956475ab08659e : SUCCESS in 23m 10s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/4e9fc08e1a514c81a08504d971148273 : SUCCESS in 18m 16s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/953a2bfa58c84c6a8c4c41118a1593b0 : SUCCESS in 19m 16s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/d5f7dcb1a0754ff98687ae8f38e9080b : SUCCESS in 15m 58s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/7f69507152934073acfa37fdab40306a : SUCCESS in 14m 55s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/edb3eaecfddf41918595cc0a6381ee0f : SUCCESS in 15m 23s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/8af3d7aa97534b3884f33d9ecb5263b1 : SUCCESS in 15m 06s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/d2d1c015bad54bf0b5a0ee65131df353 : SUCCESS in 5m 42s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/eb7180db7e0c43b4aac8e63e364de2ca : SUCCESS in 14m 13s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/b7043a655e8c494f8c750ec440c258b1 : SUCCESS in 5m 57s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/9a315cd3a7d24da2bbfa53eb58ce7903 : SUCCESS in 49m 53s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/15810372b3f34f608143e93a4a9eba37 : SUCCESS in 39m 37s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/db5529f6903843c78025b2780cebb3b2 : SUCCESS in 39m 59s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/75739c9b57504ba8b1b8e065d86a1994 : SUCCESS in 15m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d48e63db4147478f97bf45a94e47a4df : SUCCESS in 3m 34s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/fd3d27ba169748a6b6bdd10bf9388f5c : SUCCESS in 16m 32s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e24636170c9142649a99006d5cfda564 : SUCCESS in 1h 00m 11s","accounts_in_message":[],"_revision_number":12},{"id":"cdbc7de0cb9d135ee6b59fce3a297f290b75ac9c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-10 03:49:57.000000000","message":"Patch Set 12:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dc02ba80b46745fbb3fd15949aacac0b\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e8e95815c3b54b77b40ddbdcefed2ce7 : TIMED_OUT in 2h 00m 56s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/942e37c7d4df45e388fc93b3f7bbdabe : TIMED_OUT in 2h 00m 59s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/088fae9acf744ed099299ebca39dffd6 : TIMED_OUT in 2h 01m 01s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/34abbd1a164643debd5de6b2014222b9 : TIMED_OUT in 2h 00m 54s","accounts_in_message":[],"_revision_number":12},{"id":"e35415b582993373a36bf6ec367133f6bc3d7369","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-10-16 16:42:08.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"f0d428d9c46098b73548255602478508a377bf07","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"real_author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-10-16 16:57:19.000000000","message":"Patch Set 13: Patch Set 12 was rebased on behalf of \u003cGERRIT_ACCOUNT_15343\u003e\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"_revision_number":13},{"id":"bb0c6daa63f8626581a5b719bb355a985f1ed854","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-16 17:52:58.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/afd3a990152248f08f3ad581d7eb5ba6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8a3b1336ae9540529c537df2742d8c59 : SUCCESS in 6m 23s\n- grenade https://zuul.opendev.org/t/openstack/build/df25c9b340744ae19f119f708ad7849c : SUCCESS in 51m 39s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/3ba93898124f422abae00a3e0d48d7c7 : SUCCESS in 47m 57s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/ba4860cb60d444ca90c16fbb4299dcd2 : SUCCESS in 47m 27s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/214ea457c09e453a9132cee953cf798f : SUCCESS in 52m 01s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/02c00d6896f9485fa27a264856d1f9da : SUCCESS in 18m 39s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e7e141d717ff405297059ab7b0db6a3d : SUCCESS in 19m 02s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c15f7f300b7b4d568ec6c54466b43cb4 : SUCCESS in 16m 17s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/96577daa97c8406fbd961d0619a57f9d : SUCCESS in 5m 25s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/8a5b4604b65a422ca5d06b2b25672383 : SUCCESS in 19m 38s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/fc8384b767bc402e9cded8b4494f0406 : SUCCESS in 18m 41s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/9c3d0b7b51fc4223a2fc3e147215362b : SUCCESS in 20m 59s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/09caabe9dbd84e68b1eb39d20b48707f : SUCCESS in 20m 18s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/23cf985fbf114fa4af83ad691381948f : SUCCESS in 34m 53s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/4da03dce872a40528e5427b027e00d72 : SUCCESS in 16m 07s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/f31f2b39df424510ba4fd035cc1d04cb : SUCCESS in 16m 22s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/f7ac2037ec404d5b96afff6f8170912a : SUCCESS in 16m 22s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/676b628608624532832dbfc3a7ae2768 : SUCCESS in 16m 10s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/b2b79324e5044ffe837a5acf7d49cb6b : SUCCESS in 9m 10s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/071ad235bafe4acbbdb5832474e21fa3 : SUCCESS in 13m 41s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/7dd62acb1f094ae5a8cb02a2586b1741 : SUCCESS in 5m 28s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/b6709287f6f34671a61c3cb27b95a98d : SUCCESS in 13m 05s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/3205d7c5dc7d4121ad487f9ae5630954 : SUCCESS in 4m 39s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/4e0209811aa14698b4827e9509db1541 : SUCCESS in 48m 25s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/e2cdc4dec18143e1bba88d3e6aa2efca : SUCCESS in 43m 28s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/5c24c4ca520a460f82ef62e7465bbacc : SUCCESS in 42m 09s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/101b83e4bf19405e8916436fac15ab32 : SUCCESS in 10m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/974d35c2673e45b7ac90d375acfd7fd4 : SUCCESS in 6m 57s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/8adfd34a66b24e9caf54222885dd811a : SUCCESS in 22m 23s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8ba3f4f7c6bb4daba274811d51374336 : SUCCESS in 53m 29s","accounts_in_message":[],"_revision_number":13},{"id":"720d96c21ad44d6b98c0cdf0f16bb8ab2752d204","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-16 22:48:22.000000000","message":"Patch Set 13:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5cf34d043eb943688914bb0334dd0155\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/c6a70bd1c99e471caa2493f6a0113031 : SUCCESS in 2h 16m 53s","accounts_in_message":[],"_revision_number":13},{"id":"7624d1192d897389a7940fc4cffe637ee946643b","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-10-16 23:40:56.000000000","message":"Patch Set 13: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"46dac82769a23b2213336abe2474414c96eb302c","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-10-16 23:41:47.000000000","message":"Patch Set 13: -Code-Review\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"eeed477f087b2dafdaf3e7e62777fcaa97c00174","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"real_author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-11-04 18:21:40.000000000","message":"Patch Set 14: Patch Set 13 was rebased on behalf of \u003cGERRIT_ACCOUNT_15343\u003e\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"_revision_number":14},{"id":"2e5e33924d0faa3e4bd32e000c1676d1b074de05","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-04 20:30:55.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/fb024cf31af042d58dad80c7ac2b8015\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bcb8e5cdf890445583487de27977df4e : SUCCESS in 7m 12s\n- grenade https://zuul.opendev.org/t/openstack/build/a5e379e0a6ea4dfb88cb0ccf94788a83 : SUCCESS in 49m 19s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/8702b62796d04959b867bb5e739bec5d : SUCCESS in 44m 50s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/468da6aab3e44d8f9ddd97138d5c7184 : SUCCESS in 32m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c7b90d2a57e443be94d9c1d78725ba29 : SUCCESS in 47m 43s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/fa5c278cac064b4c8cf80f0e9b0651d2 : SUCCESS in 15m 42s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a0cf8466555445c39a4fd46378ea7235 : SUCCESS in 20m 53s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/67a0140eda494595bf5e503833b2acf4 : SUCCESS in 21m 34s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/d7fa2880255b4fd3b63970fcaebbd864 : SUCCESS in 6m 47s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/425a723aab174b2b84fa290d11faf0fa : SUCCESS in 19m 10s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/f48103fd690f4eccb2ec86f7fd6fefeb : SUCCESS in 19m 28s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/bb78d689667744bb85ec84c7b798d3eb : SUCCESS in 19m 11s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/6573280b0f8040fca2aadf96b9a42828 : SUCCESS in 22m 09s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/03e6e216191e46239477789e234e9359 : SUCCESS in 15m 18s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/b9d7241ae2ed432585329e972ed1b60f : SUCCESS in 16m 58s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/68af56d30cfd463dafb63a4ad8efa817 : SUCCESS in 9m 59s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/df78935b18b84e548d4a71eb50dd8bc4 : SUCCESS in 15m 42s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/c9737ae000c44a009e91ea7f8eb850a6 : SUCCESS in 18m 16s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/a6b36c844fc0413099617fa817028005 : SUCCESS in 16m 59s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/62b3450203cd49398249c3603856a7b2 : SUCCESS in 6m 11s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/28361737b85e413ea7b87d357c7cd06a : SUCCESS in 9m 07s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/7e5eee54c4724ebb8d9623b6708c87d0 : SUCCESS in 6m 25s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/f3c56b4a61794878917912c46849ad77 : TIMED_OUT in 2h 00m 47s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/8dd12ea0b62e452d8db4107b35b732d8 : SUCCESS in 52m 37s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/ec43e9b4c1ee4939931e9593c5b64539 : SUCCESS in 54m 46s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5e4c9e16210a4a5b88416e92a7e2916a : SUCCESS in 14m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/360fa9a3699348bbae520072616b24f9 : SUCCESS in 4m 09s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/fa70157013e64999ab224e0d092c6f9a : SUCCESS in 19m 03s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/128d9b45583248c69ea5848adb22c82e : SUCCESS in 40m 46s","accounts_in_message":[],"_revision_number":14},{"id":"c5c52c3af90981de9bf137886a8897fd97867be0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-04 20:46:03.000000000","message":"Patch Set 14:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4b4c87fee2474a048717038be66bc6b1\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/b18b22e54b5d4bb09780020034dd2829 : SUCCESS in 2h 16m 20s","accounts_in_message":[],"_revision_number":14}],"current_revision_number":14,"current_revision":"4c3046cdbda0cea743744109850ddaca8d3245c5","revisions":{"2bdf2489e71a75730adbe5905aa62075c0cbbbee":{"kind":"REWORK","_number":1,"created":"2024-01-23 17:25:27.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/91/906391/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/91/906391/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/906391/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/906391/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/906391/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/91/906391/1"}}},"commit":{"parents":[{"commit":"4a321be2853d2838069d8f0c28419fe8e3bd6e66","subject":"slo: part-number\u003dN query parameter support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4a321be2853d2838069d8f0c28419fe8e3bd6e66"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-01-23 17:10:14.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-01-23 17:24:56.000000000","tz":-480},"subject":"slo: Support range requests for part-number queries","message":"slo: Support range requests for part-number queries\n\nChange-Id: I8fa3588d0ce92914bf27df5e58c81d6b5a07bdd5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2bdf2489e71a75730adbe5905aa62075c0cbbbee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2bdf2489e71a75730adbe5905aa62075c0cbbbee"}]},"branch":"refs/heads/master"},"86ebd5b868babe5bd3b25134971b3168dda6a067":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-02-01 00:01:08.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/91/906391/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/91/906391/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/906391/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/906391/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/906391/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/91/906391/2"}}},"commit":{"parents":[{"commit":"6c8d82d6f4b03f684841d309173d6119ece1cc8b","subject":"sq: cross-compat test single-part obj partNumber and Range","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6c8d82d6f4b03f684841d309173d6119ece1cc8b"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-01-23 17:10:14.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-01-31 23:52:49.000000000","tz":-480},"subject":"slo: Support range requests for part-number queries","message":"slo: Support range requests for part-number queries\n\nChange-Id: I8fa3588d0ce92914bf27df5e58c81d6b5a07bdd5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/86ebd5b868babe5bd3b25134971b3168dda6a067"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/86ebd5b868babe5bd3b25134971b3168dda6a067"}]},"branch":"refs/heads/master"},"4dc17adcecea8c383bb596840f2bc648dea69f70":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2024-02-20 21:16:24.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"real_uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/91/906391/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/91/906391/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/906391/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/906391/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/906391/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/91/906391/3"}}},"commit":{"parents":[{"commit":"c36e6197e9774888d6e671fb0dde8e7ce7f7da2a","subject":"s3api: return 400 if partNumber and Range sent","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c36e6197e9774888d6e671fb0dde8e7ce7f7da2a"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-01-23 17:10:14.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tburke@nvidia.com","date":"2024-02-20 21:16:24.000000000","tz":0},"subject":"slo: Support range requests for part-number queries","message":"slo: Support range requests for part-number queries\n\nChange-Id: I8fa3588d0ce92914bf27df5e58c81d6b5a07bdd5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4dc17adcecea8c383bb596840f2bc648dea69f70"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4dc17adcecea8c383bb596840f2bc648dea69f70"}]},"branch":"refs/heads/master","description":"Rebase"},"bc273636b3a25040deac916744fcd11a0e872645":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2024-02-27 18:44:37.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"real_uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/91/906391/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/91/906391/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/906391/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/906391/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/906391/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/91/906391/4"}}},"commit":{"parents":[{"commit":"a37f0bc6c174412e48a5255e47b7e28bcff1e984","subject":"s3api: return 400 if partNumber and Range sent","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a37f0bc6c174412e48a5255e47b7e28bcff1e984"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-01-23 17:10:14.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tburke@nvidia.com","date":"2024-02-27 18:44:37.000000000","tz":0},"subject":"slo: Support range requests for part-number queries","message":"slo: Support range requests for part-number queries\n\nChange-Id: I8fa3588d0ce92914bf27df5e58c81d6b5a07bdd5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bc273636b3a25040deac916744fcd11a0e872645"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bc273636b3a25040deac916744fcd11a0e872645"}]},"branch":"refs/heads/master","description":"Rebase"},"d5eb5d63edbc6c9600f0ef5032aff853af3d0c99":{"kind":"REWORK","_number":5,"created":"2024-03-13 16:11:36.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/91/906391/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/91/906391/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/906391/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/906391/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/906391/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/91/906391/5"}}},"commit":{"parents":[{"commit":"891d06345e337189fdbb213fc91528a20e3f4c23","subject":"Merge \"s3api: Support GET/HEAD request with ?partNumber\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/891d06345e337189fdbb213fc91528a20e3f4c23"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-01-23 17:10:14.000000000","tz":-480},"committer":{"name":"ASHWIN A NAIR","email":"nairashwin952013@gmail.com","date":"2024-03-13 16:11:36.000000000","tz":0},"subject":"slo: Support range requests for part-number queries","message":"slo: Support range requests for part-number queries\n\nChange-Id: I8fa3588d0ce92914bf27df5e58c81d6b5a07bdd5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d5eb5d63edbc6c9600f0ef5032aff853af3d0c99"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d5eb5d63edbc6c9600f0ef5032aff853af3d0c99"}]},"branch":"refs/heads/master","description":"Rebase"},"56714662aa9a5f85d256b7d81c2de1dcbd32751f":{"kind":"REWORK","_number":6,"created":"2024-03-14 20:52:57.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/91/906391/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/91/906391/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/906391/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/906391/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/906391/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/91/906391/6"}}},"commit":{"parents":[{"commit":"b6dc24dbc0eca46141544dcdaabe7cdc281c9ce7","subject":"Merge \"s3api test for zero byte mpu\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b6dc24dbc0eca46141544dcdaabe7cdc281c9ce7"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-01-23 17:10:14.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-03-14 20:52:48.000000000","tz":-420},"subject":"slo: Support range requests for part-number queries","message":"slo: Support range requests for part-number queries\n\nChange-Id: I8fa3588d0ce92914bf27df5e58c81d6b5a07bdd5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/56714662aa9a5f85d256b7d81c2de1dcbd32751f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/56714662aa9a5f85d256b7d81c2de1dcbd32751f"}]},"branch":"refs/heads/master"},"851222b4277a7c3923ca802207176e4d88ffa01d":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2024-03-15 23:36:52.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"real_uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/91/906391/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/91/906391/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/906391/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/906391/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/906391/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/91/906391/7"}}},"commit":{"parents":[{"commit":"27ef11ea14240d8611dd1cbadaf8412727c85f49","subject":"test: implement cache expiration time in MockMemcached","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/27ef11ea14240d8611dd1cbadaf8412727c85f49"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-01-23 17:10:14.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tburke@nvidia.com","date":"2024-03-15 23:36:52.000000000","tz":0},"subject":"slo: Support range requests for part-number queries","message":"slo: Support range requests for part-number queries\n\nChange-Id: I8fa3588d0ce92914bf27df5e58c81d6b5a07bdd5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/851222b4277a7c3923ca802207176e4d88ffa01d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/851222b4277a7c3923ca802207176e4d88ffa01d"}]},"branch":"refs/heads/master","description":"Rebase"},"037b3bba5d395ede67fe54e3826c4f0a96855b0e":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2024-04-16 14:52:56.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"real_uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/91/906391/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/91/906391/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/906391/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/906391/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/906391/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/91/906391/8"}}},"commit":{"parents":[{"commit":"6387949a540e871e19062de8131503a8ac2ceef7","subject":"lower-constraints: update to mock 3.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6387949a540e871e19062de8131503a8ac2ceef7"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-01-23 17:10:14.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tburke@nvidia.com","date":"2024-04-16 14:52:56.000000000","tz":0},"subject":"slo: Support range requests for part-number queries","message":"slo: Support range requests for part-number queries\n\nChange-Id: I8fa3588d0ce92914bf27df5e58c81d6b5a07bdd5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/037b3bba5d395ede67fe54e3826c4f0a96855b0e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/037b3bba5d395ede67fe54e3826c4f0a96855b0e"}]},"branch":"refs/heads/master","description":"Rebase"},"999ef44f31b34c28251dcff79da935bd20a9a8d9":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2024-05-16 22:24:50.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"real_uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/91/906391/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/91/906391/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/906391/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/906391/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/906391/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/91/906391/9"}}},"commit":{"parents":[{"commit":"a1916855c204e0a18d57d855d2e684f1a291709c","subject":"Merge \"bump lxml to 4.2.3\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a1916855c204e0a18d57d855d2e684f1a291709c"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-01-23 17:10:14.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tburke@nvidia.com","date":"2024-05-16 22:24:50.000000000","tz":0},"subject":"slo: Support range requests for part-number queries","message":"slo: Support range requests for part-number queries\n\nChange-Id: I8fa3588d0ce92914bf27df5e58c81d6b5a07bdd5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/999ef44f31b34c28251dcff79da935bd20a9a8d9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/999ef44f31b34c28251dcff79da935bd20a9a8d9"}]},"branch":"refs/heads/master","description":"Rebase"},"be609813b5949386d972a6a38111fd4f68b73545":{"kind":"TRIVIAL_REBASE","_number":10,"created":"2024-09-23 18:59:53.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"real_uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/91/906391/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/91/906391/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/906391/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/906391/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/906391/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/91/906391/10"}}},"commit":{"parents":[{"commit":"94dc4cad090390b1826bf6a2fbe5986e1516219e","subject":"Merge \"Quiten boto logging in func tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/94dc4cad090390b1826bf6a2fbe5986e1516219e"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-01-23 17:10:14.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tburke@nvidia.com","date":"2024-09-23 18:59:53.000000000","tz":0},"subject":"slo: Support range requests for part-number queries","message":"slo: Support range requests for part-number queries\n\nChange-Id: I8fa3588d0ce92914bf27df5e58c81d6b5a07bdd5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/be609813b5949386d972a6a38111fd4f68b73545"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/be609813b5949386d972a6a38111fd4f68b73545"}]},"branch":"refs/heads/master","description":"Rebase"},"48483a82e36fda62a6a2af51eac8f2aa2c762f25":{"kind":"TRIVIAL_REBASE","_number":11,"created":"2024-10-02 16:10:24.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"real_uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/91/906391/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/91/906391/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/906391/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/906391/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/906391/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/91/906391/11"}}},"commit":{"parents":[{"commit":"29b0e478a737655251e595a81c61548933abdbf7","subject":"Replace deprecated datetime.utcfromtimestamp()","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/29b0e478a737655251e595a81c61548933abdbf7"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-01-23 17:10:14.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tburke@nvidia.com","date":"2024-10-02 16:10:24.000000000","tz":0},"subject":"slo: Support range requests for part-number queries","message":"slo: Support range requests for part-number queries\n\nChange-Id: I8fa3588d0ce92914bf27df5e58c81d6b5a07bdd5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/48483a82e36fda62a6a2af51eac8f2aa2c762f25"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/48483a82e36fda62a6a2af51eac8f2aa2c762f25"}]},"branch":"refs/heads/master","description":"Rebase"},"bbba9c60ce849c117b3cd33c65f205ebcd1f6838":{"kind":"TRIVIAL_REBASE","_number":12,"created":"2024-10-09 23:33:59.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"real_uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/91/906391/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/91/906391/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/906391/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/906391/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/906391/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/91/906391/12"}}},"commit":{"parents":[{"commit":"cfc7f99d43d83b90fe1498e74781ab7027a5156b","subject":"trivial: Fix variable name","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/cfc7f99d43d83b90fe1498e74781ab7027a5156b"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-01-23 17:10:14.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tburke@nvidia.com","date":"2024-10-09 23:33:59.000000000","tz":0},"subject":"slo: Support range requests for part-number queries","message":"slo: Support range requests for part-number queries\n\nChange-Id: I8fa3588d0ce92914bf27df5e58c81d6b5a07bdd5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bbba9c60ce849c117b3cd33c65f205ebcd1f6838"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bbba9c60ce849c117b3cd33c65f205ebcd1f6838"}]},"branch":"refs/heads/master","description":"Rebase"},"3cd7968e045c96e297abe74ca71c28b2bf226e3f":{"kind":"TRIVIAL_REBASE","_number":13,"created":"2024-10-16 16:57:19.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"real_uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/91/906391/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/91/906391/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/906391/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/906391/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/906391/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/91/906391/13"}}},"commit":{"parents":[{"commit":"8ded39bccd9192b570e24cf039a3a4c8e3ccc216","subject":"Merge \"logging: Add UPDATE to valid http methods\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8ded39bccd9192b570e24cf039a3a4c8e3ccc216"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-01-23 17:10:14.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tburke@nvidia.com","date":"2024-10-16 16:57:19.000000000","tz":0},"subject":"slo: Support range requests for part-number queries","message":"slo: Support range requests for part-number queries\n\nChange-Id: I8fa3588d0ce92914bf27df5e58c81d6b5a07bdd5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3cd7968e045c96e297abe74ca71c28b2bf226e3f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3cd7968e045c96e297abe74ca71c28b2bf226e3f"}]},"branch":"refs/heads/master","description":"Rebase"},"4c3046cdbda0cea743744109850ddaca8d3245c5":{"kind":"TRIVIAL_REBASE","_number":14,"created":"2024-11-04 18:21:40.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"real_uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/91/906391/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/91/906391/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/906391/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/906391/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/906391/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/91/906391/14"}}},"commit":{"parents":[{"commit":"38556600d7eed6ea2936607eb3a7b58a87732759","subject":"Merge \"docs: add new utils modules to source doc\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/38556600d7eed6ea2936607eb3a7b58a87732759"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-01-23 17:10:14.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tburke@nvidia.com","date":"2024-11-04 18:21:40.000000000","tz":0},"subject":"slo: Support range requests for part-number queries","message":"slo: Support range requests for part-number queries\n\nChange-Id: I8fa3588d0ce92914bf27df5e58c81d6b5a07bdd5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4c3046cdbda0cea743744109850ddaca8d3245c5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4c3046cdbda0cea743744109850ddaca8d3245c5"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
