)]}'
{"id":"openstack%2Fswift~906937","triplet_id":"openstack%2Fswift~master~Iaef72f57538295ce385145faa1444973cb25abb5","project":"openstack/swift","branch":"master","hashtags":[],"change_id":"Iaef72f57538295ce385145faa1444973cb25abb5","subject":"replicator: Add sync_batches_per_revert option","status":"ABANDONED","created":"2024-01-27 07:49:35.000000000","updated":"2024-01-29 19:00:35.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"53d62eab0b898c45cd01170e5882926ae336fc0c","_number":906937,"virtual_id_number":906937,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2024-01-27 09:50:08.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-01-27 08:11:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-01-27 09:50:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"992c139c25d85daaf6a38146c0224e42e4234b10","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-01-27 07:49:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5c20a2d846f8d31c94b3a7278af2de4fba95f4cc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-27 08:11:31.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6de36751c6e74b8a90d7c6fde3e8b471\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/106995037d564445991537d4e96c551e : SUCCESS in 17m 03s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d6403350478647e5ba9a1ac14077c010 : SUCCESS in 17m 23s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/5d0e73bfae8a424a83b8cc05751dfc30 : SUCCESS in 20m 56s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/0a7bad2b11564c09817e17e4ae7f4f32 : SUCCESS in 19m 40s","accounts_in_message":[],"_revision_number":1},{"id":"de12eb63ae4e90ac57859d3ab244a88d690b503a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-27 09:50:08.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/16ea691c57bf4497aa9fa58041b5a1e1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1a59ae98ebf64175873ed0f3fdd1779b : SUCCESS in 7m 06s\n- grenade https://zuul.opendev.org/t/openstack/build/9546869ecd9e44e5bad17e6be63c341f : SUCCESS in 38m 21s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/51afe842b73c485d850bef1894d82d12 : SUCCESS in 38m 13s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/4b6480ceb99e4c1596583da606383484 : SUCCESS in 30m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1cd02b8fa55b4234bec7ae0b583cf668 : SUCCESS in 19m 26s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7d2d75e0934946a5ad54958819efed0a : SUCCESS in 19m 32s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/87db6fb1fff447538726b666b3bfef4d : SUCCESS in 18m 33s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8a7206366f4e4064bce20cb8af21407e : SUCCESS in 15m 30s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/32aa4ced420d464bb9add1585c515c87 : SUCCESS in 11m 29s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/309fe549d01249df9568b0cad1ec6d25 : SUCCESS in 5m 43s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/74a5e95d9c7147b2b14301930c9c89e8 : SUCCESS in 47m 25s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/829b063539c746b9b6b599e36e7edd28 : SUCCESS in 19m 44s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/a4f3fc6bbb224ab493db6fc11e6f8b42 : SUCCESS in 18m 41s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/fcea31de4d2242f1b27828cbe4453aa2 : SUCCESS in 18m 34s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/bb3b962c0f7e46c3a19750777766b40e : SUCCESS in 16m 37s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/0b14a069f8194b15abe619c8a205e621 : SUCCESS in 16m 55s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/e80a69bce9cb4a0ba3636cdaec50dcec : SUCCESS in 18m 52s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/524ea157a505468b88e8f872a4d8e490 : SUCCESS in 20m 15s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/f19abe6466dd4b7b86fbf07c842d89a8 : SUCCESS in 15m 43s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/65dfe97228814bb9808724609f0c21bf : SUCCESS in 15m 25s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/572f7bcd648f45c58b05d77643b7cd88 : SUCCESS in 16m 40s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/250dd8e7add941568f8f698061777148 : SUCCESS in 7m 32s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/f9d1c33dca054a8089b47cbd00970092 : SUCCESS in 17m 38s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/90cb4bd9a07649079e16184c8da043a9 : SUCCESS in 7m 24s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/25b54631877d41a9894fcb4c340414f4 : SUCCESS in 1h 01m 28s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/bad49b84b4e548539c1bc6f3b5ee022b : SUCCESS in 1h 03m 52s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/38718ba08026465a968f7c98b3adc89c : SUCCESS in 53m 58s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/374cc5c19ab24a55a8bd9b5ed43c4d0f : SUCCESS in 1h 03m 31s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1f5a0a351795427f9ca8660e3e9e5da8 : SUCCESS in 19m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/80dab3a9787345aba64da7811e4dbd0f : SUCCESS in 4m 39s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/1d76d7c0b38b4f54bcc539cd8973e698 : SUCCESS in 21m 00s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6b58593791634b8782b96ad475899366 : SUCCESS in 1h 06m 33s","accounts_in_message":[],"_revision_number":1},{"id":"53d62eab0b898c45cd01170e5882926ae336fc0c","tag":"autogenerated:gerrit:abandon","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-01-29 19:00:35.000000000","message":"Abandoned\n\nAah! This was supposed to just be a new patchset for https://review.opendev.org/c/openstack/swift/+/839649","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d31539e960d944290efe5e7ab79f1622edfbe8d6","revisions":{"d31539e960d944290efe5e7ab79f1622edfbe8d6":{"kind":"REWORK","_number":1,"created":"2024-01-27 07:49:35.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/37/906937/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/37/906937/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/37/906937/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/37/906937/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/37/906937/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/37/906937/1"}}},"commit":{"parents":[{"commit":"7b1f7ee857b0d7019b3e6e17a75d38ebbf650c15","subject":"Merge \"py3: Quiet RemoteDisconnected tracebacks\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7b1f7ee857b0d7019b3e6e17a75d38ebbf650c15"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-04-27 23:01:30.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-01-27 07:49:05.000000000","tz":-480},"subject":"replicator: Add sync_batches_per_revert option","message":"replicator: Add sync_batches_per_revert option\n\nDefault to a single batch to match prior behavior; this default may\nchange as we gain more operational experience.\n\nStill try to revert all handoff data in the first pass -- just be\nwilling to break it up into (randomized) batches so we can show progress\n(and delete data!) more frequently. If any of the batches fail,\ncontinue to the next batch but don\u0027t delete the partition.\n\nThis reduces the likelihood of timeouts (either rsync_timeout or\nhttp_timeout during REPLICATE calls). These timeouts now apply\nper-batch.\n\nAdditionally, it gives periodic opportunities for other replicators to\nsneak in and grab an rsync connection slot.\n\nAdd a new interface to the hashes.invalid API -- if the line starts with\n\u0027!\u0027, remove the suffix during consolidation rather than just setting it\nto None.\n\nSee also: https://review.opendev.org/c/openstack/swift/+/818017\nChange-Id: Ie12dcb4e524ee3af3ddc2b70c027fb62f68720fa\n\nChange-Id: Iaef72f57538295ce385145faa1444973cb25abb5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d31539e960d944290efe5e7ab79f1622edfbe8d6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d31539e960d944290efe5e7ab79f1622edfbe8d6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
