)]}'
{"id":"openstack%2Fswift~907084","triplet_id":"openstack%2Fswift~master~I7e5336fe9ba5303cfa4ecea671e635df75cfff53","project":"openstack/swift","branch":"master","topic":"p-proxy-namespaces","attention_set":{},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2024-01-31 11:08:10.000000000","reason":"\u003cGERRIT_ACCOUNT_7847\u003e replied on the change","reason_account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}}},"hashtags":[],"change_id":"I7e5336fe9ba5303cfa4ecea671e635df75cfff53","subject":"sq: fix legacy container server testing","status":"ABANDONED","created":"2024-01-29 15:44:46.000000000","updated":"2024-01-31 11:08:18.000000000","total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"81f273d59d1522ade1479f550536f29b4858c6f7","_number":907084,"virtual_id_number":907084,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},{"tag":"autogenerated:zuul:check","value":1,"date":"2024-01-29 16:54:05.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"all":[{"value":2,"date":"2024-01-31 04:54:48.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":1,"date":"2024-01-30 05:44:31.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-01-29 16:54:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-01-30 05:44:31.000000000","updated_by":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"reviewer":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"state":"REVIEWER"},{"updated":"2024-01-31 04:54:48.000000000","updated_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"reviewer":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"state":"REVIEWER"}],"messages":[{"id":"ec0f260d1ef69b75bc0324375e98bd721405a84a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-01-29 15:44:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7bc888f28fdeeb3bd8aa1f2f26133db3b5abd14b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-29 16:54:05.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/281a6602feb24f84a081d9f08e8f02b8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4a4852b7b4a740ffb4e3a112046f7fd6 : SUCCESS in 8m 59s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e5518799c2104e44ada4fe13ba392803 : SUCCESS in 17m 06s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5a1619c306ac4c3481d4b7c672712ae9 : SUCCESS in 17m 51s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6d4372092a454998b2256155ff838301 : SUCCESS in 19m 29s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/ca5f5a85c2e34b5a8bba770ffc41eefc : SUCCESS in 11m 03s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/73c7af245f754228b6961b340831d193 : SUCCESS in 6m 49s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/e710738f19b147fc973cc54c7a9dafcf : SUCCESS in 18m 04s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/c4fe11584584445c95288ddde87df2d8 : SUCCESS in 18m 20s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/23c2f6523c2349f984d01fc8fc8e5ba6 : SUCCESS in 18m 40s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/5b900b9e8f904be4a9e313cd90cd2b93 : SUCCESS in 18m 11s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/52e3acb10888474ab6620b77671f3126 : SUCCESS in 18m 56s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/f02a0f7157a949e48ee635b56e19b36a : SUCCESS in 20m 09s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/26f42b470ec54288a279dcea2c241a23 : SUCCESS in 19m 28s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/459341392dd449f1b8be2cc90d383992 : SUCCESS in 19m 03s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/0dd3f2f87c974c8a91c0b5a83981fd56 : SUCCESS in 14m 56s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/c091917b40954d7499f4ee4a609586bb : SUCCESS in 16m 29s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/1db7d611baaf4987b2258b1635a801c2 : SUCCESS in 15m 41s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/4d37696398794732a8a78196b9e5bba6 : SUCCESS in 1h 00m 33s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/a53510e508c543c4b0e48b4a221c769d : SUCCESS in 50m 52s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2a0702ada9b54c02aae572479952518c : SUCCESS in 18m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fe9b4d54765848b0863144008260718e : SUCCESS in 5m 12s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/d176030802d842b49e00ae5a04c58fcc : SUCCESS in 20m 55s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"47325cb6aa5dd1721e8e6ce57dba53306bbcaf7f","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-01-29 17:07:55.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"13386eb317a938b16f33f798de049ab70316dc66","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-29 17:18:33.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3b3f29b7cfe44fc2ae6deeca13a4c405\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e711c4e33db5478396c952f032ffb97e : SUCCESS in 26m 27s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/63f47a78417a43628718275b6142823a : TIMED_OUT in 1h 31m 03s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/a875158d50664703a86d6129d307f5ec : SUCCESS in 33m 07s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/9da0cd9be0504f379dee476a1da5ef3d : TIMED_OUT in 1h 31m 01s","accounts_in_message":[],"_revision_number":1},{"id":"16f9a38cd487745db0fba2ab33303ea5db497bac","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2024-01-30 05:44:31.000000000","message":"Patch Set 1: Code-Review+1\n\n(4 comments)","accounts_in_message":[],"_revision_number":1},{"id":"a86fc1e3ef2f6c4b5e89ad90560226a5bd9b31a0","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2024-01-31 04:54:48.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7b55d671d30fce6718769563b781de67139d39dc","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-01-31 11:08:10.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"81f273d59d1522ade1479f550536f29b4858c6f7","tag":"autogenerated:gerrit:abandon","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-01-31 11:08:18.000000000","message":"Abandoned\n\nsquashed into parent","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0b9805536e24ded2cc1ce7b7201e31ea3d29409e","revisions":{"0b9805536e24ded2cc1ce7b7201e31ea3d29409e":{"kind":"REWORK","_number":1,"created":"2024-01-29 15:44:46.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/84/907084/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/84/907084/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/84/907084/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/84/907084/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/84/907084/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/84/907084/1"}}},"commit":{"parents":[{"commit":"b5869c93d360d5ff6109ab6a2048445758df7a65","subject":"proxy: only use listing shards cache for \u0027auto\u0027 listings","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b5869c93d360d5ff6109ab6a2048445758df7a65"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-01-29 15:44:37.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-01-29 15:44:37.000000000","tz":0},"subject":"sq: fix legacy container server testing","message":"sq: fix legacy container server testing\n\nChange-Id: I7e5336fe9ba5303cfa4ecea671e635df75cfff53\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0b9805536e24ded2cc1ce7b7201e31ea3d29409e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0b9805536e24ded2cc1ce7b7201e31ea3d29409e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
