)]}'
{"id":"openstack%2Fswift~907173","triplet_id":"openstack%2Fswift~master~Idef3e137f92f1c38f3fbb979d8778faf3a4c4015","project":"openstack/swift","branch":"master","topic":"labeled-metrics","attention_set":{},"removed_from_attention_set":{"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2024-02-21 05:16:19.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Idef3e137f92f1c38f3fbb979d8778faf3a4c4015","subject":"labeled-metrics: Add option to emit both legacy and labeled","status":"ABANDONED","created":"2024-01-30 05:32:53.000000000","updated":"2024-02-21 05:16:19.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"5a07b7673960ad503a9e11e7bbf57455f78aa2e3","_number":907173,"virtual_id_number":907173,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2024-02-01 00:45:33.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-01-30 07:04:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1b0ef99207a6fd3fd6701cf1433e628fc702a979","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-01-30 05:32:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5bf3a91bc71f15ccafc92e9725586602e63e5289","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-30 07:04:13.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/6b904b5427cc4c38a1185c3c3176e775\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/006989ac6f184281a972ae39246300ee : SUCCESS in 7m 17s\n- grenade https://zuul.opendev.org/t/openstack/build/f330def7ba67420cbe0e0a482f9a09b0 : SUCCESS in 1h 05m 35s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/f6a2cd88ddcf4ba2aa3cea7a4df98a94 : SUCCESS in 1h 01m 28s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/1fd5655d906447109785018bdb00e3fe : SUCCESS in 35m 01s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3b661c79c6ed40f6b7cbc718773639b4 : SUCCESS in 29m 06s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e75d6b1047f74b0dbd43b2f299da3b43 : SUCCESS in 18m 12s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/661c23960b864c1099ad9a28f180f5cc : SUCCESS in 20m 37s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1d9f4c95bcea488c848ddcffe5c32d46 : SUCCESS in 21m 00s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/8c2bf7deabe144a7b0dae98927454bc5 : SUCCESS in 11m 18s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/3e0b6b1569fd4b60a9eb975db84ba783 : SUCCESS in 5m 34s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/8544af47357a42ecbaf1c019b3e6eebd : FAILURE in 18m 08s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/ed3d6840da224e8e8a15b93d1fe81447 : FAILURE in 18m 26s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/fd67cd3da6cb4535b49fd8264bbf4fe5 : FAILURE in 16m 34s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/1053ae2708f04850a962b408305ab1ab : FAILURE in 16m 16s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/857ac54302cd49158a77d5818853289f : FAILURE in 12m 20s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/87ffba408eba491394886165650dc5b9 : SUCCESS in 17m 58s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/15fe8773c527465e852bdc9cbb58662a : SUCCESS in 18m 02s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/e9a7c0cc93994517b84eafe69ee1ee6f : SUCCESS in 18m 09s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/fd7bb21465de47e7bcfa11a3e61aad83 : SUCCESS in 16m 24s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/c5b1b2788edd4c6c960c634bbe5d6c10 : SUCCESS in 15m 42s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/c4ed74d20a3543a48ae0c5afd51977d5 : SUCCESS in 15m 51s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/bbe1a6ed1c5d42bea42344d92f2c5646 : SUCCESS in 6m 14s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/0e5db3fbaa8c49a6b1d04fce3640feab : SUCCESS in 19m 41s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/e8a5c1800def4f71854853087fd5faa5 : SUCCESS in 7m 30s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/f59641672adc42f18e77953398235960 : SUCCESS in 1h 00m 32s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/3e4c2b99bab64c04b4719c713c798814 : SUCCESS in 1h 05m 02s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/68bbe18cb14e4c6592cdbbc0dbbb4778 : SUCCESS in 53m 31s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/278351ce778b41d0933ccba539710037 : SUCCESS in 52m 47s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/dbb1ff27796c4e4faf37afad3a2daafd : FAILURE in 18m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b9dee4f8df7c4141a51326e9dafd2f03 : FAILURE in 4m 26s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/269f52aaf78f43b08aff0e8bd97b3cf4 : SUCCESS in 19m 20s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bdd0bdba5865409b9ba060078366d029 : SUCCESS in 49m 01s","accounts_in_message":[],"_revision_number":1},{"id":"4eaa4479fc62d30480dec3117a69c83b8bfbc956","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-30 07:05:55.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9228f4da9a634069ba7bc9fd73ed83f6\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/5d03b4091dd64bdbb33efb77b589f31c : FAILURE in 1h 07m 51s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/93a0301898ed43608123aa1fbe93196c : TIMED_OUT in 1h 30m 58s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/d928d6c7d2314f9d94411a8a10222ae3 : SUCCESS in 1h 12m 36s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/620c82efb7874b59b7d283380abc5de0 : FAILURE in 1h 00m 02s","accounts_in_message":[],"_revision_number":1},{"id":"83768ec772af3a0fbab015a0197978f4f0f69b1c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-01-30 17:43:11.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"43d89dac1d958e0456fe861de3c973e8be89cdc3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-30 18:32:19.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/05ecdbe8558f4a64bd027e1807bf9075\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9949d539c26645498a90d1df6a13e833 : FAILURE in 15m 13s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4b85e00c201e40d9b013cf3828ba2aad : FAILURE in 14m 09s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/8d26a4a573f84e09822747b7280ea714 : SUCCESS in 23m 36s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/545f08ab34ce469381a7ae8680f25106 : SUCCESS in 47m 18s","accounts_in_message":[],"_revision_number":2},{"id":"d28064916f76c7b2c0376f4b480bb266bff041e0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-30 19:04:30.000000000","message":"Patch Set 2: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9934b7d0fc364e56b4a663e9674126af\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b4545a327e274af293f4e812f7cbd4b2 : SUCCESS in 8m 29s\n- grenade https://zuul.opendev.org/t/openstack/build/2b94f5dde4054ea5ae0468f946c43e0b : SUCCESS in 1h 04m 27s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/669f862a57a641e5aac13a105be1c194 : SUCCESS in 1h 01m 49s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/6886329efc3e48a28d5f84679acea2c5 : SUCCESS in 56m 08s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/19ff1bde484447c58fb704f878f22f05 : SUCCESS in 36m 46s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2935490c30e34b2abf73b4fddc11726b : SUCCESS in 20m 22s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ed6b2f4a235a40edaae33d1ff040044b : SUCCESS in 20m 17s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1cc59858a05840f39e447c9870cc91dc : SUCCESS in 21m 06s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/7492b3c6eba348a79c01befb14db2a91 : SUCCESS in 11m 34s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/acf0a91a01f14a5cb1bbe81a958e0da5 : SUCCESS in 5m 38s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/c9dc1b523c474b51b28a83056f02b147 : FAILURE in 17m 02s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/dc50debf094d42de89512550fab4a528 : FAILURE in 18m 43s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/918b8ca230f742d6862fdaf5d597f4cc : FAILURE in 16m 03s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/1d2c8d81512a416a8459c2c2359cae97 : FAILURE in 16m 44s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/feb6b78d65a0440c95bbe1c0029a7f54 : FAILURE in 17m 18s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/76993fb8649a4748891036ffd78ec671 : SUCCESS in 18m 06s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/d18bb20bd5224795b1fe4948451a6652 : SUCCESS in 18m 54s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/601d16d7f85e4b588a490f77b3ffcab8 : SUCCESS in 18m 13s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/fdffb0893f654ce7a1f2632151db7769 : SUCCESS in 15m 31s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/8ce3581aeb35442b9fe80ed0b0f0136c : SUCCESS in 17m 05s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/98c4df5b7feb49828e921db5fa167a65 : SUCCESS in 16m 23s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/4aa5629d2c63452bb4289e8a78c3b2cf : SUCCESS in 7m 14s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/4dc42e1833fe47a584a4eee34a6e4261 : SUCCESS in 16m 39s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/8fbf461914e9466580adda2bad1692e6 : SUCCESS in 7m 44s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/56cdd31a999d48bd8dd0da47c7d89771 : SUCCESS in 49m 09s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/3035c366b89f4a538ed673eff0c2939e : SUCCESS in 1h 10m 22s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/e98d58c076da4b3081c032524bf1b9da : SUCCESS in 38m 11s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/83c51cb25c6e40a9a14040dc6e2d214b : SUCCESS in 56m 43s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/33b99be805cb41b59f5bac18026bb47a : FAILURE in 15m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2780744fd2084dcfb40513506b11912d : FAILURE in 4m 49s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/009eac8f4da549f9a90e25ac189a73d9 : SUCCESS in 20m 55s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bc980e57e5684b63b3aa2a54bcc8d732 : SUCCESS in 46m 40s","accounts_in_message":[],"_revision_number":2},{"id":"af74e965e2cfd30d315e7d845dbf2002b796f0c0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-01-30 23:35:28.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"a3ace025e396840435071d2a282a688f08dcf7ee","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-30 23:58:25.000000000","message":"Patch Set 3:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dfc88dd334ed4f328b179614aacbf6c5\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1d5eea2fa93546dd837a36111b5c2e78 : FAILURE in 15m 43s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/275c9a5944b64c7e9626341d8912d6a2 : FAILURE in 16m 03s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/d92c59d29e754a588515d4b81fb4e922 : SUCCESS in 19m 25s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/0a061326b0134512bcc324b9f4e5664d : SUCCESS in 21m 45s","accounts_in_message":[],"_revision_number":3},{"id":"7bb979cb572a8cd675f6f7c3a12c812f2b025567","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-31 01:12:16.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/611c6e475aa142eca4fcba603285bedb\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6260acbdc58046858bc46508d74db458 : SUCCESS in 7m 02s\n- grenade https://zuul.opendev.org/t/openstack/build/f6fcd4616f5f476c91e1fd2e54bfee08 : SUCCESS in 38m 14s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/790d7d514caa49f7b82dea2d88cbdc03 : SUCCESS in 56m 41s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/641ecf4113bd418ea0248e47bd82d2cb : SUCCESS in 38m 17s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b8d82610b334483e9584d7d48fc40744 : SUCCESS in 36m 20s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/61628fb3af4840499e7b843ebe3961c1 : SUCCESS in 16m 34s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/167d9f4912ce4ab9a7c2b750cbeb1c55 : SUCCESS in 16m 51s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7530ad54231944a1858649e331e457e2 : SUCCESS in 16m 59s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/a01868f5dbc8482d8b2d081d16f69dd7 : SUCCESS in 11m 29s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/b759276df3a04ab4b543fb1c4a2977f6 : SUCCESS in 5m 44s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/cfa46e6906ff4305912e4af5df7e8f48 : FAILURE in 18m 51s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/fe733d46edac4c1aaef4f0b6f06ccb09 : FAILURE in 20m 43s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/4334a4bc32964dc8856dc387ce2e7a7d : FAILURE in 16m 20s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/d79397d3ad314bb28cacd22bb94a8fdd : FAILURE in 17m 49s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/f89d95c76e3543dbbccdb2923c60f3bb : FAILURE in 18m 30s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/f4dea760b623437cadc4f61627d901c4 : SUCCESS in 18m 14s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/339beee0aebc4a74a42a85d99353cb90 : SUCCESS in 16m 23s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/2b09915dafb44b8c9206b59ee8aae4ca : SUCCESS in 15m 29s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/4bf29cf2e769462baf5f3df88e1b6363 : SUCCESS in 16m 59s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/02f42b8b08e94b91bdc2edcdde389321 : SUCCESS in 13m 01s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/5dfd51c217394eceae50ab37810b8302 : SUCCESS in 12m 53s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/bfbd9e2d7a074191b7311a948e1b1eab : SUCCESS in 6m 50s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/de195d69c8484fd4bba9560a5d433855 : SUCCESS in 17m 02s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/bc88d5d8b0624a4f895413ad1118e48d : SUCCESS in 7m 38s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/db8b4a4c4aed48e0934ff7e67b6c5ab5 : SUCCESS in 1h 01m 17s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/3f0a408c7f574e41b3e1664d8d1ca19b : SUCCESS in 1h 06m 06s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/384ef28d9f9a475084345461bcebebca : SUCCESS in 34m 31s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/a1695a979e3a4fa2a7b3d5411f779d6d : SUCCESS in 48m 54s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/afbda280b89444e69e5a10e98105c78b : FAILURE in 14m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4a16ab772c3240829b114a3c1f731973 : SUCCESS in 4m 53s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/e51ef037f2fd4520ad195790da9560ff : SUCCESS in 23m 55s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fa241c1d53c148d4a9508e9e9033e195 : SUCCESS in 59m 47s","accounts_in_message":[],"_revision_number":3},{"id":"0290c5fc5ac2755cb08a7c803c126f9f085524c7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-01-31 23:22:12.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"daa8c8ef8587a72c5d3a2cedbc031e832aac89e1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-01 00:45:33.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6f0092c38ef8450a935e65a0d73d5a73\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/62074f5f322c4d97b5f9c4a132c84fc4 : SUCCESS in 9m 16s\n- grenade https://zuul.opendev.org/t/openstack/build/f9520278332f4c5fbdd547fd7db68519 : SUCCESS in 1h 05m 51s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/60a89b8b52b24e18955b86b971d78bb9 : SUCCESS in 1h 03m 31s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/107098d63cb545da81e96a65a96f6679 : SUCCESS in 46m 07s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b2081b94d2574d9a905cc2a4d4b9652e : SUCCESS in 19m 10s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4f54b5aacdb24bb1ba4b8f7010fd7989 : SUCCESS in 19m 19s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7b4e15277f204118bdf7d91b126e1ae9 : SUCCESS in 21m 04s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7f09fb13ae6246f8b7ca43c8fd9f391e : SUCCESS in 22m 18s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/8c609db2a8094a92bdf6a4160fbcc9e0 : SUCCESS in 11m 55s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/6c9fcc03a123458c8b16a1230d30911f : SUCCESS in 7m 14s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/010532459fe545aeae92ffb62685f1ca : SUCCESS in 14m 39s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/92a9f2d62c33496fbea4dde01032606b : SUCCESS in 15m 22s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/93fe64996ebf4f0a98080fd4c2dc3b8d : SUCCESS in 13m 22s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/3dea1dc624684261b66903dd6cb6ca30 : SUCCESS in 15m 41s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/f2e19dc9179e4d4cb1a348746e8724c0 : SUCCESS in 12m 35s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/e3e1f662e16443dea63684e044707c8e : SUCCESS in 18m 01s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/836b4206c67d42bab1f0bb92a02a0437 : SUCCESS in 21m 16s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/92ebf584c37e45e09976a8256800e26a : SUCCESS in 19m 27s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/c2aa93edcde9429da8cd6c881bfa32e8 : SUCCESS in 21m 31s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/268f9c8fa2d24af6b787fb6e945e7069 : SUCCESS in 17m 12s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/d419efe9fe9141ef9a8b4f8e18f86cba : SUCCESS in 16m 38s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/d2e27933716741fc92fc38300d002c55 : SUCCESS in 6m 58s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/002d99dabc8d4dcebcdf761a751c46d3 : SUCCESS in 17m 24s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/66a53d78f37d4c11b7815a9bf64c1ad5 : SUCCESS in 7m 19s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/965cd2514fee4e0b97b78f9fb6f6295b : SUCCESS in 57m 39s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/ff40936c25c748b6b2cff4e56520b21d : SUCCESS in 1h 17m 40s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/b611cec702094eee97d908291b31bd24 : SUCCESS in 52m 50s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/a73c47d450884687ad36299d4fc31201 : SUCCESS in 1h 00m 39s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a236559845144a92aca0ee3287dbdd16 : SUCCESS in 25m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5c6ff37fd53844c3bb623564fad0dac1 : SUCCESS in 4m 48s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/16c261d4752a4b7481f5dcc2dd46e486 : SUCCESS in 21m 42s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d18fae1fc406405080f388d5dd03fa82 : SUCCESS in 59m 40s","accounts_in_message":[],"_revision_number":4},{"id":"a07dff7be90b991f3a1538e12b242cf464551d00","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-01 00:55:38.000000000","message":"Patch Set 4:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ce52df7711274dec8b4c742580e29113\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/6e4fa5b24aef4c1aaec976c2443906cf : SUCCESS in 15m 42s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/129626a2378240d48da37b7578505a14 : SUCCESS in 16m 29s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/b06b6af839e8402b89bd2c8d39241da8 : TIMED_OUT in 1h 31m 02s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/189758ac7c0642bea04a272e4efa6de2 : SUCCESS in 17m 50s","accounts_in_message":[],"_revision_number":4},{"id":"5a07b7673960ad503a9e11e7bbf57455f78aa2e3","tag":"autogenerated:gerrit:abandon","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-02-21 05:16:19.000000000","message":"Abandoned\n\nSquashed into parent two patches","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"2ed862a99350b566d1c31c3f48a01a7c23d31f23","revisions":{"7d2eaeb3804042ee14f45be87b9ec50c9ba59cbd":{"kind":"REWORK","_number":1,"created":"2024-01-30 05:32:53.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/73/907173/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/73/907173/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/73/907173/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/73/907173/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/73/907173/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/73/907173/1"}}},"commit":{"parents":[{"commit":"e43acc77f06639426e0eb51b768a447833fe5cee","subject":"Move proxy-logging to labeled metrics","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e43acc77f06639426e0eb51b768a447833fe5cee"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-01-30 05:31:27.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-01-30 05:32:36.000000000","tz":-480},"subject":"labeled-metrics: Add option to emit both legacy and labeled","message":"labeled-metrics: Add option to emit both legacy and labeled\n\nChange-Id: Idef3e137f92f1c38f3fbb979d8778faf3a4c4015\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7d2eaeb3804042ee14f45be87b9ec50c9ba59cbd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7d2eaeb3804042ee14f45be87b9ec50c9ba59cbd"}]},"branch":"refs/heads/master"},"84de998b8b662fe9f505e6dfb8f1af9c81809fa6":{"kind":"REWORK","_number":2,"created":"2024-01-30 17:43:11.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/73/907173/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/73/907173/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/73/907173/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/73/907173/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/73/907173/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/73/907173/2"}}},"commit":{"parents":[{"commit":"e43acc77f06639426e0eb51b768a447833fe5cee","subject":"Move proxy-logging to labeled metrics","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e43acc77f06639426e0eb51b768a447833fe5cee"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-01-30 05:31:27.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-01-30 17:42:35.000000000","tz":-480},"subject":"labeled-metrics: Add option to emit both legacy and labeled","message":"labeled-metrics: Add option to emit both legacy and labeled\n\nChange-Id: Idef3e137f92f1c38f3fbb979d8778faf3a4c4015\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/84de998b8b662fe9f505e6dfb8f1af9c81809fa6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/84de998b8b662fe9f505e6dfb8f1af9c81809fa6"}]},"branch":"refs/heads/master"},"40172db30b27e541f04ad0160f23f5b4133a934f":{"kind":"REWORK","_number":3,"created":"2024-01-30 23:35:28.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/73/907173/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/73/907173/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/73/907173/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/73/907173/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/73/907173/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/73/907173/3"}}},"commit":{"parents":[{"commit":"e43acc77f06639426e0eb51b768a447833fe5cee","subject":"Move proxy-logging to labeled metrics","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e43acc77f06639426e0eb51b768a447833fe5cee"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-01-30 05:31:27.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-01-30 23:35:20.000000000","tz":-480},"subject":"labeled-metrics: Add option to emit both legacy and labeled","message":"labeled-metrics: Add option to emit both legacy and labeled\n\nChange-Id: Idef3e137f92f1c38f3fbb979d8778faf3a4c4015\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/40172db30b27e541f04ad0160f23f5b4133a934f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/40172db30b27e541f04ad0160f23f5b4133a934f"}]},"branch":"refs/heads/master"},"2ed862a99350b566d1c31c3f48a01a7c23d31f23":{"kind":"REWORK","_number":4,"created":"2024-01-31 23:22:12.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/73/907173/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/73/907173/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/73/907173/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/73/907173/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/73/907173/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/73/907173/4"}}},"commit":{"parents":[{"commit":"e43acc77f06639426e0eb51b768a447833fe5cee","subject":"Move proxy-logging to labeled metrics","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e43acc77f06639426e0eb51b768a447833fe5cee"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-01-30 05:31:27.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-01-31 23:21:58.000000000","tz":-480},"subject":"labeled-metrics: Add option to emit both legacy and labeled","message":"labeled-metrics: Add option to emit both legacy and labeled\n\nChange-Id: Idef3e137f92f1c38f3fbb979d8778faf3a4c4015\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2ed862a99350b566d1c31c3f48a01a7c23d31f23"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2ed862a99350b566d1c31c3f48a01a7c23d31f23"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
