)]}'
{"id":"openstack%2Fswift~907622","triplet_id":"openstack%2Fswift~master~Ia86cb421047c2c04428acf8948cc271a17b04a9e","project":"openstack/swift","branch":"master","topic":"p-sharder-deleted-2","attention_set":{},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2024-02-06 05:37:32.000000000","reason":"Change was abandoned"},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2024-02-06 05:37:32.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Ia86cb421047c2c04428acf8948cc271a17b04a9e","subject":"sq: no \"Ignoring remote osr w/o epoch\" false positives","status":"ABANDONED","created":"2024-02-02 16:51:33.000000000","updated":"2024-02-06 05:37:32.000000000","total_comment_count":8,"unresolved_comment_count":3,"has_review_started":true,"meta_rev_id":"6a76afe418c624a26d34696cf24ceaf4a99c91aa","_number":907622,"virtual_id_number":907622,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2024-02-05 16:29:13.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"}],"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-02-02 17:13:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-02-02 18:13:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-02-02 19:26:03.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"},{"updated":"2024-02-05 06:23:29.000000000","updated_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"reviewer":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"state":"CC"}],"messages":[{"id":"718af61c117c5fa8e629501d30b7caf1bd8e2457","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-02-02 16:51:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1fa996e772054bccc01f503c10a2c2137a63f021","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-02 17:13:11.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f4a38990042345b68ca309af046a2042\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/128ef777eeac4b9bb21a1d83d37f6a6c : FAILURE in 16m 16s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/130abd89135e4b6cb82d0cfa5bffee53 : FAILURE in 14m 44s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/f82c96e777c3405bb2c07471f3a970da : SUCCESS in 20m 27s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/d49da474473a4ea1897675987dac849b : SUCCESS in 18m 24s","accounts_in_message":[],"_revision_number":1},{"id":"5c213e5c0e49d3c6d6381ffeb96fa93faf0c6537","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-02 18:13:06.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f6bc73a2b68f437cbeb66fb925b0a830\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4733b0ac8b054d74af467e2d723cac56 : SUCCESS in 10m 50s\n- grenade https://zuul.opendev.org/t/openstack/build/b91801dceed84a89abfc59a1a92c1875 : SUCCESS in 55m 13s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/9a28d050b15c42e590dbe9735604021b : SUCCESS in 53m 12s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/ba0a52d8d67d43218b1436dcc953f85a : SUCCESS in 43m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/dc4779a5c37f4ebab65bb206decdec1d : SUCCESS in 39m 15s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/34cec0b914914cae8f02e21f2f49f837 : SUCCESS in 19m 27s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/07257e6a69744a0ea71550d45bee4e2d : SUCCESS in 23m 47s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c6ca70bc724d4c4784c1ea20f8d54a4a : SUCCESS in 21m 21s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/9e31a46bea1b4a3fa9718fdc3dce0ba7 : SUCCESS in 10m 59s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/d239fbce76534ea7b3572e6358cd10ae : SUCCESS in 6m 15s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/01bb4f7fe6b24d04b5df0e3039037056 : FAILURE in 18m 12s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/4f282d5681374f70a5386367a17ae5c8 : FAILURE in 13m 14s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/d2b0aef5a8804f5f835104c9f58a380d : FAILURE in 16m 07s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/9e9e198ced794eccb4ee68bb87d3cd2c : FAILURE in 19m 18s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/5e7259e853ce40ff89296965e7130130 : FAILURE in 13m 59s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/48cc359bcf9848d08665aeb38a0d5aae : SUCCESS in 18m 09s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/70d1b3f536024a60b3189e71737d8473 : SUCCESS in 15m 35s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/e7a63434332f4218abcaf9b87bd0b3bb : SUCCESS in 19m 49s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/3a22ca12bc164c79be924456088c4916 : SUCCESS in 17m 48s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/19324612328242bd862478d4a3e3ae71 : SUCCESS in 17m 54s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/a6e1992c086c4a59bb42a5d87bb7275d : SUCCESS in 15m 50s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/1d6348a1ce7041128cd9404209d65eba : SUCCESS in 8m 11s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/22a90547a2f04c63831781c38287d8a3 : SUCCESS in 18m 05s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/4a5f12b8f9c843e98bccf0ef355d7105 : SUCCESS in 8m 59s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/2538a0c0ddff47bebd2f2a704ac5710e : SUCCESS in 58m 07s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/8fdda8746f3846628933d6d2b9cdb22b : SUCCESS in 1h 03m 11s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/ae9b7e77d78c4285a104efc1828d1561 : SUCCESS in 57m 39s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/96a9e5ec5e9147d6a85696b6b4fd81d9 : SUCCESS in 50m 26s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1efc97b959164e03bb53b04efada8ebb : FAILURE in 15m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cd4810fd02184982a15157b9ada2b0b2 : SUCCESS in 4m 45s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/0e5fb7da5f96479795d846cee195b50e : SUCCESS in 20m 33s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c9bffc3bbe1c4011a23a31c02ee29c11 : SUCCESS in 1h 09m 00s","accounts_in_message":[],"_revision_number":1},{"id":"c78779d7ee59d29bd4ecc4a4084382e3a18df760","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-02-02 19:26:03.000000000","message":"Patch Set 1: Code-Review+2\n\n(4 comments)","accounts_in_message":[],"_revision_number":1},{"id":"f5cda69db488b8457051a3212e8bf556eeafcbfd","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2024-02-05 06:23:29.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"d95ef500c84e910ba12b52bee0134fc9c73be96d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-02-05 10:13:20.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"5e669ceaefd0e26ab2d9e86363d26922be1ff695","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-05 11:24:18.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8d30eb4c96ec4fc7945957860f739bc7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c9a7263ab33c4f6caced767413aca7c6 : SUCCESS in 9m 26s\n- grenade https://zuul.opendev.org/t/openstack/build/1c97ab5aa190441b9c116c6c70bfc43f : SUCCESS in 55m 07s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/dba89b426ed045f69d8392eea921f667 : SUCCESS in 47m 32s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/258ad54a2ad44599beba3d23b85b737b : SUCCESS in 42m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5d23ce858ff24fe399ab0c8b39d0d11e : SUCCESS in 25m 21s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f50815a18f23437ba3683f8b8faff161 : SUCCESS in 19m 25s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/fc30040f1e64466dbd8c924c42cd7f97 : SUCCESS in 19m 45s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1090394000f248848c3aa55fe9cebfca : SUCCESS in 21m 31s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/5565348843e94eb285c4c3250415a2c3 : SUCCESS in 10m 13s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/1ceeaea85a3a45068ded41cbf0c12857 : SUCCESS in 5m 29s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/52a1cc7c98c44e8aa5dc2b82daedd6a5 : FAILURE in 15m 21s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/2a8983c227ae451a83de7fd33630c940 : FAILURE in 13m 51s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/8a526f41180e4d1f9a8f0a7c613ebcc2 : FAILURE in 13m 59s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/b6de2685ab0e40f291cdb351c222627c : FAILURE in 13m 42s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/346a99ca8d8e44a482a332284b1bd8fc : FAILURE in 18m 09s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/c71e4299b80142d2b0bb46251e618f8b : SUCCESS in 17m 54s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/5bdb6dd095974abf883563240f7bd482 : SUCCESS in 19m 34s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/adb0cfed5ba64f10a63c88c57d5f9191 : SUCCESS in 20m 38s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/6b613427166a4e83875289b649b1fdb8 : SUCCESS in 16m 03s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/568df7a67b6b4d2e96fa73b873e617f3 : SUCCESS in 17m 57s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/ad7653aa81ec4af7b856d69c8a954b1f : SUCCESS in 16m 53s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/f2c29fbe649c435994cef7af6a20e223 : SUCCESS in 7m 12s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/d661b6fd5da04c1a806dbabd4ec16813 : SUCCESS in 17m 13s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/42dd40a4845b41169b1e5b9b5794f502 : SUCCESS in 7m 39s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/a9c4d36971ce4c00827327a857af56f3 : SUCCESS in 1h 07m 37s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/25ad178a49194ffca529f6f89da4c15e : SUCCESS in 1h 09m 36s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/ca06580febb949628c7847dc039e9e04 : SUCCESS in 51m 57s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/0d5656a4bbe147f3bc0cafa8c7c00533 : SUCCESS in 49m 32s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/179f2d68bb09481f87979937b6bdd136 : FAILURE in 26m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b68cf9ddad0a487ea327675bf82df2b1 : SUCCESS in 5m 49s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/a668268185b047e9bfc721264e66ec9a : SUCCESS in 28m 33s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a85261b1128e455fbd0998fdc447a047 : SUCCESS in 52m 55s","accounts_in_message":[],"_revision_number":2},{"id":"f21e868e8f96e54f3e06b59cc60677f164018b23","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-05 11:26:11.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2ce4c912103841c8a0e6597ebdb9e46d\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ec5279966c0a4b0d92a1cadec8bb812c : FAILURE in 53m 10s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ea34583c3aa34d0f9c23466209b246ab : FAILURE in 15m 18s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/6c72ff6450634861ade7b7ddbcc0c9bd : SUCCESS in 1h 10m 56s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/8908468b8dda46ecbd49b84b48af7fcb : SUCCESS in 20m 40s","accounts_in_message":[],"_revision_number":2},{"id":"91e1233e96ff0e1384e4c93594596792270062ae","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-02-05 15:14:38.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"66783113f1ae243b0f9454dd52e747a6f603b840","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-05 16:29:13.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0d5c411b86864520934ed67401343f51\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a4a52ddcf26344f7bebb5967246a4450 : SUCCESS in 8m 51s\n- grenade https://zuul.opendev.org/t/openstack/build/e28e223016b44eec955207d26cb7a515 : SUCCESS in 1h 08m 03s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/ccb6a18bdf9142edbce2ad0c88dd0126 : SUCCESS in 48m 46s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/a63d864ff20f4f59ac826541d0146efa : SUCCESS in 37m 07s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4ac0d189140744299170427b7fdc405e : SUCCESS in 42m 08s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/767f19e8a68f406aaeba1bbb132267bc : SUCCESS in 19m 45s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/009843afc744488298caf66f40a39a25 : SUCCESS in 22m 13s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/16df217a9c33448e8f20900d69aa2339 : SUCCESS in 19m 34s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/d5da3c486c58444882e26c3f16620f31 : SUCCESS in 11m 06s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/c0a1cec6148748a3a5409853307b0bc1 : SUCCESS in 5m 08s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/8b37a5a3796742f5a944c06eee0833da : SUCCESS in 18m 10s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/2a60711b0b82416da4a3bc99ac4e21e9 : SUCCESS in 26m 49s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/dca416493d794c979245cea0f8cea1ce : SUCCESS in 26m 20s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/1831d85c25ed4fda9cfc9aef3da0990f : SUCCESS in 17m 48s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/653b0a6557354c8a8cec989f2d8a499f : SUCCESS in 26m 05s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/69aa667816224d2f836e120f1169886d : SUCCESS in 15m 44s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/283c47c8e99b4c2fae46a1874b25eba7 : SUCCESS in 17m 50s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/135960a23fd047fd9ffe84a68181d79a : SUCCESS in 22m 23s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/38f8cb0f24c849f19bb1e5de549c47a9 : SUCCESS in 15m 09s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/41567de3f46d478490bbce7aff3e251e : SUCCESS in 16m 18s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/1a11786e93d5450d8f9a4d3c3e83076d : SUCCESS in 20m 38s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/669dcfdd3e82408492f69ff70205395f : SUCCESS in 8m 10s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/592fcd5f15f24c2799987195a5712b78 : SUCCESS in 16m 03s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/04b48a1031154a3e9495c054d19305e6 : SUCCESS in 7m 51s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/92b07659e3c6467a8e5c70c8dc2d9d17 : SUCCESS in 1h 07m 52s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/6ba5f94b8ced408981a7d82a76460fb2 : SUCCESS in 1h 11m 32s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/eac374499f884e1396b8b373112a6065 : SUCCESS in 50m 12s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/cee1eee77f9f44ad918f48905dec5268 : SUCCESS in 42m 58s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/720db3f9e854425f9d0d22243fba0dff : SUCCESS in 29m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dc275a71801e40f6afc492954baec750 : SUCCESS in 5m 32s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/b0569a390811482291d7aa1dc7b708ee : SUCCESS in 18m 23s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9aab049d1e734b589b35209fd2c00058 : SUCCESS in 47m 24s","accounts_in_message":[],"_revision_number":3},{"id":"00c1a583428ffaee199e0480b67c9aad19769dbd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-05 16:48:20.000000000","message":"Patch Set 3:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2d7dfbdf0db74f16a3b95f5fc26de8ae\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/34b958e2da39495fa11fa4c4919d79c6 : TIMED_OUT in 1h 31m 11s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/eab9d9a9e22145c59174bc58b7dc9d68 : TIMED_OUT in 1h 30m 58s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/f26d2e54ae2c4c8e93ee4ca5354cd07c : SUCCESS in 25m 29s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/43ca96d6dc7745c9b80f00ce873c7060 : TIMED_OUT in 1h 30m 49s","accounts_in_message":[],"_revision_number":3},{"id":"cebe98545a044275fa06874b725142a7e3f65758","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2024-02-06 03:47:04.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"10b6eb4cb259f02a35ebdbb1be9ffad0a1b0e3d6","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2024-02-06 05:37:14.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"6a76afe418c624a26d34696cf24ceaf4a99c91aa","tag":"autogenerated:gerrit:abandon","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2024-02-06 05:37:32.000000000","message":"Abandoned\n\nsqaushed into parent","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"3cf0e1cca58f14bfce0c3ef74cafd73f841c5e1f","revisions":{"00c8003a0b509fc8819a0a6eb79b55ee530371df":{"kind":"REWORK","_number":1,"created":"2024-02-02 16:51:33.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/22/907622/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/22/907622/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/22/907622/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/22/907622/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/22/907622/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/22/907622/1"}}},"commit":{"parents":[{"commit":"1607b25175586c6e0055610b85915e4ddeba40b7","subject":"sharding: don\u0027t replace own_shard_range without an epoch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1607b25175586c6e0055610b85915e4ddeba40b7"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-02-02 16:49:15.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-02-02 16:49:15.000000000","tz":0},"subject":"sq: no \"Ignoring remote osr w/o epoch\" false positives","message":"sq: no \"Ignoring remote osr w/o epoch\" false positives\n\nBefore warning about \"Ignoring remote osr w/o epoch\" check that the\nlocal osr epoch would actually be reset if the remote osr was to be\n\u0027merged\u0027. We see a lot of suspected false positives whn one replica of\na shard has sharded and others have not.\n\nChange-Id: Ia86cb421047c2c04428acf8948cc271a17b04a9e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/00c8003a0b509fc8819a0a6eb79b55ee530371df"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/00c8003a0b509fc8819a0a6eb79b55ee530371df"}]},"branch":"refs/heads/master"},"518c829666515efbca00382db7792b0be7b822ce":{"kind":"REWORK","_number":2,"created":"2024-02-05 10:13:20.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/22/907622/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/22/907622/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/22/907622/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/22/907622/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/22/907622/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/22/907622/2"}}},"commit":{"parents":[{"commit":"1607b25175586c6e0055610b85915e4ddeba40b7","subject":"sharding: don\u0027t replace own_shard_range without an epoch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1607b25175586c6e0055610b85915e4ddeba40b7"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-02-02 16:49:15.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-02-05 10:13:04.000000000","tz":0},"subject":"sq: no \"Ignoring remote osr w/o epoch\" false positives","message":"sq: no \"Ignoring remote osr w/o epoch\" false positives\n\nBefore warning about \"Ignoring remote osr w/o epoch\" check that the\nlocal osr epoch would actually be reset if the remote osr was to be\n\u0027merged\u0027. We see a lot of suspected false positives whn one replica of\na shard has sharded and others have not.\n\nChange-Id: Ia86cb421047c2c04428acf8948cc271a17b04a9e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/518c829666515efbca00382db7792b0be7b822ce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/518c829666515efbca00382db7792b0be7b822ce"}]},"branch":"refs/heads/master"},"3cf0e1cca58f14bfce0c3ef74cafd73f841c5e1f":{"kind":"REWORK","_number":3,"created":"2024-02-05 15:14:38.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/22/907622/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/22/907622/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/22/907622/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/22/907622/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/22/907622/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/22/907622/3"}}},"commit":{"parents":[{"commit":"1607b25175586c6e0055610b85915e4ddeba40b7","subject":"sharding: don\u0027t replace own_shard_range without an epoch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1607b25175586c6e0055610b85915e4ddeba40b7"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-02-02 16:49:15.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-02-05 15:14:28.000000000","tz":0},"subject":"sq: no \"Ignoring remote osr w/o epoch\" false positives","message":"sq: no \"Ignoring remote osr w/o epoch\" false positives\n\nBefore warning about \"Ignoring remote osr w/o epoch\" check that the\nlocal osr epoch would actually be reset if the remote osr was to be\n\u0027merged\u0027. We see a lot of suspected false positives whn one replica of\na shard has sharded and others have not.\n\nChange-Id: Ia86cb421047c2c04428acf8948cc271a17b04a9e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3cf0e1cca58f14bfce0c3ef74cafd73f841c5e1f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3cf0e1cca58f14bfce0c3ef74cafd73f841c5e1f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
