)]}'
{"id":"openstack%2Fswift~908206","triplet_id":"openstack%2Fswift~master~I8536788460d6e23b6bee38b66e75f20b077c3f41","project":"openstack/swift","branch":"master","topic":"expirer_grace_period","attention_set":{},"removed_from_attention_set":{"1179":{"account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"last_update":"2024-03-06 18:24:44.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I8536788460d6e23b6bee38b66e75f20b077c3f41","subject":"test for bad config","status":"ABANDONED","created":"2024-02-06 22:33:23.000000000","updated":"2024-03-06 18:24:44.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"41e11e90fd8fb3b0f21f29240893f46d7edfc040","_number":908206,"virtual_id_number":908206,"owner":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2024-02-06 23:25:08.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-02-06 23:25:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ead95a2c359a1051abe12d1e54459e61917d03f9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-02-06 22:33:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"df5e201e6948039ad930638e310afdf580ae436d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-06 23:25:08.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/bfbc05d5583f4da8a2d9c4d6107f6ffa\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/df557cd2458a43daab82f2de016249d4 : SUCCESS in 8m 36s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/34523ea535664f429286c8589a5906fe : SUCCESS in 18m 41s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5668684672e347d4b2a29b6ec53fdb28 : SUCCESS in 15m 52s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c59d7097adde447f89f367f25302e50e : SUCCESS in 19m 16s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/ae021a267a8d4c5fa991c401c7c14e73 : SUCCESS in 12m 53s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/b66419401f8549c29564c2d3ddb29837 : SUCCESS in 5m 39s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/8841ae7eec93495780e18657158e40dd : FAILURE in 18m 46s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/d3a867a1b4be4cb2a84384b99b6ab076 : FAILURE in 17m 34s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/62f98aa7f53e4ec48c25dccfaeffbb97 : FAILURE in 15m 37s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/ce60f94b1f084606bd02384f205999fa : FAILURE in 13m 51s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/0f06bcf857284e3f93e9ca4ae18f8479 : FAILURE in 17m 37s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/8ec66a3bac824fc0a8ed6b29b840112f : SUCCESS in 17m 44s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/f62916b7931746018a400a5d8f744366 : SUCCESS in 20m 33s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/e5e57c00db7c4f539a0720e0197a1ae2 : SUCCESS in 15m 58s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/b2bdd59847524fc89c64253ed764213b : SUCCESS in 15m 49s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/8c32294bb7ae4b569ac5c9007eacf6e0 : SUCCESS in 14m 30s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/c11ea9a1b5bf4092935e084513716e90 : SUCCESS in 18m 52s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/5b09a9b0f4d44e5a9096f28b8d19b580 : SUCCESS in 49m 34s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/1c755b94bc8d45d4bb6dd831c6193c97 : SUCCESS in 49m 41s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1b1441dda8d14b9cb167f4cb143b06f4 : FAILURE in 13m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/244597495e314ff584ddeba4a4d3e679 : FAILURE in 4m 44s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/31199e12e0234910bdb7e7b143f89a30 : SUCCESS in 24m 30s (non-voting)\n\nWarning:\n  Comments left for invalid file swift/obj/expirer.py","accounts_in_message":[],"_revision_number":1},{"id":"218e330efd06a72f86b9117d22c3ebe8f0ef8209","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-06 23:50:26.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9b73cea642d74d8b96ecce4009bd08c8\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c7f4082715d645b3b1b60537a9c3e0b8 : FAILURE in 51m 20s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5c2af5ec8f6545168e38c58e381efad9 : FAILURE in 50m 06s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/7062419ed362429487857b4511a542ad : SUCCESS in 1h 14m 06s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/407395f3764148638452853846f95211 : SUCCESS in 39m 26s","accounts_in_message":[],"_revision_number":1},{"id":"41e11e90fd8fb3b0f21f29240893f46d7edfc040","tag":"autogenerated:gerrit:abandon","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-03-06 18:24:44.000000000","message":"Abandoned\n\nsquashed into 874806: expirer: per account and container grace period | https://review.opendev.org/c/openstack/swift/+/874806","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0f41535623c5e4c598d684bb7b4df1f952bef773","revisions":{"0f41535623c5e4c598d684bb7b4df1f952bef773":{"kind":"REWORK","_number":1,"created":"2024-02-06 22:33:23.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/06/908206/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/06/908206/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/06/908206/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/06/908206/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/06/908206/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/06/908206/1"}}},"commit":{"parents":[{"commit":"5cc1216dd273737ab7c8a08c68d8b44bb639b9e3","subject":"added parsing container level grace period through config and container level grace period in expirer, as well as tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5cc1216dd273737ab7c8a08c68d8b44bb639b9e3"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-02-06 22:32:45.000000000","tz":-360},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-02-06 22:32:45.000000000","tz":-360},"subject":"test for bad config","message":"test for bad config\n\nChange-Id: I8536788460d6e23b6bee38b66e75f20b077c3f41\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0f41535623c5e4c598d684bb7b4df1f952bef773"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0f41535623c5e4c598d684bb7b4df1f952bef773"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
