)]}'
{"id":"openstack%2Fswift~908244","triplet_id":"openstack%2Fswift~master~I8a3e1d4c1806ba1bf7f2a6fab3cf18923a0ccb53","project":"openstack/swift","branch":"master","topic":"tmp_labeled-metrics","attention_set":{},"removed_from_attention_set":{"1179":{"account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"last_update":"2024-02-15 23:44:53.000000000","reason":"\u003cGERRIT_ACCOUNT_1179\u003e replied on the change","reason_account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}},"36606":{"account":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"last_update":"2025-03-25 13:59:34.000000000","reason":"Change was abandoned"},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2025-03-25 13:59:34.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I8a3e1d4c1806ba1bf7f2a6fab3cf18923a0ccb53","subject":"labeled-metrics: Alternative approach to emit legacy and labeled metrics","status":"ABANDONED","created":"2024-02-07 03:53:31.000000000","updated":"2025-03-25 13:59:34.000000000","total_comment_count":10,"unresolved_comment_count":9,"has_review_started":true,"meta_rev_id":"e7c61c81549408927d49e62252ace89132c2140e","_number":908244,"virtual_id_number":908244,"owner":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2024-02-15 16:09:25.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-02-07 14:36:20.000000000","updated_by":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"CC"},{"updated":"2024-02-07 14:36:20.000000000","updated_by":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"},{"updated":"2024-02-07 14:36:20.000000000","updated_by":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"},{"updated":"2024-02-15 16:06:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-02-15 16:09:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"62213ef28d29f537df8b009decad81b1068e6303","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"date":"2024-02-07 03:53:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6698ab76e233358cadfe7246b8c12255c23a3980","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"date":"2024-02-15 14:55:31.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"92f7e32e0e327f3c30f6b8544c895fb8f87ad4f3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-15 16:06:23.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b8def83e682d4f28a79ba2fe50d2dc8a\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d09ddce4d04e4fb883658f9d51d2af79 : SUCCESS in 52m 25s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8163e3451bba44f39b959ed4e8f73350 : SUCCESS in 30m 35s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/fd9c5afd47494d4da27d0e6f1b0e2016 : SUCCESS in 1h 08m 39s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/d722621aa754449f86c9144b3c7e8523 : SUCCESS in 34m 57s","accounts_in_message":[],"_revision_number":2},{"id":"11847f555ca033702b381d2175c02cc917656943","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-15 16:09:25.000000000","message":"Patch Set 2: Verified-1\n\n(94 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8ed7eef4109d4ec4a0b62b6b680e095e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1ff7222f5a954e86876e6c37da0ce840 : SUCCESS in 8m 51s\n- grenade https://zuul.opendev.org/t/openstack/build/1916ba154a84446aa777e69ac69b83b5 : SUCCESS in 1h 00m 55s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/a142dcc7b34c4e30a21448d7af93a413 : SUCCESS in 41m 48s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/8461919533b24769a33be295ef16bf09 : SUCCESS in 40m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1e1832907aed40b698cff9068f81c168 : SUCCESS in 25m 52s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/bbbd05e86fa64a918a6800649a892d7c : SUCCESS in 18m 05s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/70c85956036b4a39a3f4f8a71d262155 : SUCCESS in 16m 51s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f9bf60eaddbf44de90b312d65455ef47 : SUCCESS in 17m 25s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/000aa56e784545f794f82db4d4724aa6 : SUCCESS in 10m 37s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/e6131e2437f249cc9fb2e41ec7c38f40 : SUCCESS in 5m 31s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/eb2f69241cf74848a10fa2ef73dedc13 : SUCCESS in 20m 31s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/082b1622c3f8464e93b6cca9b5d93d24 : SUCCESS in 17m 17s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/b3fd582fa66d4925814dc3098dc78fd2 : SUCCESS in 19m 21s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/c7e3552d5aa34da0a1705e319850b72f : SUCCESS in 15m 43s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/748476ed4ca745e9ab89d4ec9aaa5179 : SUCCESS in 17m 12s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/96f6678304c9476887f6d809bfa072dc : SUCCESS in 17m 28s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/f0e0af6db147437297b22d62c3ae76d4 : SUCCESS in 19m 22s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/df26e56a3fca4c5ebf89cf161d2a3273 : SUCCESS in 19m 13s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/89cf88faddcd47f48369d384d1d58c43 : SUCCESS in 14m 23s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/686f8dc612b74b0cb63c0b8c473ef437 : SUCCESS in 16m 47s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/6a91de7e4236453ba65142eb10f26d6d : SUCCESS in 16m 51s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/ab1933f3f3214d29afeaf78f407091d2 : SUCCESS in 7m 08s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/40767e1d474b427093e0f3c2cfcdb46c : SUCCESS in 16m 49s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/0996259708a84b7d8ba45483a8e9a999 : SUCCESS in 6m 52s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/dff4453164754e4e908e5b3262ac604f : SUCCESS in 59m 34s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/81b90e24fde64ef1ad936165a78986bb : SUCCESS in 1h 07m 33s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/80c81dbfde0c40f7b9b1c4f06a4d0199 : SUCCESS in 58m 47s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/2921e2bdc20248c098972f6613c4a3aa : SUCCESS in 55m 11s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d19767a799b640b29d2100fb61fc69de : SUCCESS in 21m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/729d6f30acba4e778c06c84c32803957 : FAILURE in 4m 31s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/3ddadfb413334d82b04d8664cb4dd6fe : SUCCESS in 20m 50s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/884e13de61204d41bbe399e55c79789c : SUCCESS in 1h 04m 54s","accounts_in_message":[],"_revision_number":2},{"id":"678d20323972a96a8838b72f343fad7f63994278","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-02-15 23:44:53.000000000","message":"Patch Set 2:\n\n(10 comments)","accounts_in_message":[],"_revision_number":2},{"id":"e7c61c81549408927d49e62252ace89132c2140e","tag":"autogenerated:gerrit:abandon","author":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"date":"2025-03-25 13:59:34.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"c5fcbc64518445de7f0cf11f4cf9096f0dc3c30f","revisions":{"5cc61c2ed288317a81aef75b6efd35afbd6d01e0":{"kind":"REWORK","_number":1,"created":"2024-02-07 03:53:31.000000000","uploader":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"ref":"refs/changes/44/908244/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/44/908244/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/44/908244/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/44/908244/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/44/908244/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/44/908244/1"}}},"commit":{"parents":[{"commit":"40172db30b27e541f04ad0160f23f5b4133a934f","subject":"labeled-metrics: Add option to emit both legacy and labeled","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/40172db30b27e541f04ad0160f23f5b4133a934f"}]}],"author":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2024-02-05 20:55:36.000000000","tz":-300},"committer":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2024-02-07 03:36:37.000000000","tz":-300},"subject":"labeled-metrics: Alternative approach to emit legacy and labeled metrics","message":"labeled-metrics: Alternative approach to emit legacy and labeled metrics\n\nChange-Id: I8a3e1d4c1806ba1bf7f2a6fab3cf18923a0ccb53\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5cc61c2ed288317a81aef75b6efd35afbd6d01e0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5cc61c2ed288317a81aef75b6efd35afbd6d01e0"}]},"branch":"refs/heads/master"},"c5fcbc64518445de7f0cf11f4cf9096f0dc3c30f":{"kind":"REWORK","_number":2,"created":"2024-02-15 14:55:31.000000000","uploader":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"ref":"refs/changes/44/908244/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/44/908244/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/44/908244/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/44/908244/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/44/908244/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/44/908244/2"}}},"commit":{"parents":[{"commit":"2ed862a99350b566d1c31c3f48a01a7c23d31f23","subject":"labeled-metrics: Add option to emit both legacy and labeled","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2ed862a99350b566d1c31c3f48a01a7c23d31f23"}]}],"author":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2024-02-05 20:55:36.000000000","tz":-300},"committer":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2024-02-15 14:36:09.000000000","tz":-300},"subject":"labeled-metrics: Alternative approach to emit legacy and labeled metrics","message":"labeled-metrics: Alternative approach to emit legacy and labeled metrics\n\nChange-Id: I8a3e1d4c1806ba1bf7f2a6fab3cf18923a0ccb53\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c5fcbc64518445de7f0cf11f4cf9096f0dc3c30f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c5fcbc64518445de7f0cf11f4cf9096f0dc3c30f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
