)]}'
{"id":"openstack%2Fswift~908252","triplet_id":"openstack%2Fswift~master~Id935de4ca9eb6f38b0d587eaed8d13c54bd89d60","project":"openstack/swift","branch":"master","attention_set":{},"removed_from_attention_set":{"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2024-02-07 20:26:16.000000000","reason":"\u003cGERRIT_ACCOUNT_15343\u003e replied on the change","reason_account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}},"hashtags":[],"change_id":"Id935de4ca9eb6f38b0d587eaed8d13c54bd89d60","subject":"replicator: Rename update_deleted to revert","status":"MERGED","created":"2024-02-07 07:16:02.000000000","updated":"2024-02-07 22:58:13.000000000","submitted":"2024-02-07 22:57:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"908252","meta_rev_id":"195bf2f5d5be136b0a97602a9708acdc5db42a29","_number":908252,"virtual_id_number":908252,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-02-07 22:57:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":2,"date":"2024-02-07 14:41:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":1,"date":"2024-02-07 14:41:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-02-07 08:30:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-02-07 08:35:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-02-07 14:41:20.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"}],"messages":[{"id":"436e848ac9e9a086bf6f8cf48a8321476d09e4c4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-02-07 07:16:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bac5e27ccef5b35cd1b282df4d862e71c49caed4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-07 08:30:46.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3eb762408dfa4dae93e5e5471015af76\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/3764b1c659ed4c63bc5ee9e872a9b8fa : SUCCESS in 16m 06s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9934e47537dc4659a2e0da3a74fb9d16 : SUCCESS in 17m 03s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/2825e40a961b4a3e981549a94ea0014f : FAILURE in 1h 12m 50s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/9ead5c32e4984c019983be6579174bd8 : SUCCESS in 18m 14s","accounts_in_message":[],"_revision_number":1},{"id":"adf5a2415d9bcbb28b139f8467afe18d86a31bb5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-07 08:35:47.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1c9b5d7f270e400f8a8a585bd13fc2d9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/704019321b45434e9767edb034fddf3b : SUCCESS in 7m 31s\n- grenade https://zuul.opendev.org/t/openstack/build/763d79f68e7e4aa0b2e1a0aae584c191 : SUCCESS in 45m 27s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/b6c77a94e79b4addb476cc24c4e5241a : SUCCESS in 47m 04s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/30bfcc40f2fe4f0dbe1be10196a3eaad : SUCCESS in 42m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a7efc1bd07224492b09133e9a12f79fe : SUCCESS in 26m 41s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4855dca6edf14977b32dae9fb945baa0 : SUCCESS in 21m 58s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d3bd94180d26433fb79ec2fea101bc32 : SUCCESS in 27m 43s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4b03cbe189724bfcbde8e2e513cae109 : SUCCESS in 28m 42s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/0f5b171b931748b6a729011bba6f6b04 : SUCCESS in 10m 39s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/5f8779b83cea485db33fe22337885536 : SUCCESS in 5m 59s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/9ff58d3d100f4d6a95f4d55330ef9420 : SUCCESS in 17m 15s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/eaece99b0f3f4b7dae82e599728a5a1d : SUCCESS in 18m 29s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/28ec057bba2245f3a8659cff3771277c : SUCCESS in 21m 32s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/e60f63e080a544cfa6f1854d8862e37d : SUCCESS in 17m 33s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/8dbf0e1843124d7aa75541a8246f0488 : SUCCESS in 16m 22s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/6f2e4b7337f54dffbbb06b23ba9df3f9 : SUCCESS in 17m 49s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/391b14d85eb249799850f39dfcd1f620 : SUCCESS in 17m 01s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/f17c8a77c6874510bc168337224308c2 : SUCCESS in 16m 39s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/c83d59f966be4065a15f05dfd2147be2 : SUCCESS in 17m 20s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/93957e7e9b404834a0e53b844a5aa328 : SUCCESS in 15m 00s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/154c3afc20774b9fb7d5f3d6152ad0cc : SUCCESS in 13m 20s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/188d3b78a69f45d5ad81b09d59e8e827 : SUCCESS in 7m 58s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/c0ba41f529074668b80ec00fef229b1d : SUCCESS in 17m 02s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/f6e3c48218eb4633828686b49a82a81c : SUCCESS in 6m 38s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/5d868cbab47c4335bdd05e85e3f244c7 : SUCCESS in 49m 47s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/613be17b9c604693b0699c921aeff9a1 : SUCCESS in 1h 00m 20s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/7953e3058b8f46c8b395cf7ab26fb88a : SUCCESS in 54m 11s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/315ddee177844f0eb13e91b6e9aaf055 : SUCCESS in 50m 36s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/40bb20be031b44a59d60384f89b5e1a3 : SUCCESS in 13m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2fb5a5a20af844ff96407d234becf0c5 : SUCCESS in 4m 38s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/0a567e69a58540cdaa5fcb0c6ded8ea8 : SUCCESS in 20m 15s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6cce99c616954632bb4bb2cf2fc08052 : SUCCESS in 50m 15s","accounts_in_message":[],"_revision_number":1},{"id":"a7605490331da95fb08d71db98a3a28a3bbb0707","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-02-07 14:41:20.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"819a92e7fe143aa53437ae7667d28155311bd572","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-07 14:42:06.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2993c4d5f38454eda9f4f857f61d49ee74df6e99","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-07 19:16:34.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c29c57536df7471fa0cd59d2cb9c2956\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0a98fa6689944a488457ab1dc546df9d : SUCCESS in 9m 46s\n- grenade https://zuul.opendev.org/t/openstack/build/75deb37d62764f9c8f70414711c28d82 : FAILURE in 30m 18s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/03a4e36814f846c1ab978da599cd759c : SUCCESS in 1h 02m 36s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/b005b7ff08024a09aba54fa8ee7c453f : SUCCESS in 51m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7c75683d578e4755be301df48e7280bd : SUCCESS in 24m 32s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/1b378b72c9de4ae0b776bbf045fa9fda : SUCCESS in 11m 18s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/b4b7f855813b4d66a550849d43693b0d : SUCCESS in 7m 21s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/a8913da91fb0452696d4eb0eea434f5a : SUCCESS in 27m 20s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/b9bfcf2bb05c4100899de8d3baa2f784 : SUCCESS in 17m 30s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/10399e1d15554887a6ad2067ece265f4 : SUCCESS in 24m 47s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/e8baa66ae8db426e8ae23bf2fe098904 : SUCCESS in 17m 03s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/314ae019e67b4c18b718fd5e017fc743 : SUCCESS in 18m 20s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/bf5b8401a7914e979e04ecfcf0b3c730 : SUCCESS in 16m 23s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/f502bb34463b40f1bcb059c13f4a66d0 : SUCCESS in 27m 17s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/02ac7ffab1fe4d6dae1aa6cc79c88ea9 : SUCCESS in 20m 52s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/8ea2efc07f044bc399961abcfed55cf1 : SUCCESS in 26m 13s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/77ce51124f544f91a54842650a295109 : SUCCESS in 17m 42s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/54361a6b64c643159f48fb2e34775afe : SUCCESS in 19m 44s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/2e49418e74ac4ce99ad940410a61d95e : SUCCESS in 7m 20s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/fcc61d6feba344ba905828f4526bd71a : SUCCESS in 9m 15s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/34b3442718264cb39171dab8b43e53d0 : SUCCESS in 53m 31s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/99c111442d1b4852b8e6d7f7153fdac5 : SUCCESS in 1h 03m 47s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/9a2e8abb89fe4cf4bea4997012d7189a : SUCCESS in 53m 07s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/201c1184b31f4f01924521682da0970d : SUCCESS in 49m 16s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2a078bcc65fe43c5a659c7406c7e151b : SUCCESS in 18m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/367e342938c54392a4e29bddd8bd53c4 : SUCCESS in 5m 23s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/981e718e1a6e4914bd292f2daba9f715 : SUCCESS in 47m 26s","accounts_in_message":[],"_revision_number":1},{"id":"71d63e24ce75021f602da8a935cd92964494052b","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-02-07 20:26:16.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b09d9fe3db0b2f0d323b702bf9edee4f57cf52bd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-07 21:15:39.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0154ef6e84a84b238ca92f6175f86325\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/8706fe9c394f4f5f9fd4dab3a7a628b7 : SUCCESS in 17m 25s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/15100952329c4aa2bcae9ca6778baf58 : SUCCESS in 19m 07s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/f8dc5614b5de4a14a563ebb140edbf94 : SUCCESS in 47m 32s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/6872ae3dc93c413393a9e687d62bd37c : SUCCESS in 24m 07s","accounts_in_message":[],"_revision_number":1},{"id":"e437d61a9392ae5bf557fd75f5b579e82a0d1c1e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-07 21:52:08.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eb3c1b329298411ba13e3890f595b512\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/59c97ab8edca40b08d1f0ff23dbf060a : SUCCESS in 9m 51s\n- grenade https://zuul.opendev.org/t/openstack/build/90e1f551f7ec4c109929a0e00caf346e : SUCCESS in 1h 01m 16s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/4ee1c39c703a49c4979894d3bbe38bd4 : SUCCESS in 51m 15s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/0a8b600d6ff44f7eaaa22c467074bdfd : SUCCESS in 45m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d6ad3220654b447ba4fb8901332d4ab6 : SUCCESS in 29m 29s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/11af300b0c5b40eea7235e0935612a46 : SUCCESS in 18m 35s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f854a2e2e07648f5ac9a7c47b045a5d7 : SUCCESS in 19m 31s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a80cef6dcf9e47ed912fb0a0af226c2b : SUCCESS in 23m 22s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/5118ec9d23d24e3696f3325ac91c9262 : SUCCESS in 9m 23s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/6dd9818adcfe48db83e5390979eff5b9 : SUCCESS in 5m 25s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/0d40d832af654b7a875fb8ddf1345283 : SUCCESS in 17m 38s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/3670a7e783f843b9b795242b3b3695f3 : SUCCESS in 13m 17s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/da2f28a084d645e9b6b2d9afb2f18126 : SUCCESS in 13m 52s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/498174669a0145f699ffb38ca4ea3c47 : SUCCESS in 18m 13s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/6b2fb0d0b7ef413583af29df6f09bacf : SUCCESS in 18m 40s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/97cee24fa411494293c22928a53f1312 : SUCCESS in 18m 14s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/43edb0754db14ca7b329ab7b338dfdf1 : SUCCESS in 19m 02s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/cf33406ef0554a0790c8c09d4a466338 : SUCCESS in 25m 01s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/2c7e6f4da5bd45bc9f15a0b01d29dfb1 : SUCCESS in 16m 32s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/6ff1bbbc2428404bb92efd3a5dec39fa : SUCCESS in 17m 25s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/8e648aa355cb49b5a3e7d41dbbe7fddf : SUCCESS in 18m 41s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/03a9f09233684eb4b18aa7792e40fc1b : SUCCESS in 8m 45s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/45b401eb303c424d84fa5c11b5da15d8 : SUCCESS in 16m 09s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/cde5ebd2b2944320947ebc1f06b0ab6a : SUCCESS in 10m 11s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/578a6d9c0bf9408681b2e791bec196bf : SUCCESS in 1h 02m 15s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/10cdb761858c449d9795fe8c63dce50a : SUCCESS in 1h 08m 42s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/fa46788c1f8d484bb972c15670f3b82b : SUCCESS in 37m 09s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/17592af5b3734b66834ed343190c9a08 : SUCCESS in 54m 39s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/886a21820e5b45c4ad6ad9cde61a8ac8 : SUCCESS in 19m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/97466edd7b594f7a9ba481435b15bd73 : SUCCESS in 4m 45s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/a8803e92d635428c916cc2c4cdeaedb2 : SUCCESS in 21m 43s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4f1a554d064c4d93bd1b518b06b7527d : SUCCESS in 47m 43s","accounts_in_message":[],"_revision_number":1},{"id":"077a66678012933d12e3c2322e8a5271c8e3ed21","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-07 21:52:41.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"bf34877fbd71e4a080dc7c0d67a743ed3c2f0e53","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-07 22:57:11.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a152959115eb4322a0f21b32d3e75fae\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c5e980de14ce4a3e8694c243d9c90b88 : SUCCESS in 8m 08s\n- grenade https://zuul.opendev.org/t/openstack/build/1eb08133087a4dee8898135efd1c2e08 : SUCCESS in 48m 26s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/3687cf68d00e428dbd3204cb2bff0fc8 : SUCCESS in 48m 10s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/d41f241f962847cab804bd7a486f3ba6 : SUCCESS in 37m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8d1ae2c2250c48bf94d1d6ae600e7d93 : SUCCESS in 24m 15s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/ff892042eb5545a38a9f954ed8647000 : SUCCESS in 12m 29s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/36f64323da74475a9edea58b245fa134 : SUCCESS in 5m 52s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/797c75b1fc56429985b0915830251b8c : SUCCESS in 18m 18s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/9d610c5ca02843659c879da4ebca0994 : SUCCESS in 15m 21s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/6ffc23b9de8f495099a404255a890b0f : SUCCESS in 16m 51s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/93d2d21b18d646e196e72196ea3528e8 : SUCCESS in 16m 45s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/63ef6884913a431594484e1695f08697 : SUCCESS in 16m 21s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/0b3dbbfa79d644518387ab624f86d804 : SUCCESS in 17m 36s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/efadb16548dd47ba96c9fc6710e34909 : SUCCESS in 18m 01s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/b565468b6d4d431886425c3a79a1e50f : SUCCESS in 20m 24s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/3394d7b8c6794143b38d52a118dcc791 : SUCCESS in 15m 24s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/7ceb029f32894fa58205c0b8dd8cd0cb : SUCCESS in 16m 17s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/80d8c388cfed41478b6a2f7bfa1fcd20 : SUCCESS in 17m 16s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/9e12f5cbb9934224b53f0c7987ee7e6e : SUCCESS in 8m 29s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/a61245cf89fd47699e813c78dd573323 : SUCCESS in 7m 11s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/c13ea6f76c3a4153a3cd226133de6d38 : SUCCESS in 54m 50s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/f7037144cc794a0698762c0814c5a812 : SUCCESS in 52m 45s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/638db2c4ace34b7f864bb6e4760d1117 : SUCCESS in 36m 10s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/6a526bf36e924a6b973983b84ec530d3 : SUCCESS in 34m 49s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c09d85840fea4ba5ac0fcc1c07284d90 : SUCCESS in 29m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e5b34557e0b24f4182d2bebc4a7f97d7 : SUCCESS in 4m 29s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ebaf4205d72340efa0ea0187316f78ed : SUCCESS in 43m 30s","accounts_in_message":[],"_revision_number":1},{"id":"4e44e6e49ae9abc4dce2d06c026e83346e1a55bd","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-07 22:57:12.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"195bf2f5d5be136b0a97602a9708acdc5db42a29","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-07 22:58:13.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1e8f1ec742be4a8cbaf783015cd91377\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ec30150fa65f488989b01064dcdd6711 : SUCCESS in 42s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/4619687948d840088249e05b3cc54ece : SUCCESS in 31s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/a5e605da255c469189fd3c0172b3e13e : SUCCESS in 34s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1936f6735cbef8b3aa9bc8ca2e222c394df797e0","revisions":{"1936f6735cbef8b3aa9bc8ca2e222c394df797e0":{"kind":"REWORK","_number":1,"created":"2024-02-07 07:16:02.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/52/908252/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/52/908252/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/52/908252/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/52/908252/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/52/908252/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/52/908252/1"}}},"commit":{"parents":[{"commit":"afe31b4c01881aa761f7ecc8942245301cadc476","subject":"Merge \"tests: Fix float expectations for py312\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/afe31b4c01881aa761f7ecc8942245301cadc476"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-02-06 20:22:06.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-02-07 04:24:03.000000000","tz":-480},"subject":"replicator: Rename update_deleted to revert","message":"replicator: Rename update_deleted to revert\n\nThis is a more-intuitive name for what\u0027s going on and it\u0027s been working\nwell for us in the reconstructor.\n\nChange-Id: Id935de4ca9eb6f38b0d587eaed8d13c54bd89d60\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1936f6735cbef8b3aa9bc8ca2e222c394df797e0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1936f6735cbef8b3aa9bc8ca2e222c394df797e0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
