)]}'
{"id":"openstack%2Fswift~908254","triplet_id":"openstack%2Fswift~master~I3e7bc837b1f978932cd47cd94b03e2a0e3444bd0","project":"openstack/swift","branch":"master","attention_set":{},"removed_from_attention_set":{"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2024-02-07 16:22:02.000000000","reason":"\u003cGERRIT_ACCOUNT_15343\u003e replied on the change","reason_account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}},"hashtags":[],"change_id":"I3e7bc837b1f978932cd47cd94b03e2a0e3444bd0","subject":"diskfile: Add API to remove a suffix from hashes.pkl","status":"NEW","created":"2024-02-07 07:16:02.000000000","updated":"2024-02-07 16:22:02.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":1,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"f313ca0d22c8ab8982bd54b0ce8736ac9a2d3f20","_number":908254,"virtual_id_number":908254,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2024-02-07 08:52:38.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2024-02-07 16:22:02.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-02-07 08:19:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-02-07 08:52:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c23ea7a700eb4f5b85e0d0e25377ed0221b2e1f8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-02-07 07:16:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b37e5e43fc7b54a05822d586cd200f451fedd987","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-07 08:19:04.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/33c2d2154a534920886a4a250cb06181\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/620675431eda4d9d9f8cf17c70aaca42 : SUCCESS in 16m 03s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/aa316e1f374448b7a2b5cc45e3b5e89e : SUCCESS in 16m 27s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/ca6681b7b46945db80218cba881dc0f3 : SUCCESS in 1h 01m 27s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/f6c33aa6cf2847a78313a6ea46618d33 : SUCCESS in 18m 36s","accounts_in_message":[],"_revision_number":1},{"id":"5bc05f5bef58d9e025f86e9d56d36a5c6b35adf6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-07 08:52:38.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c930f0cd7cde4255bd328157815d8352\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/98a4653542a342aca7c70a6f576f9bbb : SUCCESS in 8m 53s\n- grenade https://zuul.opendev.org/t/openstack/build/fd900635c0194cfcb95cd2180a02dadb : SUCCESS in 1h 02m 05s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/cb7a155c7a494738964c8c3b328ef65c : SUCCESS in 50m 43s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/9b896f951d2a45769afe6e2d05774c48 : SUCCESS in 30m 05s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3e5f38b2071c46ddafdb271810078ffa : SUCCESS in 34m 29s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5c37e123f722424d96e5471de6964691 : SUCCESS in 17m 29s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e43a2cf5b0154148af85595d1ceb13cc : SUCCESS in 18m 55s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/114b5cce768540bc81b7411e6a9f032d : SUCCESS in 30m 51s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/ce274182c5e44894991d3890da6af8c8 : SUCCESS in 11m 16s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/a80eafb29d9b4562be0f68742e315f30 : SUCCESS in 4m 46s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/adc7158e3e544e05b3677d42570d6f71 : SUCCESS in 15m 08s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/0625a6acc2404b74b34f29cc32c6f516 : SUCCESS in 17m 37s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/176c813db5394f1e8e34429935fe62ab : SUCCESS in 17m 02s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/58a0547ef2494181af9bb1bd2c453901 : SUCCESS in 17m 01s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/ff6abfb092e846c99220d5e97a072f33 : SUCCESS in 16m 24s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/f42d694ca859479dbbbc7a1e066a1cb8 : SUCCESS in 18m 57s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/89628cb791ff42ea9128ead33d989f36 : SUCCESS in 18m 08s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/09855332ed42464baaedc90a8d71fc08 : SUCCESS in 18m 22s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/978bfdf6a3ba4f058f59148a4be479a9 : SUCCESS in 15m 39s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/4f7df686f9864b7b8d5ed4db048f1b82 : SUCCESS in 17m 17s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/559ca0c861984c1eb7c6fa48b607d2f7 : SUCCESS in 17m 48s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/0428e8c5b793467a8cacc84dc30fb583 : SUCCESS in 6m 31s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/cd075e4300364b2aa188a0e13df97f2f : SUCCESS in 16m 48s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/d858ff9f1821409ea7ffd72fa710fe59 : SUCCESS in 6m 48s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/a7317f1167e74c038daaa0322bf63578 : SUCCESS in 57m 21s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/ac1bc52e16344d75ae0092ed2a201e8e : SUCCESS in 1h 08m 06s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/9db9da4cf0544fb3b3eca12659ce9aad : SUCCESS in 1h 01m 01s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/e487bd185a6f4834ba2c3e4add175afc : SUCCESS in 52m 01s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4f1351a5dc6e44e9a02811c91c00ed3a : SUCCESS in 18m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4581a158c8d54243af0c0f0b31c86fd1 : FAILURE in 4m 30s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/fd216b6238894e0f90de65961e0d6cd5 : SUCCESS in 19m 37s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/eb3e3ac7fccd414d90868a4ba1b817db : SUCCESS in 1h 00m 01s","accounts_in_message":[],"_revision_number":1},{"id":"f313ca0d22c8ab8982bd54b0ce8736ac9a2d3f20","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-02-07 16:22:02.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"608183c3249dea1dcdd0bbf13d9f232027bb2ab9","revisions":{"608183c3249dea1dcdd0bbf13d9f232027bb2ab9":{"kind":"REWORK","_number":1,"created":"2024-02-07 07:16:02.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/54/908254/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/54/908254/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/54/908254/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/54/908254/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/54/908254/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/54/908254/1"}}},"commit":{"parents":[{"commit":"4353dc6e513fff9853879d12be88ca01e693610d","subject":"replicator: Add sync_batches_per_revert option","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4353dc6e513fff9853879d12be88ca01e693610d"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-02-06 21:01:58.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-02-07 07:15:47.000000000","tz":-480},"subject":"diskfile: Add API to remove a suffix from hashes.pkl","message":"diskfile: Add API to remove a suffix from hashes.pkl\n\nThis keeps the state in hashes.pkl and the state on disk more-closely\naligned when the replicator does progressive deletes.\n\nUpgradeImpact\n\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\nIf the object-replicator is running new code while the object-server\nis running old, some old versions of swift may see a spike of rehashes.\nConsider configuring\n\n    sync_batches_per_revert \u003d 0\n\nprior to upgrading to disable progressive deletes.\n\nChange-Id: I3e7bc837b1f978932cd47cd94b03e2a0e3444bd0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/608183c3249dea1dcdd0bbf13d9f232027bb2ab9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/608183c3249dea1dcdd0bbf13d9f232027bb2ab9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
